)]}'
{"id":"openstack%2Fswift~908300","triplet_id":"openstack%2Fswift~master~I199d2a31035610862421a3bf8c6ff80e63f31d3e","project":"openstack/swift","branch":"master","topic":"p-coop-iter","hashtags":[],"change_id":"I199d2a31035610862421a3bf8c6ff80e63f31d3e","subject":"CooperativeIterator: ignore period\u003c\u003d0","status":"NEW","created":"2024-02-07 12:16:30.000000000","updated":"2024-02-08 10:57:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ea65df4e5f3b7471c64656986d8460fdd9f68f07","_number":908300,"virtual_id_number":908300,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-07 13:49:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-02-08 10:57:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 13:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 13:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a530e0f2adb90de1b6126e50dabc151c23476709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-07 12:16:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7c961a7c083a6bd6f3f73e4fa2ed1467ffdb425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 13:29:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c403aa62f14920b54527f6cb49de39\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4554ef07434a41f3ac3eefa8c35369d8 : SUCCESS in 1h 10m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adba16d1fe834a3498a50438dfa63cd1 : SUCCESS in 57m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/998b20188aeb4c0c82e81dd3581f1122 : SUCCESS in 40m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/961d9eab697f49be9400c33d1ee6127c : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2007cef72430a3f7ecf906fc756bb3b62f713ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 13:49:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38a53439ecd34ff09986e08162ea2c32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9596f82084b94561a1e55c737ee844e9 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/259073565ad44dff92225480c126357f : SUCCESS in 1h 02m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e4cb9b8bd80348559e1d4ff290ae62ce : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cea1b13b09d949f898627d235982ffd7 : SUCCESS in 49m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3271b7b342b04301bc91e2400841802f : SUCCESS in 36m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a4bba3622c545afbdf1faabb1ce2891 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5af6a1393203490585e4e074cd75a083 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c5e57a4fc9845678bc906d2483619cc : SUCCESS in 21m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/216ba483b12b4dfcb4a1fb8af404bafc : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5b11468048a480cb70c48f6a1b11d78 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a530797eb084d708f1b69aab9cbdc1f : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/095c2cb65471440aacffc022d4ab0521 : SUCCESS in 17m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66a40cbb1e1a400a89d63d4e49318a7b : SUCCESS in 14m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8fb407d0cc7045939b5c1a90172b29c4 : SUCCESS in 16m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c2fcf985c4074d8299f6af3f485038d2 : SUCCESS in 24m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31f123f20ca64b1f86b4c5de96947143 : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60d2952910a147b1914ced2d336b1f7e : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20b2cab3f3d446a186eac0bdc0cd859b : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08e08ef7646845619fb271ea20f3793f : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d7085d6962d4c1d9fcfe7d7d7acb300 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4292e1e063c441f9c1abf5df0111601 : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aae98568e30346d48d175d7e0468b7b9 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edbec001339b4c8fb9918957c844443e : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17c56a439a334214825020659b7a3ed7 : SUCCESS in 6m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fec42e3b3fe44d1b8e13952b322b742d : SUCCESS in 45m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ded438fd30b24179ba9ea36ad46119f9 : SUCCESS in 1h 22m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5b77871ed4d49c2ba11b90aa5d3427e : SUCCESS in 57m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c3ee53f456145d4a5993bba0ec40d3c : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d5d74af048f479e93073c432328deb1 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83585463294d47109aa2abc1d3b885e4 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/872ac3dea8d14e22ad9c79aa7bfa7c5e : SUCCESS in 24m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb5fe382b45a48f2955a550290a56aa2 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ea65df4e5f3b7471c64656986d8460fdd9f68f07","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-08 10:57:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"456eb3f2f3f0fd5229e93e5c6f5a7c3e2ad42402","revisions":{"456eb3f2f3f0fd5229e93e5c6f5a7c3e2ad42402":{"kind":"REWORK","_number":1,"created":"2024-02-07 12:16:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/00/908300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/908300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/908300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/908300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/908300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/908300/1"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-07 12:13:28.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-07 12:13:28.000000000","tz":0},"subject":"CooperativeIterator: ignore period\u003c\u003d0","message":"CooperativeIterator: ignore period\u003c\u003d0\n\nMake CooperativeIterator tolerate being given a period \u003c\u003d 0 by\nignoring it and never sleeping. This saves callers having to check the\nvalue of period before constructing a CooperativeIterator.\n\nChange-Id: I199d2a31035610862421a3bf8c6ff80e63f31d3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/456eb3f2f3f0fd5229e93e5c6f5a7c3e2ad42402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/456eb3f2f3f0fd5229e93e5c6f5a7c3e2ad42402"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
