)]}'
{"id":"openstack%2Fswift~908337","triplet_id":"openstack%2Fswift~master~Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278","project":"openstack/swift","branch":"master","topic":"p-native-mpu-backend","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-07 20:13:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278","subject":"DNM: WIP: test jig for native mpu audit","status":"NEW","created":"2024-02-07 19:08:31.000000000","updated":"2024-03-01 22:29:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3156102ad3388c1c20cc1df2722788d3091664f1","_number":908337,"virtual_id_number":908337,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-01 22:29:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2024-03-01 21:16:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 20:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 20:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fd0b6d7cb8458ff1d347383ae348b5a4c42e1b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-07 19:08:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"412410412f568c5d12897aa19efb4d0636bdca07","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-07 19:08:50.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"f1cd59065b8d6ee76f5bd5d8cb0c5a6415ba0d59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 20:03:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7431ebceac4ce3a554a7b1297b54da\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d54600a80ff046a9870746da4fc48c0a : SUCCESS in 13m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e820b1c317345a098df61ed8b8b3efa : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1604b900fe1499b908a876ab6a0e206 : SUCCESS in 18m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbc7c0f291c549ec9b1e1d47054b5010 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6f2d03053923a24b21cf6019bd22835bd78c3458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 20:13:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23687c9283b14bcdbb4b8467e3405820\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e2b02fe86146acb89591bee1df4fc3 : SUCCESS in 8m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/552b826d29e54b13adb931eb632e6f0b : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d8ccdc721f34e19aea28c4bbb740cd1 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b421646792674b7eae09590ab51b4448 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d1528a4cf7846718d907add53265270 : SUCCESS in 11m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/938f40d5cf634896a89b8c92927558a3 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/626d017f8d404b0db3d8af4e59f20603 : FAILURE in 25m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba3d148f2f6846a6bc9e799929bea4eb : SUCCESS in 17m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ee9e03f5fa5b44ef898fdf7b81a0adaa : SUCCESS in 16m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf5c249aa464ba7b0a8af7bda8a7832 : SUCCESS in 23m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca5034c3f49a49799fdeaefa97fcdec2 : SUCCESS in 14m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4589f0d540ce456a81e07de878586e4c : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75a2c6fb35db4882aca76641ff9e607a : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfe28a22e4b84b068285397284bea0bd : SUCCESS in 19m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/59866e8ef261443480f3bc16ef5a7206 : SUCCESS in 20m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d158d21e65544404b90a35f1890efca1 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45b30faade9748eb8bac866102d0fc84 : SUCCESS in 20m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34cb5c70331b4197bed1a96365d5b44e : SUCCESS in 58m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd702bed286b4a148b604f9245d1f99f : SUCCESS in 36m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d1d1877c66c4695a799635c9b653142 : FAILURE in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ba4be42b2e4a5486f9493ba9c92f41 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/300cba5b557a4b9aaafc65c5b25bd609 : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"796b217ad726e740638a8214d2ed9e96c039fb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-09 17:28:15.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"29f4a0b59b5de053dd39bf5a34d6319bd211c826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 17:57:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac0cde55def4a1484952a8ba82dc473\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c4aa17af8094704b82ec8e23a55f402 : SUCCESS in 22m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9de42d5729e94599aed67ba0f4337d11 : SUCCESS in 22m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0460b15c6d534e9c923739853ac26ad0 : SUCCESS in 28m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/48345504d09249409b0fbfce00aa1314 : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":2},{"id":"4fd8da9ba53624430ed293d0e456750ede498665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 18:30:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a54d09937144daba1e6ad33837af69f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629bc5826f2142d0b148e739b138167c : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76bfb8fe5239466bbe3c4b03b367aaab : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a08f520ef3c42ebab2766e57df96ba6 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4f69fc942144d049b953f1111ace633 : SUCCESS in 16m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0b0797da462471f88495a10c3bac3ce : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc0a676775034b808ef9b61dc34b87e3 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac5d42132cbd42b19d4e30136bb32193 : FAILURE in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5da6fc380e944545b28d62e545471d65 : SUCCESS in 13m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b58ebc4c705d4ad6b2ec7dfc447e93a9 : SUCCESS in 17m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a86981f8bacb4e02979fc2feac92c3d3 : SUCCESS in 14m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/afb910e97e9e484387cf21c46a04d88b : SUCCESS in 18m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd5499038579448e8e203d12b9e9f83e : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ff1a0ba27344aa2be72f23b235c2940 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/962dcfd3388d4de48a3b34f3413f20c4 : SUCCESS in 18m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cbb6c1bf0ef04fd9bef1b284ec0996f5 : SUCCESS in 22m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/414ffc76d53b4e7e8ed59f4f5bd689eb : SUCCESS in 24m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5a9f98b2c5fe49c894e24c29da24fb11 : SUCCESS in 24m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/413e650405ac43149aa91d258afe1dc3 : SUCCESS in 54m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec1e49471d454ec78acabfeb00813975 : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f72378d8b12340c399beed0fb764b56b : FAILURE in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7964e62464034bf99ed81f52644645d5 : SUCCESS in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e751e560721c4cf882e36e1aeaa03e2d : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"088d54f672a35894e8693adfdfc87bf8e924d23c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-16 17:14:01.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7973a877b2bff2e5125c73434fea5a622fd70ee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:58:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3046f7d39c7b47c5836d11f69e1db418\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbad2a4a527345eaa1a214e0f8fbb657 : SUCCESS in 20m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfafbec01ff346ec9e32c04c060793da : SUCCESS in 23m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7753e674aa5f4b8392741cb66a087971 : SUCCESS in 24m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea50618ae32342d682c0db36be3184d1 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c6e6cfa49dbe16d12e4c7f9baee426e3c588442b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 18:20:52.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ceaaee0698764073ae0407d237d5a998\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80fd40aabc64d1e85e60f29d813ce26 : SUCCESS in 8m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de8fcd58551c46ae969418cf78d11e83 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ad3199dcd524ca8ab3a05f69dcfd392 : SUCCESS in 29m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84117d1deb5e42348fec21c200aaa913 : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a85fd17f3b494f7ba809defbb22e198d : SUCCESS in 11m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89b8788ba176425bb215c4da03823f46 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f25fd28d9344270b37098d1b351e799 : FAILURE in 18m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5685d7e5475143d38495ab7306e2aa14 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/93a993f2943a4e02b0fc0ab2d993c35a : SUCCESS in 18m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0033cbaa22fc4731ac73517fc810c688 : SUCCESS in 19m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/94a0052ea49243bd88bd3e74bc5d1024 : SUCCESS in 17m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53be43605f864f7a998a1fe7e911d0ba : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a4c2e24b3a740c798645a41f35e5efa : SUCCESS in 21m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f36259520ae448bca7cdb0f64455db3f : SUCCESS in 20m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be7e2c7caf694f8d9c19a67dc088df24 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3139eccf50b24d219cded60cb76bd040 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/076a211a57354dab9bae57255fc1619d : SUCCESS in 19m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a555b0500c243a583da7daf058312e3 : SUCCESS in 55m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1413856856f24b58ac9223e0ef8d76ba : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1bb23eedf77476697017736a9a5b9c7 : FAILURE in 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b44463ec747459b9a25bf61156982aa : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee3bac3b052b4bf29b904b1d85155194 : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4804f8726963f4856657821b303bc82dfdc57d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-01 21:16:38.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e8bb2895001b25d68b43c55f4b0095eba61dcdbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:24:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304ddd4537ef48d293a8930ce28fa38a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ee1a0e94491469b82b77c9d31a06d38 : SUCCESS in 17m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf0925a37e1b47409aa3d707636881f1 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd896ba20f7d4a25befec710d71de079 : SUCCESS in 1h 06m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0cc8e7a477741b4a399e3efe59b9136 : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":4},{"id":"3156102ad3388c1c20cc1df2722788d3091664f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 22:29:58.000000000","message":"Patch Set 4: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d4aa582ff2e495fbd663dab23c0a3a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99abdf8c14d4d20a1dc8908ef7c7312 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/bcf568322fec4b909fabdedd0a16560a : SUCCESS in 1h 12m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/095213380a0546faab0611bdf786b71c : SUCCESS in 1h 08m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec1dbf68a32e4258b8c901e998ca4c63 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bc624db13b248f5ba4e41233c27befb : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d6ecda674c44212975ab184f9815895 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/819e4bf8928d4194b62c746999923432 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f291cc0eec4141d989ee954c815f5d6d : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/deb500df9f4540bd806e454153c864ac : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f587f92b4faf492ea1d691555e2f618a : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e04c739491e6485eba8826f7945cdfca : FAILURE in 5m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3678a9f9746406cad80e9e298ab05ce : SUCCESS in 22m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/453645b00e344804a9445f5cc34edac7 : SUCCESS in 15m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/076254e66f0f449591613752a05597d5 : SUCCESS in 23m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4082beae6a5e49699577fec4c63f2fa7 : SUCCESS in 17m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96da164e258e4146a85eb5f22e261e66 : SUCCESS in 24m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b471f209e433423aa7eb683cac2f3c3c : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e2101589c7d47f49fe4126dfed7e612 : SUCCESS in 23m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d8288778db0144b78630f38f93919557 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d8a05cd022934018b3abe5769608271a : SUCCESS in 24m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/026c6b467b7640b7aef6ac73cf8278a1 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f7c01babd3d413eab497cffc7e52cae : SUCCESS in 8m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/855b2d5b6cd44d888dc1466b8d6aa6a0 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/805b68e65641426d971bb5b9f3953d5b : SUCCESS in 8m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d3887e8341448abbab731d6ec717455 : SUCCESS in 1h 07m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/02b3ba6cc326471eb68cf5d2d79d1d74 : SUCCESS in 1h 02m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72a9e7614f5f4409a4a9681cf3f57c58 : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e791f522fc84f378e373a988019d9e1 : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fa1a2a2a5724fdbb32819002488a80c : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a94b538e4d4956a34d2e22424b29d6 : FAILURE in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3129a1963aa4463cb55634303622868d : SUCCESS in 25m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39275ed0a6d34c5aa11fe5ba3862064b : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9bf088fa59022d2fd2dd8f8ad841689c5e047090","revisions":{"fbeb43eff4a5561546e31f7d9eb95bada74f66e7":{"kind":"REWORK","_number":1,"created":"2024-02-07 19:08:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/908337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/908337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/908337/1"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-06 09:30:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-07 19:01:03.000000000","tz":0},"subject":"DNM: WIP: test jig for native mpu audit","message":"DNM: WIP: test jig for native mpu audit\n\nPlaying with some ideas for orphaned MPU cleanup.\n\nChange-Id: Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbeb43eff4a5561546e31f7d9eb95bada74f66e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbeb43eff4a5561546e31f7d9eb95bada74f66e7"}]},"branch":"refs/heads/master"},"1bd882e4b1827c67167b0589c7337d5e3ff4cdf7":{"kind":"REWORK","_number":2,"created":"2024-02-09 17:28:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/908337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/908337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/908337/2"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-06 09:30:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-09 17:28:05.000000000","tz":0},"subject":"DNM: WIP: test jig for native mpu audit","message":"DNM: WIP: test jig for native mpu audit\n\nPlaying with some ideas for orphaned MPU cleanup.\n\nChange-Id: Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bd882e4b1827c67167b0589c7337d5e3ff4cdf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bd882e4b1827c67167b0589c7337d5e3ff4cdf7"}]},"branch":"refs/heads/master"},"4eb1c94ede46c160fc88546ce4ee8fe412e6589d":{"kind":"REWORK","_number":3,"created":"2024-02-16 17:14:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/908337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/908337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/908337/3"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-06 09:30:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 17:13:49.000000000","tz":0},"subject":"DNM: WIP: test jig for native mpu audit","message":"DNM: WIP: test jig for native mpu audit\n\nPlaying with some ideas for orphaned MPU cleanup.\n\nChange-Id: Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eb1c94ede46c160fc88546ce4ee8fe412e6589d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eb1c94ede46c160fc88546ce4ee8fe412e6589d"}]},"branch":"refs/heads/master"},"9bf088fa59022d2fd2dd8f8ad841689c5e047090":{"kind":"REWORK","_number":4,"created":"2024-03-01 21:16:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/908337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/908337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/908337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/908337/4"}}},"commit":{"parents":[{"commit":"afe31b4c01881aa761f7ecc8942245301cadc476","subject":"Merge \"tests: Fix float expectations for py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afe31b4c01881aa761f7ecc8942245301cadc476"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-06 09:30:07.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-01 21:16:23.000000000","tz":0},"subject":"DNM: WIP: test jig for native mpu audit","message":"DNM: WIP: test jig for native mpu audit\n\nPlaying with some ideas for orphaned MPU cleanup.\n\nChange-Id: Ib5442f2caf3b7dbf676bf84c62d6ef0458b56278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bf088fa59022d2fd2dd8f8ad841689c5e047090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bf088fa59022d2fd2dd8f8ad841689c5e047090"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
