)]}'
{"id":"openstack%2Fswift~908953","triplet_id":"openstack%2Fswift~master~Ia0eff5d435e390a73fc2888a963a889f510d318b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-03-01 16:56:36.000000000","reason":"Change was marked work in progress"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-02-17 07:08:48.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia0eff5d435e390a73fc2888a963a889f510d318b","subject":"Get basic write support for mountpoint-s3","status":"ABANDONED","created":"2024-02-13 22:23:52.000000000","updated":"2024-03-02 01:03:09.000000000","total_comment_count":2,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"038ac8a8cf220ee545d8bae6c48fccd2cb3200d9","_number":908953,"virtual_id_number":908953,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-01 18:59:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-02-13 23:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-13 23:43:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 21:38:32.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"}],"messages":[{"id":"c3518e6b29fc9ef4fc4ff3c64ace3f174f9c6f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-13 22:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e85cc08a2e7fd4ffe016e536eba1e49f01e6419","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 23:10:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d219559971e4903ae00b1e6929045fa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f17bc934e4e462bb1aca4b0d85685bb : FAILURE in 17m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f21b81ccd33494b8a32bc320ac5cd98 : FAILURE in 15m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bb2b4b7ecbf4acab38c084511748478 : SUCCESS in 45m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/944af1d5794a479f8628c479b6fb0820 : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9311a43e47f933bd031fb2c8c26d8d527487a0a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 23:43:47.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cf83e22d56b4ea6aa5b9faf09d259e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d955e32f80b54375be40d196a32881ae : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/5306ced649524ae7ad348d042f71468a : SUCCESS in 1h 09m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/11455b47c77b4d44a92ed3bc11e34d27 : SUCCESS in 54m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/141f3e5df7944bd98bd16c8dc8a45b3a : SUCCESS in 36m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87eaf2691b4943cd85eebce14f155150 : SUCCESS in 26m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b03055b53a14e0cb2216b7c245e45e8 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1790f9b03db340e496ae6cb3673ffbb2 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6eab021362254ba0b5fb9ba7c9c3927f : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd49b24c3db44b08b15485419df085af : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3a6998d2c854e84bd8d9212e38e107d : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc2b2a16d98f4d0089fc803f1aff0175 : FAILURE in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7c3713ef9b4d5fb0baf1ff93ef8f99 : FAILURE in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b437cc1e87404924870b76338b3f5672 : FAILURE in 16m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bb900d0e31174469bc75779e293de695 : FAILURE in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d10f4795747d47c2a3fff71214afd002 : FAILURE in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5cafb609609246e080cf4039ef6695aa : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d60e015edeb643e583c1798ee065d874 : SUCCESS in 20m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fcfbdf9075564ef98df86fa806c67463 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5bf8aa1941c04d3085d720238faad359 : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/554f95de2f2f44519215ee2cf688b965 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8ca9037860bf41c898ba3c639a9ede8c : SUCCESS in 20m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f85a3646e3a4fb98d158fe7714b8c01 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a016f163a0c48c3815b765b772eb68b : FAILURE in 17m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a3a819e4a53452282578e602334633c : SUCCESS in 6m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/21c297cd8aa54682886515cd9110e919 : SUCCESS in 56m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/268dfd4e489b49a3b53d6576caab7d82 : SUCCESS in 1h 05m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa9fb6b06b864b1c9ab8bcae08a0c038 : SUCCESS in 51m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a49806583a5b472690423d8323842b6c : SUCCESS in 53m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c3c2ab66c16443f89745edc13690b0a : FAILURE in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8484da2be554d1aa4258baade7717db : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a87e36e5a3ce47c682f1ce6b8dd4e3a0 : SUCCESS in 22m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f703f1ffabe401287fe2ed268f71f75 : SUCCESS in 47m 26s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/s3api/test_multi_upload.py","accounts_in_message":[],"_revision_number":1},{"id":"127992bfcf62141ccef2185a09c01593b0b9203d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-14 05:23:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77fcbb5430ec424f2235b0438c6706de2ba66c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:51:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e63c56fd2b5d4091b5b777b40198f4c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cbdb94a62264c9789cb00e108cee283 : SUCCESS in 23m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e21d20462b344e4c9dce6dadf0da36e3 : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/088eccbcb95b4f3e892cffeaad0885bc : SUCCESS in 25m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca81d57e7706484395dd917651d1124a : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6656255463620ff149c20cc93dd8e8bd231b079b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:38:12.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/857ab40805af445186a619fb2541bdfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bda4e3f09bf4638974497d155ef8e03 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8ad2dc8c6d294237bab096a3e5a11e99 : SUCCESS in 1h 03m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aeb8168e62664547a802f7312ec6a7de : SUCCESS in 1h 06m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5239ac1fa0874e2d86260338b24c3c7a : SUCCESS in 47m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba0d1f7e43a4647a14aac2d9d79a8e2 : SUCCESS in 26m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c93977188e4c4bdd8dcca631488844f4 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df81a3bda62648b783697596f4330088 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0263824b0404901b588616c095306f5 : SUCCESS in 18m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed31dc95ce404c0cbdd49b0fe76da8d1 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c7f2f4958da4d45839666646b19086d : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/acc9b95ec7554454b3e90d2a83851787 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19ea2fb1839e45488c42222c9365ce56 : SUCCESS in 15m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba96b38c2d6a4968a6c4d0c87154bab1 : SUCCESS in 15m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/22b60e2d38894d9c892958775819ee5e : SUCCESS in 18m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/47764b24c738431bab15588f1e597448 : SUCCESS in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d2a4b88e9794d028d7f84c981bbcfd2 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae32fc9374c340918f04cb56b4c2a0da : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8869c302458f4d78b215ad24cbaa7045 : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e417f8687b804b4088a1bcf3ed4bf00c : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c9488e630a94bca8275fd4dfb0f76d1 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cbc4f956dcc64a6eb47dc12260a632a1 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c30ec27b8e534f1385fecb9a3c567096 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0094820625ad4a8fb943eaaa81212030 : FAILURE in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57ba66a4a7ac4f06900526836d769e21 : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/66aff34d23bc437f8f2e295580784a3e : SUCCESS in 1h 01m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/11b1c601577943139900d5727c58c135 : SUCCESS in 59m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86016c8e486e4f12850b5f981c78dc8f : SUCCESS in 59m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2353c0d2c0944f285cabac6fc17bc92 : SUCCESS in 38m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e393b82d164b27a7dd1fcb5265a57b : SUCCESS in 22m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f685cb30ef39416099f9ad3c46b78123 : FAILURE in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a379afdb28f6402d81cd149e1b801ea0 : SUCCESS in 19m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d6576bce5644f76be730b50d6d09910 : SUCCESS in 1h 08m 06s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/s3api/test_multi_upload.py","accounts_in_message":[],"_revision_number":2},{"id":"0ac410f6d4756a861f405c2b0b5b1c64414853b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-17 07:08:48.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"638bc8f6436f8be45974872c2c453b860bf3c6ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-01 16:56:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* doc/s3api/rnc/complete_multipart_upload.rnc\n* swift/common/middleware/s3api/s3request.py\n* swift/common/middleware/s3api/schema/complete_multipart_upload.rng\n* test/unit/common/middleware/s3api/test_s3request.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d766b4c55b42e5f51ef2f85693a2637957cd4d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 17:16:10.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd51188ff08f4fc483c41d37e467b3d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b7d098afce441ccbaf4203e18dc14c8 : FAILURE in 8m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78bbbcec21f9405aaaeb5db265a2c214 : FAILURE in 5m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9a7e3bf2e3d4aa8830042bbd69b3b7d : FAILURE in 6m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c264477986d43e4b266ee5e1c500b72 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"51cb9c93c1e4b94657fd6c9499bcd092bd459cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 18:59:37.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81c27beee1e1428cbd90d3d0b66cee07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4daf7fc7e0f840378f7143de37ad59f2 : FAILURE in 5m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/d2c9509323d2433889e5dacb842d458f : SUCCESS in 1h 09m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbbe4c91fc8a4be49d80610444614117 : SUCCESS in 57m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c01d21b431c4d238e8962de92e211c7 : SUCCESS in 46m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb666d3a450840b08a7db6e91dfffaa5 : SUCCESS in 1h 00m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e19dc4ae53894d44bbb9f2b8053c0316 : FAILURE in 9m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c551ad18334f4229aa8d1753eeff85f2 : FAILURE in 6m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3650e369506246ce9be9d463408244cb : FAILURE in 7m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e64e573ddc64742873f9cef58eb4f69 : SUCCESS in 11m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1583770a27854bf395799b90b62ad669 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59a31e6c83fb4f839f24b47c1c4414a8 : FAILURE in 5m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77c02dfefdaa4e57a2a5766df248f9bd : FAILURE in 4m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d29492872d194a9bbedde919f25683b5 : FAILURE in 5m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2a9f42182e4b4a53aec8f40c78d628f7 : FAILURE in 5m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e13a80180804c7eb0783cbf5a3b0d7b : FAILURE in 4m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8f3d81d05d84106ba8c62b2ae3459bb : FAILURE in 6m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d08d01ee64c431aa2764287a8d43ff5 : FAILURE in 4m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b8da51f77d34fbf9d698d145913b8a9 : FAILURE in 3m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/93fc561f96f04b6e89847f740fb38ae2 : FAILURE in 6m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ba20340f20ab487e8d280527075053fe : FAILURE in 4m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/87b91eeebfb8447ba23de745d67beee3 : FAILURE in 3m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd8a2542b35d49098b5793d61a0c8f0c : FAILURE in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c8c35403fac42eeae789f0f73526615 : RETRY_LIMIT in 5m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2552afc1be7c49a6aac5ff3d7c855a24 : RETRY_LIMIT in 5m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e378164cb0fe4322b38dd6609336c951 : FAILURE in 7m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fc0c79ba1754a39a7bd648a8e83033f : TIMED_OUT in 2h 01m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7299d3aad2184655959dcd6c42597b5e : RETRY_LIMIT in 13m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74074fc0239d498d9cb9fec3c69d6f24 : RETRY_LIMIT in 12m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9b1eb0090ab4c0fa7e228e25307ef89 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd51c81cc47747ae9fe36e837975592f : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35421c8cbc5441b2b3ffaa497ddc416f : SUCCESS in 19m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7cf3fd081834275b0ade7294d4b9952 : SUCCESS in 1h 01m 21s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3response.py\n  Comments left for invalid file test/s3api/test_object_checksums.py","accounts_in_message":[],"_revision_number":3},{"id":"038ac8a8cf220ee545d8bae6c48fccd2cb3200d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-02 01:03:09.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1d4ee842f7044f99f6ff332df56ebed139b0ae22","revisions":{"1d6f3b8df7bb4dae4a5b953cf520f9f74c2b05ae":{"kind":"REWORK","_number":1,"created":"2024-02-13 22:23:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/908953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/908953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/908953/1"}}},"commit":{"parents":[{"commit":"2236731d7bc66fde44a63684a6a977fb08a9609a","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2236731d7bc66fde44a63684a6a977fb08a9609a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-09 18:26:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-13 22:21:43.000000000","tz":-480},"subject":"Get basic write support for mountpoint-s3","message":"Get basic write support for mountpoint-s3\n\nStill no trailer or proper crc32 support, though.\n\nChange-Id: Ia0eff5d435e390a73fc2888a963a889f510d318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6f3b8df7bb4dae4a5b953cf520f9f74c2b05ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6f3b8df7bb4dae4a5b953cf520f9f74c2b05ae"}]},"branch":"refs/heads/master"},"e9c99b50829906ff74898824044c7f79ee70f341":{"kind":"REWORK","_number":2,"created":"2024-02-14 05:23:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/908953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/908953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/908953/2"}}},"commit":{"parents":[{"commit":"2236731d7bc66fde44a63684a6a977fb08a9609a","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2236731d7bc66fde44a63684a6a977fb08a9609a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-09 18:26:20.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-14 05:23:24.000000000","tz":-480},"subject":"Get basic write support for mountpoint-s3","message":"Get basic write support for mountpoint-s3\n\nStill no trailer or proper crc32 support, though.\n\nChange-Id: Ia0eff5d435e390a73fc2888a963a889f510d318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9c99b50829906ff74898824044c7f79ee70f341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9c99b50829906ff74898824044c7f79ee70f341"}]},"branch":"refs/heads/master"},"1d4ee842f7044f99f6ff332df56ebed139b0ae22":{"kind":"REWORK","_number":3,"created":"2024-03-01 16:56:36.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/53/908953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/908953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/908953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/908953/3"}}},"commit":{"parents":[{"commit":"3ebf7cf1d7c76bcfe0eaf05bfec74593c4c4adf7","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ebf7cf1d7c76bcfe0eaf05bfec74593c4c4adf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-09 18:26:20.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-03-01 16:56:36.000000000","tz":0},"subject":"Get basic write support for mountpoint-s3","message":"Get basic write support for mountpoint-s3\n\nStill no trailer or proper crc32 support, though.\n\nChange-Id: Ia0eff5d435e390a73fc2888a963a889f510d318b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d4ee842f7044f99f6ff332df56ebed139b0ae22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d4ee842f7044f99f6ff332df56ebed139b0ae22"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
