)]}'
{"id":"openstack%2Fswift~909206","triplet_id":"openstack%2Fswift~master~I7338c3a47ddf9336a506a4944e0271961ed772f7","project":"openstack/swift","branch":"master","topic":"p-native-mpu-compat-tests","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-16 15:26:41.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7338c3a47ddf9336a506a4944e0271961ed772f7","subject":"DNM: s3-compat test: overwriting unversioned object with version","status":"NEW","created":"2024-02-16 14:47:36.000000000","updated":"2024-02-23 00:06:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de9f46dc7de77c8d543a44ae3a94898e46cd278c","_number":909206,"virtual_id_number":909206,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-22 23:26:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 15:26:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88cb345d893b085bc23425bb4937e68615e926be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-16 14:47:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cc159c0d6d1ee8c63447daa321fdad0e433308f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 15:26:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3db124664aa4822b28ad64feedec9ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8913c283464243ac0e10169e93dee0 : SUCCESS in 9m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/513abfbd6e874ac0ae5e73dff0552b7a : SUCCESS in 12m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4ca4bca8ac37476fa82c9d49881a93ed : SUCCESS in 6m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12122a60adb94e9d8759f0b1a1c18516 : FAILURE in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22a3a73fb78460daf4bc02786878649 : FAILURE in 4m 03s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/slo.py","accounts_in_message":[],"_revision_number":1},{"id":"7983a02210fcfd40ee67337b6df3dbc5e1164522","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 16:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb558dc24d04fa6b99f7c44f91d2bab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9efe1342b1524f98a483d772341064a3 : SUCCESS in 21m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06793afaa5a04f8bb3f7a4e17dcc48f5 : SUCCESS in 22m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1de80656656b4922bec243266a05a65d : SUCCESS in 29m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6811677e37b84b5c88feb805af09a7ff : SUCCESS in 1h 21m 18s","accounts_in_message":[],"_revision_number":1},{"id":"72914d9800c1ef775a7add295a272f75a172bee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-19 15:19:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c02453b5acfc945eb11e7b068f382012a62948ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:35:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/85986b806bc3440d99be502e7093f1ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57cfadfceb8f4854bb166b876ef88d9c : SUCCESS in 9m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b531fbab497045b2ac7a4e755eaa8c58 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bc728515b2f040faa7e4db1d2ae48c73 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2801b428d534454d9374e1bae67dd278 : FAILURE in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba9a684cc7143ed82c4395b8a334dbe : FAILURE in 4m 37s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/slo.py","accounts_in_message":[],"_revision_number":2},{"id":"fe10c0b95329267d05cc94d8d95fdc395c80f306","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac7f492d55949369af82a9091acc811\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be4c204dc2b8455abdc59d99576296e7 : SUCCESS in 16m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57f5e626ec734d3eaf82390d2b231eaf : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/034f91458203462b95e0c8a23b4260f7 : SUCCESS in 1h 03m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/960cc2dfd1f24d6db5b0eb95b984a970 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6a9338d2a7df644400fd204d7258c5ba90d06f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-20 13:23:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f32dd16057b129e1b31dba60c1bc95b0f1053b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 13:40:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/287acc9b116c4a65bb498e3b592c25a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8096eff2c04726a30dc68c99ff7a32 : SUCCESS in 9m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f637b966da5640cabb4792e3eeb6f58d : SUCCESS in 11m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/727f47102b2346e783a5ee91e50b1517 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5135f2a0d8834762a5e4303f46d33cd7 : FAILURE in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9839c9dac5a42a2a672777d06827018 : FAILURE in 3m 50s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/slo.py","accounts_in_message":[],"_revision_number":3},{"id":"cf62fc4c0e9600d162b0f63211c4da6393113ec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 13:55:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4b8ca6ed5049e69a72737d7dc5db28\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6b9c59058ba411e863dfa738a6c4183 : SUCCESS in 19m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3fe1a1aaf4f465fa71064424af867dd : SUCCESS in 21m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c23e913d789449b89fea39bf9ff0c896 : SUCCESS in 24m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21baf4d1e20a4374815ebc28254360cc : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c8993fbfc797cb5e4f09b5ee919fdcd91a0aeb27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-02-22 23:04:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":4},{"id":"ecddac2d0f1bbb0b515b077f48e49cd3fa97eb70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 23:26:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f8264174d7846e38dc240566434a3ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd79d89df7e411cbacdae9c41897de7 : SUCCESS in 6m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e5f9feb99a14186ae6acf12efeb0fcc : SUCCESS in 11m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0be4490a043b44378a32a796c892decb : SUCCESS in 4m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3c836d5727de4159a9b70135404a7b80 : FAILURE in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b2c9f48500409f897eb88c8cd05b3a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"de9f46dc7de77c8d543a44ae3a94898e46cd278c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 00:06:36.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e901de39e68043e1aa8d264f61f1ed30\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5516ae56b4b84754b3ef5233b26c85ea : FAILURE in 16m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f629b1227234672b726693f5959de22 : FAILURE in 57m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab7d0e129a474470887b905613441563 : SUCCESS in 19m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9bd17d8ef4d4330ab9e542c80934463 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d1c6bc0ba33721fa552a08872c984799ae2f88b8","revisions":{"3e346c4a7c1ed0cd9fbf9ead0248d2f356abb18f":{"kind":"REWORK","_number":1,"created":"2024-02-16 14:47:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/909206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/909206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/909206/1"}}},"commit":{"parents":[{"commit":"7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd","subject":"WIP s3api: add more MPU cross-compat  tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 14:38:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 14:38:02.000000000","tz":0},"subject":"DNM: s3-compat test: overwriting unversioned object with version","message":"DNM: s3-compat test: overwriting unversioned object with version\n\nChange-Id: I7338c3a47ddf9336a506a4944e0271961ed772f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e346c4a7c1ed0cd9fbf9ead0248d2f356abb18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e346c4a7c1ed0cd9fbf9ead0248d2f356abb18f"}]},"branch":"refs/heads/master"},"6fddc79f964785bf6732af60550a9e49748c7a73":{"kind":"REWORK","_number":2,"created":"2024-02-19 15:19:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/909206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/909206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/909206/2"}}},"commit":{"parents":[{"commit":"7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd","subject":"WIP s3api: add more MPU cross-compat  tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 14:38:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-19 15:19:30.000000000","tz":0},"subject":"DNM: s3-compat test: overwriting unversioned object with version","message":"DNM: s3-compat test: overwriting unversioned object with version\n\nChange-Id: I7338c3a47ddf9336a506a4944e0271961ed772f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fddc79f964785bf6732af60550a9e49748c7a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fddc79f964785bf6732af60550a9e49748c7a73"}]},"branch":"refs/heads/master"},"dc575aba3eda5691014ccd26412f9f4cca047372":{"kind":"REWORK","_number":3,"created":"2024-02-20 13:23:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/909206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/909206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/909206/3"}}},"commit":{"parents":[{"commit":"7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd","subject":"WIP s3api: add more MPU cross-compat  tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb4eb8bf72029d3c1ba4a1ae1d9a948834e6cbd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 14:38:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-20 13:23:12.000000000","tz":0},"subject":"DNM: s3-compat test: overwriting unversioned object with version","message":"DNM: s3-compat test: overwriting unversioned object with version\n\nChange-Id: I7338c3a47ddf9336a506a4944e0271961ed772f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc575aba3eda5691014ccd26412f9f4cca047372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc575aba3eda5691014ccd26412f9f4cca047372"}]},"branch":"refs/heads/master"},"d1c6bc0ba33721fa552a08872c984799ae2f88b8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-22 23:04:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/06/909206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/909206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/909206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/909206/4"}}},"commit":{"parents":[{"commit":"65e8ae005410d9ac533346107ff0699e3b6bf120","subject":"WIP s3api: add more MPU cross-compat  tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e8ae005410d9ac533346107ff0699e3b6bf120"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-16 14:38:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-22 23:04:06.000000000","tz":0},"subject":"DNM: s3-compat test: overwriting unversioned object with version","message":"DNM: s3-compat test: overwriting unversioned object with version\n\nChange-Id: I7338c3a47ddf9336a506a4944e0271961ed772f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1c6bc0ba33721fa552a08872c984799ae2f88b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1c6bc0ba33721fa552a08872c984799ae2f88b8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
