)]}'
{"id":"openstack%2Fswift~909674","triplet_id":"openstack%2Fswift~master~I2f698c25ed65b236e851e5a307d48a12cef62b33","project":"openstack/swift","branch":"master","topic":"p-diskfile-race","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-05-20 00:07:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-05-20 04:19:35.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-20 04:19:35.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-05-20 04:19:35.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-05-20 04:19:35.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-20 04:19:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f698c25ed65b236e851e5a307d48a12cef62b33","subject":"object-server: return 503 not 404 if meta or data file unlinked","status":"MERGED","created":"2024-02-21 14:39:04.000000000","updated":"2025-05-20 04:20:28.000000000","submitted":"2025-05-20 04:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":113,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"909674-p-diskfile-race","meta_rev_id":"a4f9e1f2fd4b5ac6a26f74207a53c9966b88bd6e","_number":909674,"virtual_id_number":909674,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 04:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-05-20 00:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-05-20 00:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 15:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 15:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 16:00:56.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-02-22 17:32:33.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-02-26 05:50:48.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-03-04 19:36:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-03-05 19:28:36.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"65c8607a082cef4dc01904925116a21f941d7063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-21 14:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cc92d1f01afcf785388ae6f25fcce8f839beb9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 15:30:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639869bee5ee4b459f545ca6a5a5d2ee\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/296de475a64c4b77abce23e8eaa0fae2 : SUCCESS in 22m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3440dae3a9154889a9757f08bcd96a1b : SUCCESS in 21m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f83e209bcab49cb8cfe32c9ae34c9d3 : FAILURE in 49m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2648e66b4624318af2ba913c5c21e21 : SUCCESS in 23m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9fee16f0612b61fbc27e6644c9100c753314eec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 15:41:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5748de8bc1124c30a45aeec4c9c1ec90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6653c445f249a4bf6ca899a0b56479 : SUCCESS in 9m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e861db06a01540a288a8af76e2202747 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bbd5a72a82547838f4af2ac47d3543a : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c641a7cc682a4d7b817d9c1b80b452b6 : SUCCESS in 31m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08f7b2d959f54e98a2775032d31409b7 : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21090f57251849a0b2d65783ad18c788 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd64ef969d094945a146a51efde6bcb1 : SUCCESS in 21m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ce9771d8e43425b88d309557d0592c0 : SUCCESS in 26m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a11636218ab34093a62eafab1146387d : SUCCESS in 17m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1a3df9908b0a40e48769be87cbf5e8e2 : SUCCESS in 17m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9212c6fecc4a487882cfe07bc8fc4d69 : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0da054a83a3c47d693f8be51212e4988 : SUCCESS in 27m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29714d0d16a84b828c20b3aad8237bfc : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d1794ba9c86452e90d63ebfaaf4374d : SUCCESS in 30m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f6d65b167d64cba865029b36d858139 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/847cae29db06414caa630c0e997e13d5 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/91d923fc618349b8adc680a743c6661f : SUCCESS in 15m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/131377287dfa4f818f3ee034f7840efa : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5adbf188124a4537a620928f0dc2098f : SUCCESS in 45m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a920fe46b5a3424fa640e955d4a8b83f : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cc686048b641fdafc6cb7e22714228 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1cf14c1556964928ba5363bf14ab0994 : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9593b30a41c6d61cd8620a51db45d6f4a78b9e6e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-21 16:00:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e2bb425e98082f046672e400d842733214f31000","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-21 17:27:10.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d84288260ebd0b1dcfe39916e2a4f5e77cc3bda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-22 16:52:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c4b35d98d7a9ebc8800e113bf1958fd14a6d6aca","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-22 17:43:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5c5f2082ea989cabc4e8d4745482c2d84153371b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:10:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bc957fa2614d8bb3fa730f798ddccf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ce758d13334b2199dc0dc6ae3e5a0b : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/eb4f2f0de9b8432eb2f51cb7a2702a99 : SUCCESS in 1h 02m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e15620bf5aab4277ad2dd5db25eb2e1a : SUCCESS in 1h 06m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/550e77d7226548c2a53fc80f8333d15a : SUCCESS in 53m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df67599dfbf9405b96f34dd68485bb5f : SUCCESS in 55m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a92808a284f459f85850d3b864db44b : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8691245308e4cbab4d2d6719ec6d2f3 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a62cb187fb447718dadd544e47ca3d6 : SUCCESS in 14m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90defbc2c8f74c07acbbc4d169255b37 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e9289bed34d4f4bbaac27bb943ea654 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd0aadc38432419e84b1573e623cb5c4 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/adba7f6b948d425d8da157c6fe66ca57 : SUCCESS in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0559ae3cd24648bca8bbdf226d161ede : SUCCESS in 20m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/25d4336ea1a24c3caa3e67d31d591c84 : SUCCESS in 20m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/91434de05f1c4ee18ae164c63d6d5897 : SUCCESS in 17m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6245abba596449fb89d127bc958d8877 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c556222e814d40af82ed21fc5282df3c : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d755cba41be4f87911c599ef9b97f67 : SUCCESS in 22m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e4df47ac88b94f5d968a24d159f04dbd : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a135df4b0ddf4dac95eb01c4de66a9f3 : SUCCESS in 21m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/60446489f6fb4188901117cd9f3abd07 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ac6a450939f487a947c5294b7729861 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dca114dacef24e70ba7f0eadbc2abaaa : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d923395d869f4d8ca3fbb7e1d12dc9eb : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6b72f4625a364777bb82230cc8c5c4e5 : SUCCESS in 1h 01m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b27c126b3ad443eb2043739a7dfb6a8 : SUCCESS in 1h 07m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/722a5a7dfd034e698a460c97564cde7a : SUCCESS in 51m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c1df41460a844b59f81799ccfb12cbb : SUCCESS in 43m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/815d2eec3ba14e36b994365f649e023a : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a7355fe98b462a9f788d13673f8b77 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/751e2714c993415a8052af894d853d5c : SUCCESS in 31m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c78065b7e1c48ad8bed6abb885fb198 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":2},{"id":"381cb45e90e01c3e9bd6aaeda8caea620c40e65f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 18:25:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c3d4766563949419c98236b21e07b32\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f92fdf9e2369407985ef716e9efa1505 : SUCCESS in 1h 31m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc5875289c174a93bc65af85db04940c : SUCCESS in 49m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/da4ffeee5e43411ab07a4e9ff8d14ed6 : SUCCESS in 22m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4bc10e168814c519088715397e940e1 : TIMED_OUT in 1h 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8f0235b4255392e7460824b734094017a4fbf45e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-22 18:38:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b11156b6a73534a031d6ded50a62f822a1b74dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 11:00:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"62ff80829fc6103649376237e28491c11f007bd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 11:50:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3335ae79864c448696dd766d671974e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0676d6a8b9dc4229a02980adff81febd : SUCCESS in 20m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0af96656af4443639e0330b9faafd436 : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b55f7cb76a24e178c64579dd4d8feeb : SUCCESS in 48m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1ba6a3f14524a7dbf49f1bb6b5ee991 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":3},{"id":"2a6c2bc275b3683c1721b78617b102c096f60a56","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 12:13:15.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0e10c13706ebe9ff44fd584789e5c8c119238380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 12:13:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d84907e3d30e7e8fc7434abbd002662d096954db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 12:21:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"04f7d5255b5b348ce5bab7e5cd884abbe4d8f79d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 13:07:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"03fb03b66e5e0a58a8fe9367b743bc35c32cb526","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 13:11:08.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"27d35996ad1cbb45652d2debfb12f0073cc60100","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 13:35:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb8ad0c3fa041e09c51916fc0311b4f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1d3e0194ae24065a9cdaa0b0067fe0e : SUCCESS in 19m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bcc771f41c944188f88b2c99abd2d32 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/97e1549079de4e8b84eaca5a7505b649 : SUCCESS in 25m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/42e6b23c76704b0284e49c67e1274056 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":6},{"id":"b8387c0157b00a5607363aab6bfcaa21cce275c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 14:21:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d0a04cade64b3d9183710ae695728e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071255be066a4a5a926414cdb27b604f : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/41a0e586989a4cb3a2d5faed222970af : SUCCESS in 1h 05m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90f396911ce2449b9a0a913bd5ba26cc : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9ad67958e4e40f79fd45dc928227a7c : SUCCESS in 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1d27938148e48c18e3c39b7126f2025 : SUCCESS in 56m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7314bad7d0b14d47881dc8b9dd46427f : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a79aab35a24e483298ec76eb46ec4d01 : SUCCESS in 23m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71441a250c8d4c0abca74b4f64d0413a : SUCCESS in 21m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b64b76f78bf344d2ae863faefd927b30 : SUCCESS in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bdf1135faeab420a8c2b8e90ac15153b : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73943cdfb27d41b89a15b4aba331a90a : SUCCESS in 22m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c14c8eca20f41f6b75a85167c4963c2 : SUCCESS in 25m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11ca8f6455054a6ea2f1bdcd501ab11a : SUCCESS in 14m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c699c5b572f429caf626db8eae18d53 : SUCCESS in 17m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7611782328104bc7a317499564e8bfe0 : SUCCESS in 27m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7089604be0c42e1a2533d8008452555 : SUCCESS in 26m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/636561e5a44246ecaa54f464ad9bf9af : SUCCESS in 19m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50dac8a546a840758c6a6f5d8d545a41 : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eeac9bb491304c64b1eb843efb6b8388 : SUCCESS in 23m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9b971c532ad4c5fa1d0aeb976495382 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9157a5fb379c41a4bb6234490b7dafa3 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b2c7f4872e24fd09015c8106d0fea9d : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abf5b21de6e5478a9ce8b16e93c1630b : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27c5fb7b95c44c9e89ff6c8dbdcc909a : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2bead998c2f47afa8dff7aaf970e6cb : SUCCESS in 1h 03m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/caedcd58a67f449097be16b8bf233bf9 : SUCCESS in 1h 08m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e6c6c068aa6459194559f04aaf31b01 : SUCCESS in 57m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8995149d6a3e42c7943e2cfa7b072685 : SUCCESS in 58m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30cc4afeb4d74d8b8ee4d4f4dde250b8 : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a25900e379d484b872a70f33bc2a44e : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd33f9f05a8c4f668718554dd212f714 : SUCCESS in 19m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5948e0f130cc407f8ee630841680bda5 : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":6},{"id":"82478c9cd57e652443d69b273d03239f21239697","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-23 21:08:58.000000000","message":"Patch Set 6: Code-Review+1\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"04381368e95dbfc801c087c341259762038d4ba5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-26 05:50:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"962001988c0f5f448d4ae1e7443ac062b9bdf275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-26 15:18:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ae4471f6869fce47ea78951b72f1b3adf2dda7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:12:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195975066f704942b79f7d2b961dba7e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01bb1fbd24664ec4912de72876a02dbc : SUCCESS in 19m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d95b6cd1df948218570072a663adf42 : SUCCESS in 52m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f26ba6cc210d4363b1db37b91d8941f8 : SUCCESS in 19m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbc491eda3224c75b8e3bc8e76066ddf : SUCCESS in 39m 36s","accounts_in_message":[],"_revision_number":7},{"id":"8b59067b6f4ca5fdd1ebc0b9557fd43e3e29401c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:36:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f45e055fd8445e4bef27986b4b682e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e19c48f3aca454d981312f891de335b : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/76af72a0fc594471bdf6d194283e21af : SUCCESS in 1h 00m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/14290c131ae14f5888aa5528bc050947 : SUCCESS in 50m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed20adbb6e3b4addb7ea7ef8a240d693 : SUCCESS in 45m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e01137fbb9794a5992b6e15aeda50a6b : SUCCESS in 1h 02m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/471c0cd9db9d4acf96affe40779966ac : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42e582ccee294f8a826a6ad9048962c3 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/499bcb4d9a014045867e1cc9e58359c7 : SUCCESS in 32m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c2e5b01e76b4669af5f282a13027020 : SUCCESS in 10m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6febb8a053774ec5997e9f2c02106903 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec40eb9bfb864e008c3c732e438fec4f : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99632779f0e8472a91d24c9d05242563 : SUCCESS in 34m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/975cf2d64f1f4078a9d483547d670bf5 : SUCCESS in 18m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18b1169984ac47eda2194b1d9e2e1a20 : SUCCESS in 13m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f0c725bf8b304cea96122ed2bb7d2c48 : SUCCESS in 13m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86ae2a74e0b94ef59247fc4e7bcde000 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c580b2a651e34899afd3f727348e6057 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd932e88663f499786efb5d53d426123 : SUCCESS in 35m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/36131b75f2964d508c93ecc92ebbf49a : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ac790e02e8b42a5b3b49a7aac515a61 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77b7573c81dd497a85bea4e3e7a70cb6 : SUCCESS in 13m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d526edbe0f942d28dcc4520e6ad9651 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6f05cb8845042e8bb86295b4880d414 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8587adaaed984880a9b09f894d5b5b8f : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/704e7f88e48b451ca63d71ab56610e54 : SUCCESS in 1h 16m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7936ea042af04ac59fc6ea3a38d3b412 : SUCCESS in 59m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53178d097b944923a4aa8b102ddc03b8 : SUCCESS in 1h 03m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efa0723bb87b40caa8002d9fd1c19e7b : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c50e5559324e46089a2b4273707b6d6b : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02bf03dddc2f45098efd03459ded28fb : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd6d97d72ce14d65baac38742a64a414 : SUCCESS in 34m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23a1531fba604837a45b0263634d4dd3 : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":7},{"id":"c524c5dbe540b0490d63aed2d5d31dab6ceb1ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 15:07:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"4c6f8a179f2cc3ec5a910d7192dafd5a08b52263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:32:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a851b850f9c4687a4b22bfbbfdeda8b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1233e90ca8c4de78e90084b90473c16 : SUCCESS in 19m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46f4c00b18a34c11be554a9a58eb712b : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4bdbef08f21406085f1ed3291acaa1d : SUCCESS in 24m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/879e004f57b545a6933c0055903d8166 : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":8},{"id":"26aad063d838fe5cda60e225f28ff1a60b621326","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-27 16:21:02.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9f5124c39ebb75b2d3a3966d8efe0bbd732bf715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:27:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5cd984bfba47329392b706289a89cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f43d9ca0b6946cfa4fa7da7542f25d9 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/686f9e01b6d84d2eac978962f62060ff : SUCCESS in 1h 03m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/901f0372284a43119ef966904e34e5a5 : SUCCESS in 50m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea4fed11cad04e4fb7e99e9086f30609 : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a87964c3d86402181e5668a03204b5c : SUCCESS in 1h 08m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff37fe1e72a44274b60bfd852d0b304d : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eaf39ee357a46bb8a2585611f9e7bb9 : SUCCESS in 29m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e0019167b6495fbdb218b56acfda2b : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d24ab8afe0164ae1bdfd8535bf64093d : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/260f021130d1453f9f2d088689596cc6 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9dbe8178b0da4d33b6695e50794c8013 : SUCCESS in 27m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/376dc8ef296a49dab13d2cc809c7c0b5 : SUCCESS in 24m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/475a1421d5c14281816b66a3e306d254 : SUCCESS in 18m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/218f4035fa5f4094a25c3365d807921b : SUCCESS in 30m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01440ceb69b84cfc86fccb73272bd707 : SUCCESS in 17m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e92c0fa617f242a1bd01f7a050af07f0 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2564a8227df449c0a7bc4be65d7f5904 : SUCCESS in 29m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c674c0423a7c4232bf096fbdc6e98508 : SUCCESS in 18m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ddf389ae8e3473abd4b7cb74cb830d8 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f55d338d1bf24adaa4cebaa0e12b95fe : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5be52659b64446dbb87cd46aed141141 : SUCCESS in 26m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f57494121bd48f8a81ff6c5f3120ed6 : SUCCESS in 9m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ce1a057b353b4a98bbae6f3c5d2fafef : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a24022764f024868a09e921f2d3fe5ea : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4d879a9e2e594b20855f6ba0da7351d0 : SUCCESS in 59m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/731dfdc140b346b88e86dba978c96d72 : SUCCESS in 1h 10m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd43e6319b9b4f1d99e9d8f44ed0113b : SUCCESS in 54m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca14d24334bf44028c377ff9cb744a9f : SUCCESS in 52m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45db03fc66484d3ea0e60421f2341b8b : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5aa195e3ae248c0991fcb8c4622f0b9 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/952c38cc01084a8e9e675ea2b4cdd270 : SUCCESS in 28m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1318b6dde87f49f0abf120771cf9e9e1 : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":8},{"id":"6f5f9e98f2f8c21570a5bc0eb7a941d5c01a4386","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-28 00:02:13.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c1076bebd37aaf4a9c3bd32db2787829f4417d37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 19:36:23.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0407cd4c4fd725db62a80cd730fd69c4ad03eef8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-05 10:23:59.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3a2e2cff3f1f08ddb660e66c6778c6f0632b7aa4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 17:41:36.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c5cd8af3ec6dfe1b178cfffedb584dbeff181a5a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 18:12:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e0199ee01569e4cee0e7a97dda3fbf1ec3a211fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 11:02:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2311de931d684f39a2d8a2a04015e60454a56100","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 11:11:02.000000000","message":"Patch Set 9: Code-Review+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0d23f06feb9ad1516f4f536f5bd3d4566ae2bfd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 11:54:14.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae534fe3bff641128ee2006d790ec261\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70eae7954acc43a5b84ec354b3e9583b : FAILURE in 40m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4122b1f03b624f54a8bd1b61fd11363c : FAILURE in 49m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2787e532b76f43bbaddf9c5b46b5e2d9 : SUCCESS in 27m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bab38195eb3c44a184d8a6dfe35fce19 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":9},{"id":"e1b03fe6031649aa117e78b32ba64a93c91dd6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 13:16:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0108bc236154e068d465283de5c54e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd52d23b40bd457e995617d82c472474 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/cc5e7933e17548a09e0409057fbe441a : SUCCESS in 55m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96c57ac5c356481d8289b6e2a17a3953 : SUCCESS in 44m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29066efd80c44c03ac1e959fd5883716 : SUCCESS in 55m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08c727f869d342dfbc4053a23a18fd88 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1525d57ed8c14f9286cf7dc73f18378a : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af22e5e84b0c42d0a6a4358e40bfddbc : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d79cf63debdc40bca733637fc0e03b84 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/62dfb5dd27244c559fc797529373e4f9 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa898cf3c7c4485299bd289fde9b6dca : FAILURE in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dfe6010c3a174fefb2bd8a047bdae7ce : FAILURE in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df997ac8e5434a3db37b7185529f93ac : FAILURE in 17m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0d9ed72ec26c4e93a7502c3fec877c8d : FAILURE in 18m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f784a848567d4a8a9ab611f1102ab270 : FAILURE in 24m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cb26fa60deb4d17aa0307779c9748ff : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02f38ed767734eb3b141a398df918dda : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/103f1b57908a4ee0a0a763e6fc2e8341 : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/115657267f5540e3a5fcf9cebdc6e152 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ecf86b30bba44ac6ad49390b5d198b06 : SUCCESS in 20m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/decef622516a4e4d9e7f59d373ad718e : SUCCESS in 21m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc4e9e4af9a34f25852b52fbe364ebd1 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/736cf786b96e46e9a77edc38dd248846 : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d525ce23d3a4e1f94ee6ac181608f53 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/66f6a02ae6304f7dae0a86b4a0108ddc : TIMED_OUT in 2h 00m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd038ec7dcb84b88b9ca770dddd11429 : SUCCESS in 42m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46732a31d64b4e1194262e61f2556f97 : SUCCESS in 44m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/838f4d11d6364b10af41e7619111b1e0 : FAILURE in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0183a28fd04309a7783387a3fb1379 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6ea4faa680b4eaaa7b0223283190db1 : SUCCESS in 26m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2747e09e90f84f0f8767cfe05071e7a7 : SUCCESS in 50m 13s","accounts_in_message":[],"_revision_number":9},{"id":"7cf52f2069c924d2e0219b23b2be12baef95e3a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 13:29:58.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"ae359ec77fe87878f532b97b9c68755ad27c0ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:17:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1425a0cdb0c940bd859806c33d5240c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c992ba6a9674c85af1f038038b8fece : SUCCESS in 17m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/522a3879408f475bb90dab63c78f18ec : SUCCESS in 39m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a185abdffdd4c189fb535d4dd1a92ec : SUCCESS in 19m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aed040e803584a719edeef2b047a4c7a : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":10},{"id":"c6f1a5677ef1d11da55b1e46f4779f25b0264e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:40:11.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a65f4637c4b414ba85e2f8261cd7820\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62807fafc1c4400c9d74591b83feb8b0 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c6029fed378e46309ccea1ccf9463a18 : SUCCESS in 50m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed5317de173f469691d5878193981956 : SUCCESS in 46m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d9fdd7d66e94932a980555f66f7ec6e : SUCCESS in 52m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/527cf8708104416394d17e53e5f86650 : SUCCESS in 22m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49d1f5f699b64ee5a9b0e224fd72bc53 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb2e50b6fda84260ac0314e8f5030f0e : SUCCESS in 24m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4851223da03f40b0a3ac7cf59fb9699f : SUCCESS in 10m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7da642d105e2413c82c4dc7d114b0174 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5673d226409248c19258235ea687f9eb : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3c5afc949e247038ad25f68843dca3f : SUCCESS in 20m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10df56a11dc64bf7b3f9826ac7843b97 : SUCCESS in 21m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5abd4fc9b643451bb336748c51f92e62 : SUCCESS in 18m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/446b65864ddd443ba3c88bdd877c97aa : SUCCESS in 19m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a88cf0eca454dc49b8b36a114e6533e : SUCCESS in 19m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ace4208d380d46ff917efae0641cd253 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f738e872c26343c581c7d5e1db0f590a : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e456417011784db0b75da04117f912a3 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fba99abfcefb40399cbf10ab7c2299fe : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/42873a3692e046b4900d05cac5135011 : SUCCESS in 20m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8bcec91e419841c08c9f8d23de1727eb : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7fb1712a75264dcc8806fb939c212442 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1bea3df55ef54bcfa3294e62ce502f79 : SUCCESS in 10m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b647af519114b719d421d202cd737d1 : SUCCESS in 1h 01m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/903a2b7d319743d1bf2c122ba87c17e2 : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c18ef59a4cdf4a0da9f520ebfcf6f8e6 : SUCCESS in 54m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e697b8b175472d897e0665570a5dd5 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e46958e1b564745b2bdd684d7301399 : FAILURE in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79b83fb6e54f4548be4cb7f3263313f0 : SUCCESS in 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bd8957d310b4ad4b43f24bab41ebd12 : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":10},{"id":"953cba75bab25e808b462c61f7c951fabfdc976d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 16:03:12.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7868ebb7ad650329c3ce4cdec8e22f74a0bbf32f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 16:14:33.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6ae5641eda6288886f2efc70d466ead2860e4c2e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 16:19:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ace66d907ac6dd68f3e3485399bbb962c9b64382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 17:03:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7014ade1364a4793847786112753863b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f5006cfe73b4ee1a0f933080064ec74 : SUCCESS in 45m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7b70ca0d31549658a81ee546474f3bb : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1645852b3686498da44aba746dc82073 : SUCCESS in 58m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/586b0f60127c4b70b4a4e2d50d1697c9 : SUCCESS in 21m 47s","accounts_in_message":[],"_revision_number":11},{"id":"e4dd46c66bb7b4c29430eacb65aa208faa482943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 17:18:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c60a6818dc604af39d52c027912262d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba3ccd60b5949e9a270e7b5cc11657d : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4735beb1f745462aa7dcd3ed26866f24 : SUCCESS in 54m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f2d79101bdc46d5a329a10741368dc7 : SUCCESS in 55m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dcb42b5cbb34ce48ee988020c801b9e : SUCCESS in 1h 04m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/756af40741ec4693b8470af2715c165e : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92380ad0c2f946a2962d492cc6cf01e8 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d81d3409efd04613b861c8d3c6a71ebc : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4b967981d82436a8d9113fa75295316 : SUCCESS in 12m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df0f313bc35d4e87acdcb9b352e0c7b8 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3469b408b614ad9aa1627d14af27170 : SUCCESS in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1dce1fb223b948acbd184fa48de1e7de : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2959a482da2b444c918a3512e108680d : SUCCESS in 18m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1c6a4e40cf5e45ddbbaa5a885e3c9065 : SUCCESS in 17m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7142c412f39b4b2eae908742039d694c : SUCCESS in 21m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea485fd89557471687c7e2c7489586d2 : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/772cbcc0d48b40de8914e41cc8507775 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5d14e687898406dbeab94483600b581 : SUCCESS in 18m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5329c0e593714f2bb5a9254e79ca67b2 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0dff4cbcbff34fa18f1976b87a6793ba : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6a3f21fee6a4475a83f832eae1c200f2 : SUCCESS in 20m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84a44bdd472a4a64a1589d75435acb17 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74ac116732e146dcb9024766f5f5bedd : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f62fa2de3634419a1ee94bf2672a1e6 : SUCCESS in 9m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d00cc5a0328478db1d2256eb0c6dc36 : SUCCESS in 1h 02m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8339e11f3044e4a8a0d5a6db67c4bb9 : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56f80426287c4c8d8d9f664681def936 : SUCCESS in 55m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0c43abf27eb428487f3dfa2a44027a5 : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0307bf8f73b34ad9be30685502bdcd99 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34e0e0e60773461b97e4241454f3d769 : SUCCESS in 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928fe6068a3f4b49b5297b6b0e25a06e : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":11},{"id":"e2fb12cf9610215afee3ee721566fa2a2e8384ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 17:26:32.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"24a177335dcd0a277838c8bd570a02bec48a3f67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:11:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19956bddacb473ca2d4682a1c1bf82a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/716c7eeb866b4f718cb63517fbff86cd : SUCCESS in 16m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef06adc3e5ec428b99da64eace36c0d2 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc0ab1948d9c4effb8c5441e21d18408 : SUCCESS in 17m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8c42efd10584f05b15580adb1f58ae8 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":12},{"id":"6109f760e96646bb49231b38165b447ee5c7ce9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 18:22:05.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f2e44be5c5542c6561d44245507149b52d108246","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 18:37:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0d68da51843db57ab490d8eeba765c0806ea1eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:39:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b09d9c19e7c49c7b6afb788d5edd6e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d975c4185d4947afca9c93a10ba022 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a37d25015bb84ea2ab82129155bb89c4 : SUCCESS in 50m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ded4940835d43e29931939f92760044 : SUCCESS in 38m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bfd88adc1254b5da4ab3204eda77d30 : SUCCESS in 47m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94b6ac2a598442c2bad44be5b726992f : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb0e9f89ac094e00b3b331e3a50c8614 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd0fb246071d4f449c7665863ff23b1f : SUCCESS in 21m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fb1b7e548814c29a0926ebb7c7a24d3 : SUCCESS in 12m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df0612e2d76e4392bfbb109828f9438b : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/247617b287c54128bbc3550998aa08b6 : SUCCESS in 22m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c80c2c19eacf4ae78efaf9c15c6a3c0d : SUCCESS in 17m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1225ea7d6d44a2981f40bda9e3ba305 : SUCCESS in 17m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cb30fc706d954cf3980b355aeeb7f6d6 : SUCCESS in 21m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3a33a65e46b498baf6bf613cba9afb6 : SUCCESS in 20m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/584d378ded1f4a758cb9f083f653551e : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8a48793ca784c8c88d1cf9af7332449 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8e5a42687fc48b3a06d2ef3555ee832 : SUCCESS in 19m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56d5c0d6ba7b462f9321bd0e0ee94ce5 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b641f48824d0490792d6d83608538343 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ce116bf434c34ff1a7f279ab8dc94508 : SUCCESS in 21m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54fc0d3098e3425ba426d1de215ee936 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4daf57dd81a14356a53fe5955b040802 : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/113190ea75c442bead4b234b8ca9f404 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/955e91dbc5784747aef380f1de167f39 : SUCCESS in 56m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd65f997e94748848ce3bae6e0bc862a : SUCCESS in 51m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9e3b9a7256d4030ac8020958f55b303 : SUCCESS in 37m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13ec67dfe7e840b283c460cbba47c9a0 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8791da5b24fe4d2283e8b58892e2c894 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/130efbf16b5a4a93a1c0eeda30285150 : SUCCESS in 25m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f285b0274f76448a8675d9575a0a9485 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":12},{"id":"1e554aafa477faa26df4cdea0180e786be1060f2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 09:13:38.000000000","message":"Patch Set 12:\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"88429ebae88ee244b55c4c144975401697d6125e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 09:17:49.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3b2309272657a072f34ab527e24e91b7fb3f386f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-10 17:57:29.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"879048b27a11c7412bc7fababe0c832483f98d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-10 18:02:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0b65b5a981f39e34ad5ea113c4da7758a4a0379f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 18:11:05.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a593ae95d4b04399bba97c0ccfda3657\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2a35b4e12d840af81646613218d8a45 : FAILURE in 3m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e43342f590204487874dc60c1297f50a : FAILURE in 3m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc45a83238c643bc836f6ca923350650 : FAILURE in 2m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc5adead132e4f7a856e2691355664bd : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":14},{"id":"c29c07da7e1bb4086cc19563c01c94cfa9fd51f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-10 18:13:47.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e8a2cc7f15e2289d0825c9c189ba6a9a9b2c5435","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 18:18:29.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5198e30f32254e3dbde6f98141704266\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c0a8bb0dd6a421ea52912197f961409 : FAILURE in 3m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e896bd8f4c24f0f8765f7ce8c56b729 : FAILURE in 3m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b4c15bca670476aa0ec89210962928e : FAILURE in 3m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2afffcafc55441ca9303a8cfc942bb6 : FAILURE in 3m 11s","accounts_in_message":[],"_revision_number":15},{"id":"9c81d17dbb3db122c261f3fd6b6d01a77c0c2eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 19:22:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ad9e0d4b594387bb674f66e870d2f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9069237d3641c18bd7fe7a1b7900d7 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/08ab963c427947388c8eb848e7ea083a : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/308b448b6b964929be7aa932f0724b50 : SUCCESS in 48m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdaf14ab6a9140f7a1ea9330fc5b889d : SUCCESS in 56m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4a6d42a2ed44df691534c922927e07c : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/964f75a6b598450392dbc9ba35620f9a : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8023b3ae392a49f5a294e4c00c4e07ce : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8863b366db0e4952a68bc6d0dc712380 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/de6bb5d2dcfb4a4fb07fe2e5f975b646 : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/68110911c29245918a964d35840df225 : SUCCESS in 18m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fdaf9f116ed14bc28976ebfdd90fa749 : SUCCESS in 16m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e7313da7b76a4133a02340965968fc2b : SUCCESS in 14m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/708b80a2cfbc4490a4f356e8cddac29f : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e2d22ddbd1541d9ace014322899307a : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f7850faa9cf4d9eb91274b354b0012f : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0eae15533ee54a7cb716aa1e7825b865 : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/90b026ad2ce44b7082c4fafa364cf964 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d307e00acd4d43b5bc3480633b9af820 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3cd15406a9142898b76044241aaf588 : SUCCESS in 21m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e535549d05d441fa878db53ea7f982d : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b706ed36d0243ab98579ab013c1aed8 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/238fc529d90d4be4b5c6152063a4b2fc : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e38c1f5f630149e88104d82c8859498d : SUCCESS in 53m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04333f58dbec43589d3fb990a8f136f6 : SUCCESS in 49m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1080f6ac2d9409683b1a50b1c88835e : SUCCESS in 38m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97cf8679cfec4723a5d54ef6ad4c82d2 : SUCCESS in 26m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd8f1551bec46259fc139634a816da3 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/beeca8b24b0e499eb66c515fa71d8e72 : SUCCESS in 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c51079a062244f53bc8312d68ceb45cd : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":15},{"id":"49d09d7e8609bf13a04a123120a215ffed4f9561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 15:35:17.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"9748b562ab4afec8e444c528bb0d7fc5a7eb4e23","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 15:35:29.000000000","message":"Patch Set 15:\n\n(7 comments)","accounts_in_message":[],"_revision_number":15},{"id":"0c01f720b0002254a4d949fedb9beefb5a533ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 16:43:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd02e6742a642b3a316b5442e4b471c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5756777dffa4c629f7a3a98ca79d297 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/a0fc94c96b3e4833a93372fbfb2be373 : SUCCESS in 1h 07m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8955b23f9314884b20fdbc7fdeb6297 : SUCCESS in 46m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61cf43f8b17443ae94e33fdffa9e6eb1 : SUCCESS in 1h 00m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2307711b33c4425cbae1792c27f6b26c : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a6ef5dbb6b94828ac1b9c99b1ec02c4 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/588bfb4e9f45401793d6971d13aa5e63 : SUCCESS in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9df628750b934c3685341feb5991765b : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd61d833505f43dfbfa1e73d8dd169d4 : SUCCESS in 21m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e28af3ade4954bd69de8200d5f17d603 : SUCCESS in 20m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3c4adaf05f804fbea721631f908a52a3 : SUCCESS in 21m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7b558862961e46d8b4704221d097cf19 : SUCCESS in 21m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6368252529e94928a331c2716d21538b : SUCCESS in 20m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/84d87447ddc3497b926eef141754b494 : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d10de945eb4048cf933db687ec1894f8 : SUCCESS in 22m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4a9059efa78454c8f5f2a9b362cf0ee : SUCCESS in 24m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/898deed274fc40b6a36f176a5f8505ce : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/85e0a58081ec402cade8d9719b97599e : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b3d8fdcb81f4fe9ab3ea7cee5b51fd6 : SUCCESS in 22m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd8a02e62cca4a839ceaa3d26f153f3d : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d8bb4c0ade54a67837047b31df0796b : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/76d03f3c22fa4d08901ee2bf4fab805f : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f495639708e849618c661a6295c94519 : SUCCESS in 50m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/187e20a9c2e847d99f1b2d19d3e0d776 : SUCCESS in 48m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e390af5648243eeae245d36e7290a48 : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87e7f3d3cafb4a07883a2eb6a049f877 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1039ca88bc2543d6be4f40e5e1c47409 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/103d9e4c507d4308820b973451eea049 : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/280a2370a01849deb5572f94a79d96cf : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":16},{"id":"baa8e3f10a426b7bf24bf0da3faf0a3eb12c8c71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 17:08:01.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6abc644f77a94456be5b850445a9d809\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d08e7a164bb54e48be44a505635703ff : SUCCESS in 1h 24m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c825de64eb6148f29c9c94583887fc5e : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/88b7212abb8e4459adf8ba652e68dbe8 : SUCCESS in 1h 30m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/79cc4053c642421392429edc80c7ea18 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":16},{"id":"515335d4e6dfb2440f2895ef85b5e9eff4893a2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-12-23 22:59:24.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"df6a5d25710ece3c1ab1cf7e2937c4c300f16224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 00:13:33.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4eb1e4443934039bfe9c538594d1864\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca91825bfe542dba09d554f7eb2908b : SUCCESS in 5m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/aacba509a96a4c6b80c902145706a88f : SUCCESS in 34m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfe1cab4dfa244138781cdd956214aee : SUCCESS in 33m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3c9b686c7024212b1ba305bf72be861 : SUCCESS in 37m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e21948010244d2899dfe5db5bdc78e5 : SUCCESS in 47m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f199c4e379a484d946747ac1c363cfd : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc0bdcd27ad34192b1b7de754b06069a : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b1ce04dda18462bb9569fc5d1446a04 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c787bf7033e141f6bcad58a42c0245cb : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/585d1be9b65f4565b2f85e7f9bdd69a1 : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa63dc2ea363499f9ea835f9ee1cde7e : SUCCESS in 15m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f04eb0e7857d43f8a2cb5228b4a67c94 : SUCCESS in 16m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/046f4c96587d41d7a6fe48390bef7279 : SUCCESS in 18m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4ca96fc2141401dad034ff61ac93290 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/159a5fb61db44576ae1fde4731351467 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8673399459e4681a548fa1338bd2a08 : SUCCESS in 18m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eabf8bdc432e4fd2afe61a1295d08999 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/649dbca2f1494c9c8c3998dd2d8928ce : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f8c827dae88c48cb84e668ab1d6f4e72 : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eaa14900abeb42379b40ef79e2cdb228 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d8b0527a4504ca39d29963781423d8f : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/81c3b568a87d4ecd95604f0d1edf72d3 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e29a124b1e0143708aaac5fa4256436e : SUCCESS in 38m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/176d64231e654ee1a4b58f0d2309457a : SUCCESS in 47m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1865bfcc84be45f485c380d73c3055b0 : SUCCESS in 27m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47bbdc4d223d4b9480a4302a542957e5 : SUCCESS in 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1dc0b71edc428eb0ba0b8be2def812 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ca7b6472088444b89462d574d09e246 : SUCCESS in 11m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c5ab23ecc8f4d73a31424307e2688c7 : SUCCESS in 52m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f600a116e21415d8ad5168565b26a52 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":17},{"id":"2c139089bc5219bc8f5f92a47f360b4a1a5bb766","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 00:31:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64fe89e2608c4a5fab8b6b6fdf0734d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eac0f971d6094df1910b4b4e2645aa67 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":17},{"id":"247f59877067d58ba8563d0be80c6a9e5f542c94","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 11:44:39.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0a1f4e8b888c1e007c097ee52aadd22590f4f6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 12:50:12.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c041eda38149ef978ef73264e1a7f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f93bbb164d14a57ac771e9c262cc176 : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/155fefd9becd419c808646c006bdc8ff : SUCCESS in 59m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9970c88fa4434dabb5f6d82f4c52451c : SUCCESS in 48m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9dd85cc5457d4519845c77e341e06c0e : SUCCESS in 39m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2935c0b963914a14844b3c5b0ca02705 : SUCCESS in 53m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12aee890a1664e43b9231999e6c3aee0 : SUCCESS in 15m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caae5d7f994e4cacbe43de10b931c71c : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cade0e3e45694cbbb72a2aea4c1d4d5d : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4892bc4e0d1d414ea4e20784c1b5765c : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49322e0366974819882cf50893a44762 : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/baacf62ae4d44c4bb7c570c3d797b810 : SUCCESS in 22m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/048f154bda624e4cbb6b22ee14850d99 : SUCCESS in 32m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb60f2b2e4014bc782cde607a4cd357e : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d095929a74764b42a94b878775a8942e : SUCCESS in 13m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/33e485263b4f40f4a20dc72a5a0c9c72 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/403a0764113d4db5a977a2b40cc21f76 : SUCCESS in 11m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94653b64b1984ee7ad674f77bbe7816e : FAILURE in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9895af04beac4d87933caad5cb1826a6 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f8fade134b0444d95ec0e984a83b7b6 : FAILURE in 57m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76fc314c86c643e6bf0b93e08ec8e94c : SUCCESS in 49m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbb488b25efd4ab1b3c10ebd7d7ef799 : SUCCESS in 49m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f17894611ec49eca3b8c11735087b6b : SUCCESS in 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cba8ed053eb474285e0a4aab222bf77 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30ee9ab73ca64636bfbcaae0cdc7cd99 : SUCCESS in 19m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8c95a8bac2648f5bca86fd489d7e6ad : SUCCESS in 46m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21849570827e4e4ea1722aa1376bd73b : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":17},{"id":"d1a8b4c015e58371ac0161e5680c9ac8db42af8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:58:41.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1145d733034abcbe4e2e8f5f0cacb7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69b7cb6b169f404a92c3a3ae3ca0fe02 : SUCCESS in 2h 11m 36s","accounts_in_message":[],"_revision_number":17},{"id":"fc4f3b99f36a9617081f1c80ed77eced03680f8a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 15:02:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"810dfcd6e08dfae83bf3f04e62cb0aa18a5dbbda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 16:18:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0090a03495c446eba8f6ba92ae33ada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62cdcc1e0a2493abe24e12005342040 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/18d61f6c7a0c47288b746f1592fea093 : SUCCESS in 1h 09m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75db7d3596f7420abf2f199a1970a1a5 : SUCCESS in 1h 02m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2cbd58287f8d4034a17c7b731d09913b : SUCCESS in 36m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d575c8c1976749eebb94bd8b58b23f9a : SUCCESS in 56m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b90bba92aae40d5a274803fee943b65 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a8c9fbc0647475aa1517ca02cb516bb : SUCCESS in 20m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4854d8004504c27a76abbe48f3642e1 : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b61db906a6f947a796c787a16918ed94 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87ad3b2fc3e949fdacbd75bd044f2bf7 : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/182b7de6f1c24ff4b06849af212a02e5 : SUCCESS in 22m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c899fb18b7354055a0121996c21f31b0 : SUCCESS in 31m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2c7f095efca040a3a7909a2708b25296 : SUCCESS in 13m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/225570b8087d4bdc8b0103587d288295 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2c70fc67da0449248789132007c1dabc : SUCCESS in 14m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f3c7639a84c460da6e7385288d560ea : SUCCESS in 4m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ee89f3718324ef3bdb8b443206bcf8b : FAILURE in 13m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc4297b7ad8e4766899b1128510579c2 : SUCCESS in 10m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd90981cb0944b009a37874c5520b504 : SUCCESS in 58m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/829f01e82991417f8d044e056f6f4c3e : SUCCESS in 27m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45e7c9ff5a2a4e68b8c4dc4801b2309f : SUCCESS in 58m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc94e46225fe45b49b6081c0ef05f461 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5367dd913e249e285c1ef4c71abedc7 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19b7e66fc6c2467b9d8f2c661aa04921 : SUCCESS in 17m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08c5cb3022e949488c7d7d25e3363a34 : SUCCESS in 45m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/560f9242cc5b4a53828b1bcc2e308d55 : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":17},{"id":"0db3f69ef031f5d14b5e123da4a20c81b4f91b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 19:54:32.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73649f9dccb146c0b2b13e988cba577c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff66103657ff4694a8d64d54f0548089 : SUCCESS in 1h 47m 50s","accounts_in_message":[],"_revision_number":17},{"id":"5d157fbdfb78358e6f1fc4b42a9f64b7f9c981c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-11 00:09:15.000000000","message":"Patch Set 18: Patch Set 17 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":18},{"id":"ad15340b0bcdbd31cd9735ec7d02366623475148","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:52:21.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7795dcf305f54f6b811dc08ce7979948\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebf37c4becb2440ab368729a38aea9ff : SUCCESS in 42m 33s","accounts_in_message":[],"_revision_number":18},{"id":"70443218d9584bcaccaf727f9c3eee46b0c016f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 01:13:31.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6760fe6d5b5460ea094adb24f536288\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51fc22d87c54901817e670e207482b6 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/eaa0f9095a19442b8c578e72f62b8500 : SUCCESS in 39m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66c6b24c96c34c77a21bd47264cde73b : SUCCESS in 43m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b6c36cf686c481faa34b5c24cc212d3 : SUCCESS in 36m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85a399c1dcda4f4fbcc59e569571b6e5 : SUCCESS in 1h 01m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/216a4a62b6b4497d802ed7af18b9cea7 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0d7ddd09c0b472a8a06e9c2a19d0c37 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcb83c4c3f6842b0878fddb67f3500e3 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fa39a2ea251408385d41c36c8eda118 : SUCCESS in 3m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db9fa0606eae43d2a26277c9796ab423 : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64002e78ef4c46a8a995ace5aacf9288 : SUCCESS in 13m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f67d32983e0457f8a09f1bf4acf544c : SUCCESS in 26m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a7e4c81ac11466783e6a03b6da52450 : SUCCESS in 8m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40e517e14cd44a609a74f52d99cf9cf9 : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eb52228ac0ca49e498b64c17432f3842 : SUCCESS in 14m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b89374bb853f4f4292f1779977c10602 : SUCCESS in 8m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c578051051a24eb7bce7cd1521aff461 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4880d1a94774ecc8ce476b2cf98071d : SUCCESS in 5m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c406a703e8a4f5cab03cddd84e71de7 : SUCCESS in 59m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8aa0e06fd60490caf7879c99b4b9cc7 : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c7f3b6989e34406ae71acf2e409c0ff : SUCCESS in 53m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25fc11f6bb1242b49eddc614dec2282d : SUCCESS in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00d109778bd47bcb16c073836839e8b : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10bbf0ef87204933997238dccf4a335c : SUCCESS in 20m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d9b95616f749048dcb8858d76f5244 : SUCCESS in 49m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27ee711f85814234a0dd61a8178871d7 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":18},{"id":"9132fd66ed8c18d5d1d7f8b8ce3ba4961c4f76cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-19 23:26:06.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_34930\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"_revision_number":19},{"id":"e5325f515b69f2e07edeca960232ec2f5fb1afff","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-20 00:07:16.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"5ecf2d1c502f8fb7d73e68a2f46ba18760475ef1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:20:58.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f220c644e524140bd5298fa19892818\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/46c0865ba2264819be2fea59db8c49c5 : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":19},{"id":"fc3755a39fa8d881902de0156203c4d3b65fdb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:24:12.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5724ddc4864a18bb68b13e7a720976\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ebd65c15b3a4311b916d611dfc68b76 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a924e2d798704a0cb91ce4a1f356e900 : SUCCESS in 27m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d6ac6a045d04a30bab0cbd54d8c8487 : SUCCESS in 31m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63c80c73de5a4ff69895d4031869df0a : SUCCESS in 37m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f655ca36a51f4f839d60d0c77623f2a4 : SUCCESS in 39m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74f9bbdb2eac4d64b10cfc8c7e39c5e9 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d4748d4ebe34dae8dc8ae0f84074165 : SUCCESS in 10m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a981e460911419382acbdf8b37583c0 : SUCCESS in 10m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e1989d162ac4418af8d580254451e53 : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a7f900280b44746948a229815afd9d9 : SUCCESS in 20m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c807e1499d914e3dbe418cb188372205 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7935bb3e229a471f8ddce89e82a0808d : SUCCESS in 27m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b406c5d13eec43359cd0fd29e635ea71 : SUCCESS in 24m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fcc0eaab07e645679e8bd68f5ca750dd : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/063135766b234ab48c23bf87dc0d53e7 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/683998dc273d4036b0015f778ddc5157 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/771b8acaf3ed4153888cf7d52e3f94ee : SUCCESS in 7m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c58173b5e45948ccad3c357c63f2bc72 : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ee20e48a8974626b646eee490488e3b : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5389662cd62c49f68758d4a50014e864 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc701e7554434335ba2970b5b1fe7710 : SUCCESS in 47m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f26032da0d7f4cb0b38ee608a3fb5ea6 : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c92ba45c42e4d42b80740018f5cc4f6 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c8498b10804851b2b0e69b234ba69f : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5b8c971584643a3a9221404fa5a232c : SUCCESS in 23m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/124b9bd5c85f4209a8515faf7bc76d84 : SUCCESS in 44m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f849f5988222484fa15bec5068b39056 : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":19},{"id":"3baaa2c336ff5eda932dcca9f0fbefb67d5c49bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:24:32.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"e6d21636ae17d89442f19159bf3445d4efcc008e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 01:29:31.000000000","message":"Patch Set 19: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90d5ef9d5918413b8f95346b919fd6f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafff0da94914a4c9ad5b13ef8c6aa0d : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/4f19a69a779e414e81f4662666d8c075 : SUCCESS in 33m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe4ab54099074d75bf4f280ffde13a3c : FAILURE in 23m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/179a52ec4a0f49a2bfd1288b82c0da08 : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d183a8753374008a6c6834c7a4eda9f : SUCCESS in 1h 04m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1c242dbf1a6d490e923eb9f018bad365 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/95f9a7261caf4770a1a19114c49a48f6 : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fc301f7f97c4fd180d5d55be2180d27 : SUCCESS in 21m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40d6afb90d4f428bb38d82b4a988ecac : SUCCESS in 27m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f15e3e9042a34f3c8b1c1a18c9b9c247 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/308e6d9966364c63a866c86e853757e9 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/223b0d2c804743cf9ae343c38ad644f0 : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5985b885f83f4fcabd8e678b2ae85e47 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fb6dbc50cf94dd1bd87cc7f732a89d2 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ce02dbad2e44de396e2a24f6b74aabc : SUCCESS in 52m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/720251c48e134e28977baba57cd3dede : SUCCESS in 40m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8d704ff96da4084aed6f90a19c810d5 : SUCCESS in 28m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28cd29225aae4e8a970dd57a2e314e2a : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4849ed8f2af54a0eb9c35fd3fa71d9a6 : SUCCESS in 3m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d79277f026814adea8e3806b34ce66a9 : SUCCESS in 13m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/992a2040a0374efa8d7d5a639de52730 : SUCCESS in 51m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb30c32df16744aea9d78c5bcd3f1849 : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":19},{"id":"119ea233078675072d3b2a4b57be11cad5dc30f0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-20 01:50:59.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1504eaad15faa5a647d277420fba0aeda2f535b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 03:08:32.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f833437d862b4adfab989c868302c769\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64b45a3695134925ad0aec4a66ae97dd : SUCCESS in 1h 14m 36s","accounts_in_message":[],"_revision_number":19},{"id":"71a6dc63fe0da0d6304c5370128923b55f005ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 03:12:07.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73806721b03444e4aa5c58a489936c2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c7fb47bcdb43c4a46908e8d9f89f38 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/04d6e5f0851f4cb99416eb15ec8f0411 : SUCCESS in 53m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f82a6b8c184b43bbaf574e0ea726fc07 : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a7e9318d738494384b3b2c08548e277 : SUCCESS in 42m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46b4abee0b12423f8ece9030d40194f5 : SUCCESS in 1h 07m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88b5afb7e154471c8ec1585bce1a8563 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ff714a3be674d999121f5f88d838a77 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e5719105b514b4a87253bcac3aa52f4 : SUCCESS in 16m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de5fe5ee36ca418f8804f4a23dc0ea4f : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9be2f92ebbfb4cb0b079f93f5c652d99 : SUCCESS in 23m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9011b998718244ee8a7ebd594d0b7c84 : SUCCESS in 16m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ceeb23fbc9847c38cde235f4f11c25c : SUCCESS in 23m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/41eca1adc75a42fd950d7fe1b50dabe9 : SUCCESS in 23m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/270ddb90d3534b7d9679b0ba00ba4305 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eacb20bc58184eafae22af036f722b9b : SUCCESS in 7m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad6fdffeaad6415ba88fcb9559a9e53d : SUCCESS in 17m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77149d1e5cf546c99339501636404efb : SUCCESS in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3a1c963bb404d3ca66b3f7ce2be2968 : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/10afd89e4d5245dd9a270fd88c26441e : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a072ec422814b1595205a642e14572d : SUCCESS in 1h 05m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddd994d87fd741ba9e437d2d0b0059fb : SUCCESS in 18m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21a4e59ac00f4ef486f6d7db46559837 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b64b97ff33a484bbbda5fb10d8e7cec : SUCCESS in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f37ef5b4fd94a77be8b7a57eeb03dbd : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d72be7103e0b4e0abce459f333c64804 : SUCCESS in 26m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ae59e9df8354afa973fb9a3ba413f5b : SUCCESS in 30m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8cd3faee08747b3a1cfb719d8c876c5 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":19},{"id":"fc3ae994fbc5505e1a1b7174901c9d2f8a347c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 03:12:32.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"6a85de6cd6c69246f066aefec078742d80a1621b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:19:35.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f201146f57ba4eff96a83e625516ff96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eca1bab78fe4ed38771ac213200c7eb : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/018317980faa4f50b3a1e2e5e60cb344 : SUCCESS in 48m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d1a7e1127734394afa71fe3021a4475 : SUCCESS in 33m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ad557f3e64a410b9fcbc4654d505bdb : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a4437e813c4a6f8e87d53dea4dd43f : SUCCESS in 1h 03m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2ea5609c67ed40f8959fb4a56098d6bd : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aee11d8c8bde4f55888f9a5f258d1036 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f64838fa7e874ceaa336ecebc7d8a278 : SUCCESS in 10m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f39d6ad7cba84508ad7d33663a15607b : SUCCESS in 14m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9d84a96ca1ac495a9ade41471260a2ad : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bc7dca43db92479fbbe12943fd10795b : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f62b281a6a947908d3fa0a7cde151d3 : SUCCESS in 10m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3ce844205484d778070e611f9b74935 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/23e14048069f4ec7894254bc1de2e849 : SUCCESS in 5m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/356e31064a19489295660e8fa2f874d3 : SUCCESS in 58m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/41cf95d984574d4d94902adfeaa16ae3 : SUCCESS in 33m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bf84e8636ca491789399629cd21d49a : SUCCESS in 46m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec928db24ec4467980fdca28aed4574 : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe52d1510024561a8c423104ab06f69 : SUCCESS in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac7199307047480f8c0ee737aa5ee1da : SUCCESS in 18m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a560531d92a643f59bfd6cce60fba964 : SUCCESS in 36m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a2a3f7968f04dfeaca6df3f87c749c3 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":19},{"id":"b698a0248db0e69d8321a6e9b88e77cd12d5a14e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"a4f9e1f2fd4b5ac6a26f74207a53c9966b88bd6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:20:28.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b5042408574578845ead2fd7c15b89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ca4e7280964990b7d51478aee59b76 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/afd85f766b0b41ad91ee376ed792ccd6 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"b735b3d034b11e1472f7b471919649d165c447e0","revisions":{"8e3ad5997daedba8d0c9e75886d7c3ac9d7e478f":{"kind":"REWORK","_number":1,"created":"2024-02-21 14:39:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/1"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"subject":"diskfile: add test_open_meta_file_unlinked","message":"diskfile: add test_open_meta_file_unlinked\n\nVerify that if a .meta file is cleaned up while diskfile is being\nopened then DiskFileNotExist can be raised.\n\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e3ad5997daedba8d0c9e75886d7c3ac9d7e478f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e3ad5997daedba8d0c9e75886d7c3ac9d7e478f"}]},"branch":"refs/heads/master"},"1399d2b913c90821bb50e7ca14bcac2950fba2a7":{"kind":"REWORK","_number":2,"created":"2024-02-22 16:52:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/2"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-22 16:50:07.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1399d2b913c90821bb50e7ca14bcac2950fba2a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1399d2b913c90821bb50e7ca14bcac2950fba2a7"}]},"branch":"refs/heads/master"},"aacc064e04c511fc77f9c52b4d36e45690c72b2d":{"kind":"REWORK","_number":3,"created":"2024-02-23 11:00:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/3"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 11:00:06.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aacc064e04c511fc77f9c52b4d36e45690c72b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aacc064e04c511fc77f9c52b4d36e45690c72b2d"}]},"branch":"refs/heads/master"},"43c3cdd956ffd7d8270c9fc6ec27a96fdac3167d":{"kind":"REWORK","_number":4,"created":"2024-02-23 12:13:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/4"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 12:13:21.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43c3cdd956ffd7d8270c9fc6ec27a96fdac3167d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43c3cdd956ffd7d8270c9fc6ec27a96fdac3167d"}]},"branch":"refs/heads/master"},"a87ae2ebaaafbe6b10451e6fd9b3ba78b4426be0":{"kind":"REWORK","_number":5,"created":"2024-02-23 12:21:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/5"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 12:21:02.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a87ae2ebaaafbe6b10451e6fd9b3ba78b4426be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a87ae2ebaaafbe6b10451e6fd9b3ba78b4426be0"}]},"branch":"refs/heads/master"},"20da935f5c022f9aab81e6fc402313d4613e6c7e":{"kind":"REWORK","_number":6,"created":"2024-02-23 13:07:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/6"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 13:06:49.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20da935f5c022f9aab81e6fc402313d4613e6c7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20da935f5c022f9aab81e6fc402313d4613e6c7e"}]},"branch":"refs/heads/master"},"b80a986cf4c7b5e47a462c92d897669d30ee98b1":{"kind":"REWORK","_number":7,"created":"2024-02-26 15:18:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/7"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-26 11:37:49.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b80a986cf4c7b5e47a462c92d897669d30ee98b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b80a986cf4c7b5e47a462c92d897669d30ee98b1"}]},"branch":"refs/heads/master"},"e4d5c7d4c130319abc33e9b9b3226eb499ededde":{"kind":"REWORK","_number":8,"created":"2024-02-27 15:07:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/8"}}},"commit":{"parents":[{"commit":"3aba22fde584fe8f81feed97ae48c29142369f57","subject":"Merge \"Stop using deprecated datetime.utc* functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aba22fde584fe8f81feed97ae48c29142369f57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-27 15:06:52.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta file unlinked","message":"object-server: return 503 not 404 if meta file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nThis behavior is appropriate when the missing file is a data file (it\ncould have been removed by a concurrent request) but not when it is a\nmeta file (it could have been replaced by another meta file, but the\nobject still exists). For example, during a period of many concurrent\nPOST requests to an object in a production system, a significant\nnumber of the POSTs returned 404 despite the object never being\ndeleted.\n\nThis patch modifies DiskFile to raise a new\nDiskFileMetadataUnavailable exception when metadata cannot be read\nfrom a meta file that was listed in the object\u0027s datadir. The object\nserver may translate this exception to a 503 response rather than a\n404, depending on the request method:\n\nPOSTs, GETs and HEADs that encounter this transient loss of a meta\nfile will now return 503 rather than 404.\n\nDELETEs that encounter this transient loss of a meta file would\npreviously have proceeded but will now return 503: the meta file may\ncontain an X-Delete-At that would have prevented the DELETE\nproceeding.\n\nPUTs that encounter this transient loss of a meta file will continue\nto proceed as before: the meta file may contain an X-Delete-At that\nrequires updates to the expirer queue but those updates would\npresumably be handled by the concurrent POST that has replaced the\nmeta file.\n\nCloses-Bug: #2054791\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4d5c7d4c130319abc33e9b9b3226eb499ededde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4d5c7d4c130319abc33e9b9b3226eb499ededde"}]},"branch":"refs/heads/master"},"578b34f00f8ab228568f92c6a95fdf416d1f2039":{"kind":"REWORK","_number":9,"created":"2024-05-16 11:02:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/9"}}},"commit":{"parents":[{"commit":"bf206ed2fe2285ebdafa06058aca88ddd1e93739","subject":"Merge \"backend ratelimit: support reloadable config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf206ed2fe2285ebdafa06058aca88ddd1e93739"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-16 11:01:36.000000000","tz":60},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that it no longer exists: the missing file may\nwell have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThi issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a meta file that was\nlisted in the object\u0027s datadir. The object server may translate this\nexception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/578b34f00f8ab228568f92c6a95fdf416d1f2039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/578b34f00f8ab228568f92c6a95fdf416d1f2039"}]},"branch":"refs/heads/master"},"94dc568a7a54eba6c8f4adbee7c80e2a266d74c3":{"kind":"REWORK","_number":10,"created":"2024-05-16 13:29:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/10"}}},"commit":{"parents":[{"commit":"bf206ed2fe2285ebdafa06058aca88ddd1e93739","subject":"Merge \"backend ratelimit: support reloadable config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf206ed2fe2285ebdafa06058aca88ddd1e93739"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-16 13:27:22.000000000","tz":60},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception is raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that it no longer exists: the missing file may\nwell have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThi issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a meta file that was\nlisted in the object\u0027s datadir. The object server may translate this\nexception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc568a7a54eba6c8f4adbee7c80e2a266d74c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc568a7a54eba6c8f4adbee7c80e2a266d74c3"}]},"branch":"refs/heads/master"},"a58117194ea79ca94463648bb31da61268173308":{"kind":"REWORK","_number":11,"created":"2024-05-16 16:03:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/909674/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/11"}}},"commit":{"parents":[{"commit":"bf206ed2fe2285ebdafa06058aca88ddd1e93739","subject":"Merge \"backend ratelimit: support reloadable config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf206ed2fe2285ebdafa06058aca88ddd1e93739"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 15:59:27.000000000","tz":-420},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a58117194ea79ca94463648bb31da61268173308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a58117194ea79ca94463648bb31da61268173308"}]},"branch":"refs/heads/master"},"6be52b6451cb3f99be134a4c4d008592bc4cd8d9":{"kind":"REWORK","_number":12,"created":"2024-05-16 17:26:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/909674/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/12"}}},"commit":{"parents":[{"commit":"bf206ed2fe2285ebdafa06058aca88ddd1e93739","subject":"Merge \"backend ratelimit: support reloadable config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf206ed2fe2285ebdafa06058aca88ddd1e93739"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 17:16:00.000000000","tz":-420},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6be52b6451cb3f99be134a4c4d008592bc4cd8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6be52b6451cb3f99be134a4c4d008592bc4cd8d9"}]},"branch":"refs/heads/master"},"13d35a083f1be1b26ade80586a516701f75a0593":{"kind":"REWORK","_number":13,"created":"2024-07-10 17:57:29.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/909674/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/13"}}},"commit":{"parents":[{"commit":"9c38d756e0233eaa5f9e26066756eea9ce6a11ad","subject":"Merge \"Use entry_points for server executables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c38d756e0233eaa5f9e26066756eea9ce6a11ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-07-10 17:57:06.000000000","tz":-420},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13d35a083f1be1b26ade80586a516701f75a0593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13d35a083f1be1b26ade80586a516701f75a0593"}]},"branch":"refs/heads/master"},"c8c947ce9e225a50c14724b2f7fb41c28e6e98a6":{"kind":"REWORK","_number":14,"created":"2024-07-10 18:02:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/909674/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/14"}}},"commit":{"parents":[{"commit":"9c38d756e0233eaa5f9e26066756eea9ce6a11ad","subject":"Merge \"Use entry_points for server executables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c38d756e0233eaa5f9e26066756eea9ce6a11ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-07-10 18:02:22.000000000","tz":-420},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8c947ce9e225a50c14724b2f7fb41c28e6e98a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8c947ce9e225a50c14724b2f7fb41c28e6e98a6"}]},"branch":"refs/heads/master"},"6c9973fdac029cb2899d60272d2fba1b47b74a4c":{"kind":"REWORK","_number":15,"created":"2024-07-10 18:13:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/909674/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/15"}}},"commit":{"parents":[{"commit":"9c38d756e0233eaa5f9e26066756eea9ce6a11ad","subject":"Merge \"Use entry_points for server executables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c38d756e0233eaa5f9e26066756eea9ce6a11ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-07-10 18:13:36.000000000","tz":-420},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c9973fdac029cb2899d60272d2fba1b47b74a4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c9973fdac029cb2899d60272d2fba1b47b74a4c"}]},"branch":"refs/heads/master"},"e2c5511d19fb28b0f6e1b114c160ceed61bcd737":{"kind":"REWORK","_number":16,"created":"2024-08-06 15:35:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/74/909674/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/16"}}},"commit":{"parents":[{"commit":"9c38d756e0233eaa5f9e26066756eea9ce6a11ad","subject":"Merge \"Use entry_points for server executables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c38d756e0233eaa5f9e26066756eea9ce6a11ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-06 15:34:40.000000000","tz":60},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2c5511d19fb28b0f6e1b114c160ceed61bcd737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2c5511d19fb28b0f6e1b114c160ceed61bcd737"}]},"branch":"refs/heads/master"},"cb745837536196371bd6d0465d1dd03538d3a6b7":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-12-23 22:59:24.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/74/909674/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/17"}}},"commit":{"parents":[{"commit":"5d1dbbccbee04bf7c703538dbab7c903ea146778","subject":"Merge \"docs: Changed OS version to RHEL 9 and CentOS Stream 9.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d1dbbccbee04bf7c703538dbab7c903ea146778"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-12-23 22:59:08.000000000","tz":-480},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb745837536196371bd6d0465d1dd03538d3a6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb745837536196371bd6d0465d1dd03538d3a6b7"}]},"branch":"refs/heads/master"},"3ba6f321838c4806285b4ac2dbe464228e4ab8cd":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-04-11 00:09:15.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/909674/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/18"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-04-11 00:09:15.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba6f321838c4806285b4ac2dbe464228e4ab8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba6f321838c4806285b4ac2dbe464228e4ab8cd"}]},"branch":"refs/heads/master","description":"Rebase"},"b735b3d034b11e1472f7b471919649d165c447e0":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-05-19 23:26:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/909674/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/909674/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/909674/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/909674/19"}}},"commit":{"parents":[{"commit":"84a70769b1c56cc376a1485391f81569cd7ce494","subject":"Merge \"s3api: Allow PUT with `if-none-match: *`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a70769b1c56cc376a1485391f81569cd7ce494"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-21 14:37:44.000000000","tz":0},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-19 23:26:06.000000000","tz":0},"subject":"object-server: return 503 not 404 if meta or data file unlinked","message":"object-server: return 503 not 404 if meta or data file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed file no longer exists when its\nmetadata is read then a DiskFileNotExist exception was raised which\nwould previously cause the object server to return a 404.\n\nIt isn\u0027t appropriate to return 404 when the list of on-disk files\nsuggested that the object existed (i.e. included a .data file) and\nthere is no evidence that the object no longer exists: the missing file\nmay well have been *replaced* by a concurrent request.\n\nFor example, if a POST request races with another concurrent POST\nrequest, the object\u0027s .meta file may be replaced by a new .meta file,\nand reading the original .meta file will fail. Similarly, if a POST\nraces with a concurrent PUT request, the PUT request may replace the\n.data file causing the POST request handler to fail to read the\noriginal .data file. In neither case has the object ever been\nnon-existent.\n\nThis issue was observed during a period of many concurrent POST\nrequests to an object in a production system. A significant number of\nthe POSTs returned 404 despite the object never being deleted.\n\nThis patch modifies DiskFile to raise a new DiskFileStateChanged\nexception when metadata cannot be read from a data, meta, or ts file\nthat was listed in the object\u0027s datadir. The object server may translate\nthis exception to a 503 response rather than a 404, depending on the\nrequest method:\n\n  * POSTs, GETs and HEADs that encounter this transient loss of a\n    file will now return 503 rather than 404. The DiskFile may\n    still exist.\n\n  * DELETEs that encounter this transient loss of a file would\n    previously have proceeded but will now return 503: a replacement\n    file may contain an X-Delete-At that would have prevented the\n    DELETE proceeding.\n\n  * PUTs that encounter this transient loss of a file will continue to\n    proceed as before: a replacement meta file may contain an\n    X-Delete-At that requires updates to the expirer queue but those\n    updates would presumably be handled by the concurrent POST that\n    has replaced the meta file; a replacement .data file will remain\n    on disk until the DiskFile is next opened and one of the .data\n    files will be cleaned up.\n\nCloses-Bug: #2054791\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I2f698c25ed65b236e851e5a307d48a12cef62b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
