)]}'
{"id":"openstack%2Fswift~909801","triplet_id":"openstack%2Fswift~master~I93fb5fa33c5d4f51d73084e4ee945708fc3183ce","project":"openstack/swift","branch":"master","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-01-17 20:15:22.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-03-13 04:17:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-12-18 00:28:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-03-13 04:17:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I93fb5fa33c5d4f51d73084e4ee945708fc3183ce","subject":"s3api: Add support for additional checksums","status":"NEW","created":"2024-02-21 23:33:08.000000000","updated":"2025-03-13 04:17:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":162,"unresolved_comment_count":42,"has_review_started":true,"meta_rev_id":"1744676461190ea90cc329c3da41e2e272ab152b","_number":909801,"virtual_id_number":909801,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-11 18:57:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 00:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 02:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 16:10:49.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-09-19 04:05:12.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-09-27 21:58:27.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2025-01-20 18:48:15.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"78b4d2159b66f85fcf7c3d7e0b7662a3296c06ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-21 23:33:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2cbc3359ac6d70e4f3362392025f143ef756ed6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 00:28:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34b18f662f94d9d92e506221877b641\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/243510c6d35b4eb1bdeabdc4c511cd61 : FAILURE in 7m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/127c5e36e98547b88e5a94d6c8a09104 : FAILURE in 5m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/db043846e2e04e1fa90afe93f9b4822f : SUCCESS in 52m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c6801d859134c9596407de95e5b8bfb : SUCCESS in 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e98855a4116e6f37c84f048dddb33b54b4671aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 02:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71c78e4a08ef4af6b9a6c7e34fd1a68e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5bf3fd8d064547aa555d209f0ce7fb : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/3e8cbd5496734bd18f52af2028170354 : SUCCESS in 53m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/799111af385b4114b86c392f38ec799d : SUCCESS in 41m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62d005006ada4c6898c06e6c9c5bccc5 : SUCCESS in 51m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e87e0c9f020945a3aac3f22341fb4f96 : SUCCESS in 1h 00m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d673984d29464af3b720e3b8a81dc9ce : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8585a4dfe174bbd813c407efd9f9bda : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbaf00010ccd4f328cc102f6a2fec92d : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d09b8bb2c024d738f4c7102e0f96c87 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3fb6a26993634057aaf1f649e5a204f2 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97e94a6dc2094cccb04b8659e274ae06 : FAILURE in 4m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14a6f8de104f46249ab71366658615bb : FAILURE in 7m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a3d3527d78046b092acc07e8cac6b26 : FAILURE in 4m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3716c094e82b4887ace055f27c991541 : FAILURE in 4m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3c251125eed4471ca948002a2b24f2a9 : FAILURE in 5m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/22eea6aeeaa64a20b5316db639b0b674 : FAILURE in 4m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da24c94fd2c6409a98be373b1bb5cad5 : FAILURE in 3m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b054e47c2ee4da6872557b7a507ccfb : FAILURE in 3m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5f3df9bd4dda48aebb7bdbb00dacf86d : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d3aea273fcbf4b609300da0cd458b709 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9a4a600901804f80931e5fe9b08fe695 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6ed00e22d2c4118965e6a5c18d36656 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/471a496ec2244487b9fe767f92d1685f : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/481592d655484ceb88f438970d43fbdc : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/80e4ec59a3a6489b8e54257d1f7d757b : TIMED_OUT in 2h 00m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2979eb3f2ffa448d9d4d966d09b2d5a9 : FAILURE in 52m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63e4f2c2aacc4a98a55d1750c7ed538c : SUCCESS in 47m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08fe4e3cf5d749a4a73236ea6a120c91 : SUCCESS in 38m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/691e478804924520a27609a310e92e91 : FAILURE in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1b6e1874f84dcba0d6f0e67cc7b5cd : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a634c7b929d24fff97f4b68405bbb31f : SUCCESS in 18m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63d1c60a8bd5421284be0c0fcdcbd198 : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"490cd3624fb36ece12d6cae1cef40b8306e8827a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-27 23:42:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"435b287e783b456e5448906e3a1b9253ae6bb63c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-28 00:13:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a295bbe938226e5bfe2b4bef9dc3a99344daff6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:32:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8849b151838942659d70eba8af5e3c8d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d475609051481798a71326971b51f6 : SUCCESS in 17m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23c3f60983984c11af7a3b3e9e3dcf96 : SUCCESS in 49m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3022bf09d2d546108a49aa55b696ede2 : SUCCESS in 24m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/676bd9cfddd14492af248e95b1c26569 : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f39ea5f1a002621cd60d6cb5454b0fa549e0c3b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:37:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c02fe5b8c53a4799b76e40cea96b7267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a536b7c7d19e408dbd5da2bbdfbd7001 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/afc825b390cf4c4b8ec314ee2779cb7b : SUCCESS in 59m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7faded1e01bc44309adb58a53e5dd8ef : SUCCESS in 46m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c653754f65f54f64bbde2f6efb6450da : SUCCESS in 44m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1982bd63ed634843be59b818cc1103c3 : SUCCESS in 56m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d7148bf333d421bac5ef74c25fa0d59 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b586a913c8041ce92654d9d3b5f371a : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/054b2b9188294c8a91b26822f688035e : SUCCESS in 26m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e4744eb8ece4705a51816d5fda56a44 : SUCCESS in 12m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/921d86aacd2b4e6a8a2a161e3291a678 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ece61db5a6dd4f7cbb2a55f6f809b484 : FAILURE in 24m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93bf67a9d5014496a0061fd8aaf79629 : SUCCESS in 19m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/537e84dfe26b48c386cf276c51f280a1 : SUCCESS in 22m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7eeefc7dc6774f41a66c06f2adee25ab : SUCCESS in 24m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/83ce13f5114245caafb4dc76c8aa720d : SUCCESS in 24m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06cf9ea1e0d3497d9dc7906ff4ad541c : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b939e11d106c4a89847a830b1b83af8c : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d50492542d024daa8e7c1196dbf15a2e : SUCCESS in 17m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6e96d311cb848928635a54fb808afcf : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce40ccf9b647471dae256849655d5f38 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e082bcce3c9e45e18d91e15e3d805081 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24f649de27214b58aa569c9d6c05dc10 : SUCCESS in 8m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5af2df3731143e38845d2388cc6deba : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/752ed0877d65435bb026576bf4698922 : SUCCESS in 7m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/549540091cc94531949bac1a1fd8b249 : SUCCESS in 56m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa6e2967110149f3b1d051fde3c5dc17 : FAILURE in 1h 17m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e0f1ef8e9974b48b1c619074d0f4800 : SUCCESS in 53m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/709ae4e179474cbbb18e346a71444cae : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75e9e6f4e19348509976a1d37d2c85df : FAILURE in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd9755175594aca8586dfc03320ab24 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa146b96ea8746198faecfe7f4fbd3e8 : SUCCESS in 18m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ed9405990a446c68bf3a0e21a6a8205 : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":3},{"id":"adfedc66ba69b85a1cf5fd51d3285f789571a832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-01 05:18:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c91748905c6275a7b61215293ced3b9f1c8efecb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:40:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75feddbcb1a94a53b96a87c4e1b0945f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd6049da2eb544cdb5f2cd61a5c4884e : SUCCESS in 19m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d20eb0feca254d2198d91bc0608be023 : SUCCESS in 1h 20m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4735e84abf2a43b1a89fa58de8eeae7f : SUCCESS in 22m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/086d9b737b994ae180cbacd288a7c43a : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3881eca081c60eb3e5908e66b26885d6a9852208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:51:49.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f15582f75f9a41cd87b8bd4d240fe2ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410b3f254ebf47889c0b61d5b06061fc : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/87691fbb04de4b67b67e36158b95364e : SUCCESS in 56m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1b255b5cebfc43eb83fc7ab7d6894a0a : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33fd6b793669401bb7b74e23eb8cef62 : SUCCESS in 34m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19b93e5c4b0642ba96d77adaa1cfe07b : SUCCESS in 1h 01m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e8fa085f61d4f7f9faeac01a81e6184 : SUCCESS in 23m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c60e220170834d7aa066f275f2b3a41d : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdd9c6d1a19b4e07ba6a2896fd4bbc13 : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac810b93a39d4359933a8a50be9cbd0f : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0a4296d7e223498f8e7660ba4928594a : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/216569d27c0248f195bd5069fd14a5ad : FAILURE in 4m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76bd5158365943078c0c604a447d88fb : SUCCESS in 13m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab6ec40bb5bd403b8cd47bdc0ac9ea71 : SUCCESS in 26m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4c9869b8acff4e788956ca11175f56c7 : SUCCESS in 22m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/27bb3029710c40f68bdbe8d6daa1a803 : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a844d892364f442d9adfb4d75ecdece5 : FAILURE in 12m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/164e9459a73f4656a129919df563d1e8 : FAILURE in 12m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b85a9a726e5e4fb290c950f71c276d2d : FAILURE in 12m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6bc16b62227b4659b9c94492bac472ce : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa39a7931cb0491db7ad0b361e32f449 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dc727775263e4d0aa10f69b97330d584 : SUCCESS in 23m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d38e806366594692b307174f84f89b4b : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/894590a64f5e45c99e74d4f43308713e : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d2223d0c7ba4393915346d6a5efb504 : FAILURE in 8m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/73314e0e1f6842ecae88ebfde282338d : FAILURE in 5m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7d1b74a0e3346b98b6fa476dd07fe40 : FAILURE in 1h 31m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9028e907002e4aaab88f680fc1ab1338 : SUCCESS in 58m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4272197273d14ee7aecf04404f9247ad : SUCCESS in 54m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84a3de0618654f989ed27faab3c34cc9 : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec061993de545409579886f591c2d22 : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b52ff84399445be81de9aabb6682f8a : SUCCESS in 26m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/445ccaee1d254187938d6602e0379893 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":4},{"id":"098a4bbd5d6e7c8c7fc2a81af833b0abb80de9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-02 01:02:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e1c3e29ad3b2153e7aa5009403fe050d39ea8f75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 01:24:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d988bc8773734e0484d93d104beff621\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02d6b5316b8e4e7b9d781deca97f3829 : SUCCESS in 17m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b90e2c97e884468b8f961554598f13c : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cca919fffcf4cc88f3f347cc7dc0445 : SUCCESS in 18m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93a747863547419f96dc535f4442779f : SUCCESS in 21m 09s","accounts_in_message":[],"_revision_number":5},{"id":"92df878eaa66fdb4ff159fa6143b97f7b30bf587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 02:44:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c67e8bd859a049379b8afefaa781ee97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16ec4e776e04e39b17bd72274b09a4c : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/0e9c7518f3bb4957b4f3008e05b61a26 : SUCCESS in 59m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b11af1afe7dd4e11a9c6d01b67afaf34 : SUCCESS in 1h 04m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0412fad6d6354cfdbb0391e252be537b : SUCCESS in 49m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0054ecab18d4088aad7c6c15237c114 : SUCCESS in 46m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6220cdce8d5940bdb2e2576a6c6786f9 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64ade5f557a44155b5151c3d2038514a : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94509227abef437abcd1cbc5c6744fda : SUCCESS in 23m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6999d08556024c55afe03c78906bea9f : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63f663e70bc24fa394d7610b647e7ebf : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae2133258ba94116b8ad36723e667b8a : SUCCESS in 21m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0423ee5f4436452a8a5ae5dd85f5089d : SUCCESS in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f424e6ea09f487c9b0a220cb9ab5a18 : SUCCESS in 18m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/98bf1a9570184dcd8d6f4571277808e5 : SUCCESS in 21m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a43e47a1c05e4852a8bbccf6ef4dae9e : SUCCESS in 20m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97bdae44d3974b02ab849cae206f3094 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd9909a1430b402ea2231d67366f5079 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a84797dbc7e48dfaafff95ac85831d2 : SUCCESS in 20m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d2899024b9fa4422abee0f179092da19 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b21dac62d1934706951f6907a128bc39 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/587833016b81418a8473e59262527b7f : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5376e7c3e647410c98715ea56b480aae : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1d360eb9bb24140894a93fbaeccb68c : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/50ae2cf9cd864bd5a55cc9961779b41b : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7fddd72b92f54fc7939c5c6ea2d8653d : SUCCESS in 1h 04m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/57c565c79a204037b1c9cdde405409f2 : FAILURE in 1h 13m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1bf142914044e9b91be4dbb5ef8c44f : SUCCESS in 58m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f2e29442e334ca790b27bc0af07eef1 : SUCCESS in 50m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b54b348ff06344df84e35691123f4ef9 : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c20830df06d4102a181bea2a4739b7d : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06ba5c99488241f3b1dac8710de6862e : SUCCESS in 20m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5fb13f414d74817b1a57de8b880cc70 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":5},{"id":"d836740727dbd08be61d6b2f848506781b9bcbdd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 22:59:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"960f559e45dab00546e8eaccc5d2ff655a43f463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 00:18:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6417c165dbeb4744927c1b83dfe1d192\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5326cd7fad804033b2e9f7e6b8e4472c : SUCCESS in 56m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2713a93bd36443d3bb1d662592d17fbb : SUCCESS in 1h 17m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4823f8ac69748e283002cdfaf974b87 : SUCCESS in 22m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb132d4ea87c4bc4947a82a4b44cb67f : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":5},{"id":"810f2832b9e2725a197614d8be7ce3b58f4c3c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 00:24:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f5c84baa45f4a5683935f6e77d8d49b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8da72ec8664e7781e85c2fdb3741bd : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/6aa63bdfbd2b4e8885dff8b0ae3516c7 : SUCCESS in 1h 08m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c74a86b78eb646e7b0ddea71dc28284f : SUCCESS in 1h 12m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45f5f67d18584213a055c1719def2a27 : SUCCESS in 50m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37e69b554d1c4d7cb87ca8b26b0605ac : SUCCESS in 59m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3543a64af404e1cbfe5b62f9ce8a8b6 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bdcf3e4c145487797a37d158121bae2 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fad84d15c5d341fdafb5c486604f4182 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ea17735509a4cb69c274cca3129b1b0 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4feccc0c405e416c857d4b272e5ca7af : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78f1ae2e2b7949ae8e3e1ac01e63f04f : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b79d205e15f46e3b7149814d7d84083 : SUCCESS in 18m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c042b3a23ff244c8bc51a53428cafff4 : SUCCESS in 18m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ee0744cf2d854a66bce51c3c9ed9a39d : SUCCESS in 20m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8c0822638a134aaa8c116a8161908862 : SUCCESS in 19m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb9c23050d594f7984156a08aca47e67 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85eb1d9bac254521a60a62cd35770637 : SUCCESS in 20m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eef1868bec974bf7967f466d933882b2 : SUCCESS in 20m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e56b9702667545c6b669d085481cf9d1 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/41979020387a4174964e5b488b8bb56b : SUCCESS in 18m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/695d383e68864ad6ba0c678d184e596f : SUCCESS in 18m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6e97077a15c14ff0bd674d36295ae7ae : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7605cb8233334ef6bdd59c47d752494c : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fc90a8ad0064f7eabd6ec943dbdb94b : SUCCESS in 8m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9133077195fe4aed9fcbc0747bafafcb : FAILURE in 1h 06m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/396c3e56d26d41ffbee2b2a01e80115d : SUCCESS in 44m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/237aa75f806a41b399af26d6bc8e9dfe : SUCCESS in 53m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/428cd974419846b587de9354a44bef1b : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d8e047f1d6466990fa852dc93f8d88 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae2130caf7c44cb5abdb5b5501c14b18 : SUCCESS in 18m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0198040b4564fe19bd29676b94930eb : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":5},{"id":"fee08ee749c227bced9dbe0f38a40ba8088f1664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 00:39:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"39fbbcce36d81bbfcfd778c8a6a2fbf3719c1dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 01:53:13.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03dd831f8cd748439889ad68bf0b03ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e112070f6cd4e2987d42da3f2c8c681 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/c961a55df103473a9226b51e77e2308e : SUCCESS in 45m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/43b74244ae8f4100bc13e752ab923a50 : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b024af1443c6442999ffe87ec993c7d6 : SUCCESS in 55m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a69d191232c489ebb3a6a90ba86e6a8 : SUCCESS in 47m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d815f0b3a664359a95405392301db39 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf916966a6f74af498b3d0a5f0ee7e90 : POST_FAILURE in 17m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee8e7c9364ab43a199a312be8d5d7bc4 : SUCCESS in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a4ded1b1f5f4486ad3daf653e65a30b : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3aafdea6d381428b8ebaf064e1120451 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84a676edb8b54be5be20ae21868b457b : SUCCESS in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71c33fd82a6c4b94bb8da39b9e2103cb : SUCCESS in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d7dd5ac4306643bd87dcb6296063ef12 : SUCCESS in 18m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3ca8fdd2a1d2480a8940accd453af0fe : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/60f73f170cbe48de9bd067b16c7dbb1d : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/815c65f910194b8e836062d712156558 : SUCCESS in 22m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da9e203f879f4619933c2b102466f7a7 : SUCCESS in 18m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d8ea2901f0e4e9ab1bf3dda63794e47 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7281ad7253e43de8a2a98a01aee46bc : SUCCESS in 18m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6ac925416af84f6da01a86eae11ad6d5 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09318af86c1c40ffa743e5ea212ab3e8 : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff2c68ee81de4ed7a4adbad343aae8d4 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e944d2c67d34e09acf4d145c5233d06 : SUCCESS in 8m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f32e2237e7924833bee82919d9858c88 : FAILURE in 1h 11m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d2b8605ef7f4db8859fefb71b5cad2e : SUCCESS in 45m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bb978919efd478d9943a039779ea900 : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/467f9ba116f941519be7b67ce290c8e1 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/717464fa1cde4e44bc9fd12d5cdaad4a : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b78097b5843648498c774f2ad62e7479 : SUCCESS in 17m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e13f4be023a45bfaeec4a045de150d5 : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":6},{"id":"6cf3d45ff20b2affe501a128947347ca0d7c0ece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 02:18:37.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8733c46b19f54b12aa046660f64ee3bf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fe6b7e0f7a845118af199a929bc26e2 : TIMED_OUT in 1h 31m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b44cad033a541b89e6797aa5a8115c7 : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d06e6d18a3e43cfb03a49ae5daa4cbe : TIMED_OUT in 1h 31m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa02765c34ef41faade8928141edd982 : TIMED_OUT in 1h 31m 05s","accounts_in_message":[],"_revision_number":6},{"id":"4046976ea72eb92741a6324f843ef4437cc73be0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 02:39:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a7ad0608b79c5876829497e3cb8e23185010b433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 03:57:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ac52818259467e8726327884886747\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a6cca7fef0b4212bcfd65d2af3e5fc9 : SUCCESS in 18m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb12d96213784c8c84a0565dd4e81945 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2007f9c16de4f91bc67496a604e7f0c : SUCCESS in 20m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb9e829752d84a5882b3e131db511098 : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":6},{"id":"4980b06355d5067158f894f9618c7eb0d276c776","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 04:57:59.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0622c08fefaa4d0789dc9f50a15bbf60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a6535af4f943fbb41eb993114c9a12 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/475a9c0275654446806715de3ddc66c5 : SUCCESS in 58m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/892b8f8f8e324049acd0e97de56a01af : SUCCESS in 46m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2b00d563aa2347888ee638035713e092 : SUCCESS in 57m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48468f7a45e740b6a11af0307ba17635 : SUCCESS in 1h 01m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/812d2cdc4ca947409bd9b427dc9f7125 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2373e49bcc894a3e8819030dc7730969 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05fbeac003ea4c80ae2c891c213230b8 : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c206bb0199e4bcf84870ea4fa4f3387 : RETRY_LIMIT in 2m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/287802071e6e499c814629357b4d7940 : RETRY_LIMIT in 2m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf5d9be80d0f49bb9e2a03e232bc989e : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9788303f0c844ea9a4ca53854be70d39 : SUCCESS in 20m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0fcf2c6e81734b5cb68c2ea9d235fcf2 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2cad24c21cbe41ae9eb0ce00516842e0 : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c92f5b06cf34c5490f6d10c3d30db8b : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc9369c479b34344b0578a7bb98b0708 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ed3dbce1320d416691f6b5a46c53f233 : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b297ad8712946cca0ff122e8f3fc00f : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5c82ec3e893c4dd79a6ddfcdf48e0c12 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26c6ed8cc2bd4d298e37cc2277fef4ae : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/445639b191634d9b978b29dbce9b65b1 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8577739dc1434402bb5d2c130e40b595 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fbb81b34cea54bb3aac26014ebd5e96d : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78025600e3f6464989d28e061c0c633a : TIMED_OUT in 2h 01m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4c42a84ceaf4330b74332e3aeef4d22 : SUCCESS in 40m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6956db5269214e8f8ddf878bd1e0a111 : SUCCESS in 39m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9cd137ffd944c62977184d04402f600 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a50be4d86b347edb94f157ab832b97b : FAILURE in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/448c84ed24e84b648a98870b29d504ed : SUCCESS in 20m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1db919fb257d4eef9f06e263732268bd : SUCCESS in 41m 31s","accounts_in_message":[],"_revision_number":6},{"id":"d36ab01fb6ccf0327a1ef76a0aa6648cf6e20821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 05:46:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"2f0e5570984c18faf4217807a4e91850a322dba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 07:00:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50b96d8f5594f238788cf429a2de791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09487bef28494130b30ba3f37dde7e19 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/dc2bb4a9a0054dcd94b11fd7efd0bbd6 : SUCCESS in 53m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/351acbab89de4584a80e1522372a15f4 : SUCCESS in 49m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8343c8331ddb4e698ceab23a9cce5034 : SUCCESS in 37m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36a4045efeb845e9b19e1395d4c2acbc : SUCCESS in 57m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa9c6829738544c3bcaafbdb86eaee08 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2587d7174def4653be364c99bc3b3bb8 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/907d7faf80cb490f82fed55dce2b66f3 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/840977b1f1024fac98f4c031e55cd0d5 : RETRY_LIMIT in 2m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/830620d02c214c56b58fa573f042137b : RETRY_LIMIT in 2m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c39b963920849babf57c57f3b847879 : SUCCESS in 26m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fceeace42c3f4996b9ea5004c30b28ee : SUCCESS in 23m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0161bfa9f6f04a5abb80569a92ab12e4 : SUCCESS in 22m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca133ff513c24fa38337bc32094dd823 : SUCCESS in 11m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fa2df279feb4b4b886e350f4a8dc48c : SUCCESS in 25m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14f8788caf544af2981f1b412bfc65e5 : SUCCESS in 23m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1fd06eed7c5e4b4099677e5244d25ba3 : SUCCESS in 26m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/60ccaec2da194efc8405a1219ca40608 : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/caeb73040cb846078080b5aa05c476c6 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a08066d825ed40838314b2b1ceed7b87 : SUCCESS in 13m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f983a2b6f0b24be2ae262b385318b13d : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c2703ddc4b64b7194c3767dbf5d820c : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90fffae9d39648a8b9e9bb14140f21b5 : SUCCESS in 7m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2160bed6e77b46fda21d1ac5d1ee03ff : SUCCESS in 1h 00m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9a9a0ed42af4e96b95f4ee53567a084 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90e8590c7f644c3d94e5f0d93f56202b : SUCCESS in 43m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c316a6b61a84448f8d617e86e1cc590f : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651796d08fb74c049cb014eba959cf3f : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49d2c0d9e7f945cd87aae97b133b7724 : SUCCESS in 19m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/437ae810041e42aa9496728aff3c47ce : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":7},{"id":"f5beb52c6700adaa272fafc1f4e34fa3152896fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 07:01:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21dafc525b5a41b38a1a2b173edac2ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acc7dd0ca273405f91e45d659927a330 : SUCCESS in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bec1d01eeb0c4744b991f7a6345bbf83 : SUCCESS in 1h 12m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/83270f574db94c9b9dcc127b19dc7b6c : SUCCESS in 19m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/48782562e5a14bce8515008b788ec05b : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":7},{"id":"c8ac67deab60aa3f8b985984b4adc840bb24fafb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 22:49:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"fc58026c9cdec764887cfcb9c489917caa648e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 23:29:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f482700c83145d7a0d51b96859cc6e1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7ac269222424812b92a9535659efe00 : SUCCESS in 19m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c357cd461e8a452baa68737d0bc856e7 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0ca17f8a89547ee8cb78dc0392446e7 : SUCCESS in 37m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3325240b13304813ac0d6cb6935fb0d7 : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":8},{"id":"5f2d06cad20523fa38e9c106e25c8730576ad859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 23:58:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52e904ee63ff47b68c94d88437c03841\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb4065b9bf046308d8330573b98fc96 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/1d5a52a143e84c09bad4e6588dd022e5 : SUCCESS in 44m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dac946df52384727a0b62a9b07dacca7 : SUCCESS in 44m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a6ec74b0620452084d435d51c3e9e66 : SUCCESS in 36m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc0e0e223af4b65bddc5556f14a3ad1 : SUCCESS in 45m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5028857f234d426a9cb842f1ceb6cb88 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dce66c1090044f2a11b4156fdbb006b : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98140b439e274d8cb49108a47fdef8fe : SUCCESS in 22m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9e858976ca54a03a0db42e392369608 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a7b351f590f437c94ebb60cea8a2314 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c74ab6ff973348d0b2f9fe0e58af5228 : SUCCESS in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79b10fc591e641efbd5cfade70f96ec4 : SUCCESS in 20m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c70b35caa1484f12832e65bef59ec035 : SUCCESS in 22m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f2bea1db9a6745699f46935717a6f893 : SUCCESS in 21m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bfa04eca0a4649b0a813867868a77977 : SUCCESS in 22m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eaa4d5c34cfd4a7b93f076c2ead7b729 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78d3f9fd28d94d60ac586e0b5b197461 : SUCCESS in 16m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f65211581c9549ba90eee5118bf08636 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1915942d85647c8bdcbc491bde9d141 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9e172fda063442daab347b6e831fcc02 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef5416f1c2a3444986e69c461e3c7d83 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e4e90bfde294b8ea0d3afaa2f36a399 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9d73954047047cea41dd3c2e2c0c99d : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f8f9d141daa4e1a9257543769490416 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d545d4a08f44bf1a9ddfafb21a7404b : SUCCESS in 52m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89fd8518a95141a28ba459468196ffe5 : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43b0604c14e043c88139d5d91b53a6f0 : SUCCESS in 24m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6fdc8a61a00411b9d660b32ba9c5c39 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0693b261cf34f199a9c36839edb40c5 : SUCCESS in 23m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7275962ae744a26bef938f755d6ee4d : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":8},{"id":"fa55a3f809ae1d44130e7d6a57e62bfe8a314915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 18:47:56.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2746be190531f1d69bd98347037356024e21b4fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 20:02:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a7f316604474a8b8c1a26c971835df8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f1c180849a0439482ed108467719763 : SUCCESS in 58m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90bbcef079cc4c2e894a80617257d6a2 : SUCCESS in 1h 12m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8cf59e965d24fdf809c18deb5a3b3a9 : SUCCESS in 50m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c7d88b79a71463db5522d2ea50aef89 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":9},{"id":"6aca79730b37287edb457db7e321736b4e1d5981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 20:23:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6cffe88a25c43f990c3fe11443dedfc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9daf98df3348442a922164787a6e2d57 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3347408170574dabb84db97efc049e49 : SUCCESS in 53m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/569335fd85f94790be3f6fd2186197b3 : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/770d098349884a37ab3df251a062e214 : SUCCESS in 1h 01m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f7bc0bc3c824462be4b439e50f29e7d : SUCCESS in 1h 04m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/465e2e0a36ca4f419fbf9fd53249c984 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6a71e69dae5460a91d641fa9449746e : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2f53213a79e4304b6bdf9e650324bed : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45a414a8e1d44896be1e052320ea7492 : SUCCESS in 12m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ed7c981c60a44508254aba81b778c5b : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f61170524cf4974b8980f26166341da : SUCCESS in 19m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/974f607367154a1d8c213df9ba8e9bb2 : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/028865d0dc6d49dbb4ca2d7d97f74e15 : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/51e3a4fc048345e3ae3ea747481510b7 : SUCCESS in 17m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3359652a0e724d67b229b37ba527ec0b : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc38717d98f3466c8ac6e148e614f3d9 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2371acd889784232b129bdcc77b4227f : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97755970a54d471ca8d30a36b54ca6a6 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f416a39a35964c8384872eb3044f0494 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f332626e97d64039a81c9b6bab775261 : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/616fc57a183349faaa1be16c4323add8 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bcbe125a44b54b64970f8ac14c13f690 : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f17b84f04b1841fd9daa6d587502e24c : FAILURE in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaee176a7481412e84c1705e7038c89b : SUCCESS in 56m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2be798afe456446abd6785cf21b4c66b : SUCCESS in 51m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c8f76ef108a41a78d65a2d1d2ff9240 : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/813d173751e34778b8db0cba5dc2a483 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d77857b9ee14aa6bdaea70f0b039c0e : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/959457af2e6c424d8c18527e1be4712e : SUCCESS in 19m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1537054f71f44759ac066d7046e01dd6 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":9},{"id":"b504dd51f1dd8c3c6fd62b27638a69e5ae8f7b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 22:06:43.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"54e50573e483639101a42226898b819f4e6e3e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 22:42:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"6dc0f081b2ca3707dac9ea493072d12bfe930e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 23:09:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ed56f64b8e5520667d961ba112064ac59b746c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 00:20:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1fefdb2a39346a3a366a6dc30de6110\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e453d87bc820498ca3e8865c19bed952 : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f5010a7dc7464b29ac10c4e6d3cf2905 : SUCCESS in 49m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/93d5579f24e2481e93cc9400e6426ed7 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0e580419b464b26b621505c782bc039 : SUCCESS in 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7be009a00421444ea65e5ddeb18169a5 : SUCCESS in 1h 01m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/804b6b9b95cd456584df7ebe86ef3f34 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f6ad2d7fb23421b87934bdfbf43a08a : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f0ff1648a61488d9b77b88033bf4032 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3141972982b4ae4ae4dc30f567847c6 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/989caa98a3df48d48f4cb60ccaf52b9b : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45e77fc620294a258859a8b97dd993eb : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/390ad64391bd4a21b912c44e75f85005 : SUCCESS in 15m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bbcc682598874f94af7d4345660f95da : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/317fe9d7dcae409a959b997124a1f652 : SUCCESS in 17m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24e43e4b05f2487681aab861e589b0cc : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3d733e4838e4cfea3454f1dd3adc416 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33012391982d4032983584c0cfeb029d : SUCCESS in 19m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/818ccfdbe36944dd950e3746f1d11929 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e8c5c7a18c44d11bd7c93e5d041ce2f : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1a3a64145b5451fbf24d5b5d71dcee7 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87095ecdbc0f4b06b93aa652fec09a0f : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53e2afea29d041a48b0f19b214768449 : SUCCESS in 14m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b3e7cd287ef4dd4ad2390af28128760 : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abb8979a2e7346f6a142139fd144a13e : SUCCESS in 46m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/894799f9bc3d418a8adc5d6cc17a1eb5 : SUCCESS in 39m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1b622b047f84fada1f5750541f4d57f : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/777d892718944959a6cff5330baa9f78 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6553c48b19a54858bfa047fe5840c8dd : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fbda15d4fbb4c65888e1bb6b944c29d : SUCCESS in 19m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af832abb15864a178d7285479d20248b : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":12},{"id":"c88ba5f63002920c34975d11deff0a27e288f31e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 00:34:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24025e997a834c68822a9bf4d7de7d91\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b089005dc4942e6be7ad30edbeebd4b : SUCCESS in 17m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/321a9231946b429e952173ecbd08c91b : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/490a85764a4a4da49e7093a21d5562ec : SUCCESS in 1h 22m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f70b760f8ce435d9ef6928d730ffe19 : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":12},{"id":"a1170b786dcab17daafedb00325363d9f470a595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-01 20:21:48.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":13},{"id":"002dd78c44699a0a557fc8c70ae83c8937908f94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:38:16.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04f31dbb32c4d63ab43214214acdf4c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcd4e521ce33419287a1c41c9d8db76b : SUCCESS in 1h 04m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1969c5ea5aa4012bf5672dcc7b4f01a : SUCCESS in 1h 08m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fd7c6bcc78a4ecba79911b6864e159e : SUCCESS in 23m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b8d9a2cb6324fc5944ae9df393484fa : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":13},{"id":"f2aa259619aedb8e1c724414895f610dd027c2cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:40:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a965c4ade1bf42499ac4fcdc3ba73d3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fffaa1e73942b0b5b5cd4e44374d19 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ca476599ca4945beae10c1c3b4f568c5 : SUCCESS in 1h 01m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0efd14a43aa84aa29dde9b0a53cb9a06 : SUCCESS in 43m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e04c6f5568940a19e939768df508447 : SUCCESS in 47m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4e2fa4e557845d1a87860a4ebf26958 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/059adce7b7644986b9dbb38789648cec : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0e082f644ac4433a4c94f40bd37fda8 : SUCCESS in 23m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/edc419ada8ce4f268980870631f835fd : SUCCESS in 11m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5713551568164bc2984e474aad6f520d : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/294d002811fd4d1cbd138986bb6b04ed : SUCCESS in 21m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6052c7a930eb4a0483aa70d16902c44f : SUCCESS in 19m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/555aef59e2564921be04460e48c70228 : SUCCESS in 18m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/54a16554a32c44658f97294e1d32c37f : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c11c82c81584d5d9aef3993ec385d5c : SUCCESS in 22m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fdd0073214814415b2af3a7a270ab8ab : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/376fe4ac4e7f473f81ad4659aa2717ba : SUCCESS in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed991894dcf544259edf5f2065aaa6f6 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a5d3633814f42378513381ccfe9b848 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4402dbcaa5dd41e9a9fc84b9c1dcfaab : SUCCESS in 14m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4d2601c27df410cb526d9a9ee811da7 : SUCCESS in 9m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7460dec8c93c4809a55bdd4371de07ce : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e6ca8cfc1154fc5a76fdbfac7bffcb8 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a6a4d7431cd45218ad8a5632166822c : SUCCESS in 1h 00m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fb95bfad35b4cccbfb36f0749bc9fed : SUCCESS in 37m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89e7ec56d6a74f32b1b2378f390c132f : SUCCESS in 48m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dca71a449a9d4440b94b97df1c528685 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43046d0c74984e40a3c796d0817b0808 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/233b25fe07d84d219ef4c2859b3796ec : SUCCESS in 18m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9d4df7e3de84ae2a584e45131090f59 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":13},{"id":"3927ebdc50ca092ea35d62e94ef778592bd988c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-28 22:36:50.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"85251e1d48396c2d2889d6e18c37acd949fb4f74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:42:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf6c28e2b744b1287a57e14c12a908c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03593ed1d5924f018fe1ea31e56a8842 : SUCCESS in 56m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa0081447d684cd7b2b4d1863e6c5882 : SUCCESS in 22m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f54afd4a8c444df191371bb174eb1af0 : SUCCESS in 1h 02m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7272af97aac14b5791ec1b636216cf97 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":14},{"id":"5a589f7f9a756c69c9d493af5f23fed5a85c8081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:59:25.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c55f4f5d1494d7982808685e24152c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed29a3a084248bc8cf35c8548024588 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/e3339c0bd7c44bb3bf6955db326e5406 : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52248d980c0b49a5ab9f96ba05e09e12 : SUCCESS in 45m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fc62d87c1044d4ba1f7cf928644c593 : SUCCESS in 1h 00m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9de831b1524b4f38a1ac282d1c8a1166 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/710a051c75194ad0824f9a769550032c : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8183822367545b2b5bc9e2bfedee68a : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dec8c2d51aa9455da03796ac12b09acb : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d3315418c5aa4f98958663edc778b98e : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15c2d257aedc4fdcae858f5a22e42966 : SUCCESS in 19m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8c9042da4e7443d93c37b62c5a54e4b : SUCCESS in 16m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60c3566769f0435a9c990466b2685d07 : SUCCESS in 19m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d75c3baef07446159967eb16c3ebb9ce : SUCCESS in 23m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a60e032a43bb4a4d9a0c9a30ac6e44fd : SUCCESS in 22m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/890fb2f38acc44719293dfd22bb740b0 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a063c5f28a114777ba4ff4bae2f40c69 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0655711f96546c7bf02592f6c7b627d : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/62e96277ee6e41f7b07344101b343d91 : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/48eee163d3ad4109921103c8c1006ac0 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa788d116e6c4c9fb7cb1475204d1bb2 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/94b739960bb5401ca91529514405167f : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22476ce4952549f5982b70d587c91875 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a1aeded3e8c54467acefddc1d6c92c57 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e7512180e9144a2b7e5d31697a18e2d : SUCCESS in 58m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/181ec467127444149769b04c1d2d026e : SUCCESS in 51m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac26762ed2d0474bafcf95c683268618 : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9310ad06ebd4fb2a2a6dc26cb58f130 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c39a3843e2425a8a7d53aead1d7c5b : FAILURE in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a3039df147342088b82da6cf1eccf92 : SUCCESS in 24m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7344552e417047a08652fceab798100d : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":14},{"id":"9b828bc331d5eb141ff5f7c8d7129c8b68f7fcda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-29 21:03:41.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"3c23a7246797efbac0e6a6d1beca5ea06390bd07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:11:40.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b5253497024cb0bd3a2bc77588331e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25b24805b2fa40e8a6523ff0437469e7 : RETRY_LIMIT in 1m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c9c7395d96d4ca0a6697bf771c0b457 : RETRY_LIMIT in 1m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/03b9d98611af43818b604ce5f3590581 : RETRY_LIMIT in 1m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/71db546df1084605acf2f35c71d239a6 : RETRY_LIMIT in 1m 06s","accounts_in_message":[],"_revision_number":15},{"id":"ef4c30cf7e2f7b1772c32cca7c334a61ab944421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:53:36.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76a2a4a559ca46cb8a6b5970fb2a551a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498133d12df44899b107523c5b4175de : RETRY_LIMIT in 1m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/305941097a204be388df021d52d9a5cd : RETRY_LIMIT in 1m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ef0523ac5454abfbb3905f112498d28 : RETRY_LIMIT in 1m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f81371614274fd6aa2a9d43e5bbd6ea : RETRY_LIMIT in 1m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7af32f7987ac44cfaec291734ea89262 : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ede2e594507490cb4d532d2574ea073 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6267761ed83a4e9eb70acf08ac380c34 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f291eedd16c47cd868fb7c4422fa747 : RETRY_LIMIT in 1m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/29a11b8f8bad477f9ea091659b141f8d : RETRY_LIMIT in 1m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9159a4657b024ae78b67db77a99c525b : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57f0bfea09d542db8205942413e3d0a9 : SUCCESS in 19m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/12f9b3667eed4086a0afdda1b304d08d : SUCCESS in 21m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/24611c9ed08f4bec94d477a86a141205 : RETRY_LIMIT in 1m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/67c4e576778f42519cdfc2f40901fd76 : SUCCESS in 24m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61e949c8c9834077b702d5d558c0b1d9 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0489096d27e4acc949ccadc4888cadc : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2b271a2ce6242cbbda12b4daf285cf4 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/910e294168de426b9b99cdb7c89e93c8 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8f9a3be10eb4647a377eb55c48a7fc7 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e315b71554c4249b8edfa3e257fdcce : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69c4f209512047c38812fb8a27d09c70 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c31a521de934085a51d14a7096670cf : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/55e296294143447e992c9b06c35f538a : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/583ecf69ffc94c6d8831d59e22829c3d : SUCCESS in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfa97bee4ebc4af8acf3920e06150532 : RETRY_LIMIT in 1m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35a3f3c07bfa4bbbb3b572b52ef9ab6e : RETRY_LIMIT in 1m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d01a546a424738a7b035ed1f2be5a6 : RETRY_LIMIT in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce8eeba23f241a7b90893a624d95af8 : RETRY_LIMIT in 1m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5423f9886b5746ae981b7f64912b0986 : SUCCESS in 23m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117450c52395483ba5f62faa6e816b88 : RETRY_LIMIT in 2m 29s","accounts_in_message":[],"_revision_number":15},{"id":"1752587b48f986731b17e459dec5a72cf9559337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-02 00:25:00.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"18aed27eb20185e185d9f7a6f4e81e00d7fb791c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 01:52:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a907c2daca6746ceb41a45ad155500b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473ea5971c554c1c92a93b22d6b1edb2 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/96e53c392d5b44d1ab1d155a47ca143a : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65ff5e7e196d4aab849126dc5939c56f : SUCCESS in 51m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/045fda958a4d4160ba66e47dd6af7844 : SUCCESS in 1h 02m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a5f0f7f29c74e3bbe6a794ce2a602b9 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/625a2e7feccd4b8e96800061da47e226 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e8359883d2141e489fc5ef849528a2d : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/793dee9e444f41c4a11b74555ff1d6b4 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bc5a697c756451ab20ba0885f0623a7 : SUCCESS in 20m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9493939d824b4907b9e837ee72288493 : SUCCESS in 17m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a438e41007d41aeb1643824dcb42822 : SUCCESS in 16m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9c7b73b53ecf472fbecb24391041d4a2 : SUCCESS in 17m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8e14dc2ee009429885e60c36e98e649d : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31d17de973874e289f21a88a41fc1ddc : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6664c6e99f2b479989f9177a1f4a14f4 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d057b81d39a34dcbae740a282ca83910 : SUCCESS in 17m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/212e1a07c6fc40ca994848c3fab49bdb : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/84318180f2ad44fda8829fb31571808d : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c79fa913384b4e0d875a7ea1d3e56caf : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52cdbce123924a848d1bbd01dda7ff51 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5a58ba4b15042ce821e7e8d330d4091 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/48ce3ef631fe445099ec4820753b158c : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/02f7018d74094441859d3727ae9e758c : SUCCESS in 1h 17m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8842cd702d9f4cc6809b7ebf85776937 : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/518d3228ac454f08a4081c2735e343aa : SUCCESS in 52m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/656696d5e735424b9aa3c026bdf28f44 : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8ec9809caf472680e763ec904ff098 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dfbe1d2d5224e8d82ec8d538f52b8c9 : SUCCESS in 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a6e97c1e9ae40399187c7367c9f8a4f : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":16},{"id":"c8126a00d5b470f1f47321700d1886e342267bdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 01:58:54.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7689082eb09841db8d8f9cdbd3330d58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/103805624dee4a22a96fe8d4f3dafc8c : SUCCESS in 19m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31993006e4ef4059887044b5d4c64bb5 : SUCCESS in 1h 00m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2138c04c53774dbab66d685a55f38536 : SUCCESS in 25m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/71f29004e8a84674a46008298a443d1f : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":16},{"id":"670aa73401aba6e04f2634889d09313656760d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-16 00:11:36.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"354148f8a02a9f345ce5b44d6a50dd94bfa71fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:15:03.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0676d2c368844f5cbaa33fe72273e242\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19e282967044fc7a63cf0ac5b3557b3 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/44deef46baca4fba95668cb034c2a672 : SUCCESS in 46m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f642624ade04bd096ae76b6f06eac2a : SUCCESS in 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ca68ee94b6e460aa28d32064276cef2 : SUCCESS in 56m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/713a078413644763932332c13a437b2e : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04a10314f71b4452a08bc2f7ac8a79c6 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/117adcbd1a5f488d8569a94a4ebc7c75 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7b7f9280c4a4020ac0e2479a2c7f345 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85febfd64b9d48a98052e431b6f30ad4 : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5af8c17368ab4dd9a17f8862ad1428a9 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ae259049ffb47cea549c76fe93cb5f7 : SUCCESS in 22m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c9b1cb94d09846f488134778dbc6f02e : SUCCESS in 17m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8043033cdd4a4e028557428fc8a4c1b9 : SUCCESS in 23m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/645e53ae3caa4f93af1eacca0534630d : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f39414f3c44e46a58c0a19c135003a2c : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb5502555231466bbc5711cfb3c4486e : SUCCESS in 22m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fe642f655fc451da9036236bdd79b8d : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e70d6a307b85499abe2dbf1da78f2912 : SUCCESS in 21m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f295217253a4dc1a0bb4d254dbab4bc : SUCCESS in 18m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9af1fec68a0d4b85b76e0d22e802aeb8 : SUCCESS in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a415021746a04aa0b38ee01db697e905 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef4d8b0eb9a74e1dacff0affe94d8f8c : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6e199f41d054da29c6b155701a95a1b : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d4beeb712e542f4b9e4b4da19d519c8 : SUCCESS in 39m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a88437a1e6b4457b99a362c0ba609311 : SUCCESS in 50m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b919857c4754a1da901818c9eb47316 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01213ae339a421b91f5ca6eef713618 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/931a9dd9588c40d8a8cdda0bc07edae5 : POST_FAILURE in 46m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b0609353467492abee8e52fba95e244 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":17},{"id":"118548643449ed218f55cca1e5e58e7e75f9765b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:45:30.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/966963c939a3452fb940ab0c209d2bdf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71f24a6adc094132b427f0730c0cf2db : SUCCESS in 45m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e9305a19e374463a02031c5b43d1ea0 : TIMED_OUT in 1h 31m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdd231ec7ae948f5acaf974ab8a793f4 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eaf0e1418aeb42b48d1cea98eac8749b : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":17},{"id":"1d3f9ce20c40300ceaa96dbe0815103fdc2f3dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 00:06:36.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"181bf28d434c5195c53a6350be7cb5bc4be4529c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 01:14:18.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3fb9d966034272a17996f3906f4a02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9398ea8dca4b4b8d84b1a42be42a514e : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3a62107a211e4806b3f0c9526bb7d8ab : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7383cd19ffb4229a7e7eba9ffe6daec : SUCCESS in 50m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb2fcb7c17244d2d931bd32315d2675a : SUCCESS in 1h 01m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1d4acb0cc1a42a19417b53104d5b7b3 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34aba3b1af45454a95d0b444936bb66e : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/077f3927634b4b54830d78f702fc771d : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf78f7db2c9e4a879451bea061428be7 : RETRY_LIMIT in 2m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbcb83ed08454a4c8eec1cd67eca761f : SUCCESS in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a2d100ca7684562b2685cb42f390dbb : SUCCESS in 18m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/893c9c108ae34ae1adf4bd2c5ae18da9 : SUCCESS in 22m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3669266b2eeb44ecbfac1def437d6fc9 : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50c14c2c489147ca8000cc08ca84bdd6 : SUCCESS in 20m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00c57ada78e04328a48e0e2bb07c4d74 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7fe108a9eadb4bfeb17fdcc62f68d005 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f66fc08d39f040918df46ba24d2ae560 : SUCCESS in 19m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65c9a275d8e947088cb20d3156b7250f : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/91f0850a6e1549f888dde20f89ba5e9a : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38c690b05d024c85b39ea2b2b6ab5bef : SUCCESS in 20m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c9dc7566f0e49b7a53c7deddf9b4d9d : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f64a1b878eb4070922958032caaa405 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06afb11e838d42ea91cf2ef4595c2761 : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a061bd8fb8745979a9962a098aa11a4 : SUCCESS in 48m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16099350e9f94901b6c4db333b72b6aa : SUCCESS in 51m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd661b4acfbf4faeb0d4c9b8dbd9b2f9 : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6574e9b8ac904b399ac405bfe6c1af33 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4644b5b6aba0452ea6a52185fc15fa5b : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad5d5346f0d94593a453c31ed2cf0147 : SUCCESS in 16m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/002a40419613494d838c4665ac9b4dd5 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":18},{"id":"2acbf39cd34fa1c2d973f6dfcb10ae1189e1ea93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 14:31:34.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a715b5a1ae540e2b033759a126458a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50ccb8d9b7794b83a6b0cea4d5e76d28 : SUCCESS in 1h 30m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1405184dbe6947b090d65d98128d17dc : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c022efdc09c4fabbb166d77aaf840d3 : TIMED_OUT in 1h 31m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93e3d1f007674dffbbfbb960f05b6a92 : TIMED_OUT in 1h 31m 33s","accounts_in_message":[],"_revision_number":18},{"id":"9034ce5b5ee06e65975c2cb612d7974704a769f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 22:20:50.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"5b312cc78cf4f25b2731da9042aa322224bf6211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:37:01.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca7094f6b8cd4f9f81ffd5e51292a2cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20765333220448418424521d058f49ad : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/0c42bbd11aba46c7a515556c4b4d138d : SUCCESS in 1h 04m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb69ed5245064aa898be8a24c1f6c8c7 : SUCCESS in 1h 09m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db3096bf01ef4a3285af301cb28a5ec9 : SUCCESS in 58m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f20f977c9685455f88326c93fbf18288 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/461bbae234d84495b7ec29c6481c4673 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0548a3611ce34b2c9453feb7b603c997 : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ea172579b624952980481fc4c2713e6 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0b470626b0c49a4b2a84392685a226b : SUCCESS in 15m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cad6fdcd70db4634b51a08d86f76558f : SUCCESS in 17m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afe0fba48d444554a55c51dcf25e36c3 : SUCCESS in 16m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/68e205cf35ab401395b33a9df0266e9f : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2c686d49a62f400687be56d032f3b357 : SUCCESS in 22m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2de675a3c0e041c6abaee7f851ac90dc : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0eafaa505a8e432bb1ec5d2b9d4da328 : SUCCESS in 21m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ceae801df884febba01b1cc1767cc97 : SUCCESS in 16m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/96652779478b484593563e6113f777e0 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce6511c4821449299709833810b6494d : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/027e82bc99774500b1bd43502d08451b : SUCCESS in 20m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5ecb806f0044d188dc51c102123be2f : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/895acfcaa32f42529152e1fdd55d2095 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74cb38e014bf4bf9b56d0f263c25677f : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/441c5fa987a94e86b5db15907a1e2565 : SUCCESS in 47m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35d9ed90ad664b1fbb86636e59d431fb : SUCCESS in 38m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac1c1ff07c514206931e56e0ff62c589 : SUCCESS in 51m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b385bf0fa3bf4a709b1f35a8227dc39d : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1147087299a444292829e1df77cd2d4 : FAILURE in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de4a382c9a7e435f9c5fc0000ff96275 : SUCCESS in 17m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca504b7da29f4a7a8c4a3a1c2a16a3a9 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":19},{"id":"828dd78ca541b7356e82c47289c63d4d735ed06b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 11:04:58.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed35c927bcf14f548f047c721b562070\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64097ed5502e415ca80679bc4bca352e : TIMED_OUT in 1h 31m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6d33bc41bff489e8f7b99aef0afd757 : SUCCESS in 1h 28m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c733eee9df8446b4aa83393ba421bdbc : TIMED_OUT in 1h 31m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/87574e05d1f4499da831b8d26b6894a7 : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":19},{"id":"9fbd6ded99199ca45ed4964c10da17aa4164ce11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 19:22:17.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"7ec3ae3953314c7a6856fb5af3212b6ff6177606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 21:24:21.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38d0a4546d2743a79541d6e38f8199c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0506fccd65247a7861119fdf363f702 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/dca630add6e74541b1e625e619435303 : SUCCESS in 1h 00m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89843fbde624454594e531ba8859233b : SUCCESS in 35m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f71b123afdb4203b1ea3a088220df6d : SUCCESS in 59m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/294d8c73a0124f8d92d4ec2fcb6f223c : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89fe3062adf7413b89359d171500899f : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9890ffa59d7042f980b20d1e5d38d8ed : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ecd01e727e4462db2f148b1084696c6 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c546b1e015b460d855e4bef9a25b288 : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b708c79609f245a6bd3ef186f00016b3 : SUCCESS in 20m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5a90f8d7d3ee4393a4cff1bbdf8f1989 : SUCCESS in 25m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3319261483b346a9a280454f2704394c : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1ea08fe3532e45479cdee64a6fd29c58 : SUCCESS in 30m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ae7b3ac72734009be2005fb8b8ec983 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0104e76dd3f40a49670c418baa660eb : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b432d1d7b28f4c82af15af3ba2cbb0c3 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a4d19fb732a247e5921b0edb62142dcf : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1807d9f7a4545a9a1d451778f606ba6 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b425c7213d5745399de31002b757f5d5 : SUCCESS in 21m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba3a56e553e3465c9493ec11cf8bdb55 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/953b3ea4b12d481081ae5ad4a4dc991a : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4854717abf604d338302cf3202f115e9 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa081129d07c419cbbe85580b47de8c6 : TIMED_OUT in 2h 00m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c77f73b2e3f74f40b0c35a5b56658d31 : SUCCESS in 36m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31763910bed84f1897c267133afcc6ce : SUCCESS in 48m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b95f9a11ad784a4294124ad698aa5ff4 : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e1a9be6b6d469f82b4c55caa832aac : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8e95f11736f431e82ffce480d5093a7 : SUCCESS in 18m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba147c7d07704c5283af5316c20cd220 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":20},{"id":"f1296a8b46ec3ed132c17f841eff59e2a8947ffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 23:42:54.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"f68c686982124d22cfa6d6d7f5c04b37b4b8380e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 01:45:00.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3e0f4b6a438485984f27301bfa5299e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c72754d002149809f730d5fdcdb8490 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/42da68c339d54098aacba7e8c46ab490 : SUCCESS in 57m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab03a64f7b2a491f8fb321a5809830c2 : SUCCESS in 35m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23cc4b941d7a4462b2f382fc58fbde25 : SUCCESS in 57m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c382735d02364133a9b1db4e9820e873 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c383c20df9be4f9c9922bdebddd089c4 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d055f44aa1f4d9e87cf42c15d121fba : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/643025d40ce846bf8a3d7162955a5c43 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6840cf9829a9488c9d9a0eac1964162f : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/910159bb0baf459aaaca8f826324ac43 : SUCCESS in 18m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff3635fe578140889e3afc225174d48b : SUCCESS in 18m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/993cd1c6eecb4c90b4b2da0801b7cf68 : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/76d0020183c34de18cebcb368864d1bc : SUCCESS in 28m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ba40d7c17134648bad55c05e6a15fc2 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5472a2146714480b6b40f134dfbcd5a : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/baa9898035b140d49d0b145e224e3960 : SUCCESS in 19m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b3d89943e7e44789718317cc4503f03 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afd897b02eee44eda50b6149b2821935 : SUCCESS in 35m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c66fa07e68c8431f9f9911d57d1a89d9 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4898986481424cb391aa38092c9cce94 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f03c514ce8b46d8b8549bcea911ad63 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3db3180981ad47f6a782f071b89a97d1 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/38f256de1e0b4f928a00d7968b60853b : TIMED_OUT in 2h 00m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfdb879df44e497bb76499f4985bf461 : SUCCESS in 32m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40b08e3679104b92a5b95b382153a638 : SUCCESS in 37m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc9d79305d06434cbdfd12db63c8abad : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1a5c324de545c0a777ec28e425c26a : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7d70e246b4740fc950293345d4c97e1 : SUCCESS in 16m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c13c6298bbbe41609bfa2e82c1203f45 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":21},{"id":"107b28f8ae08585be17d52405b494dd660b5b0a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 12:58:45.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2dbd2b80d0d4176b1e6472c1bb36082\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8be838e265a4f6e8a157d1eb81e0440 : SUCCESS in 49m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df5432282fd44d3f83032bfb737e6b89 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ccd710218d84c3c932e95d7562b4b82 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd193c5bdb58424e8900e91621ce4e9b : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":21},{"id":"4a4bfdc52bbea15a1b98efefce00a3228592285c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-26 15:45:49.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"59055132b8c7c6b76def552edf0930c6eddff851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 18:11:13.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b0747c46a8947eaa9cb85d2f91d385e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d20a6b7a2554e6290657e7753af8832 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/3efda3d4e7ce455b807ffa2150f1ebe3 : SUCCESS in 53m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c972e05903084dfd983813e781df6662 : SUCCESS in 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8401f5191dc4f64b69720989667d968 : SUCCESS in 58m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6741f222b5cb4201ba8777642521a134 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeb439373f024184a69ce7c48db4eb5a : SUCCESS in 23m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9163319fad3c4912a257c6946ded3c64 : SUCCESS in 23m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b912284797644b59920c074133d43c6c : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81d50cb082dd4cb8a4e5ffb291576175 : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c752992d24204ea4ae8caac47bf0ad67 : SUCCESS in 17m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79d61459a86e4c909daaa60f7c8d4bbd : SUCCESS in 21m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/746c831dff494dcf8a45a1fea8641aff : SUCCESS in 31m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/83133fd5c3be45fc8a2ab3465236d47e : SUCCESS in 34m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8601427f59994f42bd4918593db8b9ad : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bafe7c0115d44e0a4e41d62d61b00e5 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7b9c7e0cf3f4587b85163056543b5db : SUCCESS in 17m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ecd937b532549a08198c9325d85cd59 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d661a9cef7cb4d4da3bec1c04536969e : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4935439cb96b485eb1f2b1d5ac4e34b0 : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/937fe097e0cf4ea6801e93091c50d561 : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0adc51a821b54bd992d18f9c71af3e18 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6882a2ff98e74cd0a8e69558672c137b : SUCCESS in 9m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39d899315b6d44e088f7755806d2324b : SUCCESS in 45m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0eadf2500a974982a722b88364b8739d : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/897255214bcf4f928be03529f49ae799 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fde114c08964fbbb4757d6f2663e38f : SUCCESS in 21m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b2d744f0024f3282db1f188a139ad8 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7edc14d449a463580a595538af2ab28 : SUCCESS in 24m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f64fe8500f694b539415e81bf3691e62 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":21},{"id":"d5e513dc43bcfc19e25d53297e367a185600c759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 02:02:59.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79ef3543f314aafb5aa90753e0b39f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/752daf0fdbdc43429ba718af0e5077bf : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcf28f9e2b9f46bd9bb3f68da55a2cf6 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/886652ca819f43b7be108ac95a815584 : SUCCESS in 1h 07m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5ce3c5af9bc474fa4c510156023d854 : SUCCESS in 1h 27m 32s","accounts_in_message":[],"_revision_number":21},{"id":"dc539ec7f497091c468df28278917afb7cffe5ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 19:24:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"fefb7ef2993fe7982d9688d5023431bc908ddc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:36:22.000000000","message":"Patch Set 22: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1bce0dbdc5a4a3ca5ec5ab87489360e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3e6a63c7524fee81cce262f6c201cc : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/93bcd6054b4c4f93a59c27d5badaa249 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48e32b1149664c8588f1762c803922f3 : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb93bd4ff90342469c71f8ce671151a1 : SUCCESS in 53m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b3c99a68a3d46a688644b812d38e7aa : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cbe4cd83e2a4e49823af4c78cec0d19 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86bf80be32464ccc9be08a8370378be0 : SUCCESS in 18m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/99116748172a4bc18af85c1a60856708 : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/242e2f6beec04af1b96e6283c99a7173 : FAILURE in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b078c2515c354ec498289fffdc01575e : FAILURE in 15m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd32f7f1df8f489cb7b0f3e5bc3be9c9 : FAILURE in 19m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/67871f6b6a53436095df23356c3572d0 : FAILURE in 18m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d3814a21254437c8f7c80cbda592a77 : FAILURE in 25m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b30215c967c41f2a3c9e359028a65b0 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b75bb05d7799438984a29be380cadc9a : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53ffba4ce5e2420a8cc09bad7a2a25b0 : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6c05788b06344344b0b659e5edc9e229 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dd59af4a83e742b8b38109a97658caef : SUCCESS in 15m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fbd996cfcc2447d5ba6cf111667894d9 : SUCCESS in 22m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69704f18e3af4222a4a564360995b322 : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6935d94f0e814ab69628272594660a35 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33983a76ac1a40cbb7b308f37aaa2bd8 : FAILURE in 9m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8721636c91ad4bf2972e5248200359f1 : SUCCESS in 46m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c543f3178ee343e799909fa88ab0cbd2 : SUCCESS in 53m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4a62fe058ec495899b70229bb19e035 : SUCCESS in 47m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/193791d43a4c4cbba87365e2115cae27 : FAILURE in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35df1b21ce9e4489856bc50be55a8b1a : FAILURE in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f10a9f3a4e7344e5a639657c7662289c : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc7f533b24e1476fa52148be716e1842 : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":22},{"id":"9cf89aea79581033d9ecb69253449002f00ce5be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 21:10:50.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6c0d0d23cc4b11ac3719f3002c25f5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/936169b45d0c40fbabe9357cade86695 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b41f705ee53440068dc492f39b6e8cf0 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b6077e09ad44936a5969f3f60944ec8 : SUCCESS in 1h 20m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d3fce51750d4adb9ffeb73b3d5beb7f : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":22},{"id":"b2f696a42ce2aab874dc87fe5219563b464956f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 00:27:19.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"d5efa580cf250017d21cafcb43945c13b18ea46a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:41:51.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e3e73252c424f749af9dfb53f62c5b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9449d9c6edd647d19c9ab9613f546cb9 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/9f37dce07dad4eda812efea29f4385ab : SUCCESS in 1h 04m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/808c2b8ccc0a49d28c3b250a82df4f21 : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8605884d670d4a948afe189bd7b27ea4 : SUCCESS in 1h 02m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bca42621b11476990e7d272015a08b7 : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a1b24cea92e47bca03650a344e09a88 : SUCCESS in 23m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42f8248c3ac140bf9662129fb89ce6c5 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/296364c727124a74a9885c9e3b6cefed : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79599613cfbf499cb8b07678c1367328 : SUCCESS in 20m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbdc8e6e0454933bdc59b7bc27a8289 : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a9eb05037043019cd4518a7ae3697b : SUCCESS in 21m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/45a77c580ab84495b8e1d4d6283ec1a3 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/47bc54aa91f647ca92c30cdce0fb4471 : SUCCESS in 20m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d49f436e68da4f0b97e83bb1620ada78 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16ff3f6cd75f4ffa9c695c6466807b8e : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8941d39b26ad426ab90f88492ba3c233 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a8feddf88b8417881e9a20a6f9183ce : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0163173db3a44ea881ead0433ebc0f42 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51b1750d124742dfb4195476ff948f80 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ece1cbc842f4e58bb14aa2df442f6ea : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7ca0bdc1ded4ea6a95e8a535f6abcfa : SUCCESS in 13m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/85cf165ce49f407189932a7f31bffda9 : SUCCESS in 9m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/84f500dbd5094a0dabc7ccf5343df9ad : SUCCESS in 51m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76d7fddf5cd444bfa2b429fedf11526b : SUCCESS in 48m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07a63471f3e240d99fc69796685b31c9 : SUCCESS in 42m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8deff2964ec843f99ba82626be4a5d59 : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381ff5a694e04175b768454b2c97d041 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b31315ffb964e889a81499310545fd7 : SUCCESS in 17m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ecc8792c50845cb86e1e42e5c5f0077 : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":23},{"id":"a1f02de00e0fcc0c1014939c2f149800c274342b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 03:05:27.000000000","message":"Patch Set 23:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ed50cc42884250889613e6a4f8f60e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1582d01860df47df82ef2d24fab93b1d : SUCCESS in 1h 05m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/514b168287a64da1b23582a5aea4eb7a : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/97a30c8d085d4ae6a10fdac3207ab9b2 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c094c662b211451595c8cf674873de95 : FAILURE in 1h 27m 15s","accounts_in_message":[],"_revision_number":23},{"id":"60a7a879cbbe6c538b05050a81999b41cc7aa9fd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-19 04:05:12.000000000","message":"Patch Set 23: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":23},{"id":"0f541f74961c730b0489eef48528df8a9f2c2319","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-19 04:07:47.000000000","message":"Patch Set 23: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"489d8d25ecfe1702be4a6e9e5aaef3bf927a5c6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-20 23:29:09.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"1c7f1be92cfa58ee83c8965a246b9b9dc853d39d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 23:39:29.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d68a5d09e948a9ac2fb706b5879388\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa5c2849a6804990ba74c9af767f0429 : FAILURE in 7m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68952f23354a4e058c8bfa082285b205 : FAILURE in 8m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae21bf1953ce4f5b9696f25f3ffeee32 : FAILURE in 7m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/452565cebb554d20a9c0615adc94eaa5 : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":24},{"id":"96fef65718094539191de192cae441335b5c80aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 01:35:15.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cb13b91a92645dcaccc7c0367cf2981\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac462b7e3d234a09bfe1bbf18c9a1be9 : FAILURE in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f8348d7adf0d4645b28a4e2d4439ad43 : SUCCESS in 55m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2fa339c3ba04a8aadeaf6e8320a2ac3 : SUCCESS in 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/406ce63573ff4de7a0cb0d7cef984ecc : SUCCESS in 17m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/285ba823ed8846fb9660d9d040721f77 : FAILURE in 6m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e3741c2ca844848bc2de1a67e08d921 : FAILURE in 6m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc7b05310d6b45058974ad221f6c6e99 : FAILURE in 5m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/710a53d9fb0b4ce4b70a6e5599d9dc15 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/280301108b0c471d97fe3c9446236d2b : SUCCESS in 20m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57efa1f908ad48c8a6f0c76f164d47a9 : SUCCESS in 17m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a898b4631b845258345caed9f7a1740 : FAILURE in 4m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9a31ddbc96cc4830a09fa2ba5b3cb9f4 : FAILURE in 4m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6e263589fa54686b38b184aa86f7bd0 : FAILURE in 7m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/723f31c7a79f453f92f2b5a1f92fd67d : SUCCESS in 10m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc1c68fd9d874104b46489333479ca76 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f88d40c289f2452cada46f6348f8db9c : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b403aeb2a6e64b089694b246ce26cd1d : FAILURE in 4m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9a31c47e907a4a8093473c0bb34819b3 : FAILURE in 4m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bdd80aef5de141e9acb15a869e19b084 : FAILURE in 4m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7487157ef6044be7a957523aeee75641 : FAILURE in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eabeeedb49fa4257b54fabf419ed40c9 : RETRY_LIMIT in 3m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cc652b795ee741b181921d655d528308 : RETRY_LIMIT in 4m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/87c6976c85d94f429cf27f272984d11a : TIMED_OUT in 2h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1774832e0c0d49e49361a21701eb1e6c : RETRY_LIMIT in 20m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2e3e33f66144e9395379fca2dc14de9 : RETRY_LIMIT in 19m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3102fcd01cb24bd89b908985a83afa21 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0ca7b3d83544ce829d4e407a7d9c5a : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a49f3a4a9044ec7a191b73a1e1457fd : SUCCESS in 17m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c4b5cc368e84284b3bbe786c460e915 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":24},{"id":"3285b6566002eb2a3a7c6a7229d7e9988d53adde","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-22 23:56:13.000000000","message":"Patch Set 24:\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"7ad44ab3dca981564d4cd15ff51acab38c1573ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-23 00:06:46.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"0ff6c2ccc2886ae17a1a9629460bac22b04c6068","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 00:51:33.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21319a30469d4de2880cae7995767136\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a616790b7fb84be9abd33146a989b20a : SUCCESS in 40m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08e8ee9e3ffa4357b1319af964f2fb04 : SUCCESS in 42m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/39086d680fa945f1b8fd5cc77be5b09a : SUCCESS in 38m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/df25ecf3c41b4b989f643eac55225a27 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":25},{"id":"7bf0d12dc9ce80a709e07b0355fec396667ca57e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 01:28:13.000000000","message":"Patch Set 25: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a27009879c2044688689601b686281db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfeac471b7124d22aea9f2c890a198ef : SUCCESS in 6m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/9a3e6fe5f6274fc89c9c286c631c41bf : SUCCESS in 1h 10m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89a89c007ae74519b100b0a18bc50baf : SUCCESS in 37m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6bb3bfbe9cb4980b760931e39d17f7a : SUCCESS in 30m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/424c9124f31845d8abd88eec8786744b : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4746d6e9e2554daa85b121f3e54c4ae9 : SUCCESS in 11m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0abddc95c1ca4910a253210b40f69126 : SUCCESS in 18m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bae7cf66d3a9485ab613d7b2711b7584 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07e996ff2015432688e1a7c4c2a6e682 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c8daae5ce1047548efb4dd176662e09 : SUCCESS in 15m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e1b5ebeae14f8f88baab73b542f436 : SUCCESS in 17m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/10e1f54409174502af76b89f0e6184ff : SUCCESS in 27m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a5c3871dd53419aaaeed3d5e2749e64 : SUCCESS in 17m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7fd964c5732b4340bfb6eecac44dfea3 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be3dcfc0a9e14161b29189764f14fdb3 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0519b04d56a444fea900325df5705fac : SUCCESS in 15m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7c6ca9c594c84c7291629c96f2bc28b7 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fd0384a7dc9d4a6c966eacd22aebf55c : SUCCESS in 14m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/56de4af372ab4eed98d8096238ebd34b : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8bea662069ca4eb8aa739a8218dcaa7b : SUCCESS in 4m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e4221cfdb294ccbbbfc630bf924bce4 : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18fc468d488242d2b30142025a505cdd : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dad4382622c4950bb6c8cc567dca734 : SUCCESS in 42m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18cea2df77bd4da09cef85857524d26c : SUCCESS in 31m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6c8536e678774381ba97d3d210016c6b : SUCCESS in 32m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb5106ada76845b6b143e90bec911ed5 : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7972ea2e955b49e7b256ad3e515737f6 : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35bf28462ccb41ceb31e76bdcdaba8a1 : SUCCESS in 18m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dbc1f8023da49d7b053b60bbfb08008 : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":25},{"id":"ba937674c476559596d07533aef53549badb23c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-23 23:18:49.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"3a71867f74ef00461057a4c553f75adbdd28feef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 00:24:06.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cffa9a6b486c4cff84205f046bc27706\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08e7fe165b8469d9f820a0cb17bff41 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/ed3b99d5e3e54519887ebe0ac4e4cb4b : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af2f724754f7433591b9120043ed2cfc : SUCCESS in 40m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2335362f6f5478fba0b21f0cf096bf0 : SUCCESS in 21m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/971377678a324997a64853c62680eda0 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed2792e8dcb045d58ea1ed50cce685fa : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b84e9f9ddcbd457399ea70da0bb8ab8b : SUCCESS in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f0a9b3dd91d484e8527cf228cbbbf4c : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e03d26a0941a470b8bdaab44e5e64b0c : SUCCESS in 11m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1b32a8696fb47f88573619f51a02fb3 : SUCCESS in 18m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1af2845ab544276804b8b284ca8d74d : SUCCESS in 21m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/de313c98220a4b7fa044810f9928755b : SUCCESS in 23m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4693079eaceb495db78f0a40edcf1feb : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2a1aaef2d264a2e808b361d46f22e75 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1385bf1624f4cbd87ad789f2737f652 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d2c528301c640718cc9199f61ba4391 : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32f8c5db64f34608b8478e6e1561105a : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a2b1946b10aa4af7a67fe9dae13746aa : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8a047ec95ce3470b9fbd64c0ef9a03bf : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5bdee8cb75424e729d48c1b8fe73b504 : SUCCESS in 5m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4f0952813c74fa7b3efe5371a518c10 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efc0079cb431472a877690776a33280d : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f02775299c84b499f6d4dc1b90de4ab : SUCCESS in 47m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d3e0de8cd1644de852c26d5f1713849 : SUCCESS in 47m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4accb113659406498f66253f725aa22 : SUCCESS in 45m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ac6f9eb4f104ac5a3df94f76d7a20ea : SUCCESS in 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a352f1b1d764da09576ec6434ff5f10 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9037b9dd9bf64680b6ebbe2b02e9ea84 : SUCCESS in 19m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4b331aa1f37495ba890b5ef474ef07c : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":26},{"id":"fb7a3f781f01db9e6ea70406291cda863767ae9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 03:56:11.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f7bee3eb9047928a0e846109df7737\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72d4a1147a94450aad39f7e15c214f97 : SUCCESS in 1h 51m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4040c0400704118a3b31e5936903075 : SUCCESS in 1h 06m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/08556b05db8e465da60dc9ee1f5e725b : SUCCESS in 1h 44m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4972f45640a549839c01230cf7e2893b : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":26},{"id":"f687e11f2a3bba57ab2c4d1825a7ed02f3bd56eb","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-27 18:59:47.000000000","message":"Patch Set 26:\n\n(8 comments)","accounts_in_message":[],"_revision_number":26},{"id":"b102334cc79deab8600539f40bbd558d9575c11d","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-27 19:21:50.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"1abb82cc4130103962a1c6244603bdadc650ecbb","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-27 21:58:27.000000000","message":"Patch Set 26: Code-Review-1","accounts_in_message":[],"_revision_number":26},{"id":"5b3ec17b057783664dab1504bc7043cfdd833fce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-27 23:23:07.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"73205651696a9e9239ad262ef5f8240c94a09736","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-27 23:23:29.000000000","message":"Patch Set 27:\n\n(5 comments)","accounts_in_message":[],"_revision_number":27},{"id":"f25918135c31fb7a0bd151a39f3605178661246a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 00:33:25.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6176d68a33e4c0d94b9ab5e1f93f259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa95dab00624396947ba8386aa6dd16 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/bd68e6c039f04b5c9c84cb3d8ac45327 : SUCCESS in 1h 04m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b87cdb7d04c54fc08ebffd512615f845 : SUCCESS in 39m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/213a7991df224b4eb9e31f6a15690d78 : SUCCESS in 43m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc7a3adb084a4d9388666bec920011b2 : SUCCESS in 18m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50a55ed31a9b4a28b75a22dae44a4267 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d33fd8167eb1448d9cd0f7f850c44ae1 : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f300c9e930b747f4b4e2464dddac8ba9 : SUCCESS in 19m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4ce977885ba41a49ad946caacf74a6f : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bddca968168c4cedb3efe2253285c259 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/837af0fe69b24715aa2c8bfa49ee113e : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6c0570ad08c4d25b98aa396e3babfd6 : SUCCESS in 15m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c17caa62e1304c0db8958e8a97550a16 : SUCCESS in 15m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c110838dfea74b2093d84189a667548b : SUCCESS in 24m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64fef22edf334290b86ee1754aad5e63 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/160e2c3798634d0cbf264d4e645e9621 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/880ff9f779c54307a384165f4f03e8f5 : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2be51791d9524aa8b77b74d74971e982 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12408d149dbb449690d0e63f5d086a94 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eb319f2f2834469f9625abe0ac23ace2 : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/14fae58986624840bf72492f3ed366d7 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e43c1ae62a5d462e8b6d8859db070ace : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7b6adbbdbf614b98a603b60bdd31aae8 : SUCCESS in 6m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea648a247b644f499bbacad57bb2d311 : SUCCESS in 51m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d1fe85760ae4e1884dfdead7bd931a0 : SUCCESS in 34m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a2fd1de61cf40d9b32765cce34970f1 : SUCCESS in 33m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29047f705609487284b35ac8ef52b263 : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0969284bc2074832a331c41adaf39a62 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08be897882904f898271c9c67d27a013 : SUCCESS in 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a7a1b4c50c4425cbf4c6fffa26d7e0e : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":27},{"id":"147267e9b39f06801deca52a128677d330215bb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 01:45:16.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf31e0dac2af4c2d96e39d40a22095b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d67d9ee861c3480f8e1aea74bee492b4 : SUCCESS in 1h 16m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14df3dbb2cf244548a0bbda6ed4d463d : SUCCESS in 50m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7749c7ca07954dc194502ea7f3c5bee1 : SUCCESS in 1h 25m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/50a78d2ea1534a4997bca2a1b744018d : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":27},{"id":"d899001525149505b545d40a138846fd2caaddf9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-30 16:04:10.000000000","message":"Patch Set 27:\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"ba2bb8d51cca76951e013e84c6de9660fa104020","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-01 20:37:34.000000000","message":"Patch Set 27:\n\n(4 comments)","accounts_in_message":[],"_revision_number":27},{"id":"583eb2fec49a7293e3e76251d8582cd4e527304f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-01 20:38:16.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"6c1df16607ee680c4b2ca09b8966c2b6dc1f6193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 23:18:11.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"f7bd16103dc63616c82e112500730e49fea532a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 01:24:39.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45f4fc63e27346aeb3bcf9e38b2a0aa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95a4241ca57474695c7f0745bca4dd2 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9aa49823c8ca4ae89ed09c32feb7718d : SUCCESS in 44m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/822799bff50f4195a61484eec0afda49 : SUCCESS in 48m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a9d1a8025054647a7120ef22526f1ab : SUCCESS in 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/661a8fe82cc34f0fa4b7fe0b1ad7e35d : SUCCESS in 19m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54acd3f1322c4227b80fe1cb66223402 : SUCCESS in 10m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b6844174ff741789a837f91f3644c46 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba1717e030934db3b33cdcbc6c6073d6 : SUCCESS in 14m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ad4f510b18ed4743988ac7eae9a61969 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a8b738d8f5243f1b434319df5c12f2e : SUCCESS in 18m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24e297f03b81403c8c16a617e2b28e83 : SUCCESS in 17m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/617c888fdabf4bdcbbb6ac6533d3dc8d : SUCCESS in 21m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c1dab8d003b94485b257139f03d9fbc7 : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f662f0f9bdba4990ba7c7041542d790a : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7534d315124f45fba629b8bdfbf44fc8 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/471691524ca64280b6ccfba30643dd66 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fcba2f36b38f4cb296a1330d4fb5e8a6 : SUCCESS in 20m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/804dfca9755444eca2d8976370820403 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92838425fc71477389cbce10a3affe27 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/675c7ac2a90c42219d9cd1b8361ee74c : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/527ac7279cf24508baedbaf7f54beb7e : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0824a9d407d4f0399de1a6db3895fee : SUCCESS in 9m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04cd505f2e7142179d144e8e3d69d361 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4813d88d022475593eb0466b6f89d81 : TIMED_OUT in 2h 00m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b11af6ae9137476a8d8421fc4fd48201 : SUCCESS in 44m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/167b2907988f4d43a672a4a017f1d238 : SUCCESS in 35m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a50aefc65b54ac0a887859078f7612e : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d14af8cf3e54a57b0be1d4e21e06f0b : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e9597fcfea24bf8a74f8baeb6cf232e : SUCCESS in 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfdfd1a60ef445579d0c066560316173 : FAILURE in 38m 47s","accounts_in_message":[],"_revision_number":28},{"id":"0037a64aee23bbbbcb2c40ce1faa10562e9b0440","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 03:23:59.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef72652b934545e992c127e85955b387\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2b8dfe17a53492d8a900e7fc8b107af : TIMED_OUT in 2h 00m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1702bed2a5994063b008740d818059f2 : SUCCESS in 1h 19m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b26548f0dfae4153a2e57e1262a32d7a : SUCCESS in 49m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/70752f358a24400cb7e2a3d51fb2a8ca : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":28},{"id":"c354795d8046f6ee70bcc8cbb67fe2af03f2431f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-02 15:37:07.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"fc33739b15650fb339dac0d2e21f5d1971509a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 17:03:45.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1ce4be7e8d34d96909580992f78eee2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b4ea32e9ef47d3afabb127ea1a8ea0 : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/a40a1c9c4688413f95a0bdeac19ca687 : SUCCESS in 59m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa808ffca8644ee49f9a5cfdcb32ba82 : SUCCESS in 1h 12m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c793df41a1c74cf19a7e10b711d17c28 : SUCCESS in 1h 02m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d5aeb922bf64ce9941651439801889a : SUCCESS in 27m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d219f8a8fad4828bf321a7186dd6ebb : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60c41b3ffb9d45e9a7b7715f268cae42 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34b46190c93f47e0976af4ecbe30866a : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5401845a2e65422d858cccfddafaa0b9 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f2d428515694cd08c4f55da13c6f536 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c812a96e145048a68ed7c4d71b7a59f4 : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b33950fdd3824955af76a06c6b7241a4 : SUCCESS in 22m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e8290215e244ddb876a15816f3a4a4f : SUCCESS in 22m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3be557e23f1241b8a2a790e39cb006dc : SUCCESS in 35m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e08767f592c4294b8d29a4ee9c954b2 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5f6ace55d0c4dc6a5ff4b708839a158 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac4dcc5162b94100945525e3edcc7d49 : SUCCESS in 10m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/89e51048c28b4395aad8f2996b63480a : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9354f2b31d9249929808494dbbd8c1d5 : SUCCESS in 21m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5bc5c1384714cfdaa7768443b62f177 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e04626d852064d05b03f7b112e9dfabf : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a19e7a9ae6b4b21b0c5ecb774083abe : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2487dc3eada0477fa362e4ad92784c36 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f641e4a9d2ee489c8eda3c722bbd315b : SUCCESS in 55m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/797212dc2e3a4a3a9bb4193cb4c3118d : SUCCESS in 54m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c72e9f6ec10b44dfb51e104745c13192 : SUCCESS in 1h 07m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0d9cbdbb2b84af3a3d94bb82b38fc2e : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487aaa77c1774055b58898abaff16af4 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbe0051b9a734480ac1a526ad0092d4a : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b49624ad37b4bdabea20f90bbf1dba5 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":28},{"id":"f84aedd99be9b1ed050b99b8e7d82067ebfcc6c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 21:57:32.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634a5b2ee8304a84a3643d533e74de2e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f84a7737002479c81a30b5ae8f8577d : SUCCESS in 1h 36m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bef338b6d8c147a28e396506716524d6 : SUCCESS in 1h 09m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa475be35a874dd7918bc54b78fc7d73 : SUCCESS in 1h 10m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/20dc283a1ce642518293e944f13f3000 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":28},{"id":"23e3c8e84745568961c2aa7c6efbb4ed75c5c5de","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 15:13:44.000000000","message":"Patch Set 28: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"78b691ec3360d40dab984f7be7f106ab7bcba719","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 15:14:40.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"5bd189e92785256f7c8ec893aaa5a652899f28c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 20:09:51.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"2a3495fc81cdaab15154ffdec9df721545c29f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 20:10:30.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"5a0593e6c952f0340e5f28fc62105609cb3f9700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 21:32:27.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86877d2098bc499089b2bda0c0660bf5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/adbbff6199ae4ef8acaaa9089f608acd : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfe9099e28b4469a15037e6a0ce734c : SUCCESS in 9m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa838a75811a4688b19343c7ec6da3e5 : FAILURE in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9192c50fce04a15b8862894cd197b4b : SUCCESS in 5m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/bd1e2d6177b148799c831e0972ab872b : SUCCESS in 49m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8986276a0c35467cb83e7863a9a10f0e : SUCCESS in 1h 07m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b845fd967ecc4c3c89f4031c5f95f840 : SUCCESS in 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6e7409c45134b088350af7080a88442 : SUCCESS in 46m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f098cb8713f413292d70dde025d5782 : FAILURE in 15m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/990f57ed08d14b7c806d26ca3f2b15d8 : FAILURE in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9d876d03a094d2eb1a20495b09f0b2a : FAILURE in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbc0ac19f3eb43b2a35ac0e50df25804 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55d622e1a62c4b33a229e31c32c288e0 : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ded9ce93c4b44b4abcf3c7409378dc7 : SUCCESS in 17m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/882d746eaec74aa4ba6953ee8d2a54d8 : SUCCESS in 10m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9191007db3684ec28790ee2b59135d01 : SUCCESS in 21m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/000483a58bd241c582c76e32d3440289 : SUCCESS in 13m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15c67af3d4ed48c48c42d9e8fac3335a : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74c8e132bfb443a099f07dc99bd31117 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1352a4a69e9e464d880a35d21f480e9a : SUCCESS in 11m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75e554631c8244318bcd3558e0ae7c10 : FAILURE in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/49d6d836872b4e7bbe20d3ded210c252 : FAILURE in 13m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6e93fcafcf04d39b4dd8a31d4b48b5b : FAILURE in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ee2b55cbccd444ab3f9d5a7ac05c62d : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ceef4d6aa7a1483a97e56ac6d1ab5cb7 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd50018f240248b895b2ab9ecefb6857 : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4f3ab8b69fe49cab2838bd12fd7a3f6 : SUCCESS in 47m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e771bb0c4e1941a0ba4dfbcb0727a6d6 : SUCCESS in 38m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1d885e9a6c4464fa0d0267c0be9b0c1 : SUCCESS in 53m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2ffb0f6bd654af884aa86a18a928c37 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bab5620db45477995ff224b270e6cb1 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c75aee19d444a21890e0946a456344f : SUCCESS in 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/322712d9306b450e980d6e1026269fab : SUCCESS in 38m 21s","accounts_in_message":[],"_revision_number":29},{"id":"12183cc75403df0b941b565ecf7524b04b01e9dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 21:35:48.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e182a5c89689443daf7e85eb8920e8d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25311d2fc0894f2caee7262ee37f5fdd : SUCCESS in 1h 23m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17a235652da145488bd338ede231a37f : SUCCESS in 1h 19m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c37c805951a84dcf898f905bdcf52a39 : FAILURE in 1h 22m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/747611f3b7df4e9493fd139b88914444 : FAILURE in 1h 09m 48s","accounts_in_message":[],"_revision_number":29},{"id":"251824def997435b575884223cbf7c00622054ff","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 15:43:06.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"6b4f1f01dbf8a240c133d47d8bf8802cf0c244c2","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 20:29:12.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"e15270cdbad82fc14d767054c3cfd2afef4d29c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 20:29:50.000000000","message":"Patch Set 30: Patch Set 29 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":30},{"id":"e8bd38439a0bcc766d28f46d83b39ed57e8c88e4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-09 20:41:39.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"4a403080dc0aa817957422544ef0dfcd7fca9ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 21:39:23.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0ccc6d4606c434083c2fc55b61609da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c473c5cbf19341599cb4130cdc02e447 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a08a8ce09e44b09add8be7a66f47f6 : SUCCESS in 9m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4533807041024c3181b36574719d850c : FAILURE in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f2861a513fb41cc985d4d6f45f986ba : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/454da2c3de8a4331a9638f1029b4ee47 : SUCCESS in 1h 03m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4f7bbad64684d14afc4445ee0de73fb : SUCCESS in 50m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5026150d35874f8a8717c18902f95836 : SUCCESS in 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4df13298fe284466a6c6ee53c60fc9e4 : SUCCESS in 1h 01m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e6756b6a1d7401ab5505ccfbf7bbe51 : FAILURE in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c28fe11d73d341878e9f08819809eecb : FAILURE in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f6b7527afcb4dedac781d8c8d7f33ea : FAILURE in 16m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6ddd201579145c4a53e2c32139b2ac0 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b15e670f37645988a43233d07d36db6 : SUCCESS in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5d086f06114a4f862ebbae5ed45267 : SUCCESS in 15m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ea011d5317479b877dd2e9699c05fb : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a54ff3a0e1564a4088536eade686c90a : SUCCESS in 18m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74a66a05a19c47c18cfccac3521fe4b1 : SUCCESS in 25m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2e85875b0044995904788beee0c05c4 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e4e5dbd2b6804bf2b09e94a97a2f9bed : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2727f8a3eec24980a4e0322216156ea1 : SUCCESS in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8f7840b441e64652801cdd6d4cc5041f : FAILURE in 13m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/52eada6be6de42558923d1b35a25f509 : FAILURE in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f70674886034c698ff0dd7bcf5288a2 : FAILURE in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/658569d4810b4bd79d6453327a78e1ec : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93d9d75147a148aa8ce71efae8940da8 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0bc1d1d78184fca8ab32fc69c8eb234 : SUCCESS in 7m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2a50d68886f45a6b66dcb7aa9e52631 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0964f1e912d4f68bde3afa29f93ba0f : SUCCESS in 29m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f570e321d8e945dc8391734e31fcc7a0 : SUCCESS in 58m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ac62c66f5324c0cbc8145bc342840ac : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae1d8a3cbfbd453b885ed4eb9f049e9a : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ecb24d9433a64bd59ab757db63671ce4 : SUCCESS in 19m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ea774de6e3f46d7a689ecf2ab341afd : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":30},{"id":"56a21b1e1d1a84c77736d91d64f136b883b2f4aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 02:51:01.000000000","message":"Patch Set 30:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39dcfb33aed4f8ba1ac3459f4253431\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/966ec983ed9e4751bfab26357c967287 : TIMED_OUT in 2h 00m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e94fa443d69449dfb8429ae5e452887b : TIMED_OUT in 2h 00m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63346ff6e558497e99ae433c9439fdcd : TIMED_OUT in 2h 01m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7aa1407688594e7289e728a24ee8971b : TIMED_OUT in 2h 00m 53s","accounts_in_message":[],"_revision_number":30},{"id":"52319526d88105fde797bf284e5e0d6924b52772","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-10 21:38:27.000000000","message":"Patch Set 30:\n\n(3 comments)","accounts_in_message":[],"_revision_number":30},{"id":"3ac3b96bdd2b8a18899e226b17f907e45c0d73ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-10 21:54:21.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"4acdeff85a283671573279d519b80bd0bb265a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 23:02:03.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b4b9d62f67418dbf9186aff519fcd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06519e7559e149f089f8d9ccd2694fb6 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/728fdeeaa6c24149a28d3883a91543a7 : SUCCESS in 1h 07m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2018de74d95f4dfd903827fcdc996dec : SUCCESS in 44m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02a3173fbc6b439db0fecc3c89b33403 : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/478a807cd7f74ab8ac068c935836925c : SUCCESS in 47m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b06147d5a6334443846dff4ebb7c5d26 : SUCCESS in 13m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e949901e5ef849388394d223c6d8c3e2 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f922ddc898042b0b5482e15b702dab1 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/272627fdd5bf442f8fec0898c5074a24 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d07eaaed65a4c57a43c30f54370940e : SUCCESS in 22m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3e054f9ad64f3f8e42890ee25cb2ba : SUCCESS in 13m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5d5226318ad84239868afb1524ca2504 : SUCCESS in 21m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/641f505bc68940e3a2a760f059e6dad1 : SUCCESS in 13m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/799825a26a5247289d5d2f57436aafc3 : SUCCESS in 21m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c872d6a496341ac958d6ebe8eb6a043 : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96bb2ca15f6f4e38881c04a56e79dc78 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1920f4d08eb3407b84438bb852d2ee43 : SUCCESS in 17m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/36f290bc067d4399b89fa20802dabf86 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71e978e19f1e4f8db3c3c277cd612590 : SUCCESS in 21m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2aa0bf2dc6bc46f09c0a119dee892a09 : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2daff43f38f747358c65edb9b8f3bc90 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/752beae3cfec407ca538e06a96f9bfcf : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4fd586793c947eeb6df2bee9b0c0325 : SUCCESS in 9m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d5cc66f653e457b9ede6122ab3f2f98 : SUCCESS in 54m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/581c91b7689a471ea795fbc91c33e57b : SUCCESS in 45m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fa20091d37d4f97b81f43d1e970e908 : SUCCESS in 49m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90aa58c7503b45d08fcda3757d9c3156 : SUCCESS in 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f681a502e0d94ad895134b803b68dab8 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be5af9d2bd9242069bca38bae3a4db1b : SUCCESS in 15m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff1d7ef3713496c8316ce6125d4f3a5 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":31},{"id":"55714894c76de41b296d2422f3f2f7e52589bfb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 00:58:08.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2663dc36b4c642e8b924946645a8b87d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f514728231034ca1abcdcfd49772545d : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":31},{"id":"4db1aef7b05ddae71ec0ccb0b6994ed48a4cd9ff","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-11 15:49:53.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"763bc2bbc2c6fe7ac41666dbbd5350b0280ca5c1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-15 21:01:05.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ca0e3b9fd32a7bae60d5fe2d7edf933cfee3538a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-15 21:01:37.000000000","message":"Patch Set 31: Code-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"210374c95ed3810f7c8a713598c6ae42f4bf6e34","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-15 21:51:21.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"1c61f7a2af28ab33832c06fce8962e0773cc9566","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-15 22:01:32.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"04eca628ac84e6f0e2543dd798f0c9e3211f86ec","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 20:27:54.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"66edcdc5b9e8cc01eb4cea957d0ca2bdff049c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 19:28:21.000000000","message":"Patch Set 32: Patch Set 31 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":32},{"id":"5c3d1799b1388b284ea4b77a1b06adae554ed5f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:35:04.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4588adb48b24c6688d6ce9a74325a6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52437f3491a64db68cb8b30ce4cb147d : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/fe75e6d98d63496c9dfefe16f72b09b5 : SUCCESS in 1h 05m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e54ba51fac1b44f4bd92c36b6a1dd033 : SUCCESS in 38m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/605f3476cf7e46be9c68f2e75384ca3c : SUCCESS in 53m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a16051a88924d318deb3cfc64c88d94 : SUCCESS in 59m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d0e9d8a6d3e412d9e41d6439787fe5a : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c12d06a7a65844169e0900a8bdd4c405 : SUCCESS in 25m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80fadec62d734330b1f3bfaf9801f56e : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4648a442b2c14168aec56c6cd86d03e3 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96ff100503d7407fae3ffc09517dac31 : SUCCESS in 25m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a26ed69acfc04adba067763fbcd1b49f : SUCCESS in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5cacf494d92043dcaa1c89468fdf6234 : SUCCESS in 20m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c54b3923ba04905aebf390f1cdbb2c3 : SUCCESS in 23m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34c0ab4256924c5a9df4a35d1aee9592 : SUCCESS in 21m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/459b417c4b144de9a1dc5a8335be5a61 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05ee808933404017a8288bbd8839ebcc : SUCCESS in 17m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5d05cafaf68e40af9805b004d00276a5 : SUCCESS in 12m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/075dc70a5e8747f7b765059bf440276f : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/27ea26a8d8794e4dacbe749a2536e006 : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d759bb6adc6b48e5a1f0280cb18d9459 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01b285e9d00c4cd386b84af67e6aec5b : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a1ac077ee41d4dacafceeb5e2bfb10be : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa5da321e8d94a979fdcfed154df6383 : SUCCESS in 50m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b202d9ac18a4e1598d0b2feb9d75cf9 : SUCCESS in 56m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a98a27b445c54f4ebe1a1951418689a2 : SUCCESS in 51m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a786170ac3fa41c3a95ee33246e9dbfd : SUCCESS in 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbd5124a05449b298ea83174d235569 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd9dc1aa208b47258b4439750a589b25 : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6cd508fc546412cb4d1d9c3f2481c67 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":32},{"id":"632f03099fb43b277d5fb1d9fa4ec990f1780cb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 21:49:09.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbb45a86a70945c09fcd26e5aa7203f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99a27b52624442c5986207e82e5029be : SUCCESS in 2h 16m 46s","accounts_in_message":[],"_revision_number":32},{"id":"ebc763e13a47a32f16f560d64cffdc219f5381cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-06 20:53:15.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"eb515172ad353d02dad236e91a95bffcd4d23ae6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-06 20:57:11.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"4b14c06f47448fbfe92e19450d8307a8c654cc32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:03:48.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3db50d2a0c430eb23c6f01108c3efc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92199775d7144ef8869b343cf261dbd6 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0c3bde9ce73e4a7c9b6fecd2e1501ed1 : SUCCESS in 51m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1125e19f99ef4d19bf145fae4e29a3d2 : SUCCESS in 53m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ffd104fa6104a4783994bca3df1bb89 : SUCCESS in 43m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1da8b031ab414299a63d5308809f07 : SUCCESS in 1h 03m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f29eedb612a845898826d16e774b95b8 : SUCCESS in 22m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c0604cdcacb40e490efbece27925ef0 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74664bdeb621471da6316210a2efba97 : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bdcd165a6814f75b3807e1f2aa0ae0b : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47fe92964abd43a18162933155f2ecdb : SUCCESS in 21m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5095f0d127d24b83aa664959072d63f6 : SUCCESS in 15m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2145b84de83e4ca48e1ae8bb7ce6d55a : SUCCESS in 23m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ba69176382248f7ae0bfde3bf5200a6 : SUCCESS in 23m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c58965fb83d47af9cac8ea3be977ac5 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee1177e79e2a4d03bf3e1435682ebb94 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/18c28c3fa9664c888fbaf783e498ba5a : SUCCESS in 17m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b5bfc5fbf6745cc977c4b8dd317d719 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fcbb50fbccc44830ba8298738328296d : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/361cbd6f31cc4cdd936f18d0fa16c4cb : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf36fa501ca845978d1ca5d86d8ff815 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b04e15d0e274b08ab5b92d9fac6a967 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fbe26b5cc0bc49b3ada0197b77b3f590 : FAILURE in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c37a34f6cb340b5874b11f15ca5478c : SUCCESS in 48m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a4f54c90a1f4dbd970b0a13e90cc3f6 : SUCCESS in 43m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b41ba423fc7742bd814dd7ae89cdeba9 : SUCCESS in 40m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14da7e0a5f394b779aa7afeaa61ba559 : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28327c1e73674edb84655be56e7a67ce : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd467b9f46914d55876d3e3fd4ad69e0 : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6b0e295589f4626bf6bb6ebb428f9b4 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":33},{"id":"f4fda4e945315615ce4cf11b1fece0dfb315b8fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-06 22:35:59.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"cc3a63ce9db8bdb7deb0851ae196e6562cd994fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 23:57:20.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a45ad947c4c94ddc859f9dc33f863ded\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c129598471c4171b02add42a5abbdf1 : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":34},{"id":"a3b22a126c594ae36f3bfd4f54d45b025bea1c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:39:28.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65cd0135c6124919845075fbacb3324e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f08bb6e1c56466aa153c7778341e5f4 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a76b562c0f5b4b60a0a3608d54f84e36 : SUCCESS in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/496ad046f67344109d497045269aa832 : SUCCESS in 59m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7554c3f0c8b440dca2e28e7d3fbe3f77 : SUCCESS in 33m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/920cf90bd1034d29a72018e3f50a357b : SUCCESS in 49m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/673d93496a454e9aafd7f10fec6195e5 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/657a03e240e2426f912ad233a3c0aea5 : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b8aaf836f864ff8a8097d83c5af3dd0 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e5c8553242343f8ae4390d82582eb95 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/50d2ba54628343e8abe4618372a0264c : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d59f5242196463986856fac455928ee : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22b543a995be4a00acf0eacd78afc5a3 : SUCCESS in 24m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98b29f6ea4be44ae943b06b83aed08e9 : SUCCESS in 28m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8e6f8c35be849329c7be50dfbba8113 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7a2b22ddb844f7a9b121209a303330c : SUCCESS in 16m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c74115689664916a8cf68da87e0d5d9 : SUCCESS in 14m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/82508da28ebd486db1cbe52045126759 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f797d7c4fe04af6a9ddec84c4098d9f : SUCCESS in 16m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab56db22125d4e2b9ef1b57614550566 : SUCCESS in 13m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e8250c2706a4f1ea4c66862a141c437 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d399b5442e44170aa5262aa65dc9f71 : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17972ba3d5d8496ba30b2c5b798d9917 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48bab61cbb10481f920d119435abed2c : TIMED_OUT in 2h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4af7c4bc8b8c471da9252f9c3a004024 : SUCCESS in 53m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f6d65a307b646368dc7ead630454f16 : SUCCESS in 38m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d60e293d0fca4a14b7221eafd2b6441d : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae0cdbf9a854727843b81f397bcc562 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52b16d59f77d4aeb8c11565fcda6f775 : SUCCESS in 26m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b25c857320874e83805be9477a39ec39 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":34},{"id":"bcbbac457078d15cac950265726cf3bbf6606696","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-07 00:42:47.000000000","message":"Patch Set 34:\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"96e74d21bd6b4fcfc92e6673bf50a6e1adfdca57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 01:52:22.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5039362cb61948409351dbfc49f8a279\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486045f714fa44d5a49a2b80075f9f9b : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/1dad3504a91b4b8b90372086776abd22 : SUCCESS in 48m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/426f6f5c18304c16b713d676c301b499 : SUCCESS in 38m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/083e2ba30ffa4e1bb6e86660fb8c8529 : SUCCESS in 39m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1946e9ba3750452ca4538ae6bea673f9 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c895a65ecdd44768ad42b4fca5252d1d : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b059dab6ca2d46b2a47d279c4814f813 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83e3436ecb6148a1a277e1fcf2b329c1 : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2550f160869444dd9be582d3bd71c17a : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af4b6e563e224fdba0df5c7059e99fb4 : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72fa4913816c4674a784f60720bdc2d9 : SUCCESS in 18m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8a65c54968945d09f2a898647ed3d5b : SUCCESS in 22m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ab33a53daf1a4900b32f5d0440415a1a : SUCCESS in 30m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4caf4d8eb004436d8f8c11b3852a0f47 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d19d24255f24aaea3c7a751f4533b3b : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84ca7d367ae1411ca77c9b4ab6bb89f3 : SUCCESS in 18m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b1e088ef77c348d6931cdc1390e0dad4 : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/013f9d4b40ff473a9989fd54330fe816 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6141183d4aa4139aa59cc2eae2a46ab : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6764ab69781402aa3af45d2349c516a : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84c4f5dff5c34cb2bccf3e59107d8c54 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06b963dffcbe42bea0a82be3b9d133cd : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a012852b5bfa47bcac32adf47a8f3147 : SUCCESS in 47m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1f0b80442a8461194c45a89d44d259b : SUCCESS in 55m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60ae2a1e0fd54073ac15723b5e982a04 : SUCCESS in 42m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45259e7fef044212ad0f606cdc597401 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75330db138164a29a832ad43d51195a7 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2aa1f7702f4f4159b995b7fd9744b61a : SUCCESS in 22m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b6e0bb69feb412ebe3de742a50a3264 : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":34},{"id":"65459011469f4be5c8e11fbb3cce4acdb6c5ce83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 02:41:03.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848e1867fac4436b8f5b819a31123a36\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b63fcc48e7a453c8114e4322a15a978 : SUCCESS in 1h 55m 46s","accounts_in_message":[],"_revision_number":34},{"id":"68c79a7ac69267cfa5b759815d180f228bd88fee","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-08 16:35:40.000000000","message":"Patch Set 34:\n\n(5 comments)","accounts_in_message":[],"_revision_number":34},{"id":"7986f51ba65404eb1e54c7d1e81a23f19e06dfd0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-08 20:54:38.000000000","message":"Patch Set 34:\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"ac33dd662a5544fd32b2c8e2cf1edb57616fc696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-15 19:44:17.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"94c4774047068763c7838cbe1b820d6d2e37dcc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 20:46:29.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4448a002e414084a472e6e3e957a65d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c11cd7eb4de41ad9e7f83b846652b8d : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":35},{"id":"9c3fb37513a9c8fbee1c21070c0d07f008802256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 20:47:29.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6726143100d94a21a65455616daeb2f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c0e68e279b4eaebefc66ee28690cab : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/15a9d9de7806414c84dc3946dd28a6c8 : FAILURE in 57m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e6e3773a8bc4e889a83cc711cfc806c : SUCCESS in 43m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba00312e691e46909bb7b30a8b13c632 : SUCCESS in 39m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b869ea94c474404d9d25b31a2af69c1f : SUCCESS in 1h 01m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/911ea90faf7a4263b1d0146419340a91 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53dc020d3a5b4c9fbf254c02e68f1096 : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51cdf6aa22004ce492b3ba35978b8196 : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b65c9e115134249bd9985c51c68599d : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/951b0c46c0964276b13ded44e8053878 : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6fe2903427b74850a50ab4444d6c640a : SUCCESS in 17m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/488870cabe2248d1801a19cb8cf16976 : SUCCESS in 11m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ab50336f8866498eb42ea857c7bbdc14 : SUCCESS in 20m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0be5aa5495574591a4d279ccb83b8de1 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90a186812fad48ea98f4ca243d31c398 : SUCCESS in 10m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3399a56a496e43e7a301becafc75487f : SUCCESS in 18m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1a6193f4fd6e4c3eb131f70dca3b5cdc : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0e7bb17e62864bfbb12dd69255bfea1f : SUCCESS in 13m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db5644480a66430f95f5f2f6d80aaf24 : SUCCESS in 18m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b89c30397e14bd19c350a6cadc3017f : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e092dacb29a14d5f92ce99c71d156bde : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/809a341fed7c472a992fcb919d53495a : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83b3e720cc7c47d48e26e3d42de76a1d : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f432bd1949843ffb861fee2589943ad : SUCCESS in 43m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10d2fefedb74431d815e8b03b18797aa : SUCCESS in 53m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd930ca967964fb3833bccdc6aab23cb : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c3f98b328a4b9a81812c90a0594fc3 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb73fe5f703049d099c4a9b5a1c4175e : SUCCESS in 19m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5040c3debe648f699c3dfece2e1c789 : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":35},{"id":"d7e6c391b3f3d953d3622e928951a57da88bb533","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-25 18:20:43.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"5385acfbdbaaef59dccdf5103c6d76830c2c825f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 19:15:41.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad3a7ea930574d1a8ff5b7c1f710f353\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a439841e14e64f9ba10ae33e4e9263f7 : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/201c7ffc7d554ee2bcb6b6692ea4b907 : SUCCESS in 49m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2aa255e8d7b54769861b449947adc5ce : SUCCESS in 48m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f53e2f9b1ca474c8c75f5ed9ffc5128 : SUCCESS in 53m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4c60e7e6b93489eb0be00abdc84038a : SUCCESS in 52m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03dd39cd9186493eb5fe8e8f460669b6 : SUCCESS in 23m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5e94fbe0fb6485cafd7393d033979a1 : SUCCESS in 23m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03f29f4d77c24afab132f90f8ef6d6ad : SUCCESS in 12m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5529e3e52b634bf79b01711e339fedf4 : FAILURE in 3m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b41aba4d7d784d5f827b0f477a34ec45 : SUCCESS in 22m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/65e4da657497449c86268ce31de02b1c : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72dbb1877902455fbb619d285464a788 : SUCCESS in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f04b005eddf8412187037140a884133e : SUCCESS in 31m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b13373ef0f7e436bb452e187819f318f : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c823eb1731064cd1b92958b81b1c46a1 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c771bd13f1a40088bdf87c077b396b4 : SUCCESS in 16m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ab9961ff041944d3a11140bb83a546ae : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bdb82c4f803a405889092b4dab715547 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/59fc156cf6314c22859b02fc8657fd4b : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7721124ef3d4417ba2433308a4665b0 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1d159ca4669462a8c57f1b9fa95d5ea : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2fa68b967220413f9c9ccf158de03586 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd9cacbc710b4caaa36a482be68fc00f : SUCCESS in 41m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cea56e0884094307b2fc7eacd320afe0 : SUCCESS in 39m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/135542ccf24f44ba9367122fe4546004 : SUCCESS in 40m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db6defdc67a8432ab8082b0c2784af59 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/187b67632eed46e7aa4d24c37af67b39 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10c1d2ed08a241f4be61105294e3d270 : SUCCESS in 10m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e1057080f2147ffb1a3c2b2719691c1 : SUCCESS in 46m 24s","accounts_in_message":[],"_revision_number":35},{"id":"a1120e7878b05faa59296d1f7d09667d8a6c2fdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 22:00:12.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c718fd371a1b4f2c8334c52ef824edf5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/27589156fc4444c681d4cb8a0947a757 : SUCCESS in 1h 51m 18s","accounts_in_message":[],"_revision_number":35},{"id":"152653a360c88ec6f287c78bdc76bccefadd1931","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 21:50:26.000000000","message":"Patch Set 35: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":35},{"id":"92c0c7a8468c4d6e843688e14934ec13be96cf8f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-17 03:56:58.000000000","message":"Patch Set 35: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":35},{"id":"0b3be5410111a7258f168ced3fb86942f96572b5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-17 05:17:30.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"afb4673a4c8d76af1897f8bb6b43c3daa5c75848","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 20:53:26.000000000","message":"Patch Set 35:\n\n(4 comments)","accounts_in_message":[],"_revision_number":35},{"id":"4ba898a9ba6d2a82e07286668b7027df9f740e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 20:53:27.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"8744e6bbdbeb4c05491c457cec7111743cf2411d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 21:59:59.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05d674347c8649fcab8fc1fc5607646d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36f9a0c5c86d49589af5ff9a0b9d52c7 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ad8775a315c444c5911c9e0847194cc9 : SUCCESS in 58m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25c9d1547f664778b2af4c43c051d79a : SUCCESS in 52m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c62b21c996b34715930ca7ea364e0ddc : SUCCESS in 40m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d261ce8c8a6f481399b047e363df57ce : SUCCESS in 1h 03m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fac301e778b4140b220254ae78e2cb2 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b93d499718c7437988711c54a7eae652 : SUCCESS in 12m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1292a0b11b0349c2b84353e31d603ac9 : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e251837857944bcbd5fc9fc8e0bacef : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1812358fc7b84055904fccff3435b37e : SUCCESS in 22m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/371f9e0b62e141fd882b609987e36984 : SUCCESS in 20m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2165c6384314458b7ce138dadb98ba0 : SUCCESS in 21m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0d073128386045f286238338c4404e87 : SUCCESS in 28m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a798378bd24344629124593482888848 : SUCCESS in 20m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c6050a601ca4b458b408d9d5e43508f : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/447c8c7587554c5190de17cccdb7ad80 : SUCCESS in 22m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3d703c11177642aba6e2a24138dd5cbf : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1949325e4514361ba0a90931648d733 : SUCCESS in 9m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29a6dd9cc6f1444280d67beb2cea0c70 : SUCCESS in 13m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/979a30d2322e4360bb49c4a238e65c0f : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16746691f7f64156a870dd014f93ce6f : SUCCESS in 13m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4836e02bd6d04e0a9a9b354f4bcc8321 : FAILURE in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bdd65bc070e4d0fa8100bf174e9d4f0 : SUCCESS in 46m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b76a6d2bb814ab1aeb2f69a99fb6c01 : SUCCESS in 48m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45968131dab94329a29d610caa0a9323 : SUCCESS in 47m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/894129a6f99a44ac9af7d10a153752b5 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c474b131ecb4b68badb4ce42ce0b0fc : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5742eed86a2a439784eb84a6be400813 : SUCCESS in 11m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dcca1af64e646b9ae0f55329cb276df : SUCCESS in 48m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcf6e9106ee446d5b0d96a536c01fd4a : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":36},{"id":"b783a9c86b566e5d7fb1c31e406d9675ecfdb07f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-17 22:27:41.000000000","message":"Patch Set 36: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"848d73aa236591e0539d7fd8aab6d72494bd5047","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 23:10:25.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"091278947d1e4282e0f3888421af5fc5c67b0641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 23:10:31.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"3dc9535b2610be15f887710badee09993a961a9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 00:19:36.000000000","message":"Patch Set 37: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/429ac0cf6cb1455792316f86014fe1a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ac03428ffd43ef9f5253999f8618fb : SUCCESS in 5m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c060b24736594baa94b0e0d691f06317 : SUCCESS in 1h 01m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb66085d0d44412ab830bb9a68e8a20e : SUCCESS in 34m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f9dc869b22c46a0b7da6b0fc5889fa4 : SUCCESS in 1h 00m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/711d29785236410b80c13ea7d10860b2 : SUCCESS in 39m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bda78ecaa244479a83bbcc9388b8993 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab42e879b60746f7a000eb48353ff271 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/974cb7d5afcd416e9c5189b2d2dc66e8 : SUCCESS in 21m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38f7eacd8b914f23bdd8ebdcf4f13cd6 : SUCCESS in 4m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/75b11d70c8d242e79aa2d74ee372058b : SUCCESS in 18m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/15dbb332366244efb32eceee086ab493 : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6aaf72913c4b4c3fb86a744517f8e2f0 : SUCCESS in 19m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2deaf4f90a3f46a2ac9d7582f101304d : SUCCESS in 19m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/491294115f574681863c52ce6f514041 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/259b0f78591c4a6aaddf69bb9095f794 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b1a866d67e84daa8867fe042a355cc2 : SUCCESS in 18m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e94b0f3ab0284de0920ab30f67355240 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8aeb44ed52f448c589ca542b5a70eb1a : SUCCESS in 15m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee2ee26a0216408eb99a9436d9c4fd54 : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cff261902edb4fc189af296acdd3ba23 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1181c344a5714771a5b8ff6a0d12991a : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e452cbcd57c4babbe0eedc23d3baa01 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ddfafa96534cf58837fcf6aa5ed429 : SUCCESS in 46m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f8c8c7eae93472f909008bd845c7f44 : SUCCESS in 36m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e60321f7a634fe094cef48cd03e613c : SUCCESS in 52m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d21347d4cf2b4bd5948753bcc550c704 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2fcc7f607a4323898d44f585d8ff5a : FAILURE in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b5fb26e1ad24147a32fd50a5ab5c64e : SUCCESS in 19m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd073f31ab5c4875be7051f79abe6588 : SUCCESS in 45m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26037395ce284b21beb1343bbf5d9f7c : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":37},{"id":"a82ccc921ecae5ca02f0e20f426d1cdeb8f11c6c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-18 00:28:24.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"ce55ef5c51712e944cee7ce175775c0dfd2eb61f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 02:01:19.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecbee8761e6c4146a26420acbc7333f6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3bb683f246ad46c7bbfbba4015d1f491 : SUCCESS in 2h 15m 02s","accounts_in_message":[],"_revision_number":37},{"id":"6f5ccb4a4ce41b6b33af8c630902a66de1db618a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-18 02:11:01.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"8af2700da5768ee24b4670eedf8c0fe573f9b289","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-18 02:13:14.000000000","message":"Patch Set 38: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"468357b2f59859aee4772e0114b3d176094bd4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 04:11:35.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96cfa378a024be9932ff38421ae659d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d52e42e23e48caba023f825c35e5a2 : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/22b217a6659540f1b74e9822ad142074 : SUCCESS in 55m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ccd05adca8443d8a2aeadf9fe5b04bd : SUCCESS in 1h 00m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9272d94ece54abf8fb36f10c97f4813 : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b169b34a89141c6a38389673ed02cec : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5d331e901704fe58da581cdea9a9354 : SUCCESS in 12m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd8658fef7d84f8389decb3b0e18a128 : SUCCESS in 13m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdd4668ca94a4fadb9b5ea9d1b6ed91f : SUCCESS in 13m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/57e97da7e3144baca42d362e1a3472e4 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f010ecd97f467d8ccedd91c94d13fd : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/884623151a5c4e5587209ff3a7481fd7 : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/887869ddc2584823b1703dd5f520adab : SUCCESS in 20m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c7a5431d08774195bbed44051e890030 : SUCCESS in 33m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f7a0cdb616a4e36beaffa75ecaced58 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7332e691a00740e5866ac12fa797667a : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f17851b73b04e7c94d7759bd21b7c27 : SUCCESS in 26m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/69462f657e6d422e9d9e25cd7389a4e2 : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9f5c13155a9f436ab51b93115610f80d : SUCCESS in 18m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ad9242f8fc2848e6b9d44f0c3ad6373b : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75f5b2ec9beb49ea887edd05a43b32d7 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf71f9e52241439cabf3d562bf620fe2 : SUCCESS in 13m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7c94b5061834f08b982115abddd7661 : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8201e6e4b1bd4ac7a519e5b4a6bfb350 : SUCCESS in 49m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c5566c3a4824bc4a253ab56678b2bef : SUCCESS in 38m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5ca6cfab5a44e59bede9c220a3d80a0 : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdc3a638bb844c24b23867f3a617a44b : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f1d30d24814f01aec0eb7854c2990a : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7711bf137b5740a3a7266a15c2cdaf8e : SUCCESS in 21m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e1cdf4d24ec4f0381d33eb24204d3a8 : SUCCESS in 52m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa2a4dc604474c8eb93ef991fd409060 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":38},{"id":"b528eaa5514d2c62029feb03b04ff1555552a9b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 04:34:01.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d3abf7d5484e2e8221cf25573a090e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff9c138a0ea2488783cc74f9a70f2df6 : SUCCESS in 1h 59m 34s","accounts_in_message":[],"_revision_number":38},{"id":"36f3bec739baaf27d9dcaf8d6bda5a0cfc100b6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-17 20:15:22.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"9bc120ee943d8c0175f8a0e5a585384f5de5bcf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 21:28:58.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7010621e42a4f1984d666a5caf0d0b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6258b86113914cc69c1ebded86a4618c : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/07393a3a8d004852adb598cd58437f0b : SUCCESS in 1h 06m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e01b710827c4d1aa8aff6ed34ce71d0 : SUCCESS in 57m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89db071aa8b54553b83ec67acdcb070c : SUCCESS in 37m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e278ed01e369438b9d77cc3a8b123dc6 : FAILURE in 1h 04m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c8083292cc34fffb1f3406e243d4b20 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbbd7dd0e00a422480d6c635f8e37dd6 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f639ab9483345069100fbe6c7d60820 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3af92c9c609d41f488a9148dee05190a : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8569612c285248fca57817b9b285c554 : FAILURE in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e957fd1fdd5a46b886d72934395f82e5 : SUCCESS in 19m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f2150e056c764d19a0deec944d0e2797 : SUCCESS in 18m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b4bc6ed80d214627bf2d487d4091eb94 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f88d9537a53c4530a0cdd688b30471f5 : SUCCESS in 13m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/100b8db9011f4ed5b44926e4701ccede : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f482d4155824dd3ab2a275ea764ecd0 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fca439df181c4fcb8eb252ffb3b1d78f : FAILURE in 10m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/60b9ad5b8be24b6ebdb53a221a23b58d : SUCCESS in 5m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b48fa3bc54f64fda9f6edcfe03efa812 : SUCCESS in 36m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2600dbe27ff4fa9b35c8570c22d7ae7 : SUCCESS in 25m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbc8e91bc2434db9a46ba7b89024a610 : SUCCESS in 52m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56f1d2a4f07b4d5e8a99f11e3b288216 : FAILURE in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526c1b8257064d0d9efee22d9881d47f : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3448d65b76694e0d8218eee04b13d52c : SUCCESS in 17m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/513f2e4e85dd4ace8bb2b3e716f1d2a4 : SUCCESS in 48m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f2b4428414e47afad6b23a2f8a8276c : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":39},{"id":"3bd197b705373a3a2656fcf41a4c8c9fba1fc04b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 21:47:13.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c28a6ae39d504beaa8e56e47643c3c68\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d84002d94a4a4be6a089f606fe2b930d : SUCCESS in 1h 29m 09s","accounts_in_message":[],"_revision_number":39},{"id":"019016193c4189f7355b8c9eb068c899ebfd83ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-17 22:07:53.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"d2c452f1d5b2dd53c2a0f899c27808ef1a9e2980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 23:38:45.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d40da9412404248a629d03533a0e575\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd1c5beeb1e485d8657bd8f8efbaee9 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ee6c5ce251f3412bb7690c0ca87906e1 : SUCCESS in 56m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92835c7baf444d6d9b1acf8f52433a00 : SUCCESS in 55m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8db13089c6d41b9a17054edbde82e9f : SUCCESS in 36m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6590d44117064534950ce4935b4c361f : FAILURE in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69f532fcdd514c639e72868af90cfcfb : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe6ba1a6c00d4ce09fefd53e5ea1c7e6 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99e56da329fb4fe0bd3d9e6061ae209e : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d0a0a3b152d4177baee1ccc589edc5e : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0057927238a340d89f5d5eb71473d69e : FAILURE in 19m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32748df8e0584f95a3ec09c4077dac70 : SUCCESS in 20m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3511751cc4c14a70b794b8fe5d6959d4 : SUCCESS in 17m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bbd925788444424a9d605880596f0192 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aa99473288854c598ab2e1dda4e313a1 : SUCCESS in 12m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/40d03792463d4426b5bca28077c4de2e : SUCCESS in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/652d006b295746f29cb83e7665f91a3c : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d8035305594b40389853ee67e0139d27 : FAILURE in 15m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f47d6c23b9746a998d0f6bc1278db8b : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52465a0e9ae94e0ca4ea7928b6c8f57a : SUCCESS in 45m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2533f7db46214ab7b2aac45f8e97e965 : SUCCESS in 46m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de62281ccb1747d1affcbff303a2a252 : SUCCESS in 46m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6a00b3779c74df9a81d6304ca207ef3 : FAILURE in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bd115d22a944e081966b26f8042c8a : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94b6fe02279840bf89e6e48faf7b9ca9 : SUCCESS in 18m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d65f10b6f2a4cb298a877ccedc2b405 : SUCCESS in 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed450261cd2465b8158efb709ba948b : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":40},{"id":"4822da0aff651b14d324ab8925079dcc835520d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 23:48:13.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0174015c6bb46c9acc50d23449e3758\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f4309caa5b7436182b36545566ccbaa : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":40},{"id":"9c4883162fdf7ab5a99d1bc752bee244a02f0ae0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-20 18:48:15.000000000","message":"Patch Set 40: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":40},{"id":"8de3d1e200b5a458ff19f0c81936a726cc36a88b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-21 17:10:46.000000000","message":"Patch Set 40:\n\n(3 comments)","accounts_in_message":[],"_revision_number":40},{"id":"ac4abb47d3152475c45c09dfdf4c09b91d1d6e08","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-21 17:27:15.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"ae7265d7ef5a17b9c301160e507cdb37d169a7d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 11:05:44.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"473f065d6f28c60f2b162389b0dfd064c7d5deff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 12:29:06.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d305da168304fd197eae22ee0e3f48d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b8812764aa473a95473405c93b32a2 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/21c31afd16214185a33a857a1a356f52 : SUCCESS in 44m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25de89d0e9234043a2d5d27c923af359 : SUCCESS in 1h 15m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2196e18034344089df4cd56d9a63a1e : SUCCESS in 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3638d65614d74f2a8c15c8299b2e073c : SUCCESS in 45m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f37a81a37b14e8791eb3e1116801704 : SUCCESS in 23m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c246b7cc39b4564ae1ef92fe0a11237 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d15a7081d9d480f89747940cef099d6 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d55bdc548f644fb68c86c1e318911145 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e216ef7ffc24ca3af5a5edbc69c31ea : SUCCESS in 24m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d2b258baf6f43dbb28d393fa66744a5 : SUCCESS in 12m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b3c5d902270046e18b92bc0fd0f78fa9 : SUCCESS in 28m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4248ec769f474a228414b297aa5b1a72 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4956eb6a5bf14ff9a40b7152f3869750 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a717fb51acc94566806a60e08f130a85 : POST_FAILURE in 11m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9223120592a9493492ad72a90b68d112 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5142fb19a7654829bd56f10189cfe7eb : FAILURE in 14m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/93dac285efa44031ae810b3de3279107 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d86632d48ee432ebaf480ee6932a779 : SUCCESS in 48m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5146d0c4040f4df8b1d925c926d7e6ad : SUCCESS in 42m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e5cbd6a171d4c0b98e9a029acab2d93 : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb76dc7e6dc494bbbd3599fa4538738 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d930c2471364464a8ad735e8a008890 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37b29ecafd9f4024ace706ee2b974497 : SUCCESS in 27m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84598eeead5a48e18498783d9f30a0f3 : SUCCESS in 51m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a37d5e87c04c44d3a9c77f113d7effef : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":41},{"id":"0196bed7b81b32ebc3bc414bd4d60e873395ab37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 13:07:43.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deedecfbab484603836b23e5f746fe6f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6da38571c3ad4da3b56b7feac47e3381 : SUCCESS in 1h 58m 16s","accounts_in_message":[],"_revision_number":41},{"id":"c2bfd41ba3c15d6129e132fb1c63267ecc27da92","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:46:27.000000000","message":"Patch Set 41:\n\n(3 comments)","accounts_in_message":[],"_revision_number":41},{"id":"20654c51d4143eadf573e1463e9f36ca824e3794","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:56:00.000000000","message":"Patch Set 41: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"198aa9356a8f22046cd3838726b4a72be0eebf6a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:58:19.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"3da2d4350d559e0630d73a98df0d111d2578e733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 16:37:37.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae7e6f2f0c240adbe46318dc59b1b37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0ffe8c042b42af83ecbe8e53921114 : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/658de81b2dd5451c950268eb63be84cf : SUCCESS in 58m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf2fdbcbc3ed426b9b9614093b2e77df : SUCCESS in 1h 08m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84e32c7b99d74716a7af09a753a96101 : SUCCESS in 1h 00m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fadf60b52f74be2b4a5c08193ab10e8 : SUCCESS in 42m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e3b8e35588040faab6d89ba8fdbee25 : SUCCESS in 26m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bfbf072216c4b49ae4591bc37657906 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5caab9f3828f420fbc301950d6db7bf0 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d439ac910abb4719a7d7db009c2aec47 : FAILURE in 4m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ee52cac6b9a40279d37d0ded7717405 : SUCCESS in 17m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5db83f46ff914ee28c9326f73a7a2704 : SUCCESS in 20m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6b65796a841042228e79a39d9b2d4690 : SUCCESS in 33m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f1292756c984a63a2b9e616709fcd98 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/146f6c2c4b684a7f8cbbc9275eeb2c3c : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e55b09283fab48c6a8d67ca09e986b4b : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/994ab962d55a4b3ab060ecc32f1a84c2 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b350c9e2533c46dd8b1db932d0a5b704 : FAILURE in 14m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/238520290efe43248ae81fe522b28a2d : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/29482b81c278447cb3ceca700abf40de : FAILURE in 1h 02m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4012c42254a748c28b212c34a1516d60 : SUCCESS in 51m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3838271b9bae4e4986c309e049c55296 : SUCCESS in 55m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fda61822d4f40d5a1186eb4e6d5d77b : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0be57a2a8234af7b9c793522e8ad4eb : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c875a802f2664e1b99580c9810c1b79c : SUCCESS in 18m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cc28e0088fe41bea81884457eab734e : SUCCESS in 51m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1eab135c2524ca98645760f9059305d : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":41},{"id":"5fb569be7cb990760ea91f473435cf68be23b5e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 17:19:37.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa0bdced20e41f980e9f4dfc1967f53\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd3a36308d1c4ea38b6373966e4cd758 : FAILURE in 2h 18m 51s","accounts_in_message":[],"_revision_number":41},{"id":"26d75f52808364f142a1b55d059c40cd9a2ccf2e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-23 17:03:44.000000000","message":"Patch Set 41:\n\n(3 comments)","accounts_in_message":[],"_revision_number":41},{"id":"abc8f9cfbd1b5bf3c21c69209544b9ca5bbd2938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-24 05:04:39.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"e2b5749a84bc3823b4af2beee9fd7d441a44c0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 06:24:20.000000000","message":"Patch Set 42: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a6c9efbe65e4fe783be2add9dfd954b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4d375e0ccb46e2b37315b6eefb2656 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/112deb4d18c443aa879330384258ab8a : SUCCESS in 1h 10m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6aec3a06b3d5478bbb79253796a995aa : SUCCESS in 1h 00m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a009508d7684639851e275fe29aeb1f : SUCCESS in 53m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66fcd941d9f34e12be7f94ecdfc666d5 : SUCCESS in 1h 06m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59807c3d131a4daa922b9d2c2206d849 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d6b35c4e75e45d08432fa0207ac7d1d : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/686eec270bab42aba316703ac9465d2e : SUCCESS in 11m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f93cf08f7954806b787616e271c53d8 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efa472223b2d43508c48d43c5e7ce756 : FAILURE in 22m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb00e6670a324f9db6a4661999573037 : SUCCESS in 20m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19cd35fba254426f877561e698598e73 : SUCCESS in 31m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ca2b75d1f4314911a5ab5199a8a24280 : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3bab6b5722044a5db681949f4768680f : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/edbf06487f634427a40f71a8ea629fb4 : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9a9b565add842c38c38b29565c99adb : SUCCESS in 4m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6789c7e3fc34376a8ad4856391366c5 : FAILURE in 18m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6655bd8e02b499485425f28315ee88d : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/23e21f2e6eba4a439a69940dccb12897 : SUCCESS in 1h 00m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4055f6ea985143fdb7ab84739be1aca0 : SUCCESS in 33m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5b0494cd947429b8dfb999d71295642 : SUCCESS in 51m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c079f68987840babf6834891cee9189 : FAILURE in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e72fd4511a49d6ab5e481b83d4a0b9 : FAILURE in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cdef7e9b4a7c4e03b4ba8d9e8b4bc9e6 : SUCCESS in 19m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ec209af563c49c0a3bb45073d5c8d09 : SUCCESS in 47m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef39f5e591c84b8f9d080b0de9f8d436 : SUCCESS in 41m 31s","accounts_in_message":[],"_revision_number":42},{"id":"7a66d0f485f514e328d881e110e3b1bdc065c2fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 07:57:07.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e86d3064009847dd9cd657adb0696a9d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1694fb5cf3e495dbdd072a071f3bd47 : SUCCESS in 2h 26m 47s","accounts_in_message":[],"_revision_number":42},{"id":"ff41aa03bdb00c4c5d9e2e5eceba8a309dd27843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-24 16:57:11.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"efee0834f88edf8643a68ddcc88dbc90ecbc3044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 18:21:59.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eebf8268f9784948b1ae50a642b58c95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c130987caa47fdb85214ff2f58d3a7 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/40726f4237694431925d4c4b46b3df55 : SUCCESS in 40m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fcbe11350ba54216b88028d952eb8110 : SUCCESS in 1h 18m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/857bab2dac5f4c15be52e0c146b1ba7a : SUCCESS in 40m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f8341ccf9cb485ca9db0040d557fcbf : SUCCESS in 1h 13m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed4bbf966bed449297beb0e991edc0de : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05887145d4384521b491cbc6cc4a260d : SUCCESS in 22m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a5f578ee84a433fb939127635a75d9e : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1f4a385b1fc49c6b9f0310bb9f405b5 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09a7fdce06a142d2aef488d3c70028bd : SUCCESS in 21m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e38af5c1866642c9892900a26099d9a9 : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b68c9dd731504a6686c7bbc9e4a02da6 : SUCCESS in 23m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff593150e1c343dc861be7a54853e406 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9f45221e0b1e4466916a0d06242ae0a8 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e892c9be3f914d2882227a311047c1e7 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1caa367743604d089766db7f654c2992 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/664d4e055c8d4312b2e06313013769e7 : FAILURE in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc1870a60fb542099acdd439b3d232a0 : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a4f507dd11740a4a44890b11f9e04c6 : SUCCESS in 51m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80b90a24128f4348b2e85c5a2a966162 : SUCCESS in 34m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/232de1b54d9b4365a53c5600684c8fe0 : SUCCESS in 36m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddf6b521558c45c6a2c96ed5b3a5a016 : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb87c20691d4da3b528b81e20033ec1 : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee745cf3053a420e9f1ef4579699c554 : SUCCESS in 18m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b9b63d4ee9e46c8b5b7b7ddd229c3cf : SUCCESS in 58m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6362730bafbb427c8c5c5a189fc99af6 : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":43},{"id":"be6d26d6f8d8eafd94712d992322d75b7cc73d7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:21:49.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b83adb3a64b452990c17d5bd3f0c24e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a4eb8ce2bed4f76a66952091f85ca7b : SUCCESS in 1h 46m 40s","accounts_in_message":[],"_revision_number":43},{"id":"0aacadf2f06fb510b9187e6b6fbf9a7fecb7b310","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-25 04:33:47.000000000","message":"Patch Set 43:\n\n(7 comments)","accounts_in_message":[],"_revision_number":43},{"id":"bccbc08d1865cbf3ead26e3417925da56617a691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-19 20:21:11.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"194a748b8c5299fd9408355023f3829fa646d2ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 22:01:13.000000000","message":"Patch Set 44: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab0433122b62497ea04156cea65bb68b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c5512a04994c19ada58ade4a8cd641 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0d6f018eb4bc431ebf373b1b19a3c02f : SUCCESS in 1h 15m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d04d45932054225b7fb58e76cdec60d : SUCCESS in 1h 06m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/515c1df9874040b3af9055e42c89649b : SUCCESS in 45m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1d117dd29ab440383822e3edc133ff0 : SUCCESS in 1h 08m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9e071bc877546409593b7e2dad7bb2b : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47572761e5b1481c8207425782d1c26f : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d24ba1e2b4184b87bdf6ea74a8e8261e : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4deaaa7df63a4af1a92dfe39ec888ec8 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33dc9d6ac44544c6a14e0ff14c997038 : FAILURE in 17m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16fa3884d9ce4115a6a6fb9a73ded27c : FAILURE in 23m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2adc9b1d355b4de5af260f2a6c824613 : FAILURE in 32m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94ef84b1477d4cf685246772331877ea : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e5c82092649f4d9e9a41bc66e08ef818 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fb97dba5b261490c96639440616c2aaf : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ff506e899924b239301ce8427d943ee : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22559eb27fda428783ebea12d8918bca : FAILURE in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/32d13855632d40d39e937d03b1e2527d : FAILURE in 7m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8001c8b9c074e9e99e77c7aac55b2f9 : SUCCESS in 54m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2f28b25d8cf455bbd5cdfc33b7c9ca7 : SUCCESS in 37m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fe61302ea0d45e383cd3782e7c13add : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/317ddf8000754ee882587f995590a498 : FAILURE in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce03aba29d6c404fab9debdc31a3c9e0 : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd55d667ea5441a4888d367baf2a4d0a : SUCCESS in 17m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42a8ad0208b548b7bc51c021ff728997 : SUCCESS in 1h 07m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23d0575879a8495cb2fe0dfb092acb8f : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":44},{"id":"a8c8a87a4257e78525fa89e297dc1c294d00aa3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 23:26:27.000000000","message":"Patch Set 44:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b9b568371440fa9a58c7db1ac083cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d11b5e943db649f3b8e54f12081369b0 : FAILURE in 1h 32m 03s","accounts_in_message":[],"_revision_number":44},{"id":"50857345bb3a3689555cccbabfe947c44e374fbf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 16:09:03.000000000","message":"Patch Set 44:\n\n(4 comments)","accounts_in_message":[],"_revision_number":44},{"id":"634d693de13c860eb85261fe77584caec699fed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-21 04:53:04.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"4f21545fda12650780409a3706b10f1c9e4c6ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 06:06:30.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/43f5d28ec7f4416f9177f0f8402c7fa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2fe88a21434ce684503d81ccaf10e7 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/0c995cdb57714a4c9903104e4d6c0f42 : SUCCESS in 47m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c352ed06e1ed4019859d173defd6896e : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/467e422b99184247a2dc281b66a7c117 : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ba9789d6f4d415a981847fb7f53cc43 : SUCCESS in 51m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c3d25ef97db4c639061c68488df53e6 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/275efa7794ec45aa8c3a8c8fa262baac : FAILURE in 25m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b391b2693747403fae09a2c48398a525 : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59908937c18d4aa29abc9d97603a24ed : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ff4043868a843cea44de1a4e64da62e : SUCCESS in 19m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4644c25559744eadb4dbca191cef9367 : SUCCESS in 22m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/02738db8bcd844f8b39b6e32b466ea4e : SUCCESS in 28m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d6180fcfeb5e4b25b452d2380c6e6040 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5613e4a1d3634b2ca66a761fc56b9634 : FAILURE in 8m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f93569cbe506494087f368131e90322a : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b829740a8fb44f3a8c7924a01f48e88 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ae592a23acd46b88c812d3866e95ccf : FAILURE in 19m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/039010cf5d0a494facfd06d82a0f101e : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f1504ab0c6741ef9e6be1c4130d0a4d : SUCCESS in 53m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2eef10a46f2a47ab8381d30a632a163d : SUCCESS in 56m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d9e71b2179154a3c94c07eadea378877 : SUCCESS in 52m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf143a77a2cb49628573acba5247dee0 : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1c38dbab8f4c0e8ea12cdda5a87a8d : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce4ed53d147f4caa84e5e92a5c279c8b : SUCCESS in 10m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be89bd01b8c4acf8d7d9a1ff5b96d38 : SUCCESS in 1h 01m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35c3e24fa31140efbec9456e6f409855 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":45},{"id":"28d392d5baa942130a217fbfc45744732609703b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 07:36:06.000000000","message":"Patch Set 45:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd0965d9feca450093197fb0b71705b3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/101948e880b74da485acd163caf9dd14 : FAILURE in 2h 34m 24s","accounts_in_message":[],"_revision_number":45},{"id":"84606f39effc9575a74995872f280fd697c503bd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-21 15:31:58.000000000","message":"Patch Set 45:\n\n(15 comments)","accounts_in_message":[],"_revision_number":45},{"id":"47b18bb5ec20b719d0efbb52173e21b5ee952241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-21 18:44:32.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"f33142816f6907451bb5763885869ce9a524325f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:08:13.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3845f139ebff48fb932ca751e70f38df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02da401edf3549529bb13db1125f305b : SUCCESS in 7m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/10edfa7b1f92400a848f2e95c48db664 : SUCCESS in 1h 03m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abffb5c7ac92488a8e09646e0a100d1a : SUCCESS in 54m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2df9fae041e4a66abb9a1c8411db178 : SUCCESS in 41m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ca142b2a89f4e7aaef2a6127ed47d07 : SUCCESS in 49m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be6b24c3ca534ebeb1cd51ab0664765d : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d6e7f1398b1414fb048bea046d668a5 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeb87330c56c46c0921ca6aa9f54c3df : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ce08389f74747f395699d33a52c50b4 : SUCCESS in 4m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49fdb29c620a408ca3159396348cac10 : SUCCESS in 20m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4939f7e0dbf443349708f67e9f31df28 : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d065cc2135644c4a34e68980aa4574f : SUCCESS in 28m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07f8a5e856ad4b849d60ce908225e21a : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/141c0cda1d164ada84fadfa1fe8c7919 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d031659e1f96440d906d2b945b0477b3 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd9778715a2d4593976e43f389277198 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a4b698ff7dd418b82a4d3bb2e251f75 : FAILURE in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c507cc4d48f845438bf51285b0f8ff57 : SUCCESS in 10m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22566351b8c0405da74287e19321a1ef : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ee30a10f50845bca9b3f3b2fd0fe3dd : SUCCESS in 53m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7cab2535bcf246699c8b367cbde543e7 : SUCCESS in 24m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d588093b630a490b869ffa516b182399 : SUCCESS in 20m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3c8a3c40214884bee95db04db4566e : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0341e6e7d1d343fa9382f78d88c4c550 : SUCCESS in 17m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e1a7d8eb4314ae589cb81dd5d7a94ab : SUCCESS in 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9711a8d053424221bf37f5da12eab8dd : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":46},{"id":"f8fdd59e1d6dbde4761c115a186974cc5627bb71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:29:07.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2b2184a45d416195801875f82dbcab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f248812c10a146748c07e7b4228fddf9 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":46},{"id":"aa2946eae33052454f987a84893b3c964b059db6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-28 17:57:13.000000000","message":"Patch Set 46:\n\n(9 comments)","accounts_in_message":[],"_revision_number":46},{"id":"76e237261937338f4e6cb9ee399191384877b3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-11 17:41:24.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"3ece3c40e0bc50cd91c22059f57cfe2fd55eb255","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:23:48.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/838fd2e7f8bb49ee9c6e77ada1bde081\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00ee208f7d0a480ca08469cc168f2330 : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":47},{"id":"d10b38fe6cfb2bd2f5ef46b13863808e251455ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:57:58.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475b2bb90edb4eae98fa55b42e895f9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7066e3a7ff21489f930575dafdb33024 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/418fd94bd66d4023849882bddfda6e41 : SUCCESS in 1h 10m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a29abe7b24294d71be0b2425891db642 : SUCCESS in 1h 01m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8263781963c349deaf00b33358b51721 : SUCCESS in 41m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a973c3c89c6249c39dc9ed3de3ae947c : SUCCESS in 33m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54ffbc2da4f442648f53c1c419204dcd : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65adf6bfbbd0469e816e0e7498bb9d23 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0d031a1dd7e43cd9fbb42d4e9930c9c : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cea46e031924a14bc72d2ebdaece5fc : FAILURE in 5m 09s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef328b1b6c64594bcc5569406b18ada : SUCCESS in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75906f41f73343c9980e5ce478250d2f : SUCCESS in 20m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/984971172c8645dbb39eb3489e370620 : SUCCESS in 18m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/561fb2c201f24575bdf314dfda34cd3f : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/88cd67cb0b464b1780119a8fcd0566ee : SUCCESS in 8m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c43edb37174344fe91af947063182c54 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59fa4c7b132c42b88a0783299dffab6e : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c7b6d828eee4851bfd04d15e2b5b671 : FAILURE in 16m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7585851a8c1f40b7882ffe338b9f57e9 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e58bafadecbc48ca99a0fd6b17fed88d : SUCCESS in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cd26b4eaffb4afa9fb240ace264c832 : SUCCESS in 46m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34bbd763b56349b89657b61d2485427f : SUCCESS in 57m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d445c17a167f4221ad8e4848f7f28389 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f65877fdc4246518c1ae97ebdc73c23 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83339aa04b34468593578b8287388ca1 : SUCCESS in 19m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c5e204aa0a74fe0a2599b04e2f2ee67 : SUCCESS in 48m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2128738748d2494c8c5a61e1a3358fca : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":47},{"id":"f874d368f0199912d2fcab0fbd592feddf25765d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-12 21:01:45.000000000","message":"Patch Set 47:\n\n(4 comments)","accounts_in_message":[],"_revision_number":47},{"id":"a1b2fc36c1addb2982320a9f02294873b647d42e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-12 21:28:35.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"1744676461190ea90cc329c3da41e2e272ab152b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-13 04:17:25.000000000","message":"Patch Set 47:\n\n(6 comments)","accounts_in_message":[],"_revision_number":47}],"current_revision_number":47,"current_revision":"81cbd84a16671b2d83538e74425bdf3ed8654b2d","revisions":{"05851fa080e856a3882353d2e1e436c7b24378ca":{"kind":"REWORK","_number":1,"created":"2024-02-21 23:33:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/1"}}},"commit":{"parents":[{"commit":"9cb6924549b290b9c866059bb71f985ee34c54f2","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cb6924549b290b9c866059bb71f985ee34c54f2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-21 23:32:54.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05851fa080e856a3882353d2e1e436c7b24378ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05851fa080e856a3882353d2e1e436c7b24378ca"}]},"branch":"refs/heads/master"},"e06fedbf455506e04e0bb4635bc2a4b3a7ab5db1":{"kind":"REWORK","_number":2,"created":"2024-02-27 23:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/2"}}},"commit":{"parents":[{"commit":"c5f8577620bcca871607cc7adfc5693858b2e86d","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5f8577620bcca871607cc7adfc5693858b2e86d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-27 22:10:50.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e06fedbf455506e04e0bb4635bc2a4b3a7ab5db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e06fedbf455506e04e0bb4635bc2a4b3a7ab5db1"}]},"branch":"refs/heads/master"},"70cf2b98f181f70fd0b8d454d6bb7df477440b3d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-28 00:13:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/3"}}},"commit":{"parents":[{"commit":"b6a772dc47f1fb9151f7322a3878b4fcbc917034","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6a772dc47f1fb9151f7322a3878b4fcbc917034"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-28 00:13:30.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70cf2b98f181f70fd0b8d454d6bb7df477440b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70cf2b98f181f70fd0b8d454d6bb7df477440b3d"}]},"branch":"refs/heads/master"},"43ce9d7a37a710d064e5f094ed23afe2680f9d71":{"kind":"REWORK","_number":4,"created":"2024-03-01 05:18:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/4"}}},"commit":{"parents":[{"commit":"7d882320bbbe2c921d0b5590fe93ad1f1f5871c9","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d882320bbbe2c921d0b5590fe93ad1f1f5871c9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 05:17:18.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43ce9d7a37a710d064e5f094ed23afe2680f9d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43ce9d7a37a710d064e5f094ed23afe2680f9d71"}]},"branch":"refs/heads/master"},"c76ec71c3ec97d1f9890b37c9d0d82df8edeacc7":{"kind":"REWORK","_number":5,"created":"2024-03-02 01:02:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/5"}}},"commit":{"parents":[{"commit":"7d882320bbbe2c921d0b5590fe93ad1f1f5871c9","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d882320bbbe2c921d0b5590fe93ad1f1f5871c9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 22:56:21.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c76ec71c3ec97d1f9890b37c9d0d82df8edeacc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c76ec71c3ec97d1f9890b37c9d0d82df8edeacc7"}]},"branch":"refs/heads/master"},"164252d6716e408215688c31afa5ebb06d6d246f":{"kind":"REWORK","_number":6,"created":"2024-03-15 00:39:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/6"}}},"commit":{"parents":[{"commit":"190243a66bf3b8b0df6662b7421a1c6829bdf688","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/190243a66bf3b8b0df6662b7421a1c6829bdf688"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-14 23:31:31.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/164252d6716e408215688c31afa5ebb06d6d246f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/164252d6716e408215688c31afa5ebb06d6d246f"}]},"branch":"refs/heads/master"},"978f46ffa880f0ee401feef34414590c7d1d9ff5":{"kind":"REWORK","_number":7,"created":"2024-03-21 05:46:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/7"}}},"commit":{"parents":[{"commit":"83b8ed4a723c37a817b308acd721b8db5bb95f6d","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83b8ed4a723c37a817b308acd721b8db5bb95f6d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:46:35.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/978f46ffa880f0ee401feef34414590c7d1d9ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/978f46ffa880f0ee401feef34414590c7d1d9ff5"}]},"branch":"refs/heads/master"},"ef155e80f64d738673855c23724d2e138644b751":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-22 22:49:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/8"}}},"commit":{"parents":[{"commit":"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 18:59:34.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef155e80f64d738673855c23724d2e138644b751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef155e80f64d738673855c23724d2e138644b751"}]},"branch":"refs/heads/master"},"1ba03618622451b245663d0a03e6001beb3eacc9":{"kind":"REWORK","_number":9,"created":"2024-03-25 18:47:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/9"}}},"commit":{"parents":[{"commit":"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 18:47:29.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ba03618622451b245663d0a03e6001beb3eacc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ba03618622451b245663d0a03e6001beb3eacc9"}]},"branch":"refs/heads/master"},"79dd045833721c5c37e7f65cacc0c460969e9b44":{"kind":"REWORK","_number":10,"created":"2024-03-25 22:06:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/10"}}},"commit":{"parents":[{"commit":"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 22:00:41.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79dd045833721c5c37e7f65cacc0c460969e9b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79dd045833721c5c37e7f65cacc0c460969e9b44"}]},"branch":"refs/heads/master"},"878cfa286ab779a85cc4aee02f80e039e012ca99":{"kind":"REWORK","_number":11,"created":"2024-03-25 22:42:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/11"}}},"commit":{"parents":[{"commit":"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 22:42:45.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/878cfa286ab779a85cc4aee02f80e039e012ca99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/878cfa286ab779a85cc4aee02f80e039e012ca99"}]},"branch":"refs/heads/master"},"f7dd71940b2b3c80bacc39dbcf7c5d4ef3442eaf":{"kind":"REWORK","_number":12,"created":"2024-03-25 23:09:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/12"}}},"commit":{"parents":[{"commit":"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 23:08:14.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7dd71940b2b3c80bacc39dbcf7c5d4ef3442eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7dd71940b2b3c80bacc39dbcf7c5d4ef3442eaf"}]},"branch":"refs/heads/master"},"80289a84648fc3223f4099ca2cdc5c00b5372356":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-04-01 20:21:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/909801/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/13"}}},"commit":{"parents":[{"commit":"993b2bda8e4b15b89795014306b04b0d3ac3f925","subject":"tests: Validate some assumptions in test_signals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/993b2bda8e4b15b89795014306b04b0d3ac3f925"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-01 20:21:48.000000000","tz":0},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80289a84648fc3223f4099ca2cdc5c00b5372356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80289a84648fc3223f4099ca2cdc5c00b5372356"}]},"branch":"refs/heads/master","description":"Rebase"},"9f1fe2d486503b2d1c577af363ce2018155260f0":{"kind":"REWORK","_number":14,"created":"2024-05-28 22:36:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/14"}}},"commit":{"parents":[{"commit":"dda220acf7c392a3a87987a125ed92a37ecf9142","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dda220acf7c392a3a87987a125ed92a37ecf9142"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-28 22:30:14.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f1fe2d486503b2d1c577af363ce2018155260f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f1fe2d486503b2d1c577af363ce2018155260f0"}]},"branch":"refs/heads/master"},"e7a63e10137a88f4ecf0f33510fc0a45329a8d09":{"kind":"REWORK","_number":15,"created":"2024-05-29 21:03:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/15"}}},"commit":{"parents":[{"commit":"aa67c8a726b8321006cdada76ffce133a8032ff4","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa67c8a726b8321006cdada76ffce133a8032ff4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 17:19:17.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7a63e10137a88f4ecf0f33510fc0a45329a8d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7a63e10137a88f4ecf0f33510fc0a45329a8d09"}]},"branch":"refs/heads/master"},"e104db1089c28b91e3d73bfda98371b0497e3aa2":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-07-02 00:25:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/16"}}},"commit":{"parents":[{"commit":"098b3a4aa0d42f196c82d3c8ca512972579ee5e9","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/098b3a4aa0d42f196c82d3c8ca512972579ee5e9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-02 00:24:47.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e104db1089c28b91e3d73bfda98371b0497e3aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e104db1089c28b91e3d73bfda98371b0497e3aa2"}]},"branch":"refs/heads/master"},"9f8ee8d533b71296de7d2202df8f9bb2dd52000f":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-08-16 00:11:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/17"}}},"commit":{"parents":[{"commit":"dbd0ca1b961ace55e930b694f88a7344d00868fc","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbd0ca1b961ace55e930b694f88a7344d00868fc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-16 00:09:40.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f8ee8d533b71296de7d2202df8f9bb2dd52000f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f8ee8d533b71296de7d2202df8f9bb2dd52000f"}]},"branch":"refs/heads/master"},"9d2578a59bb4fc6962eec0626d7fcd6966cd00c7":{"kind":"REWORK","_number":18,"created":"2024-08-22 00:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/18"}}},"commit":{"parents":[{"commit":"ce94a5f22c0a88adcfb28c317e60a94f86a9fedd","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce94a5f22c0a88adcfb28c317e60a94f86a9fedd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-21 23:11:49.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d2578a59bb4fc6962eec0626d7fcd6966cd00c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d2578a59bb4fc6962eec0626d7fcd6966cd00c7"}]},"branch":"refs/heads/master"},"5ca80ed289d55ccfd3c83284a379baaab3dcf9f8":{"kind":"REWORK","_number":19,"created":"2024-08-22 22:20:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/19"}}},"commit":{"parents":[{"commit":"9c6122eef11113c241c5ad91152efa1448dba176","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c6122eef11113c241c5ad91152efa1448dba176"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-22 22:20:35.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca80ed289d55ccfd3c83284a379baaab3dcf9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca80ed289d55ccfd3c83284a379baaab3dcf9f8"}]},"branch":"refs/heads/master"},"ceddce75d75ff5231e904f41b68be91f4ad188e9":{"kind":"REWORK","_number":20,"created":"2024-08-23 19:22:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/20"}}},"commit":{"parents":[{"commit":"051ffaecb4ab596314f8c752bcd4e918987d7446","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/051ffaecb4ab596314f8c752bcd4e918987d7446"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:21:38.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceddce75d75ff5231e904f41b68be91f4ad188e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceddce75d75ff5231e904f41b68be91f4ad188e9"}]},"branch":"refs/heads/master"},"1d2cfb45ce9007f08fe9263fa9de7f5efd75cf07":{"kind":"REWORK","_number":21,"created":"2024-08-23 23:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/21"}}},"commit":{"parents":[{"commit":"4d97d281977f23c0a3e7f5d6632aadb26139bded","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d97d281977f23c0a3e7f5d6632aadb26139bded"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 23:41:18.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d2cfb45ce9007f08fe9263fa9de7f5efd75cf07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d2cfb45ce9007f08fe9263fa9de7f5efd75cf07"}]},"branch":"refs/heads/master"},"9d8d40ed548c848cf903498d1cf57cc7fc97140d":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-09-03 19:24:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/22"}}},"commit":{"parents":[{"commit":"1bef736b6564e6f2fb2be5dc1b25673567f4705c","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bef736b6564e6f2fb2be5dc1b25673567f4705c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-03 19:24:34.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d8d40ed548c848cf903498d1cf57cc7fc97140d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d8d40ed548c848cf903498d1cf57cc7fc97140d"}]},"branch":"refs/heads/master"},"4427f2f3c5078aa617e06935dbe889296de4af77":{"kind":"REWORK","_number":23,"created":"2024-09-04 00:27:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/23"}}},"commit":{"parents":[{"commit":"d402e93058c9bcdaec41343899f58db5764f38ed","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d402e93058c9bcdaec41343899f58db5764f38ed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 00:26:41.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4427f2f3c5078aa617e06935dbe889296de4af77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4427f2f3c5078aa617e06935dbe889296de4af77"}]},"branch":"refs/heads/master"},"a71e75935f86665ba278020c1044353393c04c89":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-09-20 23:29:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/24"}}},"commit":{"parents":[{"commit":"a9640f4a199bf7cf74e3cb26b9821083cab6ce67","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9640f4a199bf7cf74e3cb26b9821083cab6ce67"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-20 23:28:59.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a71e75935f86665ba278020c1044353393c04c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a71e75935f86665ba278020c1044353393c04c89"}]},"branch":"refs/heads/master"},"3daa348b501482620b1431b9e8d6ed99fa07e9e0":{"kind":"REWORK","_number":25,"created":"2024-09-23 00:06:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/25"}}},"commit":{"parents":[{"commit":"e9b65fdd12f8cda00f483571302085b65d335259","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b65fdd12f8cda00f483571302085b65d335259"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-23 00:06:01.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3daa348b501482620b1431b9e8d6ed99fa07e9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3daa348b501482620b1431b9e8d6ed99fa07e9e0"}]},"branch":"refs/heads/master"},"ddf8c02713f47639ed690aacf5ee0fad7ecccedf":{"kind":"REWORK","_number":26,"created":"2024-09-23 23:18:49.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/909801/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/26"}}},"commit":{"parents":[{"commit":"e9b65fdd12f8cda00f483571302085b65d335259","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b65fdd12f8cda00f483571302085b65d335259"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-23 23:18:21.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddf8c02713f47639ed690aacf5ee0fad7ecccedf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddf8c02713f47639ed690aacf5ee0fad7ecccedf"}]},"branch":"refs/heads/master"},"041caddcd27e00cdf9c1af038f6921b361717959":{"kind":"REWORK","_number":27,"created":"2024-09-27 23:23:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/27"}}},"commit":{"parents":[{"commit":"f11ab93cd07d32ae02da1134403f4cb8599a2512","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f11ab93cd07d32ae02da1134403f4cb8599a2512"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-27 22:56:54.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041caddcd27e00cdf9c1af038f6921b361717959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041caddcd27e00cdf9c1af038f6921b361717959"}]},"branch":"refs/heads/master"},"79c60c8f04da7649a5959646877b587a28fb52f6":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2024-10-01 23:18:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/28"}}},"commit":{"parents":[{"commit":"e191d3fad730c278d3d514c24fad03ff890527cd","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e191d3fad730c278d3d514c24fad03ff890527cd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-01 23:17:51.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79c60c8f04da7649a5959646877b587a28fb52f6"}]},"branch":"refs/heads/master"},"d6194628d0cc0188e9039a5b89c1e056b1f14d5d":{"kind":"REWORK","_number":29,"created":"2024-10-04 20:10:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/29"}}},"commit":{"parents":[{"commit":"c9f6a81809f941c0ade36de5cecbad5078cdb298","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9f6a81809f941c0ade36de5cecbad5078cdb298"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-04 20:09:19.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6194628d0cc0188e9039a5b89c1e056b1f14d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6194628d0cc0188e9039a5b89c1e056b1f14d5d"}]},"branch":"refs/heads/master"},"458172a9a9baf265780feb7e8340853b1d633fd1":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2024-10-09 20:29:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/909801/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/30"}}},"commit":{"parents":[{"commit":"d85889d88387df6a2af4595917842c556dbe2e26","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d85889d88387df6a2af4595917842c556dbe2e26"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-09 20:29:50.000000000","tz":0},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/458172a9a9baf265780feb7e8340853b1d633fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/458172a9a9baf265780feb7e8340853b1d633fd1"}]},"branch":"refs/heads/master","description":"Rebase"},"ee0c1a1da16f6c6386341275fba9f14129061080":{"kind":"REWORK","_number":31,"created":"2024-10-10 21:54:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/31"}}},"commit":{"parents":[{"commit":"d85889d88387df6a2af4595917842c556dbe2e26","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d85889d88387df6a2af4595917842c556dbe2e26"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-10 21:39:00.000000000","tz":-420},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0c1a1da16f6c6386341275fba9f14129061080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0c1a1da16f6c6386341275fba9f14129061080"}]},"branch":"refs/heads/master"},"40e5c890a2b099c153dd9af09ed3443948d19dc0":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-11-04 19:28:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/01/909801/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/32"}}},"commit":{"parents":[{"commit":"ca9a47b981e06ac719d8998a82836e98b8983b93","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9a47b981e06ac719d8998a82836e98b8983b93"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-04 19:28:21.000000000","tz":0},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40e5c890a2b099c153dd9af09ed3443948d19dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40e5c890a2b099c153dd9af09ed3443948d19dc0"}]},"branch":"refs/heads/master","description":"Rebase"},"ecd7b9e05e07cc9c148469ab5fbe77771a44ce12":{"kind":"REWORK","_number":33,"created":"2024-11-06 20:53:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/33"}}},"commit":{"parents":[{"commit":"bb611fd6957f4407b912a480681abfccac89e454","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb611fd6957f4407b912a480681abfccac89e454"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-06 20:53:03.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecd7b9e05e07cc9c148469ab5fbe77771a44ce12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecd7b9e05e07cc9c148469ab5fbe77771a44ce12"}]},"branch":"refs/heads/master"},"0c41ecd6fc5e33fe44bb476463f69520ef0b9628":{"kind":"REWORK","_number":34,"created":"2024-11-06 22:35:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/34"}}},"commit":{"parents":[{"commit":"bb611fd6957f4407b912a480681abfccac89e454","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb611fd6957f4407b912a480681abfccac89e454"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-06 22:35:32.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c41ecd6fc5e33fe44bb476463f69520ef0b9628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c41ecd6fc5e33fe44bb476463f69520ef0b9628"}]},"branch":"refs/heads/master"},"bca4eb0cfb9ca90db7133999851c397514c4352d":{"kind":"REWORK","_number":35,"created":"2024-11-15 19:44:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/35"}}},"commit":{"parents":[{"commit":"84e9b6160a0fe04759fb91c358de057c75d3fbf9","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84e9b6160a0fe04759fb91c358de057c75d3fbf9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-15 19:44:07.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bca4eb0cfb9ca90db7133999851c397514c4352d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bca4eb0cfb9ca90db7133999851c397514c4352d"}]},"branch":"refs/heads/master"},"1fa4a325fb17f5bc3f90f7068c499528d9202660":{"kind":"REWORK","_number":36,"created":"2024-12-17 20:53:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/36"}}},"commit":{"parents":[{"commit":"4fbb94d26d5d60ac8c9c27013c932c575a618e9e","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fbb94d26d5d60ac8c9c27013c932c575a618e9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-17 20:53:13.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fa4a325fb17f5bc3f90f7068c499528d9202660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fa4a325fb17f5bc3f90f7068c499528d9202660"}]},"branch":"refs/heads/master"},"b950315b6ef97ecba10f29ee92f894efeb2fc8eb":{"kind":"REWORK","_number":37,"created":"2024-12-17 23:10:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/37"}}},"commit":{"parents":[{"commit":"4fbb94d26d5d60ac8c9c27013c932c575a618e9e","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fbb94d26d5d60ac8c9c27013c932c575a618e9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-17 23:09:58.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b950315b6ef97ecba10f29ee92f894efeb2fc8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b950315b6ef97ecba10f29ee92f894efeb2fc8eb"}]},"branch":"refs/heads/master"},"45e396a4696d3e93b359c53dfe02ec65d31a87cc":{"kind":"REWORK","_number":38,"created":"2024-12-18 02:11:01.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/909801/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/38"}}},"commit":{"parents":[{"commit":"4fbb94d26d5d60ac8c9c27013c932c575a618e9e","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fbb94d26d5d60ac8c9c27013c932c575a618e9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-12-18 02:10:22.000000000","tz":660},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e396a4696d3e93b359c53dfe02ec65d31a87cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e396a4696d3e93b359c53dfe02ec65d31a87cc"}]},"branch":"refs/heads/master"},"d7edd90ac6dadeefdd1300b62e7b22ebbfd0dad8":{"kind":"REWORK","_number":39,"created":"2025-01-17 20:15:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/39"}}},"commit":{"parents":[{"commit":"577c810879f166bf76fcb88ac23f7c5b46ec9a8a","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/577c810879f166bf76fcb88ac23f7c5b46ec9a8a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-17 20:15:09.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7edd90ac6dadeefdd1300b62e7b22ebbfd0dad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7edd90ac6dadeefdd1300b62e7b22ebbfd0dad8"}]},"branch":"refs/heads/master"},"f2ff609392f7b83e683f00b03a56631634a1b2d1":{"kind":"REWORK","_number":40,"created":"2025-01-17 22:07:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/40"}}},"commit":{"parents":[{"commit":"d869cd9b1322d936e2eddf3d01f4d9703950d347","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d869cd9b1322d936e2eddf3d01f4d9703950d347"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-17 22:01:48.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2ff609392f7b83e683f00b03a56631634a1b2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2ff609392f7b83e683f00b03a56631634a1b2d1"}]},"branch":"refs/heads/master"},"9fb91818710ded3d1f501a00ba5037511eafd703":{"kind":"REWORK","_number":41,"created":"2025-01-22 11:05:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/909801/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/41"}}},"commit":{"parents":[{"commit":"d869cd9b1322d936e2eddf3d01f4d9703950d347","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d869cd9b1322d936e2eddf3d01f4d9703950d347"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-22 10:17:29.000000000","tz":0},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb91818710ded3d1f501a00ba5037511eafd703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb91818710ded3d1f501a00ba5037511eafd703"}]},"branch":"refs/heads/master"},"24f3d05334431412de093d3f746c8b77ca43c12a":{"kind":"REWORK","_number":42,"created":"2025-01-24 05:04:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/42"}}},"commit":{"parents":[{"commit":"361a74365505b03e88d30b077a21f51f791f6f4a","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/361a74365505b03e88d30b077a21f51f791f6f4a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-24 05:04:29.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f3d05334431412de093d3f746c8b77ca43c12a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f3d05334431412de093d3f746c8b77ca43c12a"}]},"branch":"refs/heads/master"},"9ae00a55701be3a9d283406d86054d3395b138ae":{"kind":"REWORK","_number":43,"created":"2025-01-24 16:57:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/43"}}},"commit":{"parents":[{"commit":"a64619ad90b60d9df4ee8d7db81af613a0a0add4","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a64619ad90b60d9df4ee8d7db81af613a0a0add4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-24 16:56:57.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ae00a55701be3a9d283406d86054d3395b138ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ae00a55701be3a9d283406d86054d3395b138ae"}]},"branch":"refs/heads/master"},"9e9641d7029cd509854cc4629263dd307ff95bda":{"kind":"REWORK","_number":44,"created":"2025-02-19 20:21:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/44"}}},"commit":{"parents":[{"commit":"7909f90c629ce401f0c28a1452afbbb17a0c67b8","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7909f90c629ce401f0c28a1452afbbb17a0c67b8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-19 20:20:39.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e9641d7029cd509854cc4629263dd307ff95bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e9641d7029cd509854cc4629263dd307ff95bda"}]},"branch":"refs/heads/master"},"66103a90deb743faa73d0bff2177f8494b05b1e6":{"kind":"REWORK","_number":45,"created":"2025-02-21 04:53:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/45"}}},"commit":{"parents":[{"commit":"e970b2b626b960ebc4523e15dd9894173dfd68aa","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e970b2b626b960ebc4523e15dd9894173dfd68aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-21 04:52:34.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66103a90deb743faa73d0bff2177f8494b05b1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66103a90deb743faa73d0bff2177f8494b05b1e6"}]},"branch":"refs/heads/master"},"aa1d677caa779d3a6a90889606e409ad7bdee462":{"kind":"REWORK","_number":46,"created":"2025-02-21 18:44:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/909801/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/46"}}},"commit":{"parents":[{"commit":"e970b2b626b960ebc4523e15dd9894173dfd68aa","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e970b2b626b960ebc4523e15dd9894173dfd68aa"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-21 18:43:51.000000000","tz":-480},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa1d677caa779d3a6a90889606e409ad7bdee462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa1d677caa779d3a6a90889606e409ad7bdee462"}]},"branch":"refs/heads/master"},"81cbd84a16671b2d83538e74425bdf3ed8654b2d":{"kind":"REWORK","_number":47,"created":"2025-03-11 17:41:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/01/909801/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/909801/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/909801/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/909801/47"}}},"commit":{"parents":[{"commit":"f110ffb10f74b0689a3fb3219b44d20055e63890","subject":"utils: Add crc32c and crc64nvme functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f110ffb10f74b0689a3fb3219b44d20055e63890"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-20 21:11:09.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-11 17:33:24.000000000","tz":0},"subject":"s3api: Add support for additional checksums","message":"s3api: Add support for additional checksums\n\nSee https://docs.aws.amazon.com/AmazonS3/latest/userguide/checking-object-integrity.html\nfor some background.\n\nOnly \"normal\" objects are currently supported; MPUs will come later.\n\nCo-Authored-By: Ashwin Nair \u003cashnair@nvidia.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I93fb5fa33c5d4f51d73084e4ee945708fc3183ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cbd84a16671b2d83538e74425bdf3ed8654b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cbd84a16671b2d83538e74425bdf3ed8654b2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
