)]}'
{"id":"openstack%2Fswift~910029","triplet_id":"openstack%2Fswift~master~I072974aab22d8b698152bf2f05c7ec97844f89f0","project":"openstack/swift","branch":"master","topic":"p-diskfile-meta-deleted","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-26 16:44:00.000000000","reason":"A robot voted negatively on a label"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-02-23 22:45:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I072974aab22d8b698152bf2f05c7ec97844f89f0","subject":"fallback to .data meta on PUT if .meta unavailable","status":"NEW","created":"2024-02-23 21:07:31.000000000","updated":"2024-03-05 18:45:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"82a64093bb7cc469662cb740f8eff7ea3b8517ac","_number":910029,"virtual_id_number":910029,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-27 16:25:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 21:29:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-23 22:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 15:22:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-03-05 18:45:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"2dc6293ff6a45f78aff855886bf2d2c915d83ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-23 21:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"769058c3a4a5877e55dc638d8867e905e0ff8d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 21:29:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f179706713147ac8a8aa48fa3af30e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/119baed4928d4472b3c781aee4ddfeb5 : SUCCESS in 16m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1abb81737d7b47a4a746a05b90299cbf : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/60221be4a72046f6996c4ef0a297be10 : SUCCESS in 20m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f60ce87d6c74740909e17092974bf7e : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"095af898d52b5ec133011241d31e3a55b75d5bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 22:45:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/319c396ee12f4edf8cb4251d0e1a3df8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a8c9b760ef4813a0502431ea627fdf : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/2c01e1f98187439ab53d3b09d15a2fab : SUCCESS in 1h 02m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c6ca00e6cfd4185840a8bd2af4cec5f : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01f7db620efd4d63becd29bd8725c9e9 : SUCCESS in 38m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea02b2d185104bb1beb2ac941ccfd5c5 : SUCCESS in 55m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c4d947c982f430b97ea289c8a8decbb : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43eab07d2eb74694ba60e3c02a9a56d9 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3c505908635480faef9c76c9a1e9ff0 : SUCCESS in 55m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66c81ec49b51424a879db20e056d188d : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9955f1eff5324362bfff90f57ff33b7c : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cda4505f576d4571a2f061d71c4896f2 : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4afe7cde84d6471790ade4d55c7faad0 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e6d99b68f574578a66d4d519b6f7385 : SUCCESS in 18m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b88997064c3b451ba78831a850cbfe75 : SUCCESS in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/33347cda97d34d6dabb3735e296b59b4 : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23dffcad8c6e46e0a4504d8510d9a3be : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/497ffd1f15bf4c19ab534f372aaf7695 : SUCCESS in 50m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7878378bf75f46fa8e6099b14a123d96 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf1fbab64e6449839f8a292cd3a37ace : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4c5ac4e79efc40cfa18cce5eb581c5c1 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/979aba6930824601aefcae5c27ecd77b : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/685d011fd6ac4b0590496efddd88ff35 : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36d5ffcbb1ff48509af1a265230599da : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7903706549540cc9a86a1352ca71d94 : SUCCESS in 8m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3847735ec0ef44838da8d55ed648fc19 : SUCCESS in 1h 36m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bac24544a3504403a764fb5cbae3f5d2 : SUCCESS in 1h 04m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dea14be542ff43e39c818fb673585397 : SUCCESS in 50m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53d49ad1ea7d418b96026b1ef7408a0c : SUCCESS in 1h 18m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2365668d8af8452ea7451392e7dd5a09 : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2526efd0b5f64c12afa6aa42660ac51c : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a8bbb653bfe4621bdfb0478d5410f90 : SUCCESS in 19m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c2df4329eaa484e9b90ac7b36c1734c : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6cca0adea8bf508c4e83d36a99ce6929bb142071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-26 15:18:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"809a51f11f037149af091b9cceb07efe70162ef8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-26 15:22:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"033c5e42084be06748d4f28d629d5a9f75f4bcc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:07:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7bd2b6ea2345e197e145dc30dfe46e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d4ce71804c84b92b0e752b9bb6e1231 : SUCCESS in 17m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac85fc0ea4bf428996b725b42ddba70b : SUCCESS in 47m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9332a0704c34b4fbdcee31f9d6870fa : SUCCESS in 21m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce0bcd46eba1403f967cf5985ad2862b : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":2},{"id":"7ec8ed0f3a233e6b6ddb0acddcadd74df2c2f792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:44:00.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa0c21c879245d5978a67e778917e17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261c43bf31d84d809f9b38873779789a : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/37d1541d75ae40d4bfb68ad5bfc8aae8 : SUCCESS in 1h 12m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/401e588fb2114b4596ca3ab99965cec3 : SUCCESS in 43m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d38621c937a40a1b776b70bc89b2054 : SUCCESS in 54m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/043bafece6c1488cac9fadfcbdff2fa0 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2308ceff7d4e4958a8222114117c4067 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7697dc987f794fe38c2623624385e2e2 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/154580f6cc5c402e837e9714f198c796 : SUCCESS in 18m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a79d92ea2a7f42d9a3e8f21361fce2c9 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/821361eec40d4c64b15a7786a803e7b1 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9206296f67304d03a65ac099a45d0965 : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd0efd01e2ff49f7aa3524fd95475636 : SUCCESS in 18m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e0fabdb2814a20a508a77594d05af1 : SUCCESS in 15m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f92ea53688240f59ea0c6ca119f2bc8 : SUCCESS in 18m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0cf5a252185c43aab23020d03a8b231e : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/268a2b16dfb243049991faa4175b7f4a : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e87c3f558a5e4c4fb0e9aa4c0a2a2f96 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/78b6c5e6555146bdb6b7344da3b4c412 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8667582d40d24df5a33cd7a039b3b3c4 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e364e5031fe54bbab204216d295875ba : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c8591a435f24b9ebf9cf0199310ae5e : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6b45ab86d6142b28343362b8c056037 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4486c1dc15a24425b6dc37fac63f926d : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5c964965b3e4cb9befa9d229a46f3f8 : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e87df004a5714578949d7cef4ac77a27 : SUCCESS in 57m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/66cc9bddf1a641de9019480b76b1a3fe : SUCCESS in 59m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/226e24c148744759a38e0f49180018ea : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82796cc6e6714a0a93c66597291c4e6d : SUCCESS in 55m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/448b3fad68864577b7c606868430f527 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f4f222ff64407582b6998000900389 : FAILURE in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b92b59822bd1435fa94eb9e40a316e44 : SUCCESS in 20m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32475887f73e4e7c9883d12247390962 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ad45a64d71b9d761aece4c55528987d485130f9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 15:07:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"75a85e75c74a62d98a0010d5be5a1016aab3c5d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:18:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dedd9b6b21a4c32a51c489f89d66559\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcea6c710ac94e3793cec4348f88f8b7 : SUCCESS in 52m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3eb57ff6fa8e4f00a95880ffd1ede2d5 : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/41eaefd9713949d4844a07d8c88a29f5 : SUCCESS in 1h 08m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5015dde11714b6389fb263e6011ba7d : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8ed1b58cd98d1e2032d73adaae7d03de3915520f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:25:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1e8e9ff80334fdf8e48776613cee468\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6b11d72c894fff9398111a1f3d17ac : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/58fa4aa0ec3349738829f0b52c2eb4f5 : SUCCESS in 1h 07m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e32024596f32406bb6e769bbb93b468b : SUCCESS in 1h 01m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/827521a40fb64173822b609a615b665c : SUCCESS in 53m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e58bd8dcfe24264900002129fadb359 : SUCCESS in 46m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e29405537c04e68af930b37c7bb10bd : SUCCESS in 29m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b23b5391ac8d4fd0abcda0511284e85a : SUCCESS in 27m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d2ff1e3a54242a2b884ff5bbfbc8b97 : SUCCESS in 29m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa20169c6ba34504899b1674f5192011 : SUCCESS in 12m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3df598d1c1244d298584a3393bc3f0ff : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c118ba13e1ce4dba912d0817d26c5121 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8a0f5ff87a4b648ee73fad359b955d : SUCCESS in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba8c52b640a84c89abb6ba58b2faa865 : SUCCESS in 19m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d19812b7aadb40d38b97a5f5bc183aaa : SUCCESS in 19m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/665def396d4f4833838b3558e1597a1e : SUCCESS in 19m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de54a148616d4eac944b659bfe06c45d : SUCCESS in 26m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9a0d85b586d42979876e2996595ff18 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ccb3be76d6942c6bed6c082bccf6f2b : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f59b7a5e32f046aa869fc0fd0f527e04 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bf93c04f2f0f44b296178c5f593e9ddf : SUCCESS in 13m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0cf65e4c3179413fb09b636aa5f96665 : SUCCESS in 26m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a539d62612a9420e8a9c72ef0b10013c : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5dd65e719d44a0887d4b7fd09930cce : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18d49a938a8c4fef818f62a075083fe9 : SUCCESS in 7m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3ffaa9a0879e47af9652b9104befcd72 : SUCCESS in 1h 02m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/712c91b20cbe4dceb032d639c9a1316e : SUCCESS in 1h 04m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3c69961d45543c581d40b82b3413f46 : SUCCESS in 52m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5c290da9da348a5a0d0cc205d5848e7 : SUCCESS in 55m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed6a1c5eafe6487b9dd32e200d73ec66 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5dfb52023b24fe0b9b5449f77573318 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fea0d1ccf75647539fc689b26b72bce2 : SUCCESS in 22m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6436e1b9e19d402e99c47e4f68e9b5eb : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":3},{"id":"82a64093bb7cc469662cb740f8eff7ea3b8517ac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 18:45:37.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79444c71355436334b924702c805c0961d909084","revisions":{"2450bfcd2dffcde51b2ebca75e9590fa56c63458":{"kind":"REWORK","_number":1,"created":"2024-02-23 21:07:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/910029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/910029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/910029/1"}}},"commit":{"parents":[{"commit":"20da935f5c022f9aab81e6fc402313d4613e6c7e","subject":"object-server: return 503 not 404 if meta file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20da935f5c022f9aab81e6fc402313d4613e6c7e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-23 21:07:18.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-23 21:07:18.000000000","tz":-360},"subject":"fallback to .data meta on PUT w/ .meta unavailable","message":"fallback to .data meta on PUT w/ .meta unavailable\n\nChange-Id: I072974aab22d8b698152bf2f05c7ec97844f89f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2450bfcd2dffcde51b2ebca75e9590fa56c63458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2450bfcd2dffcde51b2ebca75e9590fa56c63458"}]},"branch":"refs/heads/master"},"f740bde10ed56ce11c4b2ffd8630ca6ed2535074":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-26 15:18:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/910029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/910029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/910029/2"}}},"commit":{"parents":[{"commit":"b80a986cf4c7b5e47a462c92d897669d30ee98b1","subject":"object-server: return 503 not 404 if meta file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b80a986cf4c7b5e47a462c92d897669d30ee98b1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-23 21:07:18.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-26 11:43:47.000000000","tz":0},"subject":"fallback to .data meta on PUT w/ .meta unavailable","message":"fallback to .data meta on PUT w/ .meta unavailable\n\nChange-Id: I072974aab22d8b698152bf2f05c7ec97844f89f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f740bde10ed56ce11c4b2ffd8630ca6ed2535074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f740bde10ed56ce11c4b2ffd8630ca6ed2535074"}]},"branch":"refs/heads/master"},"79444c71355436334b924702c805c0961d909084":{"kind":"REWORK","_number":3,"created":"2024-02-27 15:07:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/910029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/910029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/910029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/910029/3"}}},"commit":{"parents":[{"commit":"e4d5c7d4c130319abc33e9b9b3226eb499ededde","subject":"object-server: return 503 not 404 if meta file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4d5c7d4c130319abc33e9b9b3226eb499ededde"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-23 21:07:18.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-27 15:06:56.000000000","tz":0},"subject":"fallback to .data meta on PUT if .meta unavailable","message":"fallback to .data meta on PUT if .meta unavailable\n\nChange-Id: I072974aab22d8b698152bf2f05c7ec97844f89f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79444c71355436334b924702c805c0961d909084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79444c71355436334b924702c805c0961d909084"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
