)]}'
{"id":"openstack%2Fswift~910219","triplet_id":"openstack%2Fswift~master~Ibc760e641d4351ed771f2321dba27dc4e5b367c1","project":"openstack/swift","branch":"master","topic":"p-diskfile-enodata","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-02-27 21:32:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc760e641d4351ed771f2321dba27dc4e5b367c1","subject":"Fix diskfile test failing on macOS","status":"MERGED","created":"2024-02-26 11:18:47.000000000","updated":"2024-02-27 21:33:13.000000000","submitted":"2024-02-27 21:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910219-p-diskfile-enodata","meta_rev_id":"f9019bba3284fb3e668016f9f447455334c0f6a6","_number":910219,"virtual_id_number":910219,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 21:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-02-27 10:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-02-27 10:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 11:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-26 12:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 16:12:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"0ebaf97d4f2abf97e072f6d681641cec9f24d87f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-26 11:18:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57f7575ccba852ffdc0f22c227fb3fc68eb7a1b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:49:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529f4a1d46014b77828c9ea59782381f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d4406c745c64b7da8d985e4cd9f94be : SUCCESS in 21m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5562766a6274d868e048ac98ba011bd : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/adf4f6d24eb0488abfccde61d0abe672 : SUCCESS in 28m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/87caa397f7914902b0e576633a73b350 : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1a44eac706c3248ca396c9ac91af892867d64344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:19:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e03d2ec261849e8bec784e7b7923be7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a092f2347a848d08f7a6001333ccb24 : SUCCESS in 8m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8411256767e487d98a3b21d45c78c2d : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83afc0d42ac4f3c8b8725a90892b011 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96ca17ca14b04e39bc032c69bcd855cb : SUCCESS in 18m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5dc97f6bbd3e46fba01ce919bfe7e66b : SUCCESS in 11m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/51f8bd75c93d4033b7069bf9d2e49942 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c23e08440c29401cabb48a541a36a82a : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f597fa8d77794f98b12dc1b78b7be984 : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa50caba1be4928930c11fd1dbb5b6c : SUCCESS in 13m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f21106bb9ee94b818704ece9b919cb51 : SUCCESS in 20m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6ee53dfb1c404088951a038a2636d5b8 : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d941a438ee74dceaa1c9b9094e22e9b : SUCCESS in 24m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bf13d97cbc54aa794b97530beb1b96d : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd41984f664745e39afbd673ab9284c8 : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fda767fba65943bb995890712f48cfa1 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/36ab50ff2b2e4dc5b0bd82762e28b04d : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/91165d8b7d29443aa297023415062c5c : SUCCESS in 18m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52c229b9d19043828601456ab71927cb : SUCCESS in 54m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4444a10db01142c2913aac403bd4078e : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780f60f10418452684338b02023c77ea : SUCCESS in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0236dff21ae64ef385ff65f1e7806a29 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59079f0f8a90454abe313d22e985a405 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e325329c6c7bb3492426052ea88197fa8ce12791","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 10:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fd1ad6240b4c481b4992ee657b01e034b7d0cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 10:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6fc474b0f0356ad2b3a0ec626747c9a8d8476e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 12:16:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb3f8928372544ccb6881b80fccc5fd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d08ff20fd1d467394219cdeff1555c4 : SUCCESS in 8m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4d424d1de35540bc8aa188b8449a2aef : POST_FAILURE in 12m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e71b7cfb182b4d8ba4ce92e7556a7a03 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea655007c1b640fdb9824fde58e6c7c5 : SUCCESS in 21m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abb8a246767b41408d4ef75661ac5a52 : FAILURE in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4af9d88333dd420db4759b4061554d09 : SUCCESS in 18m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8ecd5e69efd44bbf8470ed6ce913c569 : SUCCESS in 15m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d9f6ac8affbc452e8529cbf77a410b5a : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/569271ede6d74568b33b673f0b66ad24 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fbd78717da2e43d4bcfd16cb5e110ad2 : SUCCESS in 24m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/158dc5d580274ad4a108a4b35d41213c : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4d2bc61d62d8429e8a69eb6b570eec3a : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/450e5c01989548da86e34b89ead6163b : SUCCESS in 20m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27652f79361944c6814999366d0e8199 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d44055b9db3c4a1ebd84ae3506e207a0 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9914bc03fa804583a47fbe0ed6e3f893 : SUCCESS in 7m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5dfab140872e46af8764c5e47ecb1396 : SUCCESS in 1h 13m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a062f20f44df45a8b13c1a1a9ef6b4f1 : SUCCESS in 55m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f4e6cf2a572421b8d9f1ff6629395b9 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2efb516d98d4902a2089bcf55a4a548 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2c575ae2e73db4330ff7e0055164caac4a7f4e46","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 13:40:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8df4ad8db35c8794fdc069dd03ee714afab821e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:35:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/884acf89376c478f8a08d6dabc175ce8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c642c6470c44f019e8c0bc646f4f7eb : SUCCESS in 52m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ce4f07048064d97afb9ff688ad56689 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce8878e6ca65455394a0207210fa0772 : SUCCESS in 39m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aea8b6644a504c5fb669bed973c1e912 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"53cc6eb1513591cab273aa4aac7dbb5a19445017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:37:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4cd222d0af64008a378702a031eb622\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f90c97a487cb4acdba321004d1409c5d : SUCCESS in 8m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86f5b33b7eb048f5a29ba4c6a3c5db88 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c721672910924d60b42f60dcf8038009 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/510c0ea239054ea08b45a2d897171f66 : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8978b649b6e54847b81ce808e85060cf : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87b2a21cf63441a79602b6eda9e1f503 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/070b0058fe534e5582f92064918be7f8 : SUCCESS in 28m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d5092114b7d4b81b59924235ede94d9 : SUCCESS in 18m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96e161c2777f4a1383a02b536f996c52 : SUCCESS in 13m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/020e7084d64548b2a98e2b8cf14bd261 : SUCCESS in 16m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ec139487216c4de8aaa9a0c339b21bbb : SUCCESS in 27m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63f462e0d92c48578e32499bac748248 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/36780279ac08463782cf420b50193478 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6371ca38063e4ffa96127fe24ebcf963 : SUCCESS in 27m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d13891399cf4d8b94d66863f17fb83c : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd6d9bb7e8134fbfae7822099b9ae3e2 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c71236ffae949489f5a3b51d3755fac : SUCCESS in 18m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1b892a2f58a4aefb188e243d583353f : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6bc0a1ba41d4173ac28508c40e94a69 : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fe93378ca3b473d922cb1e9d8bdd1de : SUCCESS in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476a3869c4ba49f6bc36c9a669a94f83 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a34e8ddd9044fbb8ea1e24a87b44b86 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcfe4b23e324c1b80f30ed044c32dfe9c13933de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:38:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a65621b702d179441cd43a6e5a693385d1dfae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:33:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1187c2a8e1b04185b77e6da0e5c34118\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fc5ce05d6b4d14895b8a301c389d6e : SUCCESS in 7m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9c526cf955a9432a989201f570753ab6 : SUCCESS in 12m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/204452849f44427fabed3573a02465a6 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f8f28e1a7034a8ba0f212e6437d1901 : SUCCESS in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3227442a55d6495a99977992247b88e5 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea2cdd81d06b40bc8ddb8a5d1a3be511 : SUCCESS in 18m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7acc9c8122af4965834ba9b9700cbd15 : SUCCESS in 18m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2e841786ee144d489dc4c6f3d7779c24 : SUCCESS in 18m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b74d1fa97e043b29e551809d0c5c716 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6bbf8356f6984550be136518939d73c8 : SUCCESS in 19m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98f9ce9a9c7d4150834a1fbc57648b83 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/371ed2765e444f42946d08c20759c53a : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f22264ee5ec4daf9bb43883b7c20ae9 : SUCCESS in 13m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e00106fd6eaa465d88f0553336edfe8b : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6dfe9809abd4f33a35a630feceebd49 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f85b97b5e754a579a8524b195f08583 : SUCCESS in 10m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c035a163f204cd9b75fc4e86f85b0bc : SUCCESS in 49m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f6e39d0810d44a29f1078a574e9eb61 : POST_FAILURE in 54m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b15e825968354476b02a8441e6af3734 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406e5da59f1c49c39693c4c4133956c0 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f61a3f6b5ab98cfd6d1f73fde3b79fa8f1eca3eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-27 16:12:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b32054faf6cf76f05163ce219d44666d93620aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:10:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceab46261d9345cd92fa452c1adb9f92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a45b26178440dc9454279b02cc2a52 : SUCCESS in 10m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd2bea170b9e45f58f073977e20b7c6a : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e43d5c5f8544e0ebb327b4e2cba3325 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b7fad2416ef460eb20d3df8bdafefbd : SUCCESS in 22m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bfa1af177bfd48f08887314e3620e86c : SUCCESS in 12m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0377d72d31ad48b7aac75597cbe320b4 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f090a656ff2d49e38eb239a9d4fbe47b : SUCCESS in 22m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6789bb2d144749b69271ee91d4587ccc : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/83426bd83e6c445e95ec313cfdd677ac : SUCCESS in 13m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bc9e62ada86843079df9058740332eec : SUCCESS in 18m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ba5568a28d1b411e8e556c027023916c : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c5a00b469ee4e77bda163879b206a03 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b79d1d3d7aeb4fbd858c310aec5475f2 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6feb1544ce6b47419986a8c17df60a7b : SUCCESS in 17m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0c9ddc3763ab4ee283cf10844fb46e0f : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ce7c3f3e7854bfd934c0283c715b89b : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6070b6478ab4a8283ae2f5a140797a4 : SUCCESS in 16m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e048472034149a1b11d3302609f542d : SUCCESS in 49m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/082cf962ab5c42ff862765f8ed958776 : SUCCESS in 40m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3ac8ada331149e4a2c85304ba3704d3 : SUCCESS in 20m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e1026a6f4574323b22f2d58c9b4143a : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7efd77bbe57b418697a1974079c0f048 : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e355439a3277c32cb3fe640422f60ad58c44e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:11:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c63a7fa521fc13559a598f022ae3847b6518756","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a027fb3598c9415483b314272e76e281\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61213e325c2b4b8abd5647d7bc1fdedd : SUCCESS in 1h 03m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1e644ac45ce4674925bd386d713c35c : SUCCESS in 1h 08m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aba91bedac65492d93067799ad458798 : SUCCESS in 19m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/054de99bf0204f67b81292310f436332 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":1},{"id":"437b63ef8235c6f5a104c115e9c5e61fed51b7cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:32:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1362084db1444fcb787dede9299bca6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f151164d8e4401393786a94e846b03f : SUCCESS in 10m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/986d8ec4ccef46ac98a619b191e04a6b : SUCCESS in 12m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/048862fe80494fc891d09f97204b5f13 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dc9588e559de4a0dba9d3fb1b7adf0d5 : SUCCESS in 20m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12581b4b16e94be69df002dbae99ac19 : SUCCESS in 18m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0fbd27cd128499e95b487de3f585698 : SUCCESS in 16m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/20ec58db7ccc47109cfc9d9348538eb1 : SUCCESS in 19m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/262c7a5f4d4d43adbdef2d90bc4f8276 : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/885d7113011e46af9887421398ab0b05 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62b78fb2915046df94ad960dcf450f54 : SUCCESS in 21m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dce6403f13af43c0b39483a07d3c3e94 : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de48cb2f08704e82ad62aee6823d5435 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e91941e6ef294898ada245165476ab7d : SUCCESS in 19m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a51dd9bc420149a3a97b65d75fc817ee : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a3170d9d76648978f3270e3ede094f9 : SUCCESS in 5m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d6d33f041f144938ee6837efc722712 : SUCCESS in 6m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/876d1953aab64178afb2068a87fca626 : SUCCESS in 42m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32dac5ce24d043849e851640fc00b739 : SUCCESS in 52m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ed7e7030f146c09f98f6861f85a990 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c172075c76b34b9abed15f267008ce41 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a5de1df69106140103af252f0eaa46215706b464","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:32:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f9019bba3284fb3e668016f9f447455334c0f6a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0296778df7414a63a3a1c8834da97582\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fcf01e6def46d5bfadf30141f84e98 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3ee17531b7c64337984f1373869cb97e : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/89c284e51ee940b689004a2dcf9c688b : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2da150b89088f845186f84d5408b46a1b1807876","revisions":{"2da150b89088f845186f84d5408b46a1b1807876":{"kind":"REWORK","_number":1,"created":"2024-02-26 11:18:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/910219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/910219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/910219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/910219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/910219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/910219/1"}}},"commit":{"parents":[{"commit":"50336c509877e7c325823498cd67bf6cf4f4bef3","subject":"Merge \"test: all primary error limit is error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50336c509877e7c325823498cd67bf6cf4f4bef3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-26 11:12:47.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-26 11:12:47.000000000","tz":0},"subject":"Fix diskfile test failing on macOS","message":"Fix diskfile test failing on macOS\n\nThe existing test fails on macOS because the value of errno.ENODATA is\nplatform dependent. On macOS ENODATA is 96:\n\n% man 2 intro|grep ENODATA\n     96 ENODATA No message available.\n\nChange-Id: Ibc760e641d4351ed771f2321dba27dc4e5b367c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2da150b89088f845186f84d5408b46a1b1807876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2da150b89088f845186f84d5408b46a1b1807876"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
