)]}'
{"id":"openstack%2Fswift~910286","triplet_id":"openstack%2Fswift~master~Ib7db2f077e343c26b49f760eb368329f1c718ef0","project":"openstack/swift","branch":"master","topic":"open-expire","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-06 18:25:17.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-06 18:25:17.000000000","reason":"Change was abandoned"},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-03-04 23:23:55.000000000","reason":"\u003cGERRIT_ACCOUNT_36763\u003e replied on the change","reason_account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-03-06 18:25:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib7db2f077e343c26b49f760eb368329f1c718ef0","subject":"add documentation for accessing expired objects","status":"ABANDONED","created":"2024-02-26 23:33:39.000000000","updated":"2024-03-06 18:25:17.000000000","total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a59b31f95a6180dd7e1a8010ec6647234e8b666d","_number":910286,"virtual_id_number":910286,"owner":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 23:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-27 21:21:12.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-02-28 14:59:06.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-02-29 11:51:07.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"34162d45b15221a64dd11c2f9f4160b69f16859a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-26 23:33:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cce06458ee1fb55d0ae28a5dc6b536a63407488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 23:43:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41d628442204f4fa31a7bf6323fb2e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c609cb7b2b4bdba69e42d34d96de28 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"21b696c39abca6e593884bf5034be4565acb2cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-02-27 20:51:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_36763\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"_revision_number":2},{"id":"ffe9b5c0d6d9de252fdaebac3102abe53690ec72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:04:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39963b1013064ba4b014386387a6bdd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238277799e524a74b37707f480ab0937 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":2},{"id":"94ce789122fac1adeaab4eb7383d385715533a30","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-02-27 21:21:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"738222d251fcd5ad71377118f9c794171a196e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-28 02:43:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7e43f1058f5c29139cd5b1b11b829c2595473e69","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-28 02:44:17.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c97e43076b3f8dda1c74d15195227fc38e22a63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:28:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e244f4359b12411b847b3d5888a867fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161af6fb08a249969441e22b40b55891 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":3},{"id":"28f03b1a3416e9ea1a39c93c3a70fe9e589ea82c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-28 14:59:06.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51f4342af2d69145c3c5dd893af9164d6dab9dfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-28 22:22:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4fcd149d554291be7e65263daebaf812aee55399","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-28 22:23:00.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"53ca45818dcb39a12c78432c0476d0e07de9a25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 22:33:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d2d09f74f34d078b2b6618808893bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad99a3d537d4ccc8f67b0ea129ed7d2 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":4},{"id":"668c65ec1a85b94ff6cb704aa8aa641665c2b62e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-29 11:51:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a600915b9b7a0c06d1f4f0079ab599266d660eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-29 22:12:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d93be439383e16a1b48f9d1f0627dd78a9f22ed","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-02-29 22:12:40.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6514c69039186da4e3803c3447bc060913a5a97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 22:26:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c58371c99c640e3aa91dd97182b8f52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35dc94690c8a4305861179b4dcdc4e96 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":5},{"id":"cf81b06582ce72679f94e7db1bff47bbdadcef0c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-01 11:01:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"30aa6d6a24f1f63e55428c5ba7defe5961c76b18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-04 21:59:14.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_36763\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"_revision_number":6},{"id":"91c35505930c21b992aae21b429eb99fb8fd13a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-04 22:03:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"399d0021069623957e9281bea626882892c1ded7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 22:13:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4481e45545cd464e83f18b34fec19dea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b582c1c5945e45e0a613fc64b07c8402 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":7},{"id":"3a2fc245f7786f36c76737e5c18dda7523764bac","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-04 23:23:55.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a450c53c37a9a10b33cbfed95285c04af571ef24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-06 17:19:53.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_36763\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"_revision_number":8},{"id":"a59b31f95a6180dd7e1a8010ec6647234e8b666d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-06 18:25:17.000000000","message":"Abandoned\n\nSquashed into patch https://review.opendev.org/c/openstack/swift/+/874710.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"462afc423884ec07a37dfc0724920fba1cac5eba","revisions":{"8dccbd6c735e7481f59ec53281813ba96d744293":{"kind":"REWORK","_number":1,"created":"2024-02-26 23:33:39.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/86/910286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/1"}}},"commit":{"parents":[{"commit":"9f804459afcd1ac55f1d0649bbc210625b5fbb71","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f804459afcd1ac55f1d0649bbc210625b5fbb71"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dccbd6c735e7481f59ec53281813ba96d744293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dccbd6c735e7481f59ec53281813ba96d744293"}]},"branch":"refs/heads/master"},"35e998db3ffb19acb008c9301a470e0d634c300f":{"kind":"NO_CHANGE","_number":2,"created":"2024-02-27 20:51:08.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/86/910286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/2"}}},"commit":{"parents":[{"commit":"536398a0796778e3971f7630c3d4452ce08f4f90","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/536398a0796778e3971f7630c3d4452ce08f4f90"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-27 20:51:08.000000000","tz":0},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35e998db3ffb19acb008c9301a470e0d634c300f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35e998db3ffb19acb008c9301a470e0d634c300f"}]},"branch":"refs/heads/master","description":"Rebase"},"ccf5182fe6363a2b4bb21a388c7dfa8a7d25e967":{"kind":"REWORK","_number":3,"created":"2024-02-28 02:43:52.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/86/910286/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/3"}}},"commit":{"parents":[{"commit":"536398a0796778e3971f7630c3d4452ce08f4f90","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/536398a0796778e3971f7630c3d4452ce08f4f90"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-28 02:43:41.000000000","tz":-300},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf5182fe6363a2b4bb21a388c7dfa8a7d25e967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf5182fe6363a2b4bb21a388c7dfa8a7d25e967"}]},"branch":"refs/heads/master"},"1c7213eb012868249b778535ecb0fd003b16b97d":{"kind":"REWORK","_number":4,"created":"2024-02-28 22:22:29.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/86/910286/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/4"}}},"commit":{"parents":[{"commit":"536398a0796778e3971f7630c3d4452ce08f4f90","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/536398a0796778e3971f7630c3d4452ce08f4f90"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-28 22:22:20.000000000","tz":-300},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c7213eb012868249b778535ecb0fd003b16b97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c7213eb012868249b778535ecb0fd003b16b97d"}]},"branch":"refs/heads/master"},"1cf942ea99217f3b33d18207c34e2370c9aba85d":{"kind":"REWORK","_number":5,"created":"2024-02-29 22:12:14.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/86/910286/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/5"}}},"commit":{"parents":[{"commit":"536398a0796778e3971f7630c3d4452ce08f4f90","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/536398a0796778e3971f7630c3d4452ce08f4f90"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-29 22:12:05.000000000","tz":-300},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cf942ea99217f3b33d18207c34e2370c9aba85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cf942ea99217f3b33d18207c34e2370c9aba85d"}]},"branch":"refs/heads/master"},"959726dd551da9f7d56842bc07ebc71ab452c2e5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-03-04 21:59:14.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/86/910286/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/6"}}},"commit":{"parents":[{"commit":"4d50eff2fe695f7e2030074a9f61176271c5d56f","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d50eff2fe695f7e2030074a9f61176271c5d56f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-04 21:59:14.000000000","tz":0},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959726dd551da9f7d56842bc07ebc71ab452c2e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/959726dd551da9f7d56842bc07ebc71ab452c2e5"}]},"branch":"refs/heads/master","description":"Rebase"},"cc4278920f76fa7077cdd50d2461f15194cda2ca":{"kind":"REWORK","_number":7,"created":"2024-03-04 22:03:18.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/86/910286/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/7"}}},"commit":{"parents":[{"commit":"4d50eff2fe695f7e2030074a9f61176271c5d56f","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d50eff2fe695f7e2030074a9f61176271c5d56f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-04 22:02:19.000000000","tz":-300},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4278920f76fa7077cdd50d2461f15194cda2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc4278920f76fa7077cdd50d2461f15194cda2ca"}]},"branch":"refs/heads/master"},"462afc423884ec07a37dfc0724920fba1cac5eba":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-06 17:19:53.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/86/910286/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/910286/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/910286/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/910286/8"}}},"commit":{"parents":[{"commit":"dac78d726c5864b2a9415f7d5fb35dac8ade38d4","subject":"add enable open expired in proxy config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dac78d726c5864b2a9415f7d5fb35dac8ade38d4"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-02-26 23:33:12.000000000","tz":-300},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-06 17:19:53.000000000","tz":0},"subject":"add documentation for accessing expired objects","message":"add documentation for accessing expired objects\n\nChange-Id: Ib7db2f077e343c26b49f760eb368329f1c718ef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462afc423884ec07a37dfc0724920fba1cac5eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/462afc423884ec07a37dfc0724920fba1cac5eba"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
