)]}'
{"id":"openstack%2Fswift~910351","triplet_id":"openstack%2Fswift~master~I7845e3b81d0e65d5674896853ffc91cb4942dd18","project":"openstack/swift","branch":"master","topic":"p-diskfile-meta-deleted","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-05 21:10:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7845e3b81d0e65d5674896853ffc91cb4942dd18","subject":"fallback to .data meta on POST if .meta unavailable","status":"NEW","created":"2024-02-27 15:07:15.000000000","updated":"2024-03-05 21:10:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"ebbceaef5e75090fbc7d5ca7d976e56e14c371d5","_number":910351,"virtual_id_number":910351,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-27 16:24:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-02-27 15:09:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 16:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-27 16:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 21:10:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"57470ac50d682fc4b5fc74f036e768e516bd3f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 15:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d166de2e4b876122d153ef4fe2cb6aa82a691e5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-27 15:09:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ed69d660225b110d33aa11f7f0cab8adc91413f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:05:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/113b0ce3ef5544a2b3fb17c148bdf100\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7749aee0a96479b80058c9e135890e0 : SUCCESS in 56m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ef8158445fe4e29a2e3592becf1ea69 : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fd6b920418840199d22ab347a617f5b : SUCCESS in 53m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5422905197c041269368ff504d97ee11 : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":1},{"id":"42a227993ae258d4d005b942420a90a5565f4787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6287af7ca90b45c1bb1d12d03ca00948\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/067e27031a5c4e3c8809ac78422ca851 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/90dff1e5aaa4479d8606f4bdfce40df1 : SUCCESS in 53m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/464ec41f52d3420ea31b65ddd034eda9 : SUCCESS in 57m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb1f7a62f483441bb46492a47c307acc : SUCCESS in 47m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81a5eba366a64e8896dd5020985869f7 : SUCCESS in 1h 00m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bdf01b6be3e4bedbdf880ed3a07bbc2 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/838afb549bdd43d19d46573e889c85ee : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1ce9b06adb54d21a59a9668a40f6247 : SUCCESS in 18m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/479c296d01e7441ba982149fed47ed7f : SUCCESS in 11m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6fd706d1701e40ab8f1200b20ca2c0be : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42b0d76d9e92471dbe812af061ab7111 : SUCCESS in 20m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af6f3f2166074e2784379c1a26893b7f : SUCCESS in 18m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aad16903e65542038c4b08cdfcbdcd8e : SUCCESS in 18m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4d7aaca925e440eaab37cb52a32ac594 : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bab029bf702c4537b4647c211e2204dd : SUCCESS in 21m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b5502eaa3eb4d9d9858efb81948d924 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ed4192c7f6d4a3c940eade56f64c8e5 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f68da5b718da4f15b009a463c34ff0ee : SUCCESS in 17m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a9489911379484e99dc00eaef7532a6 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a1387e6085474510ab81681d57bae764 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/17f08c4e05cd4983ba45c5c461fb32ca : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97d1435af35e4b07ae23376fcdbf6320 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/716d00d246824eab9703a6a8a6fd402d : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/733c0404e3c14be6bd32f2cccf4de7fe : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a1bb22fbe63c4c089fd3a8e0ff05d1fd : SUCCESS in 48m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e79e2d58839451ebb28910d65a5c7c2 : SUCCESS in 1h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6defeb04ce54483a97f53fa367fd1335 : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7293aaa1bb747c39c6e646a2813dc28 : SUCCESS in 41m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5907058d7a341b39c4e2d4103030af2 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbec40392064eaca7a36689facb2a1a : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a22e3e7526a42c883457cd8f5c1adc9 : SUCCESS in 21m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb55dfc71253490d9621a1935985c7cd : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ebbceaef5e75090fbc7d5ca7d976e56e14c371d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 21:10:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f93606ea1f21c3dc2fe0df7ef2e1810faf06eae","revisions":{"6f93606ea1f21c3dc2fe0df7ef2e1810faf06eae":{"kind":"REWORK","_number":1,"created":"2024-02-27 15:07:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/910351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/910351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/910351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/910351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/910351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/910351/1"}}},"commit":{"parents":[{"commit":"79444c71355436334b924702c805c0961d909084","subject":"fallback to .data meta on PUT if .meta unavailable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79444c71355436334b924702c805c0961d909084"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-27 15:03:02.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-27 15:06:56.000000000","tz":0},"subject":"fallback to .data meta on POST if .meta unavailable","message":"fallback to .data meta on POST if .meta unavailable\n\nChange-Id: I7845e3b81d0e65d5674896853ffc91cb4942dd18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f93606ea1f21c3dc2fe0df7ef2e1810faf06eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f93606ea1f21c3dc2fe0df7ef2e1810faf06eae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
