)]}'
{"id":"openstack%2Fswift~910443","triplet_id":"openstack%2Fswift~master~I7c8adb129b8770eee501748a378f3adc42c8cd39","project":"openstack/swift","branch":"master","topic":"zero-byte-legacy","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-02-28 14:24:49.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I7c8adb129b8770eee501748a378f3adc42c8cd39","subject":"zero bytes manifests are not legacy","status":"MERGED","created":"2024-02-27 23:21:06.000000000","updated":"2024-02-28 17:35:09.000000000","submitted":"2024-02-28 17:34:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"910443-zero-byte-legacy","meta_rev_id":"903f56e7d5f8b6002561fcadfcb6610483b1fdbd","_number":910443,"virtual_id_number":910443,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 17:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-02-28 14:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-27 23:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-02-28 14:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-27 23:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 23:27:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-02-28 00:36:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7d4ed50e049d8f05872a95733ddfdebc09dceb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-27 23:21:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f34352b549edc7d3453bc9e890111609fd81935","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-27 23:27:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24a35142662277deb7e406f16439b83ddb03ad0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 00:36:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6097375d91a41179d63289cc0eb1ebb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7ce763b6774c61b287afd51da01469 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/d42e7ad6a4014ff189f4849ee94b10da : SUCCESS in 49m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8cd125a0f5c43bdbea41667d1ca618f : SUCCESS in 43m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c533337de2f3444b95f8e2081a0ed86e : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c78d5369681f43e986ad1f447cfd3100 : SUCCESS in 48m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b7e9a21f6db444293a0c198f0d83308 : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c38dfabd4bb44fead6a55fae841d4f0 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b056a2910ff4557865b8ea1254f8fbd : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4db642721a5d4e2ea69b19e93cd30115 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8382c93b2bd4504aeca5f5cc65e2ebc : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bce6697cd42647a3bd57811fbdf47f60 : SUCCESS in 28m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d42d38afd295417781009053b854adbb : SUCCESS in 21m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab46bda387084a3396b35dc3114be8a6 : SUCCESS in 27m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/34303af9d80b4aaab5df869a250f96cb : SUCCESS in 19m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1fd4cdbc3ee0498d8375734e293c5665 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/16f0461014c0448684e24391bdd0949d : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42c14d96b833460088a8d4c7ff90148b : SUCCESS in 26m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0f8344890d9453eb62cd2e94afba603 : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98b5435850924d0fae13e30d96e48ef5 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6d7e89b379d34396960dd2bfa3979625 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8f424cab7d0f4f179a711551cf396dac : SUCCESS in 28m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6f8da1767bb406c8732de3af4a18bbb : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d2b5590f5384bfbad0ab85d175b04ec : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9eaa7f1856f4d35b864f73f974406a5 : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/295dfbe1b0fb4756bd2965a5c6a2e950 : FAILURE in 59m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c07395fe2d75442280fbed2439a4b0ce : SUCCESS in 1h 13m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67dcc41dca784d4dbb771ad457a91ac3 : SUCCESS in 54m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bdf361abdfde4c98a142085c83bfe2b0 : SUCCESS in 57m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ef6718f22f645fa84ce32ffbe507181 : SUCCESS in 32m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da0cb40be67b4246904cf2b98cff6dc9 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa4e0bc32c454a7f9259ed8e1b6e77ab : SUCCESS in 29m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96352ec084a24fcabb9ea0ea4082d71f : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"faa440298f8f1da7d29312692a4d2d2f2271824a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 00:47:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d415cdcceb497db47766f23c1c447d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f19d5e9d3ab41fcb331275c5f158621 : SUCCESS in 20m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c2ad733a9c7477092633ab9e04127a2 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a4f0ae0483c4c3f87fb532d9c016867 : SUCCESS in 1h 24m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/07eb385ea6c749d1a68b167a86b33c6b : SUCCESS in 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8242c477a55b44ab7b6fb0afa322baee4535578e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-28 14:24:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70bb4cc07f2f7f4b565cfd0a5f9937d813ae06df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 15:42:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33297d4fab54450eb8f04958fe896038\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8c2a7da0ec744fdbeee8d1dfd8becf4 : SUCCESS in 1h 02m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbe1472b710d4d42aa57d880c53edef8 : SUCCESS in 1h 12m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78e45e38dc5d40038ab02b1255b9105a : SUCCESS in 21m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4036efea04d4a2b8c38773828789ccd : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5a74c77f4c3d3231c180eb0bda019474fb6f134d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa33fb4bbdaa48989b7c421d8fdc23cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5dfc033cc84cc9aa2d3cdaced946b7 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/ff5e352a478c464ebe69eb1ed34ffc74 : SUCCESS in 59m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9144ba7aecc340868428cbe0d0a49194 : SUCCESS in 56m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/28616bdea35a4e69bbb20d4d880f4b92 : SUCCESS in 48m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20f54877828d4f14bd2b4e4dd8e06d41 : SUCCESS in 49m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d248b12032aa4c92a4466e317ca9ad31 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4494cb29d2c44eebf447869bc08fe45 : SUCCESS in 24m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b849e81927cd4d7b820f310006a2b8c9 : SUCCESS in 22m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/88a56480dba94ea4a9d0b76a5eaa8a8b : SUCCESS in 10m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efd0248b48d74d43ae33779f3ada0d87 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3041e4b1722a49dfa6eaec9b59ab07b1 : SUCCESS in 23m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35866f62a695413fa68d8fb9dc23e27a : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b1a4196e4d94a9f868ed86e9eff2c9a : SUCCESS in 16m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bcfeda85347242e69dd9d9336685339c : SUCCESS in 16m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b6b0305b811f41dda41217f5a97f51a6 : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d216c93a40a45ef863cb0041545a579 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1879a06f22b46058c2bcd7948120487 : SUCCESS in 21m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b79bf7c09aa34b22b92ab7f93afd54a6 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e7dd53ce2e94865859b6b69ea0a43a0 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03c208bf63854cc1b6c20ccaaf333946 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9dc827cd932249c4a67d467fd5c59c4a : SUCCESS in 27m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a97afbc5de44a49a2db01349bb58ca2 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc8238d8d26b428d99626871b0fd5ad8 : SUCCESS in 24m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/20b6a719c14a4e668da06f5df4f1f49b : SUCCESS in 6m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4356a83ca1cc411d8ca2bc3255e096ba : SUCCESS in 1h 19m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4643ccf3212f4692a6c82a9a4555990c : SUCCESS in 1h 14m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a27aab03d4344b18ae011c4cad989815 : SUCCESS in 51m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40527de46b324228993d61ba5bc8e3e2 : SUCCESS in 58m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a8a49fb6843434091bfb0f118b79c2f : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92816ce2e27f49e88e1824f743f83ee3 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fae1b18cad141358395a4fea86c8677 : SUCCESS in 21m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5de78304bbf747c38359d61d1df0fd12 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"748125afd556a196f075358193c4e4a77f2edbc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:06:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b4ca784c06697aaf2b83c5653faee9cb47f99a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:34:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0585d48874e4cf69e2934a7ef350fb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5bdc4bb62f4699a7a9e353cca3deb4 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/23ed60f74f3f4a0bb9377166216370b0 : SUCCESS in 45m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/de0611096c2249b2942519c64e4d039a : SUCCESS in 1h 05m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae5e108829f24a6792fc16bb67817e35 : SUCCESS in 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/330c9472941040d3bc5a284e26d4c836 : SUCCESS in 1h 00m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c819e1e57a3a4a87b866759c0660f569 : SUCCESS in 12m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/edad9902587d4a2dbea53c147ce2ebfb : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1ec772bfe4e43c694b0445c83a06055 : SUCCESS in 24m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f88394c250704e19b7caa12957e6de8e : SUCCESS in 20m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f586e56b922046b89483c2fb975092bd : SUCCESS in 17m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8aea07a200db4e37a4d3edc58fb29f6b : SUCCESS in 17m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bab911e4130c4f2e8c4734c360b90f96 : SUCCESS in 19m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec783f836b1e49cfb4840b8bc7157a68 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf0da6b1a7f8417abeeacf96d620f423 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd33c5e8c0cc4c2b9f94517322daf512 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6af45767ea48433ba5d8b4daee7a1125 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3eb53864fbb2437284504e04f760f6d8 : SUCCESS in 20m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/babcafd1e4eb4c0abe321a16836bbe5e : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8333d5cd764a4f5fbef587449582f823 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5f21e454f8a43acb18cc3410ee9812d : SUCCESS in 9m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b1024666d964aef999dd2d091937fbe : SUCCESS in 58m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc78431280b24fd49fc2cb38960ae79b : SUCCESS in 1h 12m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5cca82752cfd4689902b14df39360cb5 : SUCCESS in 46m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97f70a4bad8b4dbfbea1f965555a8bf0 : SUCCESS in 46m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98de6ce9a36d48ef8cf99993c41f1f08 : SUCCESS in 25m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93de18c20062433bbe4331aff6841dc8 : SUCCESS in 5m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e62cdeb3b6cd4bc2a6f8ac17d46e121b : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6699916967bb9df71e88ff02477dee64ec81fe78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:34:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"903f56e7d5f8b6002561fcadfcb6610483b1fdbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 17:35:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b1bf3c37c74079ad691a744bac56b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62d28cec0074d7ca36d6f4132323ee9 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/87590e0d4fcb4caab6d159786f12e05d : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8e7af8e255ec4ff79e43ffe2fc7a5a7b : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"130188b6c0f30815ffd98c268bdacf7aa836de31","revisions":{"130188b6c0f30815ffd98c268bdacf7aa836de31":{"kind":"REWORK","_number":1,"created":"2024-02-27 23:21:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/43/910443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/910443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/910443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/910443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/910443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/910443/1"}}},"commit":{"parents":[{"commit":"4c5f41cc1f7a679388e924f4ed20304a5609080c","subject":"Merge \"Fix diskfile test failing on macOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c5f41cc1f7a679388e924f4ed20304a5609080c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-27 23:20:20.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-02-27 23:21:00.000000000","tz":-360},"subject":"zero bytes manifests are not legacy","message":"zero bytes manifests are not legacy\n\nChange-Id: I7c8adb129b8770eee501748a378f3adc42c8cd39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130188b6c0f30815ffd98c268bdacf7aa836de31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130188b6c0f30815ffd98c268bdacf7aa836de31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
