)]}'
{"id":"openstack%2Fswift~910842","triplet_id":"openstack%2Fswift~master~I3516823fdacbe8fd3c2434c0de9dedd1d82980fe","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-03-02 01:04:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I3516823fdacbe8fd3c2434c0de9dedd1d82980fe","subject":"CI: fix rolling-upgrade jobs","status":"MERGED","created":"2024-03-01 19:55:15.000000000","updated":"2024-03-02 04:00:01.000000000","submitted":"2024-03-02 03:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910842","meta_rev_id":"87be8c0885a2e20c1b501b658dbd9ee6fe70345a","_number":910842,"virtual_id_number":910842,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-02 03:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-01 20:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-01 20:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-01 20:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-01 21:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00ad50322bd838529c949a78bcb6b6b50ac59318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-01 19:55:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f7031ea93c581baf39d9b53f8f316785d9fdbc2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-01 19:56:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40962d429e640349ecc4172f312d2d898d110d64","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-01 20:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fece172befc20385f61eef6c32fb5c34b1f7f2c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 20:48:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1305b9f31a224372918d3c0ed58f11c6\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eba6b39dcaac481d8d89430714d9f642 : SUCCESS in 17m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c557c2673a04bc88facbf6d816ff429 : SUCCESS in 17m 59s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/2f81457cd5a541ff9de44886eeb746eb : SUCCESS in 24m 35s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/a5b6f0d64a2d4be997415a786fb107b1 : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/cb81cdd80cdc451da9e6938cfd7e26d5 : SUCCESS in 22m 21s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/6172a862c58a4b4f8a2e8ff5d157d429 : SUCCESS in 23m 56s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1f6cbc2c5c44c318da8425ef064da08 : SUCCESS in 20m 06s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2054a7fbde24f359d88cd659d013b0a : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d6aa99063ee4889a4c4a8ed6658f602 : SUCCESS in 21m 26s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/442847575da64e5187e9e51b9a068cf7 : SUCCESS in 18m 56s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/da282f3acad5439e88e9b2fe30e9b497 : SUCCESS in 19m 29s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/41d7fd79523444a2bd898c93c0c51f9c : SUCCESS in 23m 49s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/30ab58c5a84140eda8b1c42439fffc88 : SUCCESS in 24m 50s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/f96f1d3f70e94963a2148cab9e599c0f : SUCCESS in 17m 53s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/d7fe76eb116641b38194700eb8f61457 : SUCCESS in 22m 59s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/665713d55d3542708f78315fcbe5e97e : SUCCESS in 18m 06s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/262b30c6d14d41b5a4ca29c39b0e1724 : SUCCESS in 24m 09s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/d017e86ef0d84fc0a86f6cabced13d32 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1cf83435fa28302c9c77c61029aa55d9deedd6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 21:04:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5bc17414b5d4dfaa1af7c5524a0d1ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2780c57c3377437eacc163727faf45b8 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b63b12eef6544911a2bb09c282a44f42 : SUCCESS in 55m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c63c4e492b0044e4a4891f7695c95130 : FAILURE in 42m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ddf78d6ac514fd89d5d546cf7fb706c : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ac0624774354e33aa1ec128398193ae : SUCCESS in 43m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bad9c15b3d3419baf5f0cb0c1d4ae0b : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/400b72fd049f456a8317c36e45e9fe10 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02e98b04090a46e390dac09de964bea0 : SUCCESS in 17m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2910ce6f426943788a528662060204e7 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/16c8aee6d72c4d828c73ed7f68b3bc50 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d74c991c1cd4958abe1099acef6fed6 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7a4b01811f241b4b286b0730116fd6e : SUCCESS in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8119c998e28547b1849646867ef001fd : SUCCESS in 16m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d90953c99c2543b1a18cb7e988a39150 : SUCCESS in 15m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b84c2e090404e1fa5d50bcd676f98fa : SUCCESS in 16m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1dbfa272d4740b3b5816eb37dcfa2a7 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/824dec561d264b5f8675d3c4e050db18 : SUCCESS in 22m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/467b2942b0c34ba7b2b5fadec6540970 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/73c5cbf8c66c4456b88c3fabecd3df39 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7fe8775c835b46b6b1897925ddd24124 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cbdaecaff2d6442a8632a46c6a39562d : SUCCESS in 21m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8079ccb801d34926a1965d0edb9a7e94 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc30e37642ea414fbecce9b401c5a36b : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/06ec88aa2a034b009dfe379d4406bd98 : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b76a865c717440969c918b8ba79ba840 : SUCCESS in 58m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccdc6acb7b29497789b1de2cef0fa34f : SUCCESS in 1h 05m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09672b761da540d4b1d2f9def9baa0cc : SUCCESS in 50m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae003637e6ea48a2a5fd0f599f595bf1 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65eced59a7ee4718ad59a5164f28c4dc : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5a4eb98f5c4e7097b5d1eea2eea432 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2ef7d4c8b9240ccae9687f68e2500b7 : SUCCESS in 19m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fce15dac108b42ac8c8dd1b5decae1eb : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8cea42a001d66c5c1ef3450bbb640ca62ae0cd2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 21:28:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7134d288e11414783b7a9431f94c669\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e2f55faf9d744809fdb0e0ad210a852 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efd906d4919845b0ba4318b79bf40551 : SUCCESS in 46m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d90a2641abe944dab4224c5394fd808a : SUCCESS in 25m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb6be96f50014ead9bbcbcf507c1d860 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7f2cb1979f3a7e52425188a4f8370f5a02969a00","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-02 01:04:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0eddbd6c7963c2b06d130e12fc9ac9c17d5e5201","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 02:15:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80cb8716f60847b5bbb02ba86b09308e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bf60bafb63e478699778e53f68022b5 : FAILURE in 1h 09m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be0a958a15764ceea08d69384b6d5432 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae88a1bf1bdf4157a0634f21f84e58d4 : SUCCESS in 50m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d6087dc70dc4eb5b6fab9cce18e5ea2 : SUCCESS in 19m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2f0078d77476300723d503feeb0656a47936ba47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 02:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d6b077b7d44c49bdaa68eb4d539472\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb675d11d704cb091418b03ba76f2cb : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/1ba33a41419048dc835f7e0a9a5fe4b7 : SUCCESS in 1h 04m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/653e7ee4404b47e09423cf9cf9319abd : SUCCESS in 56m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd88390ff4034f2cb774f7cccabbba24 : SUCCESS in 41m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0181a93ac9fe45e0b562eecc9b2b6f6e : SUCCESS in 1h 00m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c3b17602feb414fb8afdaef7e79efad : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4745fa16d8c44659ad76e2561114082d : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fc9b5b3b4b84eb1a6019a58f063fcbc : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ad7ee52e31d4b98ba612c84f1476908 : SUCCESS in 10m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cbfd5010951541189315e1ebea8721c5 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9de0a927c0c1456187c5c8effe9c8629 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/121336aa0b84444c89499f1532cd4390 : SUCCESS in 12m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/187af44bcadf48a8a8fec8d88c8f85bb : SUCCESS in 17m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/93dd32dd755a4b828ac83ea2071a3d1b : SUCCESS in 21m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aafccd54a4b24d1d947e71dadafcecbf : SUCCESS in 19m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e61c6231941447349008dfe0defc62b6 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0afe92db0d494651a709794251d1ff9e : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19abd73059d949a083d502349cd5c48e : SUCCESS in 21m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/46792eaec5b340b28f5b5caeda97f25f : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef35694690cd475d9ed1badf8935d39d : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/97b19f96022041d2a736f2f871d09a32 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1241aaf1dd0142eab11cf34869475ad4 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b88c8758e07b4f25bc6e1135c9c91b9d : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67b0c97c56604ec8aa3ecf0aa7cd1ccf : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15c3b8bae239453882e89d3c8fde38a3 : SUCCESS in 1h 13m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e58abcb66994cf6ac78f3fd8262c32a : SUCCESS in 1h 09m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0a28fb3866c47b68a415905f920ef2d : SUCCESS in 50m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dfeb6babcf02421ab952d50bf59eeb5a : SUCCESS in 50m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f2b1128aaa24247a12ec6d5b58ea861 : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da83e957b3ea463caeaf625d0c5687e7 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56aed11cd1d742688396db786eaef743 : SUCCESS in 24m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0962e1b231940bcb4fbb2f095dd4cd1 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"443538b4839ef6c07be138a3bf0a76831262aeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 02:21:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a5d15e7e7c7db202c3b79c5ffa0827ffd11990c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 03:57:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e6f27010a841c69c5f01eb7f772d8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1949c200066f49afaab2e8df77022693 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/54b038629ce04e1a8ea534687c3c38d4 : SUCCESS in 59m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ba2e246383b84dcea2b25f02865e872b : SUCCESS in 1h 05m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15345ecb698843ac80356ffdd1798c52 : SUCCESS in 45m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adb458a54e5c48eda5f3471a28be8a91 : SUCCESS in 1h 05m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/061e870e0a3945a0b939ee36401347df : SUCCESS in 13m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dc09bf222d6f4704829cae6455ffa412 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f33ea37d8be2497eb054c600f783ebf6 : SUCCESS in 21m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26e2283a96a645cbb63671003e64a2f5 : SUCCESS in 18m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4cfb63625e54c21b9cab59f1bf54f29 : SUCCESS in 17m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bca429ab29764aada0128f80d0726896 : SUCCESS in 19m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7fbe0e948cfd44d1a19dba0ba22db4bf : SUCCESS in 19m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/42eb02b734a94a19b2bc9302311ceb78 : SUCCESS in 21m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1730b4f2d2e4461da30b64c026eab2e7 : SUCCESS in 27m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/156852bd1a60474b976a7cd28ca754e2 : SUCCESS in 20m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f3de6eb6494f446ea755f9c0ca0b0d9c : SUCCESS in 30m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7be78233b5b487f8b6897035134fb19 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/199caf42e7a8429bb72e27f7d4e8ae9d : SUCCESS in 23m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bda174e2315443bba6c3e2aaef73a2fa : SUCCESS in 7m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d1dcd7d0068d432d8c1f5a54ed089dc4 : SUCCESS in 7m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b6f50e6e45f4aed977b4563adef5ac1 : SUCCESS in 47m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/15c9646393ee475494cb2f659471f455 : SUCCESS in 1h 04m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f70edbe45c34fe495b82a576dd1babd : SUCCESS in 53m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a5ca7846db54342aa6aa02532d08dc9 : SUCCESS in 57m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebfa96329f814208968bc80a8ea6e7ce : SUCCESS in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fa71ed55e04f5592bc3991821cc5aa : SUCCESS in 4m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb154ea626f64e2f8d7a574676fc507e : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"573f6341e2aea979ea9cad4b5fd27c3916914a74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 03:57:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87be8c0885a2e20c1b501b658dbd9ee6fe70345a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 04:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f60d758893914b319a64db90d2f45b8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3279be87a57c4914865e33807e1bddd9 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/63e6b4d52ff048b680f7ef599d0fab69 : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2c8a84558a3d4c89aed06ed0a3ffbc8e : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"275af9e008fd296623d7a9eb9320206aa536fcdd","revisions":{"275af9e008fd296623d7a9eb9320206aa536fcdd":{"kind":"REWORK","_number":1,"created":"2024-03-01 19:55:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/910842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/910842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/910842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/910842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/910842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/910842/1"}}},"commit":{"parents":[{"commit":"f32f2dd023366d45c8062985f850a063d506c05e","subject":"CI: pin python-dateutil for py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f32f2dd023366d45c8062985f850a063d506c05e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 19:52:47.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 19:54:50.000000000","tz":-480},"subject":"CI: fix rolling-upgrade jobs","message":"CI: fix rolling-upgrade jobs\n\ntrain and ussuri are now EOL; drop them.\n\nyoga has moved from stable to unmaintained.\n\nChange-Id: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/275af9e008fd296623d7a9eb9320206aa536fcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/275af9e008fd296623d7a9eb9320206aa536fcdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
