)]}'
{"id":"openstack%2Fswift~910914","triplet_id":"openstack%2Fswift~stable%2F2023.2~I3516823fdacbe8fd3c2434c0de9dedd1d82980fe","project":"openstack/swift","branch":"stable/2023.2","hashtags":[],"change_id":"I3516823fdacbe8fd3c2434c0de9dedd1d82980fe","subject":"CI: fix rolling-upgrade jobs","status":"MERGED","created":"2024-03-04 22:28:16.000000000","updated":"2024-03-05 23:12:38.000000000","submitted":"2024-03-05 23:10:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910914","cherry_pick_of_change":910842,"cherry_pick_of_patch_set":1,"meta_rev_id":"d252c6d354aefcfcfea11971d7f49e5bc10ed4c6","_number":910914,"virtual_id_number":910914,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-05 23:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-05 21:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-05 21:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 22:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-04 23:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e6cb835ec395e7c0340a253854c524f67048ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 22:28:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"802ec7dff493b62ae56ff0ee9c5600d2029af782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 22:49:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8101fff9bfa44dd96b4ba7e3f5125ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22bb224d81ce4dc68ffeda793a6ea2e5 : SUCCESS in 14m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e793691798b43a2bf28947cf626ca8b : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63b621065759470dbdc20cc70b981494 : SUCCESS in 19m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/107868f06cab4440b29232e6481a6f65 : SUCCESS in 18m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8fd23eb1571d3d89cd2e74ef98509f9708296fcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 23:34:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ddfb26ecf0f4363888bac90e38969a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36c451c2d4e4975a099a42bdb346f8a : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8f56f49090994c4c935fe5df571e6f45 : SUCCESS in 42m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b0c72c1a3d74cb9bca3cdb41af9dcc8 : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/196535c765c74ba196e5a94bc2c81397 : SUCCESS in 46m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da30ee1351304767b7298772692ec88c : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d52785bc1d1b461d943316d83723ece9 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/194d79934e3342c18f01204bf18efeea : SUCCESS in 16m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74e9e3eb34ca4607bf497305fd27a2e1 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30a1a1d5f2f8471ba3406262bf945d26 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2e22af7839e44be8fd8deace4b005cf : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72f528c2ce42448aa508c667649e0c55 : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab25b600aa5c4c28be6c88852f89f4e8 : SUCCESS in 20m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/564ec3f053a240d6b85d9f242d7c45f5 : SUCCESS in 15m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c126b4b254ee4408ba72632e90111918 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30bec7b06f33457eab6bafb9aac0f7c1 : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bcbbc04148304fe09e954dac6ce13f8d : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f96b334053e54efca2e629711c3a081b : SUCCESS in 21m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9740c690bbed471d8942a49e3186f822 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6eac44f86aa844979ee9688cf2c26a59 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b90beed925b84c59bfa08aba464d5f49 : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d06cd6282204d46b2651487cae7f56b : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f11031823fb487db0b663b67111697a : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d13361eb19e47f7858fd0abeebd0952 : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e9025c20a4e4634be1325af93da29af : SUCCESS in 51m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aae526b8a04549b996ef0dae90eac446 : SUCCESS in 1h 01m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee334c508d184f56bd89e1069d80460d : SUCCESS in 51m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36e27db61e534ff5853d605cfc5f1bc7 : SUCCESS in 46m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a89c185f6774511ba714261615ae6c2 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c486af442ea415aa2844a0f73e0a6d0 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2dc6cfb6a72472eb81addf419b59f1f : SUCCESS in 24m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/effcf248e2ad4cfbb808d54da3b143c3 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"62422ea852bbfeb3b42178d239ebf40c7c69af4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 00:13:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51c9269d0e28d541fc5ca1846629abe42056189e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 01:04:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6792cb5ec880454f98ee2346e8eeaa55\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d48b3edeafda4d5fbc15e2d9b6b7b8aa : SUCCESS in 16m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b0e799b94645b8a93a21634174470c : SUCCESS in 14m 15s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce2df74679134006aa11eb45ba8895e6 : SUCCESS in 20m 41s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c42682fcd9842a29cceb9479ea95725 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2c010a43a5c48b6b49814e9518d0cbc : SUCCESS in 12m 24s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/dac790de0c364184b863a644445e9634 : SUCCESS in 16m 09s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/fc866ef65a4f4806ae8e64f93ef14ce9 : SUCCESS in 23m 08s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/eede7796405149e7afe84351bd8b5fc7 : SUCCESS in 20m 19s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/72bbfb0d20ff4a36aeebbdd2edbf2444 : SUCCESS in 20m 10s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/df35477b7d914084b832c5a8c4755351 : SUCCESS in 20m 08s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/6ded9634954347aabf27e0771a21b1b0 : SUCCESS in 16m 25s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/108c8f261efa4a079179756532ce418a : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fe296d45c254636f99aee273d8b61bfa0122cf51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 21:33:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3094e862575f91f4b4aa24c76b924eba3fa5be68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 21:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93c67e8e581d08c90736de33fe7349d483e161ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:10:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5f7e421f884a54b99eaa2bd504f6e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0581226274cd4af9a782c3faaef1fdea : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a58a498452324f49af8a17744303a3da : SUCCESS in 37m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ba7c02095a44d4fb6d76018714d446d : SUCCESS in 47m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de764c01e94949849c7cea22484de571 : SUCCESS in 1h 04m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c29f95ddc27e4a6eb234ebcc3f193c8b : SUCCESS in 9m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/33a30800e798436fb83771997d22a380 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e5491fe548b4e79969303f4d1f6ecb0 : SUCCESS in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0edc199443df4fcfba566ef4eb14bc34 : SUCCESS in 18m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f5710641378342ae934644b3b7b12646 : SUCCESS in 16m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3e2652bf7f4a400d93f4574132ac32ad : SUCCESS in 16m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e695fc415a545d2bda7188d85d1a762 : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00841a6ef3cb48ff9a0f7ede5e2eedf6 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9810070e9444960b10a772a48ee83f2 : SUCCESS in 21m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bfb3892072c40f3a2e9ca674fa2fc33 : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d88ff543bf904da1b6e2cb0fad149bf2 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9b99cf1faef4bbeb14ef428235e3ab7 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4060fb1b977c4cf9b6f67335a356030c : SUCCESS in 12m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/966391a4ce3b4474a9767ed8ab3f87a9 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e05a976dd2e74471bb304582c7193d64 : SUCCESS in 7m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba401922b460436d99717bd2abf15422 : SUCCESS in 51m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd049ca2c6b74a8ebbddb43bbb727665 : SUCCESS in 59m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6418a265d614125b71e6a56bca2dcdc : SUCCESS in 52m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20f0c50e84ba4e08bc14d9da6208d534 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b172bcc5990c4ca6b8af1f8f6297b45b : SUCCESS in 4m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed324f643e7a40d09edea6afc6212acd : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0b16c668b0f32a1c56f31e1cb6aa09b8214d27f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:10:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d252c6d354aefcfcfea11971d7f49e5bc10ed4c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 23:12:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21524254732423e8fda17e6373de9d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba147ce013d48a2ab5134e2d7092494 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/60f6804147e149cfbfffb644168bddac : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8dbc4079b71d4b7c9eada8e6580e9efb : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e39fd8b84411be9d4ff5e6d356c8c0fd75c6a22e","revisions":{"e39fd8b84411be9d4ff5e6d356c8c0fd75c6a22e":{"kind":"REWORK","_number":1,"created":"2024-03-04 22:28:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/910914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/910914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/910914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/910914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/910914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/910914/1"}}},"commit":{"parents":[{"commit":"ff3d15bee086655f06a6f4defd78ec7218466f6b","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.2\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff3d15bee086655f06a6f4defd78ec7218466f6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 19:52:47.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-03-04 22:28:16.000000000","tz":0},"subject":"CI: fix rolling-upgrade jobs","message":"CI: fix rolling-upgrade jobs\n\ntrain and ussuri are now EOL; drop them.\n\nyoga has moved from stable to unmaintained.\n\nChange-Id: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\n(cherry picked from commit 275af9e008fd296623d7a9eb9320206aa536fcdd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e39fd8b84411be9d4ff5e6d356c8c0fd75c6a22e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e39fd8b84411be9d4ff5e6d356c8c0fd75c6a22e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
