)]}'
{"id":"openstack%2Fswift~910915","triplet_id":"openstack%2Fswift~stable%2F2023.1~I8f6c247c21f16aa4717569cc69308f846c6a0245","project":"openstack/swift","branch":"stable/2023.1","topic":"drop-centos-7-stable/2023.1","hashtags":[],"change_id":"I8f6c247c21f16aa4717569cc69308f846c6a0245","subject":"CI: Remove centos-7 jobs","status":"MERGED","created":"2024-03-04 22:29:24.000000000","updated":"2024-03-06 06:06:00.000000000","submitted":"2024-03-06 06:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910915-drop-centos-7-stable/2023.1","cherry_pick_of_change":910995,"cherry_pick_of_patch_set":1,"meta_rev_id":"b95d67e640d36d824ab94ce1bc0bf33ef3eee9fa","_number":910915,"virtual_id_number":910915,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 06:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-06 04:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-06 04:59:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-04 23:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 00:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf55a1a783715481b900f7181fddd0852dd74eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 22:29:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"529ba5c25b83e6c5a98926c97aa977c8f719e41b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 23:32:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb79687d2784353bf2ff09646226217\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6571bf03f6547da89294d2b1aef8cb7 : SUCCESS in 15m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07d055c8a94b49bba5e435c4cd018745 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ae7e4d9361949608bf6793c219369db : SUCCESS in 51m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0d87ab9fc57415792326b5e1f2b4201 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e4fc230e8c88dbe30adfd5a9a76da91b7230fe91","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 00:00:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be46447d10006332630c28eee011a08fbf7ae286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 00:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387853e0a79e46fdb02f9e7017524b36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695fb05d34ff477ca8d393a9be4ffc73 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/551d94f1981943b3929b6bb85152cca0 : SUCCESS in 1h 00m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d32dbac52f2f4e18983ad46c328c03be : SUCCESS in 48m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b91999c38c844761a776080627ff60c3 : SUCCESS in 1h 05m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24e2c3b2706242149f9034f99c00abdc : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/031b3d176ed047f58c11a383cf41624f : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f28b7b80f574f72870ac1ca8f974fd5 : SUCCESS in 14m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20a4f50d2bc44db09f06b19e3f5c996c : SUCCESS in 11m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cda34d970f104f9c908627d2279f39d3 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9371237d39c4ef2b4f377903b931148 : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f6e8969af1645b0a5f91f431ffe48ac : SUCCESS in 21m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4012d5aca6d4ee197812b92176c8b4c : SUCCESS in 15m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a74367c0800c4146a60afb4fdef2e7c3 : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa24adfafd894c978b5d51d927cefe98 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9dfb32e556f14c0bafc53825fa3dc6d9 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/093c861337a442b3a4a60c6dbca5bb2a : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f05930bbca84448784b1d2b22e799121 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e2a5eb0def174e3b907f142f00f42395 : SUCCESS in 14m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac1ca6cf67574f65bf0b7fbd01695f08 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a891e954b484c61819e9b7cd2c959f1 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97c7a97c1d9e4f3ea4ec9dcdfceddf85 : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f22abf6ad26343248023c3803919acc0 : SUCCESS in 6m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bc41276cb8f4852840bafafef781264 : SUCCESS in 1h 06m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f50ac84585543e39a352ff7f8f5ac35 : SUCCESS in 52m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0ed558ce0614354986c75d719b7aae6 : SUCCESS in 38m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/836009a9898a420782f1282101b1efdf : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba83083d92644d1b5c097e234c7335f : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6184a862f0904fe9aeb1619be287b79f : SUCCESS in 18m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45d1256327524d3a991d1cd36ebc9be6 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5bedf4a1a5164cc9c508ba4464c639a70d6f4884","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 00:37:12.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc03f70570e4a4f87ba5021f845be9f\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/abb554bec2c746059e8d554407b9e46f : SUCCESS in 16m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4094494db777495abe55f27575031be7 : SUCCESS in 15m 46s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4c22bae7d32409a8ce171494da103a8 : SUCCESS in 19m 48s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c5a1854de064bb4b49e9913f5fc0ee0 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/20cd3a1d2ff7476f8215c1ab0c12d728 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/83875a10cb2a4bc28cda88dd9786fb6a : SUCCESS in 18m 27s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/5467803e88c24b9cb5946ae4aa863dec : RETRY_LIMIT in 6m 00s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/dc69ef49d7a147ffadc001f14ec87d0a : RETRY_LIMIT in 6m 14s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/99ed6a4cabe14a878970255f30e5ce33 : FAILURE in 20m 04s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/32f3ef3fe7cf4ff592a96a5bb10c9fd2 : SUCCESS in 18m 30s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/d280f1a0944a46a192b4e19ce6676245 : SUCCESS in 19m 44s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/66f08550a33a4272b94da3e7e3fe9a63 : RETRY_LIMIT in 5m 28s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/ecda2d1b8f4f4fd8bb8f9c991f67cc9a : SUCCESS in 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e97c730c334a8f5e74a20f14f380965eb79c014f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-06 00:30:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49c1602a5e1ab441d24415c1bf47025daee3232a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 01:22:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d2706e5926448581bcb2ae3f32cd80\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c03c00238a742feb09339935c4b37c8 : SUCCESS in 14m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45622b5f6b0541c9ac86588a22e56545 : SUCCESS in 13m 19s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e9804127dca46bca81034cc115220ed : SUCCESS in 19m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fe66253b51943b383bd1f4f4a8718fa : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d847fee1e87846d1a62b5f0582462dcd : SUCCESS in 15m 28s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/6df128754227478aa995d27e56865381 : SUCCESS in 17m 25s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/b342e0adaab945cea18270f07dd9c7f9 : SUCCESS in 17m 43s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/bf8bf0e712e94f4cb2ef780a3b1cc188 : SUCCESS in 18m 38s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/69b23f82549845a3a4b772a7adb24c3a : SUCCESS in 16m 56s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/e0ca3692bfa947858fabe12febfc1512 : SUCCESS in 17m 21s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/4432985988444967bf18201bfb410110 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"21652a75b51ad147e90c90c6ed86b0ca8ea68a23","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-06 04:59:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89958093c9ae58ebae2e952fbbc13186f829752a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 05:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6264efa1c191e5528ac1adeb979c43523f93af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:04:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b162b51487b478a9881808c30983b8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6b3161a7c44a87bd887a873855184e : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/70f41ae85830492096929a21bdaa2297 : SUCCESS in 46m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ef3351e2fa04728ac754cd6e4eb89b7 : SUCCESS in 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e35fbf29285241db92b19593988072dc : SUCCESS in 49m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0c332137981043feaa46e4402a832dbc : SUCCESS in 10m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e36ce7fb4bbb44e79ddc672892845f4b : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aac265f745ad48af904a4bba47aede61 : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb4c83fe10d64334a6d9082166908c62 : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40ff96f1b2ff472b86740f4dfb1d211b : SUCCESS in 14m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/31bd9158db1b4452ba936584e2c87830 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81453e1448984dd2bbc4cd9e498cae17 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e8a3124d00b43b0b27e38cfb802930e : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84f2bdde4b484d57b282e1999a20a613 : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d5e7ac7cc0640eebd09b721123b2c7f : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e226d049e0a34d318ca72dec77b5cff8 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c59142f3173f4300814c084b723f260e : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c774dac5bf9440398d0e65329284fca3 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/77b8ac1738674a8294498712da653b52 : SUCCESS in 8m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0831c041f95f4ef2aa69eabe87d6cf6c : SUCCESS in 59m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/951303c87584494b9f442aee8af3a95f : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31d14dcf2c46490ca0eabfd464d6f1f1 : SUCCESS in 50m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5440e70224ab48ee890f1fe1ff06d210 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b1911046d84175a946bfafa6909a72 : SUCCESS in 4m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0dda2e2572a48b4a19de8b2e9ae79ab : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"eca30ef3fb5ffafd25ba49b945303d2065fb8300","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:04:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b95d67e640d36d824ab94ce1bc0bf33ef3eee9fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:06:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70a324a673904184a035e0ec36f9c509\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81abdc4d39b428caa0db655bec49fc7 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8cf321c2068b48df9e26bbe8a376eb22 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bd8af9dfaee946c5a7f1a639f2f93ce6 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c29f70b4c6aff0a23306f24fadbe67213b9db30","revisions":{"5c29f70b4c6aff0a23306f24fadbe67213b9db30":{"kind":"REWORK","_number":1,"created":"2024-03-04 22:29:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/910915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/910915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/910915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/910915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/910915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/910915/1"}}},"commit":{"parents":[{"commit":"8089925e776d0dd139d4d4a6dc8b5988b258ede3","subject":"Merge \"Update url\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8089925e776d0dd139d4d4a6dc8b5988b258ede3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 19:04:27.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-03-04 22:29:24.000000000","tz":0},"subject":"CI: Remove centos-7 jobs","message":"CI: Remove centos-7 jobs\n\nCentOS 7 will go EOL later this year, and infra wants to drop the nodes\nsoon-ish -- don\u0027t make them wait on our account.\n\nThe only major loss is py2 probe tests, but officially, yoga was the\nlast release we pledged to support py2.\n\nChange-Id: I8f6c247c21f16aa4717569cc69308f846c6a0245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c29f70b4c6aff0a23306f24fadbe67213b9db30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c29f70b4c6aff0a23306f24fadbe67213b9db30"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
