)]}'
{"id":"openstack%2Fswift~911165","triplet_id":"openstack%2Fswift~master~Ia89aef5dbc06c49a34947e50ace2365c2c59ea62","project":"openstack/swift","branch":"master","topic":"p-diskfile-race","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-17 16:54:42.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-17 16:54:42.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ia89aef5dbc06c49a34947e50ace2365c2c59ea62","subject":"object-server: return 503 not 404 if ts file unlinked","status":"NEW","created":"2024-03-05 18:11:24.000000000","updated":"2024-06-18 22:48:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"07a9d2857278026fd034a0d1fbefdb9a019e08c7","_number":911165,"virtual_id_number":911165,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-16 18:32:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2024-06-18 22:48:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 19:33:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 20:22:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-06-18 22:48:14.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"476f789e635722d0b789611798810450faf829cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-05 18:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7787d7dee2f7d0c68fc096bd7fe0b36cc5aee36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 19:33:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73861eb92a1546ea82bd6c2aa5953545\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a4e84b74864f528ae57e2180fd289a : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/761aaafa5ace48798b8c6237ba2a4d37 : SUCCESS in 55m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90725ff3bcdb4a3fbc0e2c1876942649 : SUCCESS in 49m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/322600ede65a4eb98b306a0b0c7e898e : SUCCESS in 36m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05df490dca884166b4e24cd61f4ce648 : SUCCESS in 1h 01m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ae39af9e56e4b7b98b0969dc59a4781 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7423e0ca2752408995ea847fa80d34d8 : SUCCESS in 27m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de906cb196c74d4bbbcfd2f68c811f2d : SUCCESS in 22m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cb9f67ef7611441ea8d531cd187fa127 : SUCCESS in 12m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de19bfc933d8400da0883deef5050e3e : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33f334ed36f541e3beb694399c94cfe4 : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/974e40141de540f38815f8dfee5ab438 : SUCCESS in 18m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/188dff1a52b647e29eeaf1889d328eb5 : SUCCESS in 17m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b3cd7e89a1a14912be01b1fd28d90027 : SUCCESS in 25m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0a9e06c4dcb34a02be023cf36483dfff : SUCCESS in 26m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36ef70c65a864f8da79321b049f809df : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/369f2100d7a245d28975697fa05b41d2 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43b8e561c1754a8f8bb6cb73a74260a6 : SUCCESS in 19m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e3c0857533b0464886056d8d2696db65 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/988e584e04d3412782db7ecd7971af68 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d3d205b2c2b426f8da149b0d414bcb1 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0149890d89b046f7af9d8fb2b7927140 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4191b76189ff44128adb164c49e75302 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d2e8618c123459780c1455957700b3f : SUCCESS in 6m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5925a057051a423dad7147a2b20de254 : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cda48f365a794527be723435d9e6782f : SUCCESS in 52m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af6ab5ad12884f37a5905d349ba27c54 : SUCCESS in 52m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d46b02ebccc24ee7a58bd818aee0ee47 : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aecdb3cf37a485e8ffa2afe01c55735 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1295d1089124799be7a868f021ae36e : SUCCESS in 25m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29d21b58f2fc4985bdc800fd15d0d3a1 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f3be805635d6d31b70f08fecb5f9eccd7c57c9c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 19:44:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29db02f7ed6c448b94648f90aa7a9030\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ccd309b009442df8e8257102353070b : SUCCESS in 16m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e1485f47bd14659b0ab72c6a0bd8db3 : SUCCESS in 58m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/84a672e3e5654538822a6bad76629d72 : SUCCESS in 25m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2627424a40e4d84b95e3bb871d0edfa : TIMED_OUT in 1h 31m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5a9de0fe9fd436b5ef57b62e98dcb4a725c7a907","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-06 20:22:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"743d7eabb1ce30f98514c9a59bffce25deb8fb4b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-13 12:48:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb40ab8d81d0bd460e2d636fa5b738093b3ee1e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 11:02:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"62142634d044884af2c1d719433fb276549b9f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 12:20:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95d4363334841639d74a5ccd2356846\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782d06c2efaf4dc9b233c744f677d39f : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/91ba47f267a24fd6b0ca4cb10d77ec3a : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da3d6b0e10b44f9aa4740b74e905079d : SUCCESS in 59m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5720b48a56ee4555af85041ace1e788c : SUCCESS in 59m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f26380134ae45dca9d3ae04da36dd67 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0fb2b2bec62241ceaf985e1c10f19211 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e635bb1b41947b799e216b0607c7a95 : SUCCESS in 22m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2f355220a1a4f29a339589836941e5a : SUCCESS in 12m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0a9d3f22fa1f4442a752dff3ef1cb939 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac2bcebb20294d28b8bf47152a91e58d : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b49e985a0ea4a02a215d1d0613d762d : SUCCESS in 19m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df180eb04f724d97a7ff8b357ed50f60 : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/40eca66ff0634f1bb2df55db306f6ca8 : SUCCESS in 23m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8245283dfdcc43e9a4d4b3f332039896 : SUCCESS in 22m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/640a6c5e8e3241ce86ea382d654ed55c : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eefd8317788b4e6ebd94f9c732e236cf : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/11c6dbbe2e2d4c28bcb6ed82b27e838f : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b0eb3190e274381adeede0cf5f26dc1 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8114a182661f40d68656a3af75b07513 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/460442d9f1cc44fdb9e94e7a7f6817ac : SUCCESS in 21m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9024e44b4b6742cf89d38c76cc0fb2e9 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f4529e978d349b4a413c8f54e3a0885 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8a7eef2f152457ca1506cf66fb5b86c : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ceb0524f64543df89b5cfa9a595100c : SUCCESS in 58m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bb0d08599e04228bebe329d775fb66d : SUCCESS in 44m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c31db733bade47dc94db8d4b8cf4573a : SUCCESS in 45m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/874336968dcc48debd00c3b76e97e6c4 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1e6a59c5874c69bb6a5e7885cbc938 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c25b10a1d25d4aa0aae8754babf0c912 : SUCCESS in 24m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28f8042f6c5a48859cbae38043c35204 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a50359c3d63bdf3c53fb85d7831a6d151ce88dc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 12:31:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ddfd0d9bcc4972aaf2eef270a460cc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b2a3a21d4f049fc844fdc639a1f694f : SUCCESS in 20m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d792396cfe5454d8f7e8a57772ae94c : SUCCESS in 1h 26m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5050ffa0333846c9bd9a7402bd0b422e : SUCCESS in 25m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/11df9b9928a04d05b0e709dfd7735e43 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"295c7aaaac721427c35be5b868d199d0f7a42c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-16 13:29:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e4b33e7e573dc003bbeffe4915cfbf99691dfbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:24:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd1e928b5b740caa74f7281b16e0968\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0570d5f806c04bb98713838dc7db01e6 : SUCCESS in 15m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c45d645fdc934b148670ddd37671f3e1 : SUCCESS in 39m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/257714b7f2ed44259b2c8999a39924d8 : SUCCESS in 20m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1431d01897ed402eb6de11c2ff49e658 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4ac720e07c70fc5a20a65c68518e0b50f04fa527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 14:39:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2b8004023e43c0a0e0921f6f758c02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa7cba4b7444a78ba52d6c7e56df141 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ff7f980e76684ad0a0ca27946699df2c : SUCCESS in 57m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86c97ae836f447cb9b7f3877cb155b19 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61a02f143c2d4c9eaeaedc1e0e816c3a : SUCCESS in 1h 02m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/053797c07a2b4e08b145f11049a33a2d : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f47b24dc49034bf0b326299f7b3a405d : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/213fb583749442ea85c019861b7e98df : SUCCESS in 23m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23b0c6efd20c4430ad248d59c7ea77ff : SUCCESS in 10m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/53c9156d19ce4f7a9016b2e3996239d8 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa1a759ba2c4403999f299fa07cef4a2 : SUCCESS in 19m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09bc0e1f42664672bbfb4e7c89c7cb89 : SUCCESS in 20m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f0e4746e8d44e579b89a49adcef92f4 : SUCCESS in 14m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4034dbbb1767432d88666a1a402bdcd6 : SUCCESS in 17m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1b40bdac1b904ee1a55d0add5d08f879 : SUCCESS in 23m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdff1d5e561b440d8eb88c6cd1b8cd2f : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b2d2ce3ba87420e946309364e522ace : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c94d8628dd394395acf7700c4034ee1f : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e0a7c67c3b64411296a3c277a67f6af1 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/590eb193a14245aea04ff45929d48a87 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/44a2e3ec5e294d089f39e18ed45bce57 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d75c3ce535544c05ac49e36378f5d91e : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/254a17e440bc48a8a5e4f0bfaf454f66 : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96abbb4060c8407fab55c7fcae50161a : SUCCESS in 7m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6d126af61424238bdfe2a3640c84764 : SUCCESS in 54m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5f0499cef0f4941aae5ea18eef32d4d : SUCCESS in 47m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7415aecef8244da8174c5d000cbc006 : SUCCESS in 48m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/085d9527101246e09c440e3f9cbc637d : SUCCESS in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4c79449ed541878a737e5b442e3135 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4661124b9f042d9847184369135f6ef : SUCCESS in 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66963777f5b94543907e8348c251b30a : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1fb14a4fc1cc110dc26953e74f11f3ac6e6594d3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 16:14:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68e6dea7f8cd89a550b0ce2ded3f5ff2bd5dbb5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 17:26:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5a3b0b76b4f2214058f71e80fe799191161cdf02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:15:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd732a5d348d4fc7ad35d59ab2bd432c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b21d6b7a9bd64705ac45718c3e49b032 : SUCCESS in 15m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13eb5b27ac6b402b8bbe6ec695c6b0e0 : SUCCESS in 41m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/baf0d46fa0fc4785b4133032ea328d3d : SUCCESS in 24m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be435858aab040fe8068beddfe3a7471 : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":4},{"id":"841212bcdf5a099cd99566c4ba5050b5e48537d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 18:32:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6013f78dd3ef42cd8887f1e2f2521301\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36f59e81fd941d2b4126670fc76e0b6 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/eb764ea544864166ab288fc6a7947d35 : SUCCESS in 45m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a06432e630074e3db4d611727cf7c37e : SUCCESS in 36m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/759ae9c9ccff406d82603ba353b691fc : SUCCESS in 48m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6f859e9d9914a5290547c23a7360e57 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5806b2a8ba79488f9adda270b936cbb0 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb7eb584f7c24213b3315b8fee2187f0 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e1e0526a98054c37b0bb827886d1149d : SUCCESS in 11m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6f561588f52a4a8b9b9430bf18b0e3e9 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7c9145c6a064f058691ac97561b1612 : SUCCESS in 20m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf2a9e6a0f04ce8af5059e2755ebd94 : SUCCESS in 17m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30a0fc2062214efaa769f3e2df9a6643 : SUCCESS in 17m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/57e7b36e9e574e078f3c9e0853dc1a1c : SUCCESS in 19m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d33a03c6a2fe4b23a924cc00586ba3d8 : SUCCESS in 22m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/045ddd088d6443f6b54d942581c54972 : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b3847fa3bf74142bb8e691e1c8cad85 : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4a8eb395f274b15af1849b981ab4013 : SUCCESS in 23m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c01c87cab17b480cae1e06b85161e466 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ac02dc466795405d97736c95893a4324 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c938413154aa4dc09edc495d00307b44 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/611370088f8b478e9e63971565a1af31 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfabbb0542924020bb566ee794dfc0f4 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e677cd6b694b4e858a14d5925201fd4f : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3f0be5d2d12460ab90e6293e7789af4 : SUCCESS in 55m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06934c35428d4b2994904dbdd8167f53 : SUCCESS in 38m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/376fb55720da47538417abfc2efc7613 : SUCCESS in 50m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97682082da474e7783f0d86834e871b9 : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20285d6decba4793ad5dba354e7f75e0 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/762d25a30c0c4273837434de132dd1c0 : SUCCESS in 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8941094a8a204027ba59e7496b6aec07 : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":4},{"id":"e5da9b21f7b0f9f828e0b5e5bb87b6407a7ffd38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 16:54:42.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"07a9d2857278026fd034a0d1fbefdb9a019e08c7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-18 22:48:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ac1acc2e8ebd80e9f4ca9ef232b253732f2847bc","revisions":{"c0c0eee84a8e54eaf6274a4975145323f385cab8":{"kind":"REWORK","_number":1,"created":"2024-03-05 18:11:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/65/911165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/911165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/911165/1"}}},"commit":{"parents":[{"commit":"e4d5c7d4c130319abc33e9b9b3226eb499ededde","subject":"object-server: return 503 not 404 if meta file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4d5c7d4c130319abc33e9b9b3226eb499ededde"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-05 18:10:58.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-05 18:10:58.000000000","tz":-480},"subject":"sq: raise DiskFileMetadataUnavailable more","message":"sq: raise DiskFileMetadataUnavailable more\n\nChange-Id: Ia89aef5dbc06c49a34947e50ace2365c2c59ea62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0c0eee84a8e54eaf6274a4975145323f385cab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0c0eee84a8e54eaf6274a4975145323f385cab8"}]},"branch":"refs/heads/master"},"f448ecbcfca75264de502bd5060a35582f453787":{"kind":"REWORK","_number":2,"created":"2024-05-16 11:02:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/911165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/911165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/911165/2"}}},"commit":{"parents":[{"commit":"578b34f00f8ab228568f92c6a95fdf416d1f2039","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/578b34f00f8ab228568f92c6a95fdf416d1f2039"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-15 18:15:37.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-16 11:02:28.000000000","tz":60},"subject":"object-server: return 503 not 404 if ts file unlinked","message":"object-server: return 503 not 404 if ts file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed tombstone (.ts) no longer exists\nwhen its metadata is read then a DiskFileNotExist exception is raised\nwhich would previously cause the object server to return a 404.\n\nThe .ts file may have been reclaimed by a concurrent cleanup, in which\ncase the 404 is appropriate, but it may have been replaced by a\nconcurrent PUT request, in which case a 503 would be more appropriate.\n\nThis patch changes behavior to assume the latter i.e. return a 503\nbecause there may have been a concurrent PUT.\n\nRelated-Change: I2f698c25ed65b236e851e5a307d48a12cef62b33\nChange-Id: Ia89aef5dbc06c49a34947e50ace2365c2c59ea62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f448ecbcfca75264de502bd5060a35582f453787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f448ecbcfca75264de502bd5060a35582f453787"}]},"branch":"refs/heads/master"},"5159174f77e25b68d77d3588aa00e317425b4ed8":{"kind":"REWORK","_number":3,"created":"2024-05-16 13:29:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/911165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/911165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/911165/3"}}},"commit":{"parents":[{"commit":"94dc568a7a54eba6c8f4adbee7c80e2a266d74c3","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc568a7a54eba6c8f4adbee7c80e2a266d74c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-15 18:15:37.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-16 13:29:29.000000000","tz":60},"subject":"object-server: return 503 not 404 if ts file unlinked","message":"object-server: return 503 not 404 if ts file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed tombstone (.ts) no longer exists\nwhen its metadata is read then a DiskFileNotExist exception is raised\nwhich would previously cause the object server to return a 404.\n\nThe .ts file may have been reclaimed by a concurrent cleanup, in which\ncase the 404 is appropriate, but it may have been replaced by a\nconcurrent PUT request, in which case a 503 would be more appropriate.\n\nThis patch changes behavior to assume the latter i.e. return a 503\nbecause there may have been a concurrent PUT.\n\nRelated-Change: I2f698c25ed65b236e851e5a307d48a12cef62b33\nChange-Id: Ia89aef5dbc06c49a34947e50ace2365c2c59ea62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5159174f77e25b68d77d3588aa00e317425b4ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5159174f77e25b68d77d3588aa00e317425b4ed8"}]},"branch":"refs/heads/master"},"ac1acc2e8ebd80e9f4ca9ef232b253732f2847bc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-16 17:26:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/65/911165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/911165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/911165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/911165/4"}}},"commit":{"parents":[{"commit":"6be52b6451cb3f99be134a4c4d008592bc4cd8d9","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6be52b6451cb3f99be134a4c4d008592bc4cd8d9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-15 18:15:37.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 17:24:17.000000000","tz":-420},"subject":"object-server: return 503 not 404 if ts file unlinked","message":"object-server: return 503 not 404 if ts file unlinked\n\nWhen a DiskFile is opened, its on-disk files are first listed and then\nxattr metadata is read. If a listed tombstone (.ts) no longer exists\nwhen its metadata is read then a DiskFileNotExist exception is raised\nwhich would previously cause the object server to return a 404.\n\nThe .ts file may have been reclaimed by a concurrent cleanup, in which\ncase the 404 is appropriate, but it may have been replaced by a\nconcurrent PUT request, in which case a 503 would be more appropriate.\n\nThis patch changes behavior to assume the latter i.e. return a 503\nbecause there may have been a concurrent PUT.\n\nRelated-Change: I2f698c25ed65b236e851e5a307d48a12cef62b33\nChange-Id: Ia89aef5dbc06c49a34947e50ace2365c2c59ea62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac1acc2e8ebd80e9f4ca9ef232b253732f2847bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac1acc2e8ebd80e9f4ca9ef232b253732f2847bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
