)]}'
{"id":"openstack%2Fswift~911631","triplet_id":"openstack%2Fswift~master~I41b9b228f76ab06dd70f63cc2c09497efe373b48","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-03 16:12:00.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-03 16:12:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I41b9b228f76ab06dd70f63cc2c09497efe373b48","subject":"Move statsd client into it\u0027s own module","status":"ABANDONED","created":"2024-03-07 00:33:44.000000000","updated":"2026-09-03 16:12:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cb6b490ca8977f42d38fe994ce9f9d6882523bb3","_number":911631,"virtual_id_number":911631,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-07 01:43:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 00:40:37.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-03-07 00:40:37.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-03-07 01:37:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-07 01:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2de43dbab0939a9e133126e0ad9453bbd5618a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-07 00:33:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"150f7dcd3be0ea515930aae7a2d632e2b6f6fdb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-07 00:39:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b15e1562bcf65994731e8855993c3a8976d9bc32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 01:37:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33860d596fca4eb2bb7d8b88589b5c0b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd6676efb7cb440c801a9c6639f026a4 : SUCCESS in 15m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52b29657c04b49d6885c82bde157b7e2 : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/79288a9f87ce4b03947f28a281224e91 : SUCCESS in 24m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c660df37f354b25be67f0bb3e5fc893 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8bddd5abd5077608b09f37cd456301888b082e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 01:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c67b4253f84494f883f6afe594db97d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae419b02011c449489a4bafbef7bf675 : SUCCESS in 7m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eeb070733fc9448bae40cc22c10d8123 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1ace9c274b647f2a534efb811ae9131 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd1635cfb5cd4bfcb56475bb008786e6 : SUCCESS in 25m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e424ce83d4b49c4a46e4f1b708546c1 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce82adc8ee9c445fb0f49b2262eefd10 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ced5fa247ef54d689c68c3eac45f5259 : SUCCESS in 19m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3712abc91aba424f99d5e233bf211bc8 : SUCCESS in 19m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed78ac7b502e414b96641cd3b04797f1 : SUCCESS in 17m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d04808a74a1f4a85a87d21814f2c3238 : SUCCESS in 22m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ef25d6e4793847d6bd1edf99e8243f7d : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4c8b523dbc24836b63fdad124a9252c : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43b4391b49e24577873ebe972372aaba : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c775ab9766140d1b76975d480bf29e9 : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a101115b48bc42419e630e77f478e9b8 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/364712c0bb144a14876003b7918b3cdb : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8761a370c31449c5960e4d42642614df : SUCCESS in 14m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d71b7dc60128442388452621ea48d27a : SUCCESS in 57m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e130a0d38b004fa9b61dcb077f1a986d : SUCCESS in 1h 03m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aa1f142af7647c284f965e510e0c448 : SUCCESS in 26m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064d8e3c508540709c6787fcdd5cdc71 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13602f50c9374d47822b58636c8ace60 : SUCCESS in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb6b490ca8977f42d38fe994ce9f9d6882523bb3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2026-09-03 16:12:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"755faee186bf23584520eb7e8c0457aecf955d61","revisions":{"17cd009ffd0c0e001f12693347b6bc22189cfa56":{"kind":"REWORK","_number":1,"created":"2024-03-07 00:33:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/31/911631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/911631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/911631/1"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:24:25.000000000","tz":-480},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:24:25.000000000","tz":-480},"subject":"Move statsd client into it\u0027s own module - testing part","message":"Move statsd client into it\u0027s own module - testing part\n\nChange-Id: I41b9b228f76ab06dd70f63cc2c09497efe373b48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17cd009ffd0c0e001f12693347b6bc22189cfa56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17cd009ffd0c0e001f12693347b6bc22189cfa56"}]},"branch":"refs/heads/master"},"755faee186bf23584520eb7e8c0457aecf955d61":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-07 00:39:36.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/31/911631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/911631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/911631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/911631/2"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:24:25.000000000","tz":-480},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:39:36.000000000","tz":0},"subject":"Move statsd client into it\u0027s own module","message":"Move statsd client into it\u0027s own module\n\nChange-Id: I41b9b228f76ab06dd70f63cc2c09497efe373b48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/755faee186bf23584520eb7e8c0457aecf955d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/755faee186bf23584520eb7e8c0457aecf955d61"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
