)]}'
{"id":"openstack%2Fswift~912294","triplet_id":"openstack%2Fswift~master~Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2","project":"openstack/swift","branch":"master","topic":"p-native-mpu-legacy-cleanup","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-02 15:01:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2","subject":"DNM proxy: Add list of backend response data to request environ","status":"ABANDONED","created":"2024-03-08 18:06:36.000000000","updated":"2024-10-02 15:01:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e1bd895eebb0aed444f70666aa86f00549307c3","_number":912294,"virtual_id_number":912294,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-08 19:26:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2024-03-08 18:06:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-08 18:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-08 19:26:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c95fd54bda8588a140279596b5aa00c55670ea41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-08 18:06:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3bbb2c8cc401f474e81eccfecbabb7683a47687","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-08 18:06:53.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"d3954f2d6fe71cbddf014f1daae8e4b6056ef3df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 18:29:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0364f4245f9441d5ab19d3a57257a2fd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3af6bd98e4b64037b2d4b219dd39f1c3 : SUCCESS in 16m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8490d43d4b241f5aecb37ee6fab4bd0 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9b4122064a145b8a2611049396142e2 : SUCCESS in 19m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e98fdf15eeb24779942fcd4dae0af392 : SUCCESS in 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"34f6a5837a22f800718e5ff8abd2fa4cb128bb1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:26:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd48934136a14c37a986c60d01f96419\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091c69e6514146b0b8019f5a7537d63d : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/89e1bcb9e7be4f2697dfd16f5007b37e : SUCCESS in 1h 13m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/29bf7383369147e69d1542d62f8938f6 : FAILURE in 46m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86c7c560d6af496880b65e7dfd58e93d : SUCCESS in 47m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3268ff102bfd4aaaa5a2b3c100168953 : SUCCESS in 59m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf03a3a22dfa46438273c2f2389a7f99 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59d7dc7c1e974f58b993786aa8649c8d : SUCCESS in 32m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f379bf7584f44aa2ba1ba13127de4d9b : SUCCESS in 24m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25aaa8966964481098e82bcfc890f01f : SUCCESS in 12m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5aad0c40cb4d4914aa9b1d5f131703fd : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cfcfe0b904c0447cb304d3139ed8fa23 : SUCCESS in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59047806c2ce4a96af51b9bf05f97110 : SUCCESS in 29m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/245bbb383df2433f8593e70a2dd14d83 : SUCCESS in 17m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/05c0270432b4456ba9ea0715a36e225e : SUCCESS in 18m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ed4ae75631884ab99224c9825406df57 : SUCCESS in 18m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3307bb42a2a54a7387754822c0891462 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a9ed7f016354a5b9908e9d476fb0228 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/64973388e9ea437990dbb699d07e658b : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9b2a5bdde6a4a2bbea8825f8ef42dcc : SUCCESS in 28m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05c3abe40d1a42d4b0e74cc00ee96de5 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83751fb3aae84a13a2c72ee283347d41 : SUCCESS in 29m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e8be2099c0e495d8e0ad298ecf0e32f : SUCCESS in 9m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eae1e70f3ae441ad8ee2b3a40cb9db06 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2600a25a43a48d2a1d18d86ea92bb8a : SUCCESS in 9m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6a956039d5243ccbd61fc957dafbe83 : SUCCESS in 1h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd4cd936099c47ca95906543678ebad9 : SUCCESS in 1h 01m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83a2d7a935a045cba6f1ab05ff593e6b : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/067f0073c90d4f8cbfac24b15431bfa2 : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3467b83ff5457ab7255c985a68e36c : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c95b024d719a402883b7952b935fa86b : SUCCESS in 33m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fe9adcdde4440ccbc46cbfbc5861220 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2e1bd895eebb0aed444f70666aa86f00549307c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-02 15:01:41.000000000","message":"Abandoned\n\nmoved to feature/mpu","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f359264b0c4a721132cd4d208a663745fb796dca","revisions":{"f359264b0c4a721132cd4d208a663745fb796dca":{"kind":"REWORK","_number":1,"created":"2024-03-08 18:06:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/912294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/912294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/912294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/912294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/912294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/912294/1"}}},"commit":{"parents":[{"commit":"0f6ecb641b2aca99b8bcc48d8e855d22132780b0","subject":"Merge \"docs: add discussion of content-type metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f6ecb641b2aca99b8bcc48d8e855d22132780b0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-08 18:05:15.000000000","tz":0},"subject":"DNM proxy: Add list of backend response data to request environ","message":"DNM proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f359264b0c4a721132cd4d208a663745fb796dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f359264b0c4a721132cd4d208a663745fb796dca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
