)]}'
{"id":"openstack%2Fswift~912322","triplet_id":"openstack%2Fswift~master~I3d61b93a42920087ee8f60e0b371afc87ea7c637","project":"openstack/swift","branch":"master","topic":"p-ratelimit","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-11 09:49:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3d61b93a42920087ee8f60e0b371afc87ea7c637","subject":"sq: make source of magic numbers a little (?) more obvious","status":"ABANDONED","created":"2024-03-08 23:30:02.000000000","updated":"2024-03-11 09:49:48.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"06116eb2582b507cbe565b164be4fc779cc1578a","_number":912322,"virtual_id_number":912322,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-09 00:30:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-03-11 09:49:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-09 00:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-11 09:49:40.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"e16143e202090784b45f6fa3b7d3e17e878b4845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-08 23:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c5109a4a3a2fa463be3c41278d73bf167022832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 00:30:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/febc23e9c30e47f990af000c5bcd103f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d982de1570384524955fb8a473fd080c : SUCCESS in 10m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7661b1e8cf794c5fae92990d90a24193 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c1b19e0631d4ffb8e02bf30eadd82f6 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bbf9da35744b408c8146d11d6c5b6d01 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2cab85ff1eb4e23939fdcaec2880fd8 : SUCCESS in 11m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6a4e7f23eee74e4da6c0bac18ecda0b6 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8138047f2dd94d5f869c288749918b29 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c82800be3d94cc08a50d4a6e392f7d6 : SUCCESS in 21m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a5bf1ebae3a947ae8c49630a4c580493 : SUCCESS in 16m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1259fd2d4bd242aa8ed7ea33327a4474 : SUCCESS in 12m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fea51f65458a49709c595d9da327584d : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5f02ddb685c4770ad75051642c096c6 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b0647de51dc4322bdc1e60ef1e434e0 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73eb3cebaf924886b87587fd27d7033d : SUCCESS in 20m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c2f75fd237244d51861cdbb7165dcf7e : SUCCESS in 12m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b7a91d0ae244d2cb0954bfa1ed2a487 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad52c15d710b48ec80224541baf57566 : SUCCESS in 19m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee5f7428fab749b68d3b8c69bac7e183 : SUCCESS in 54m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a321173d4d034d1e948964133145b96d : SUCCESS in 35m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6c07a7e71ed4051b2b719ca9d3c3275 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5801f5c1205d45c1bd0207d77332e47c : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11f9f07a1e8942c284b1391560cd4e33 : SUCCESS in 18m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57c5c9d4ee752bc435cce1004825b5b89166f91d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 00:51:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8ec774e2b844a187a5d3d86f650413\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7582f69141041f9824b37c7bb31fcb4 : SUCCESS in 13m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7f725f240f14712ad32ca8cbd1ba9ca : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/92b98f27907d4eaa8b6ad1061278947a : SUCCESS in 1h 19m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae41cc1c5f5a46938941a1dd34a0930f : SUCCESS in 23m 41s","accounts_in_message":[],"_revision_number":1},{"id":"94ca833c8ad9434abfbb3376d085e7ab25ea65ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-11 09:49:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"06116eb2582b507cbe565b164be4fc779cc1578a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-11 09:49:48.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8209fb58dacd443eec204a5117a85d998be119b6","revisions":{"8209fb58dacd443eec204a5117a85d998be119b6":{"kind":"REWORK","_number":1,"created":"2024-03-08 23:30:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/22/912322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/912322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/912322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/912322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/912322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/912322/1"}}},"commit":{"parents":[{"commit":"e2873285fed45aafdad7cdd6778159d49079d4de","subject":"backend ratelimit: support per-method rate limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2873285fed45aafdad7cdd6778159d49079d4de"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-08 23:00:08.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-08 23:29:35.000000000","tz":-360},"subject":"sq: make source of magic numbers a little (?) more obvious","message":"sq: make source of magic numbers a little (?) more obvious\n\nDrive-by: make tests more robust to existing\n/etc/swift/backend-ratelimit.conf\n\nChange-Id: I3d61b93a42920087ee8f60e0b371afc87ea7c637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8209fb58dacd443eec204a5117a85d998be119b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8209fb58dacd443eec204a5117a85d998be119b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
