)]}'
{"id":"openstack%2Fswift~912496","triplet_id":"openstack%2Fswift~master~Ie4b25f1bd16def4069878983049b83de06f68e54","project":"openstack/swift","branch":"master","topic":"task-container-per-day","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-06-14 20:29:12.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-06-14 20:29:12.000000000","reason":"Change was submitted"},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-06-14 20:29:12.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-14 19:04:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ie4b25f1bd16def4069878983049b83de06f68e54","subject":"add bytes of expiring objects to queue entry","status":"MERGED","created":"2024-03-11 23:06:14.000000000","updated":"2024-06-14 20:30:35.000000000","submitted":"2024-06-14 20:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":209,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"912496-task-container-per-day","meta_rev_id":"e0835a4f1117e7c2c084e28cea11b1734bc2a303","_number":912496,"virtual_id_number":912496,"owner":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 20:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-14 19:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-14 19:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 16:15:32.000000000","updated_by":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-03-13 22:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 22:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-09 18:48:57.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-04-22 09:32:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-04-22 21:53:04.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-04-24 22:32:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e268212a78bdc89ebfa62def9594f2716ca02fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-11 23:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58e58de8c57255f50c2bc3c343d796604465cf86","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-13 18:32:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":1},{"id":"30338d01f142e8e963951c1dfc6d866007a0ec68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-13 21:36:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_36763\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"_revision_number":2},{"id":"af35c90ba89ae5d4bddbd6c2de57cde69dbf09e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 22:42:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73b434ddf9cf4bcd8b5c7a1a753fe141\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3098dfa2265b4f2aa340dce89bbd78da : FAILURE in 16m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bab7abbb95764e609c8d1e069b9b332d : FAILURE in 1h 04m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ac6baa8b0c94854a553236614dae8d4 : SUCCESS in 18m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0082c8ae143407eaec7f22fb5694fa7 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6726a0ccfe7f0552276fccdf9e119dcc4c5f5d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 22:49:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4beb4af91aa74560ae12a060a66de632\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b30c98ddcc846f9b1ea0958d497125a : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/268389afd07a4ba79e0e43c95d731a44 : SUCCESS in 52m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/42937cd56e784d2e9817d1b149cd2baa : SUCCESS in 1h 00m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/620d56c75d2d4e7b8ef99162ae2f2e72 : SUCCESS in 50m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85339981d0154fb9ad180591d9943568 : SUCCESS in 56m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c32b5e5f2edc4884b8dcb29561d94012 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/366a7f9d8a524559aeb70323835faefa : SUCCESS in 25m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67486de17f244eb0b0e42afdedbbc223 : SUCCESS in 18m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ac8c351192b45fd84dff9bdca6acc66 : SUCCESS in 11m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9299e6431eda4e7c9a09dab825ee2d22 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f5af748700d4208b86d378e0794d8b4 : FAILURE in 22m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7701fc05ad894879b36cb2a651fb4bc3 : FAILURE in 19m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3f801d2ac60746c9a205be6213e01f22 : FAILURE in 26m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/346437687beb4bafa7264423011be3c4 : FAILURE in 17m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/933a4c46a5664272a162d4be40887f9a : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2d9bce0fce948d4bf08c1ae7ca44d4a : SUCCESS in 27m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28cae0eec6714dbcb434023fb34063f4 : SUCCESS in 14m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c0b51ec2c964dfdbcd870ea7929ceb7 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d613f0a26d1844fda8d9f3cc93d4ed8f : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5c86644d0e90470b8983e9e73191b68e : SUCCESS in 24m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93997c811d554030a2276b3fa3d18588 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3f8a8bb0f2a0494da9a864aafc736f95 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99af83bc18bc4824aff7d2a4ce241c22 : SUCCESS in 7m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/26fd7fac76f24396a967cb777730ee4f : SUCCESS in 1h 06m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0527a1ad0be943c786679a69515d816b : SUCCESS in 36m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccb08aa7479c423dac9c206d3eef0e58 : SUCCESS in 55m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0993dd33c3f46e1afd794ccd02df03a : FAILURE in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78895843dedd4afaab8e0a8444a7fb60 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fb2b0f8747d4b3387177d6b4136de47 : SUCCESS in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbe90b26f0794d1282614112f64fc138 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":2},{"id":"2278cdab961e6cbb73fd549a3a562b74976fa7bd","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-14 00:30:32.000000000","message":"Patch Set 2:\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"77518542f586518ee50be7dc1415c79836c7d30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-14 00:31:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a5dee72500ca6b7f7e91d887e61930448d10a476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 01:46:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a6b8ed7d2f2429b8d6f4a3e41cb9267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3011380f5fd4d17804cc074f57e215d : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b8553ebd5cf540e3bb0a62b76e6a46ff : SUCCESS in 1h 05m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/58ca94b375de4081b07e1206b81dcdc9 : SUCCESS in 1h 04m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0205dc690c74ee288b2f706816d2648 : SUCCESS in 52m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c52390a84fb841aebab385cbae774615 : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d78e4ce6543402ab1a48941f6e3cc28 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ace3c6882b54e64bed1480646a9dcbe : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed56c264a081451db272cb599e5b9346 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e23cecaeab94a16b4d37a24cb24ebf1 : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/444f3b6e7f1d42dc8d9a8da422bd6421 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fec65392a74422bbe3664d989a6d36e : FAILURE in 26m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46498f108c6a406b9e63e871cedc69a3 : FAILURE in 21m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cb08b92792274832a35f4cbfc4022053 : FAILURE in 17m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2a597cf449e64925a31fe3e4a27f9c2c : FAILURE in 18m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8db445bf435047cd95aa34db93a394f3 : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4715b287d0f74a4db41abe5534abd93f : SUCCESS in 16m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f79f8677a7a41a1bcfdff18d8256e24 : SUCCESS in 20m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a52a1fa34f7a46b883186f5de0810101 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8c913f0449c4dc9954dfbc2278e57b1 : SUCCESS in 14m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9cdbd80717e044fb8b45e3d027560a8e : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fcdb86c8f1041518f2d993e68539c32 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c7f9fdd398d432a9c1a2e24b147696c : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37ce94bd2baa49e0ad05907e9de55822 : SUCCESS in 6m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c839c2b8b9d94cf88443dccf001b1693 : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab098d3ad10947e3b9cbeb380eb95e29 : SUCCESS in 39m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/033246b4d2cb4cbdb7358a0967febfb5 : SUCCESS in 55m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d0299d9a053420c877c2565587c26f9 : FAILURE in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea84d7513d14accbd7ddc5b680d26f7 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a94197379f94a1b969419c3949af873 : SUCCESS in 19m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6205d59aaeee4b5f987b451c0f3e913f : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c1b520c99bc2f8e78079546bed84a1ec8df00e75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 01:59:17.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd653224fb30439fa9132b3dc07b9d48\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41108773eacd4c0782e7e32b67a11a42 : FAILURE in 18m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e74a1451cb242aab2d9abb8486fbb97 : FAILURE in 1h 06m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d8bc18055dd40c29080926f86740724 : SUCCESS in 1h 25m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cff20f397944f4a84713163dabd045b : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f7b7d3e6972644b39157afefac52d5e03e4c98fb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-14 13:27:00.000000000","message":"Patch Set 3:\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"091f52beebc30dc5e68cbc2707767d342709b298","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-18 22:57:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ea4ae692cd44cbde8e1092e98ffca8673c16f816","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-18 22:58:33.000000000","message":"Patch Set 4:\n\n(11 comments)","accounts_in_message":[],"_revision_number":4},{"id":"788948b5394ac02a587a093b68b009e39d706097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-08 18:51:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"78770d241abc3e7cddd0352a5451164a40cb0ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 20:05:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23b255c1d1b04c68aaffb2c80d2d50ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb3c065e2aa476b9339f00ef54284cc : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/fe422ab8f19242a48b97c424c783c15e : SUCCESS in 41m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2edd85c82d3444085819bab150109d1 : SUCCESS in 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6905bc1de29f435883605a214cd32995 : SUCCESS in 59m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9abf11d015704232a1ae7b5152aaf347 : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41aca4d02e70428c827730b0c304b207 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fed0e7970a24ed5baead858383839ea : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78e33f1993b94a75be58c4288ed0ff22 : SUCCESS in 11m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e616dc21ad014c3fa4ad3ac62766bacd : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9038980042143a1bd71e7d3e208f51a : SUCCESS in 20m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d33716277646498b88bbf1ed743399 : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b462bcbea8f43e6ad1c0a52dda74c48 : SUCCESS in 22m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d054a2e36134c9090f81ca52ea78d06 : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6098dc98ffc41cda958b98c0e24e22d : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59ed50e329b940869b4021db64be245a : SUCCESS in 22m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/383ceb6bed7f4dc797351043a8ed770f : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f96b7b7d18443579000be7f83a33683 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/456da33d9ab34c6fa24c0414fe87fc08 : SUCCESS in 12m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb9e856a8080436fa4afb290f92b1113 : SUCCESS in 17m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e38b56b3a2b4182808849da89c78d41 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44de452c62f3435aa60cd60977b6ef06 : RETRY_LIMIT in 1m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ce456fe09ae47188caa0214eb11352b : RETRY_LIMIT in 1m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/978b0d646bae49c8a80ac49eb5f35d17 : SUCCESS in 1h 08m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b635aae385e4367837037d1bbe0e239 : SUCCESS in 40m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97c0481279b442eab0d5ee1e5c1f5c83 : SUCCESS in 54m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c9e001158ce4452b059b0636db85396 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381e93b246bf461093e4aeb55c6df5b8 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7788318f20742c7bfcd107c373eb61d : RETRY_LIMIT in 1m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f50ab52581472bb9e378202697b4f0 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":5},{"id":"9b66b99f27ab8534f50153b3b076c8ffbeedaa24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 20:09:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/299c079da60548e0a2663b473b1ea876\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00a60a9b9bb64609a351a1903ed14dab : SUCCESS in 14m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc25c336d9b1486584f501b2e9dd9b27 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cc08eeb612647ae9763c3d380e98e8c : SUCCESS in 1h 16m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ec33904bdf84fb3a7ea4cb9e2ca1080 : SUCCESS in 17m 39s","accounts_in_message":[],"_revision_number":5},{"id":"4d537431c7c3400425619bd580b370640958e96f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-09 18:48:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3d08e646ffa7ebbfb01337256bd729276c55223b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 20:05:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edff4406475048d59512d2f3b265a267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d0ce2531be406bb5557762a67aa87b : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b398483ea41f42f1b1208fcb21011597 : SUCCESS in 1h 02m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96ba1daa5a2e4e82bf37b72f9cb8b4dc : SUCCESS in 40m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f39b68e405ac4a2fb1ba26604a5ec828 : SUCCESS in 1h 07m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bafc83ee4415471bbac1101c74a3c1a5 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1d4f175c4104796892483107431021c : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/964917e3594c4fe8ac0751b2505cf25b : SUCCESS in 22m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b803bad68554737805283a23dadd1a9 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4d52db31db7c4e81b5604b36f4b465a1 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96f5c0de2d7943a4a4a4146e11d754a3 : SUCCESS in 15m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/beef59be149f48e5aaee0000918be14d : SUCCESS in 16m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c5d224ceb79417ea0a9c41113a07e30 : SUCCESS in 14m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1fd201b8c18f4f8c8c12ac8d6e688251 : SUCCESS in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4a72c1680e714f338ebaf8e17c109d8f : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bff3d2d119d747059a33ffb134007280 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a3cb1f71d2c47888fb9e982545279a5 : SUCCESS in 20m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d9dcb5ba1314bf88c2b4a2114128388 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7934077e54074ec986e65d685e8373c0 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5dc4e64ba351422b95783e317ebf1992 : SUCCESS in 22m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3b471e995ba464987836e55f2893066 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c88b1c1fbc5f42e695ada3f539c2a522 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a090f94374f4df9b52041e4d8714639 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa4bbfa5fdb24f15b093f22a38a14c04 : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e724898daf8b452ca9b63a2094fe57a1 : SUCCESS in 35m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6e5b11018e44b58878eaf8ca9f3cad1 : SUCCESS in 43m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e724f012dd1d44a7bce2cb5f48c18533 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17567c11598144a2b90211b383e695fe : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/401c1d885faf4f1db5d789d9c0180a68 : SUCCESS in 18m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12a513df921b41eca6e6c861c2129f7c : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":5},{"id":"281175dc118f6d15a0d35c2986bd16f307f5b7a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 20:16:58.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c77730b58e41509ceda74ad249b44b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/631215112325496799c65ddc4e472432 : SUCCESS in 16m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/611d76f4599945698c51274ba25ff8da : SUCCESS in 1h 10m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/68a02abbd0d94216b5ae5b4a129bbf30 : SUCCESS in 18m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/38d444e969e547918cfb6bf85a2c388d : FAILURE in 1h 26m 26s","accounts_in_message":[],"_revision_number":5},{"id":"5a08fcfcecbf3a0c4b76038d70ed7a9171c19595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-17 20:08:36.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"37d8465dc172826e64971abdf1bdcb5f34048e91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 21:03:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee97b20eff524a07bf32c2f4f776ab8f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b82f5120b6984686a043f07e80835dff : SUCCESS in 52m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95e228bc1dc04f50ba295fcecb5a7611 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2fbfc29bbb445f7a8bbedfe7b9cda4f : SUCCESS in 47m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2469571b6301423bbf0c5c1e58f64c89 : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":6},{"id":"f5b4741a76af7ddbd8126d809704a6cc009e7d79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 21:19:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce0f26a17be4d29ab2e83adae4be2ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dd485bae474d38a52accd1905e058e : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3cee5178dc1d4432bbe4321553f01c62 : SUCCESS in 53m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63e14e39bc7b42ee8c5890bae186090c : SUCCESS in 50m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c62d23c3d34e70bf50301591e2a7c4 : SUCCESS in 1h 00m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0929bd596e0d491ba017df37c28e5076 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02b3eb90b8b34fe1bb4913bc8be33a5f : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/935d8c1d2acd435cb2023a51e3ab2ad5 : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55aceb86a99f4980820f0853282c36a7 : SUCCESS in 9m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24232b0d951749a3bd294409f1c06316 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b86bae6159f4565adaa2a75a990d4e2 : SUCCESS in 21m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9443b4787bc148868143cdc966b2cdcc : SUCCESS in 18m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/13500b217792428bae172df636c5ecd6 : SUCCESS in 17m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/382894e4b0684ddeaf85256b4c8180c5 : SUCCESS in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/564cc5252dd64c4ab0d3f997c6728876 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8787dc2f3b3244b28affd26b71add3a7 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1ee0c80b4b24c6ba35de783d108a6bc : SUCCESS in 14m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/49ae0531b2a840058650f89268db18ca : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f6e5e998c00c4903962b6a6ea6eb5e33 : SUCCESS in 12m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bf1db69d0a4f413dbd07ec26fdb25cf0 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4ff5b38d5944736a08dcb9d024ff58b : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a840a23b90c347b981b60de254d1569b : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/34f13d6395084fc9a6b6059ccc169484 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea7bcee831454a52897e825e4d56ecd5 : SUCCESS in 56m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/618f46d3053f466c89651f0df167343d : SUCCESS in 43m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17544ecb34f748d3ad6606d41c0e6f78 : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/291d7ab94d4e40ea8ccfc63a117f8048 : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115fc01989e5488ba295461077640e9a : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69c3ab4cffdd43daacbd3523fd018a27 : SUCCESS in 17m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28ca581036634708b5d2d03184b80eca : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":6},{"id":"6bb3ec1a3b8ba4a2d708be1fee035418852fa603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-18 16:12:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1bc2a555b4d5329bf1f4fb7de50dc167f1bb8df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:25:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12dc68db9e95464799a02b824745c32f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80450eb3651d4268b766c94cd787de30 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/9fcd58f834be411b9cc8106272956c07 : SUCCESS in 1h 04m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49e4847953e84e2a854c9004a8b62b2b : SUCCESS in 38m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a80f30b8afa4993ab7f8cf1e9f48abe : SUCCESS in 1h 01m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f552cf1aa0f44c9bb6ff936a95b9d30 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b61db9ae6db5469ca4c0f4749fbc3616 : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af9cf833be0740e7a4405376a0a55d32 : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e29f3010d65b4147b7506f1cf42ce993 : SUCCESS in 11m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d6e822fc3be451ea0a3778cb751829d : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ecd1dbac9c6447a9b2620f501d4d18d : SUCCESS in 21m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/820106c6c2604798a12ebc5a60f0a2df : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cad526bf71dd41eea87c99a674f4e077 : SUCCESS in 15m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/693d1b3f45104acb8899a493ea188963 : SUCCESS in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/50dc83d6d89745f3925810c701328be0 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9723d5a502e647518dbe5625373f9379 : SUCCESS in 20m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa31884351c44c1eb89eedc25340d240 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c707891cd7364cbaa9edd7671c7186fc : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e0c489fc9f3844a9bbbee73565b19b74 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/03f08c9ab5fb4f4289034bcb58e445d6 : SUCCESS in 19m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f4f621db0264503bc65a114e6215fdb : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ca88153046504d82b22680b7231d3fc0 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/09d804c6fec14e10aa4058f02bcd46ee : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/357cf320073747d4aa86bf17e89a8ad6 : SUCCESS in 1h 05m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/917353de53be4b6e8f30ec8e19523f47 : SUCCESS in 48m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40fc57cfed204771992f74685787d093 : SUCCESS in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4434479b629f4ab7a74dacfc0a9449a9 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78ae2b69b1348649c6047c007eae194 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c9a7464de7f4188adb98515c4f84bcd : SUCCESS in 19m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f2278f9c542488ab11a77a0f4c8238e : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":7},{"id":"f0cf0240214bf4bd48b494563cc3902b2d409429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:38:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e65567cee1934483924acfa4b532bbc4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c675764df30745ac902cc4f4804186f2 : SUCCESS in 17m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/393f4cf98b604b62bf8d09c574777d82 : SUCCESS in 1h 24m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/acef6b58996e4ebcba5eabeaa6ac5023 : SUCCESS in 22m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ceb191878984e05bca2f6b7ec6f1267 : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":7},{"id":"2492da6baa2a6b4816c2483564731cfc4034159b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-18 23:00:22.000000000","message":"Patch Set 7: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5a55e15ce110d09acf57e8c91709d13c8123d867","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-19 20:46:27.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8ae85199200a2f9a6034796439a354acf11d4dba","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-19 20:47:03.000000000","message":"Patch Set 7:\n\n(13 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b7c8652233356e9ecd3c01a8a4ae21f53d06847a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 21:09:13.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f085da712d400b8f9dd7eb1463f4e3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e5225b2d3244990a792da168569c78b : FAILURE in 16m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d43cd37b75a7461fb161fdbf6bf5d10b : FAILURE in 17m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca4595951040462491061c478442b6b0 : SUCCESS in 21m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21f705723d704b3d893bc426c526d2df : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":8},{"id":"ed33e1e54b1709befa2d01fb4ccfdb261d1bd917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-19 21:39:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"73d4b097be96a1af91e5f1eb0ec50187a19b931b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 22:00:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2be4ba1b39d41389f1908ac8a778548\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97a90d8dbb3b4f0f85bb0af7bab98689 : SUCCESS in 15m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2f1e25533994cb88b96aeed3dbd9a13 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca06eca5c82d4a4cae7222f1fb739f64 : SUCCESS in 17m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2593df412d2c4d93bdd845834f014c1d : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":9},{"id":"de0d8a11426635db298fe4a3e3731dc2c71a98b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 22:39:33.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09543e77372f413595f5f81cf7b828fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031afe00eaa246da87cf472ca9f3e256 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/391836c06f14481f822d6ed5f4792361 : SUCCESS in 56m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f93ba988f4ef49a08cfb5e39ea55e38c : SUCCESS in 35m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be4fee8a234b4d19976fa828e64ab942 : SUCCESS in 59m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58136a281a82472db65a5afe8099e9dc : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a32e133334d1404fab26eb6c5e1c6d37 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df76869eeb5e4618a0d6413708e967de : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74b50d88b30741049e3575f4fbe2d123 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/681678e2bb324fc8bed24e707a9c27f8 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5533f3f0fae342b89b90499176815643 : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40526eeff83c4682832597d84355bd5a : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dffda2bfdf89442c96333af9c1cfbd06 : SUCCESS in 17m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/34bb49fc380a42b69e263368d8cc3c84 : SUCCESS in 19m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f809aa4994e43fcb1d7d86b66591768 : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8cc75f48e19f4729860f54d912f59c31 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/130c1d4d0a744344a0513cc4f194a1ca : SUCCESS in 19m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/da8de8422f234be0b8248f7f0f2bf2d4 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7226658dc2a44ad5b9f1cc5e56033c72 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9c6089af94f4487c8488fbe8c07e405f : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6f41878f7c543e992dffef49891727d : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2683f009291454c8c4907c4723a3bab : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73af245a565a4375bd712e03cf5d71fe : SUCCESS in 10m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f08f4b31505048beaf63c9d868d26bef : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8c5a7f802a54bb3b4d98e94a94707c7 : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf4271a6a00f42a8be687919ace12ed5 : SUCCESS in 43m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d33127e5e9884a9ab84714cb4894bbd8 : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c02b368f8694f3dbdae9a996d2ffa73 : FAILURE in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ca8672171954d37af7a392295b80bad : SUCCESS in 24m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de193551ddb4479a627c5cabf1cb2df : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":9},{"id":"38139a8262fad2b3db73bcc18e8968435edbad89","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-04-22 09:32:15.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"596e4deb55475d50b5447b1cf1e11bf430cb419d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-22 16:47:38.000000000","message":"Patch Set 9:\n\n(26 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9f21cbf331b5c0b54424117397c3197d1b65bc77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-22 20:49:50.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a02602be18b5a5219f3a573df606d0b9e96227e8","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-22 20:50:26.000000000","message":"Patch Set 9:\n\n(14 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b9c83db8bdc8b303811645858bed8003db057f73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 21:52:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f253cdd2cd840a5907a3a2856bf6180\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa62195f9b7f440884df0318bf3c2cbd : SUCCESS in 17m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea5db196ed234972870e691e11087ef5 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/07ad86710a3643bd9a2b509078fff9fa : SUCCESS in 1h 00m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d31b7e5f6dc422e824ea7313de8ed52 : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":10},{"id":"b3467a47cb4478eccdeca3945b64923a8b5a2f17","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-22 21:53:04.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8e7aad8ac987b1518a0f52e637cf1caa5495fd5e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-22 21:59:36.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"73e7ca834f42acee0fe70f10e5bd47a2e74ba656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 22:02:03.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec2c81b9b5da455c99dfbaa9bc992f78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d6410c2ce84c03bc1f641dd696a057 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2d65bb1393f44892acbe5cc87bffa6e4 : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e663cbb620d4ec782dccb66849fd51a : SUCCESS in 48m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29098f15d5dc462a8626e06def21b81b : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1d95cb5391a4e708c4db53f50db8b07 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0306ad44b839481894e8d94a95ca4957 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1edf4645ddd0494db5db3c5760ac8172 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f690914d4fb54486abbb01d64ca1d6d0 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12c674125e7f435aa943c198c5871b02 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fe04380b54fb421aa14cfda8843c14ad : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/763ce75e17ab438095c273069ea8ccd9 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9dba3bc7733649cb82831e50759120b1 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/26badfb425364ceebeb3034346490692 : SUCCESS in 18m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ec9226b9077438f8c88019af358df7c : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bda4b104b35f4be5919785d98f8dce6e : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f080e20759f249fd9d6eacda95584195 : SUCCESS in 22m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a70f68a5f53245a1882997b7b8903a83 : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d69002f146b4485da6fc24dffe767170 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/94cf29736ff049de9accb33a05a686da : SUCCESS in 22m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/04e3b01e22f841d8a5a642a9a8a85b76 : SUCCESS in 5m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a6736e19556440e991b14e5c763c2c31 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f0c28aa7f694b57a7eefcd1f903a374 : SUCCESS in 6m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d56a6afeed4a49249b3ce67cabbbe32a : SUCCESS in 56m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cbc7548173a24fbd9d1ae2dcc949628c : SUCCESS in 41m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79ea3bdfca3941b392766efeedad072f : SUCCESS in 54m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70c26de9b84b4004be465bff85c7b7d7 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05e7194e2d54412a382d3d44f7eb98f : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8725e5c67b394d42b1e9e599b3f4ee1d : SUCCESS in 19m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e48aae579fb04e7197fd387caa767a28 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":10},{"id":"9192d0123789289e0dfc7a0d99906765e9cea948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-22 23:03:44.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"00f7a7b747b55c90d9e3bcd852ba36b402f79518","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 23:50:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24902df84e8148d2b72dce679d1632e6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/122f9458d0134cbcba2712a527faaaa8 : SUCCESS in 17m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26fe791ce835459eb91e328bbcbbd579 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd263c0aab0b487281c478f3d222e075 : SUCCESS in 45m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2911c7477a34dfa83991739d118ee52 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":11},{"id":"0f27888ad6bf63a6c162c60eebd640dbf7d09bf8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-23 00:05:29.000000000","message":"Patch Set 11: Code-Review+1\n\n(20 comments)","accounts_in_message":[],"_revision_number":11},{"id":"44b013979e53567c5f03d583213d39b46e5e3c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 00:58:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c34f325b7c04aafa330270ae23caafd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f6eeaa80064b34a73fdce51255ebc8 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d1bc866fb2d5432882044b4f7ccb81af : SUCCESS in 43m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e1753dc659241bb980e98b877080b61 : SUCCESS in 35m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fe4ad7b6fb24327bcda3482a7695056 : SUCCESS in 40m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adf598cad72d43e8ae0e946dff868ca0 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab2cfc76ef7e4c29aa51b49223cd87f7 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/261d73de1976470d952d72df7a0d875c : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/959b18932e124d04aebc894b9d42d669 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3070ef4fba0849898f79382dc6e01052 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4a38246f533431c9685081b030024ba : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01e2ecf0278c412a8898e2ba22d9dd3b : SUCCESS in 16m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/13a1e81480a04e4e9cbfaf1292ee90ae : SUCCESS in 16m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bdfb547879674f6cba5b96ab54f82400 : SUCCESS in 18m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/631c740bcf1044149f97740fa9cd2fdb : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71eeac26fbfe482a8227a407886b36bb : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61257cdfdb9b4a729049dafb6561150b : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f376353615324b91bd24eaa9c8dc32f9 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c3279ae4595947c3b57f428021f8714d : SUCCESS in 12m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9428491aaf34abbbabdcfec018857b2 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dca997528f43492fa7c1e94a222be93a : SUCCESS in 5m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/adc9903144c149808051f45c8bad6b7c : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fba6036794be49448f192b817a9f326a : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/016aa27fb9b648b599935660d8fd2c35 : SUCCESS in 58m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc19282481c047a4a1cd1a22fa52b942 : SUCCESS in 50m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58f9b8b4c22a497aa383f3514a24b8a8 : SUCCESS in 53m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35ab6a914db24693a1e170a5869738bf : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44196a6af5884d61a15af6cfb346bcea : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0db6042ef75142c3b365e3749c46d8dc : SUCCESS in 17m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93fad706e66b4264af01a464dd8eaf5e : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":11},{"id":"d84fe5693c7ea9c7232ca7961f275ab0aeb98d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-23 03:12:55.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"9a1ae7427b17222ed4e78122420af696ce67d59c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 04:22:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f47ce9861f4325b7875dd035a28f01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a6502653c84292a3aad4bca8f0ea8d : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/8383e7c255b2496daf32279c061a18bb : SUCCESS in 1h 03m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23429d97cf064350a4400f7be941a3cf : SUCCESS in 49m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31de80937ba6423282b396d0b6bf8294 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a64a4f2f3fa0409ea95b5d6c03812ea7 : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f5fd1b06ff8446599170f21a3d5a504 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8abbbabce1724a1bbf219dd43f1ba4a1 : SUCCESS in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c86f0e8d281546bd8117e9e85416b3cc : SUCCESS in 12m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c556046ab85940438d1f81941a7fccf5 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6c92186492a4a139a3a798ce642c3da : SUCCESS in 21m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1abacf213f014526b5f445878346957c : SUCCESS in 16m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d5346b7912134e78aeee1d3a27083de0 : SUCCESS in 18m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5b95fd2e167e49388fe3431560bce1c6 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0752eb4254464e648e51d8fe38662253 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/34978b3dd7b141c596996cc5cad23743 : SUCCESS in 20m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33381388846a4843bc35c4f75c8a22fe : SUCCESS in 16m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5224c47972b42d09d501bf20448f01a : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6e53f60017184815bcef3b3d9fcd640c : SUCCESS in 15m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d948ac11bbf14023855ad9a6806ddb30 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1ad019dad264d788e2bf9b366000b45 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e923f5e5b6341288f21216c48dd47cf : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/63d2f9403643490c9c5a2dca79e4f2fd : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae3162129bed4e168f25e69dfe744f14 : SUCCESS in 59m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ab36822726645c1b58b31c55668852f : SUCCESS in 55m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f87adb9a7794404bbbffae4cca7aa841 : SUCCESS in 41m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fea89d0247384592baf70140dc80f8d1 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56aac32cd4274df5851fbc14104db05c : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d56183ea21c4ae9bc1ddce7fd60319d : SUCCESS in 20m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4f9806d503b48acb8751c31f7c2b8ce : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":12},{"id":"2d3ddbe5a4c143470b026d4ad45031532b901078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 04:28:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dec04d845244fa98fe941f5887af6a5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5a1e1214f134c24bc6f5a4f1dfb7c31 : SUCCESS in 1h 14m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4293452e50284f48859d3b4022c71b44 : SUCCESS in 1h 00m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6d3f5758e1b4da8a3f643da35a9f5b6 : SUCCESS in 20m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/32fcdfa5e8f94a87a4d0d894d87cb14d : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":12},{"id":"14e48bab15652895b1fa741b4464751799742980","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-24 22:32:25.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fd55a0887a7699aa14a0fc0167b29e759997708b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-24 22:51:43.000000000","message":"Patch Set 12: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1d8d82602cf3bf620b77927a99d7b9315973cfc6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-24 23:03:32.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9864b04907e13d98b85eb74534a6d5938c9559d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-26 14:53:01.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"09582a38009ef84935eb32f17d14c36a4a5bae7d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-26 14:53:16.000000000","message":"Patch Set 12: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"43578a5cfb95b9310569459dadb84cf4c8f645a2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-26 15:04:31.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ae00b295459d1c474386b40a4354e0e5953f5586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:07:17.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99249fa7f1be461ca714b20e451297c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a833c6abac0e407c9197089b083bd540 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/0751045b01af4a4db4b378aa327bb31b : SUCCESS in 1h 07m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/46e9959fe1504874a930f323d892ca2c : FAILURE in 37m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16e221a4911946fdaaf94131c57959c5 : SUCCESS in 1h 02m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d07af09fa1cb4275a7f23127162846d7 : FAILURE in 33m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e96e128f0f444cc90ed9b277e417afe : FAILURE in 30m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cdb0c352a85453596da62f6e8f3e4aa : FAILURE in 34m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0068fb214354d8db26033634573bdc0 : SUCCESS in 10m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/769ab795f17a448cae5b756db9e77e09 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58d929a74a504bb4b49e649bf65d51cb : FAILURE in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ce20f46709450db3f3f130ed4e5034 : FAILURE in 20m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7cc798c8309944efb6cde70492304d22 : FAILURE in 19m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/08586bfcbc5549d3afcab51554104f4c : FAILURE in 18m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/536e4e94a2db4320a90d8bf2d1e457e0 : FAILURE in 26m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac48e55d17be4cf7ad88a9e6e2fe565b : FAILURE in 31m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9ca9b18a27f4d35822d14ccfa41fecf : FAILURE in 28m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/879dba8925ef4f42936e12836f26c7b9 : FAILURE in 25m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/295d026b7dfd481fab9052a91fb1ab77 : FAILURE in 29m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e79cf14a624e459f84f54f959323aaa8 : FAILURE in 27m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/820798965c8c4ad7bfada8577d8ad3e1 : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a171ede897be465caef3fbff3e79208b : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0d2909c5f4e434e9c449c58263aee38 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2067770ba4134edf998f78daf06a0594 : FAILURE in 1h 06m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/660e7cfccb454531a835ad44a763109f : FAILURE in 36m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0290f96f203f4c69b08cdf276a0b9b41 : FAILURE in 45m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/732766fb96ff4f85b68cf2a6c08ac37a : FAILURE in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9a50190e944768ad5d68515d860b33 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5d25fd8ea524e4eaa5b21184f7d657d : FAILURE in 27m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92ecd33bf7ec4d97a3599cc8a53cf84c : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":13},{"id":"3244c71b655059ffd0553c85646b8bf8286ea088","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 16:16:38.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab77c104f774794b619316c86e8fe2c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1afccb0da0b45aba63c74620be26a6c : FAILURE in 25m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37db942ed732444fbfe403edeea88862 : FAILURE in 1h 19m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba72b8786c7f4f119c46935089ebf84f : FAILURE in 44m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5526660f512549ec800397d571b12998 : FAILURE in 1h 20m 28s","accounts_in_message":[],"_revision_number":13},{"id":"9c368930752730b8608f3d8b389d2186f798c0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-02 01:19:29.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":14},{"id":"350c99596bdcefe7e2ab4587b8d46cd10526cc16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 02:17:01.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a1f18613c64c61837bb751e92ff124\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61d718c637ff4b06999041e978e6113b : FAILURE in 17m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aca9ab3229ad407ea90164a54007b1d2 : FAILURE in 18m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8090fff9282b4ffca16849b1a7d43bbf : FAILURE in 56m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/84d169696b6e4f4384c784c57493c6ed : FAILURE in 29m 18s","accounts_in_message":[],"_revision_number":14},{"id":"cd4c73b0212be493445b4cb124d0dee4641cb64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 02:42:36.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3258424c024492cb0cbac8eb041b3c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c99004f598a4f4ca0d4d46a3be0f617 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/a808b2d0937d4ea0a701bf92dce27360 : SUCCESS in 1h 00m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65eb2960578748b9a726ca61bc16248b : FAILURE in 49m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b81921ebd6f643fa9ea00c5de7b58454 : SUCCESS in 1h 05m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85c94319d01748c19a51c35ef087cd17 : FAILURE in 25m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d33cfe8f5954d6fbe55fcb559d78dc8 : FAILURE in 30m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/426a466d9741478a9f26729fe020b473 : FAILURE in 30m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67738ea24a344285bec96569239bfe02 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cfa3ea597ffe4ccf8d34007ed1a115b1 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6fbbdadecbd54c92aae745dc32ea17e2 : FAILURE in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d2c9517f8e14390913de119ff1a5fff : FAILURE in 20m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b202f1ec2cae4ecd90a93da6fb1dba10 : FAILURE in 16m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3324c9ecb19747888eb51a9c589c8388 : FAILURE in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc7eec616aa0477898012785d8452b72 : FAILURE in 29m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5dfd8936b477489284813219ee2cd898 : FAILURE in 28m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a08d10b2e5d74bea816353bddd2f0a94 : FAILURE in 28m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe9ab14b9ca24cc5ae0de7b3a90aa57e : FAILURE in 25m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3462d7f7467e4410b0e034d31ba883f3 : FAILURE in 25m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca5a28a4273a46029c73ecb0874446ac : FAILURE in 28m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d0bc1cd2d051435b9c5e9b5aa455d714 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e54a048db649493d9bfe1ac7ad9016f3 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/443517fcffb34d388086d3ae17f0cce6 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bb86e9889404a668de70ece7a6af498 : FAILURE in 1h 06m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69a63a4abc8047c1972d7a0569b13c51 : FAILURE in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b8a418531054149a5aa7b1359fb9f5d : FAILURE in 45m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49c74f5a611478094215f80c9c3c04e : FAILURE in 22m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466ae7e325b541a3b8704b29d31f52ea : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff9a3dd3f8fa4fdcb242ba2a3f38c352 : FAILURE in 33m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f15e4158ee83456ca16c636bd5a0ff5a : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":14},{"id":"b0baf111647ab333638b280790ffc670a5d36bec","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-02 15:52:43.000000000","message":"Patch Set 14: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4e6ce231fcb4b068395e537a15e2800ed406d0e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-02 15:58:47.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":15},{"id":"23a0a8704d2f37fd1784f5730d3d07b05232ccaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:12:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc67e378ed744a119ee1fe6ac54a2ba1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb7030fbd25418aa130e3007defd5f5 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c7287f399a1846698d0487daa70651fa : SUCCESS in 54m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69d81d9d295943cfa3b3915749066fe8 : SUCCESS in 55m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca47f3d4eb644d96915ce95691ff6ed6 : SUCCESS in 47m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb3bad278f564f0db80b8575163cdbdd : SUCCESS in 26m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/053daca6a7d8400799ed276f6b40cf73 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37df3dcfab434b60870dec9f7f7be911 : SUCCESS in 23m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b99bc166148403ebdeb7db8bd2de2d6 : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/483e1cfe1f45402590f601b171e56e1a : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/872992a0cc7a4c2dae8daa329be243e1 : SUCCESS in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0fa81e954dea411eaf5f6a3edf5386ad : SUCCESS in 17m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03dd0623400a44d0b0f4e43dd6eafdab : SUCCESS in 18m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6e4fcdf0ce12419cbe3018ae76081b42 : SUCCESS in 18m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d28d323ecbd4263a56d703c9b03248a : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2918151ade1d4ea5aba85832be19d7d3 : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98a935dc94514924ac2b432cfa8dfeb7 : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/afe85f450c6a45f09266d9251dda7920 : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb28ac455b96482d85f16ca2a14f8e7f : SUCCESS in 19m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b04680d807d741dbae620c222a5c5a7a : SUCCESS in 18m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c4d6d12443d419982787d50fd8f43c4 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9b13e71884f42d4affb07b7a9fa9804 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3bb5d744ebab4bb59246eddfa3889278 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6becdabda8c4ed4a5203c0bfcfbed8f : SUCCESS in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fd3b0427828425fb64e2060e480c526 : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bb171e9c87e4ab09a409d1f023794ce : SUCCESS in 55m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e914c4389842492ca1b4409765fffbb3 : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033b99101a9740fa9dbcee87bdb4c0eb : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edd22cd8ed14428fa5655656b4447f4a : SUCCESS in 20m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec959b648b04cef8d6dbd3fce39551d : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":15},{"id":"e54a8b6c5b99193c7d9a0eca58a316a28fcf198e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:13:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba4b55240924e27b4e4508b173fb113\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f05af0c46bf4b039810f70859fd1232 : SUCCESS in 17m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ff2f843def64e06a05073da21c2cb47 : SUCCESS in 43m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aec4b2619271484197204cde97e8c64b : SUCCESS in 20m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b24f91095ad4ccbad88a50cbe37df66 : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":15},{"id":"2c9f74f7ed179695ddb7baae7d86aa9d5c7cd22d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-03 06:51:29.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cf14fe05cc93e08cd8b84c8c30e193b8c9cf7382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-17 22:43:08.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"91e55538c9b98a180419979723d04413a0e5d69c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 23:52:55.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46c8ae4b1b83451d8fe6616dba33f6b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f64ebf1fc84008bdd8544bc50325ee : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/0e0a566008c74079ab09a75d9603a426 : SUCCESS in 41m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/659d5ed88a9e4717919dff37cbc70dd4 : SUCCESS in 34m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e3d1353b68245598526c316eef8e9c1 : SUCCESS in 1h 01m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29daff8af94441288fbe3f5dc7a200fa : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d9ba956f10a4b47bf895dd72ae21e98 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13a0c58e68754154acbf59448e1cfc14 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37e579909ac84db186b07a3a6349e985 : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5868a6a7491e423ab56199682f767ce3 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa9b65a8927145d299976d925b3c580b : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c846249c61a544ee906f35a08aecb3c0 : SUCCESS in 19m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd884097c6a46c5aa46e2d19cb6b20c : SUCCESS in 15m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1d7a0b8c1d3644e3ba159de76f3cd548 : SUCCESS in 16m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/92fadb41c6a54b2bafbb9825b1099e48 : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad276d50d66c478fbd0f0d05b2983fb3 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cfa99d5401f74182949775d3d7f2fe88 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1896c17ec7c4498297fc16460b5f9e1e : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bfb6283ec50f4b00b06a6f6a7208842d : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eb2b7207d18b4bb486fb528ad3100d74 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/576e5bf25bcb4f55a655886a72156021 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4df5d9c9c284e66b11c3aaaaa886941 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/559c0d0a5d8847b8a7e12a0ee7e587f3 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b31631a07f1b4db08ec0984e3d97693c : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7072d3150024b8386064cdddea64218 : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9901c33add4f4a4298e7b597f166b018 : SUCCESS in 54m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/868bd71248584931a4557c5d192ab08c : SUCCESS in 55m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb562d26897450bbde1691f1a3c37fe : POST_FAILURE in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ec8b636b6f443fad7452fb09d10bec : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ca9d9dacd0d4a18926d59b214bc33bc : SUCCESS in 19m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b04a9f4abab3431099788a29f69e86fa : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":16},{"id":"0e22b87722c47801dd8bd155ff4ed7e84576c332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 00:03:33.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b18dee5c48245de8ab33b54c711b4c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be36b609ea8d4dd89d21a900fb8bc123 : SUCCESS in 54m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a0964b555b34e65bd0c54027e361a0c : SUCCESS in 48m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a4d25d3a2d14e148bae382ab0dabfee : SUCCESS in 1h 18m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/907e0c2f87304d52ae2e2dbdb1f14571 : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":16},{"id":"c8055e26ebb3cbf075a3b9ccdd51eea1bb6a1eb3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-20 16:46:08.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"46728ae488d58e84641ebd803564fe1dd2fd592e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:12:59.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76f81a10a5c4e11ab11c20f463786e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cb8edbd8a5c4b72b2ca9530c9c5fcb5 : SUCCESS in 17m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b3d77cad26f458b8279becb2aa8b114 : SUCCESS in 1h 24m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8265f8a4fd1e4b97a8ccb89afdf2e47e : SUCCESS in 18m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1160e0d688094e79ba34d8d6796e3d9c : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":16},{"id":"01e981f17215163634f7176c8b7deea129c69e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:17:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5734faf497d647b6adbcf4e2cc815185\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3054e91f12d4a178a37a73e42075245 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/4daa696d705940789657ebe2fbf2cb2d : SUCCESS in 1h 25m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a21a23f1610b4287b01f54dac2189e35 : SUCCESS in 47m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63af6fddc01541fd9efada1041b72ef8 : SUCCESS in 54m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9ee5e7a278f45878c4caec0f8c18559 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce49b648bff84b44afe738fdb4e00a40 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0200be5d945d493a9e174edf0d310b66 : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e83ed0e51ca46d49341be4b8af757c5 : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/082317dde02249b8a2b814b85c7b9c51 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a198e6b011e4944890fad9d85286212 : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0551127dada144eebe83a025b06dc28e : SUCCESS in 18m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3ff84f91f34299a87a2f3587601e26 : SUCCESS in 16m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e9f4caddaca64d719085a9eb305515a6 : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/236aed548b3a4144824e902f5aabf9e0 : SUCCESS in 21m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4911e96a65454ac8a3981187c8982e71 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b46b59a90d0445eae96db3c2095d388 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa12ca6122d24047a896f96e03a81636 : SUCCESS in 19m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f92bbb88ebe94f52a3b50bfa35388876 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5fce1d33bb2e409d994e149643dec91c : SUCCESS in 16m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/86ba9b9b6a7a4d4dbed09c06a560aaf9 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95fd13fab56c4b20b3045484a695f9af : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d47906a332cd43569a976f76ec8cb37c : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb6098986f0046fe9e97dfacfbdcdc67 : SUCCESS in 9m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/66ba21f63d37464eaf37bd1d53247877 : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df77f5526c5543f68b8a378bd19dd11a : SUCCESS in 47m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/908ba3485d2c4042bf36bddc2623ac95 : SUCCESS in 1h 12m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae98176c07e9455089ff24ef137d26a4 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ab7c92b7ff4132b9c423a82e17c6e3 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa90fc8ac6d745629bf91c5969c9e010 : SUCCESS in 21m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39d46dc28c2f418c932cb94a76ab0357 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":16},{"id":"985b81424fe76c31b87c33a6cf674bcadd6476f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-21 18:04:24.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a79a324337c81da890e8b02493e82b9896b2710d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 18:51:33.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c757c11908874359ac026111f51c3958\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7224331152f45c7afd3fdb717797752 : SUCCESS in 21m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b91767de947439a85e2dbf4c4c0ae99 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad71e403bb8547c7b9cee0dc57df3f26 : SUCCESS in 45m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/32ee10370eb84e5b95d1c5dfd877e8ae : SUCCESS in 26m 12s","accounts_in_message":[],"_revision_number":17},{"id":"62a2ab1d1077184398d9fd095f40bece6dee40da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 19:16:42.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/937b838a74a34737b7126d9c3d64b51a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17906f17624a4729b85da8e8b7105f70 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/b2ac4f26d933433aa339a36bb54a18aa : SUCCESS in 1h 06m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45d07e95f61e45778171b8e594824fa7 : SUCCESS in 54m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d3fd55187174d52957d45008900039f : SUCCESS in 42m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6a316b426f54581a8253102f309704f : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b003bda451f45f39ced378eddf2a0e6 : SUCCESS in 22m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a739102cc8e9437792c73c44f734d5ec : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a21cb36781f740e19ed3c40e6210d6df : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c612a3910134430b89909198665a6b4c : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c36cc234b54b48d0ae2079de2fffc898 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f61fb58480a942d196856340552852af : SUCCESS in 18m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54cfe203d5b84b6e9e6f52a81e3602d2 : SUCCESS in 15m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2672da70ec46426398b62d1c14088c97 : SUCCESS in 16m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2d9bc8ba46bf4008aeeb2912ad44bca0 : SUCCESS in 21m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2299cd2a5a734acf8c50784bb5382338 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ee4ba780af540baaf86c06507d3889c : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a9b12a57ceb4c50b709852051e31bcf : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69b869c7a93b44acb2f57b25a829ed00 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3057ca35c1cf44dcb1d56bb0afd2e87e : SUCCESS in 12m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/678f2ddf13e44271bbbdc985e934e269 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75006de724db4a50bd0c2da20620c351 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7efcd3e94de44d43a4110a46433ab95a : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/32535d1ee333461dbe192a6f9fe26f53 : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3f089379b5341e18210662cc012e288 : SUCCESS in 46m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78fb1c2ea6604036aea96b0e16ab8632 : SUCCESS in 40m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5c4721ed8da461ab6e32419e138240b : SUCCESS in 55m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b59d8facc1b415ba5dd4064e5299eb6 : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a15289c3374672a01d251a4d97b028 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b55539e88b6748e685c3905d86612dfc : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a74137883eb4b94a7c4b039c7661045 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":17},{"id":"9032f8437d472bb7a251fd9898c91e1d6164de5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-25 13:36:02.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"d40a094fd8c739a628f9147891fa7820e09e67cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 14:46:05.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c4e9e395bb4d6f8350974ce80547c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6e244e94474e95afcd660babf4c66b : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/bd70ed94f9314f63ac1e41dac31bcae0 : SUCCESS in 1h 01m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d817bfa9af7243ccbd19c1f348cb0535 : SUCCESS in 47m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08fdb1fa959140609b6d96bff0e9a13d : SUCCESS in 1h 07m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59e8a301927842ae85c2501960a38df6 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51e078be39154f8093eb0e6c8991a16f : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/703ff69ba62741ada2b96db955ba3e6f : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12588c7884384322800aee0a08e3679e : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6534cb16226a4b48b2e8bb8fe5281ffb : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68b6dec984a644dfae13573042215e75 : SUCCESS in 20m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3432f28812f041c6a053deb921c59f47 : SUCCESS in 18m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10e6a3bb39844fa4a968f5a14f560028 : SUCCESS in 15m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7a049793d97c4cc7958b8b51f4033734 : SUCCESS in 19m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e3af592a156641a08590c430b5da580e : SUCCESS in 22m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/79d3a7abef7b4975a634cb954548b162 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43d7e22550bf44dc99139a625bbdd7ec : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd8b40908c1b4627bdc3541c280bb183 : SUCCESS in 20m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec6a866b47654fb590860cd270bee402 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bb6ea83ed7f94c56baa5cb6fa20875d7 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e31533c1961c4cc192e2d409b51b6e12 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3dcae82557d4d549206fa315b4f8189 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/105dcbdeab214ef7a3d6cefa4c0ecb76 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2c44f0e47ec4d2e8f7a8ad9e4afbb0c : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f59a53c73da4facae6b1db3858db571 : SUCCESS in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6303bfbe8dee4940ad9dc4f3c277bbfa : SUCCESS in 54m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c85434dcf0ab4792af699a70e179d3e4 : SUCCESS in 43m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d462c9294e34f4c9aadaaf2fd2844ed : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74c3cde44dc46eeae489946ce39f683 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d087f57e33248d2983333f865cdfcbf : SUCCESS in 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5287f7e3b2864592b853fb416f3255d0 : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":18},{"id":"1112905b5bbfc66a6e3c584fd4eb40717e566041","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 14:46:51.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f1c45942a644b6cac26ca2b344de8b9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcb2127e39c14394845e7845432c715e : SUCCESS in 1h 08m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c382ad131b254f0cafdddafaf53e1bb3 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac605ce30d564912b43d0935990d6f2e : SUCCESS in 20m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ead29c5e51a94ff18ae1e8758888441e : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":18},{"id":"42944ec6d7e949c80bfac38d8da653c9c6abd126","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-28 17:00:37.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"81c6434206cbe80445bc7f8457c1ea74fd60c6b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:29:51.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66641856ea0642ee96f2683811dc28a4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ade04b9e17174c4e928b1e704954ec95 : SUCCESS in 21m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84dffe05a2f44d90af0e8a52d07f2a86 : SUCCESS in 21m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ed5d76ef97844e4af289084692fb44a : SUCCESS in 26m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c838cfc3c2a4f84bc934206c565a306 : SUCCESS in 26m 41s","accounts_in_message":[],"_revision_number":19},{"id":"3fea4e2169d01e06e897995f7b636052e164ea19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:19:20.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/258fdf23f8984ec98c46414dc9b18f44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a436e3f46d4f42cb9e557f80f3b91a9e : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/a5f9f52ecbd747cdb2335d102382d48c : SUCCESS in 1h 03m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7aad366b2c224411a92d5d03d4b6a7e0 : SUCCESS in 44m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af7dfc2aa1f94132ae7c0627137e19e0 : SUCCESS in 1h 01m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72471492290442f6aaae61ac43ea6c51 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbd58e62e28d47eabed6d0b6f121fbd8 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15024d56e6964bdfadded2e8534a17b2 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5aa345a2f864e9d96bbddad368b42ae : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e97338e4a90240dc88e57e76560f37b6 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a4db2a5ba374695a1b990692555fc3a : SUCCESS in 20m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b02e0120fc014e75a171e102c474345a : SUCCESS in 17m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/081a49290864414c804d9909826261cd : SUCCESS in 14m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb0fc27419094d0488b40d0ad1378222 : SUCCESS in 15m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bc7926e864b6496ebe7a144dc4148ef7 : SUCCESS in 22m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9bb706909ec44cc5b781b5bebb4f9d67 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8fac1bfdf9eb4317a6a49c6d4c9536d7 : SUCCESS in 21m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4d66d1501b3479789e8d92e962a589e : SUCCESS in 18m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f0a1545ba2c545b7ac8749d1dc9edf9c : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cade86e448d0415c996e08f612692cbc : SUCCESS in 16m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7278316dada2459bb35910de83e79ef1 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/679618a4354242468f7a385b402ace1c : SUCCESS in 8m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32ae15b75384421993031042d35ae082 : RETRY_LIMIT in 1m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf29c4924a2f4a8baf44cdcfc098d247 : RETRY_LIMIT in 1m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1def2464c724b1998acc9b0177b4d42 : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/babd7b9e3d04430cabe2f6ee717653a1 : SUCCESS in 49m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/947fd23dd8d649478c915f43e5d9e33a : SUCCESS in 39m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07b3f0a5dce7418fad4fe440c86b1d4d : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c9d97cda9c41679eaeee9e65345b34 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a6115e715654896bca5aa836d23ddcf : RETRY_LIMIT in 2m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ed2d0fce4b14e61a6618022dfbe9bc0 : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":19},{"id":"b999ff10f95436c2c3ebaaa396b4ec766c1a487b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-29 20:00:13.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"18f9170bdfb9cd792d49439ccdac4735c1977352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 20:08:20.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5fbf5f78866497ba8fde52c2d3028c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/762e63cac8274becb723889d2a7fbbc6 : RETRY_LIMIT in 1m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bcd3c8626db482392d796913c9653a9 : RETRY_LIMIT in 1m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/31cf3790e2744d7b9bcd8eea843e0604 : RETRY_LIMIT in 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2615ebf74434d849e548cb7428e01e8 : RETRY_LIMIT in 1m 15s","accounts_in_message":[],"_revision_number":19},{"id":"eae74cd891dfcc3e1e5dc1b4d7104b7df811a66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:21:14.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dbebaece57d483ba34f830acc70da28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1435c33e32084fa7b88757dc752381c1 : RETRY_LIMIT in 1m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ae77396a611a461a9103857f45e9af33 : RETRY_LIMIT in 2m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16a5c7b5fbb04ad28ba39fbad6904a3f : RETRY_LIMIT in 1m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/555d8a3e2df64318a6b0c24ddaf3db46 : RETRY_LIMIT in 2m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66214c051519442c9be507175dd083ed : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2741c56109d4123a4e4b290a38a98b0 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92e58219a1a7439392cc8f131e30a50e : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0531e6a9f0c24f91a853275a208c3495 : RETRY_LIMIT in 1m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dad45c8839534d5b9123bbef137f88b1 : RETRY_LIMIT in 1m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ac56f62f64f4ff683b114556d48fb91 : SUCCESS in 19m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89ade967b2e64703b4185ed3eb5d4a8d : SUCCESS in 18m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2a80b2d447649ac885258e187e6984a : SUCCESS in 13m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2f09d788363c4b19a03d1e86fef4609f : RETRY_LIMIT in 1m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0f1dfe8766a049a98a966f3914d184ea : SUCCESS in 22m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c5d30d8297e41f6999bd5a48cee2c26 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7cd5669ebced4dd7933263918c93cda7 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53bb94597a90426bac6ff77a520988e2 : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e4d3445841d424988149b567e3b33ff : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/08568b9e527342528430182c6d1dc908 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9a4dbe193af4a2db2ab277ef5dd6ae3 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49aee29bf09941b89ed5cfe38cac373a : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c36b56358f5446e2b5d6d8f5ad51addf : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe52c9eada7042cb9081dfecc4eb8541 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/adf55d510c1140b9b5148687fb1808ad : FAILURE in 1h 03m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70fc494a6405483786c642428328ea72 : RETRY_LIMIT in 1m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98a345c8fdf84e8da0d398f32eb9ba7f : RETRY_LIMIT in 1m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62736cb68f4d4a5faae9f4c07e2a4ed1 : RETRY_LIMIT in 1m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e21997dbc84e6699343d7ad7603242 : RETRY_LIMIT in 1m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3df8eae386e4d5f8109a0a62faf1641 : SUCCESS in 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a96b8ae620cf4a3e989d1717f52cee9e : RETRY_LIMIT in 1m 27s","accounts_in_message":[],"_revision_number":19},{"id":"48e9d1883d8cf91776923ce683d77212097a969d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:10:27.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"703d2a49e943d15cb1fe21b40442896507b89947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 20:56:21.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd6a5b3f09944ffdaeb036808c79072b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fdecab2ba934e4dab77a6ff8c67a609 : SUCCESS in 43m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/011c6de774df4f0384282b2a5d399e6d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/830336eeeefb4d24ae647984b956540f : SUCCESS in 24m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2779a80d647546378cf2de5acef4e349 : SUCCESS in 21m 43s","accounts_in_message":[],"_revision_number":20},{"id":"90ffcdb2299c6402bfc30708fa034e24c3e67451","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-31 21:26:26.000000000","message":"Patch Set 20: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":20},{"id":"1fb1a0b0c4f63b0f445445086b1d77417c5bbe9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 21:39:18.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9680a4f68f3a4640982a333d54db22fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2d3e7dd2c540e4b3d7145b59e09086 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/c2b783f4f0ac45a3a13de15b93137ef7 : SUCCESS in 48m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da032a566175420bb2baadc00ca7cc7e : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78382c2263a240e7aba3c6a1d993d497 : SUCCESS in 1h 04m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f65d99782114c11b2b5f6c2e06b4dd0 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdc4740ff64048f59fe8fa15e2e07d3d : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d7bad9ecadf41198b0e76a04d3afbef : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ddf8f20e5684523908dc5d1951c52f6 : SUCCESS in 12m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/681a3abe74574e46863cdad5bada9667 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea050b70adc440eeb4d4d69073ecede2 : SUCCESS in 19m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a23b861715749e5ba6712e9686c6840 : SUCCESS in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e651a56dc98d458c9646aed7ac813ac3 : SUCCESS in 18m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01b96bc69efe40b1bf0eda29baaeea34 : SUCCESS in 14m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eaccf361a3c64a768c9d334ca963a165 : SUCCESS in 24m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19da3d740c8445e8b32026fa72a3cbca : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9b1adc2e1594e26970ec91dd086e1d1 : SUCCESS in 19m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce23316986064b618ee6875c55ca0233 : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eeffdaae10b747d698235083c33cbf94 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/33baac8500614879a1fb98ab700e3799 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/13303e49abba420aa16c8cf90aed25e3 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f593b801632d450984751a3b646c1f66 : SUCCESS in 5m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab29d4ea776c4c5c95507672280a92d3 : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/addb5204fe11466684c51dcf8a9ea24e : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a96e5c5c2cfd4625b39b4b004a8bc3a9 : SUCCESS in 53m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b290127aba774367b4f6e9cf15789ad5 : SUCCESS in 50m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e48c2cea28a4e899db0e298fe12c3f1 : SUCCESS in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/affa629d1a97457281a6967cd5f7c5d1 : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d541d524ade94cb2bce5a2e8755c0233 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9bae67d77da48aaa6f1daabb0f4fb13 : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c6fcd7744914d15833c0c2619909789 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":20},{"id":"60be64b2086516609e3ce27015c5f37b1b9d9f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-03 22:24:09.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"17d1aebc932337856c1d6dc7ad8bbc7a2e2183f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 22:47:01.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a320ae95200b4c6487ad57a7769e00e6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42ea68108ee54a9189d691a12768e0db : SUCCESS in 20m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9f8b5de28604c15beffdc324f6c48e1 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b11700bf2e34c2fa1f7b88530bfd0cd : SUCCESS in 20m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/100a1b7633b841b68553b3652546fd65 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":21},{"id":"13b27f1595b83282bb34a79de5e331df212a07e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 23:47:34.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f12994e251334b0b96aaed60577550a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1d3db7e6924a03a8267a5e369537ac : SUCCESS in 7m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/a842fcfaddc741468a8137b4fd444b10 : SUCCESS in 42m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31a6717491334943a787e480ea2288de : SUCCESS in 37m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18cd16445a954904aa5aa790a248e078 : SUCCESS in 1h 17m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc46b0a5be5447adaf0c1f0bb88524cd : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c58ab35ba554fc49deff508c1f544a7 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc2b731eeadb4438be64061f27e5d2f8 : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24db47e49bcb433d81f098cbe4c4f5f2 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/491c2f0badc2434d8fdfd624a692e746 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3050e5d9b27443284c890a1074bc0e6 : SUCCESS in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78be9a844e20442db8dab5fcd3c70c2a : SUCCESS in 19m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3586db9e6da74c19b42001023fdeb1d4 : SUCCESS in 18m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a37c2029a90248c78e5ae92487963157 : SUCCESS in 17m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44fdfbe6645c4c4d9cb50a55e5b019c3 : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e48b5118e14426c8da0a7d506dcf905 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c7ba1ad95ac40b481fc020a5cfba6a7 : SUCCESS in 20m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b6ed4e0765d74ccb83886a19c8926a90 : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7712fd3eece8424698a88ae807f5ff8a : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e87fae833d0c4016989a862ae850bfc4 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6544ab84893472c937a50db12bc82f8 : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1643cd2c512d4da6b1be64aa69d4980c : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2fe6783d5a744be7b3b2d05380b24609 : RETRY_LIMIT in 1m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d15020ea3c614ce9bc4852f47e4fd320 : RETRY_LIMIT in 1m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0d9a573011f49f1b12f07902b919996 : SUCCESS in 59m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed5057403ac946a19b5b83a1d62e97a6 : SUCCESS in 51m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/695f9a30c2094dfe94b1ff8cd0fda0fa : SUCCESS in 53m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6433b69ea3f449b3bdc9f212e9b53c30 : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0095177f5ec84cb9bea6b176ff0917ed : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16819e7e0d0742aa8f8531d12e6c4a5f : RETRY_LIMIT in 1m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5619c068658843cdb62ab39b1fe54aba : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":21},{"id":"3dd1e42277def5d0853179b6816ea32afc796cf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 18:11:10.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"2315c8e57dce047e4a1d5d8e8d9f93fd369b7e36","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 18:14:05.000000000","message":"Patch Set 21: Code-Review+1\n\n(13 comments)","accounts_in_message":[],"_revision_number":21},{"id":"a841bab721f5151245e12a4d4dc447f87d1d9b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:30:29.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd7fb43c64c7450089c8ded550767e4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f13b43282c4a0085c1002cbc6f5464 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f6082712ce2741a3800ce4da225d785b : SUCCESS in 1h 05m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fc84cb2b2664a1fa8b7a2f777561f86 : SUCCESS in 42m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84b8adad1d7742afb04c5a022e6d6d4f : SUCCESS in 1h 05m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eaee548686c54c9581a2d73b7f44bc57 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd1119c367d64c2684e267855a73c96c : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/786e3ef4c7ff42a1811da2fb5dc22d71 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91b431b842904ae2a94e3b8eeb5e2c41 : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efc682809a2c45cf9566071fe59ca1f0 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3df9b0c90d2f4d5198368b09049997c8 : SUCCESS in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f7e1b3f54d1440d0b6093726a9fef84b : SUCCESS in 16m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/044e5d24e28e4668843eb0375f699cd9 : SUCCESS in 18m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/92b1b3d36fef469ba66d50e4469f790d : SUCCESS in 18m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4d2baa1154a54466ab70ae8f63f71dec : SUCCESS in 24m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ca1830674c942cda38fc7dde1cd56d3 : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f812a6b70cbf449db3089177c35724f8 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0916eeedb8c5443c86e490dc6ba2e372 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4297bdc7823648ffaa7f9d09ab0ad851 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c1453c76cb54732b09c5318638f8c8b : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/906abef16e62428fb974f5908c18f078 : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c78cec6ae52c40cfbeacb02661ea73f0 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/872e842c87ab4411925489743b9078cb : RETRY_LIMIT in 2m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc2f18e313e04d1e93fa174e1d28659f : RETRY_LIMIT in 1m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab815c86bdcb43a490eb3f2a2e9de4ca : SUCCESS in 1h 00m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9dd48decd9c6467c96e0d133562b5656 : SUCCESS in 49m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a24d3a574e4643559a0870651650a313 : SUCCESS in 55m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9460343fb9cb491fb6063f04581ec16c : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c4754448d641969475d93f76111bd6 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f07df65ded0d429dbb3a357f5e20e56a : RETRY_LIMIT in 2m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fc944cd7b2242038a7199f8656cad01 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":22},{"id":"98d0876a53779d00c024f4c909911f6679674e6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:32:53.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc758bc14124134a8d25d101a17274e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/638ec06af12e4fffadf032d550da77b1 : SUCCESS in 17m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e468a866ed5847ba8f65dd9654d9e040 : SUCCESS in 54m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/52334712ad45464583addede19c13d43 : SUCCESS in 1h 19m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e706c22cd1545889968e1cb782924aa : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":22},{"id":"d8d67640f4f42bba059d70a072729af47f6152f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-04 21:05:15.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"f2165e32e897ca0402dff284a2ae7f88d83c47d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 22:15:32.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7477b0aac8e4acd92d159c6dd340211\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3f2ad0c4edf4e889040973674ef6f63 : SUCCESS in 19m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79bc4ee416ca483197be4795759fa947 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/669048c443464370a408a04efc4c33fa : SUCCESS in 23m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/772e95b5e8ba412a874008ad9b2ef0c1 : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":23},{"id":"be9dddbd9ca4aea6fe1091b428d26702499d088c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 22:33:30.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2ee2df30aa485491e353027a7abee6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a90e91a56314fc9b9ead92685380372 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/06144db8c5ba461fa406ec98cfa07596 : SUCCESS in 1h 20m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c82569ab99b9431a810e8eb1bc5b0fad : SUCCESS in 50m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80052d40ed7f4010ab032b501e02274f : SUCCESS in 1h 10m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bed66b1e235d42afaa270b3d74a2df1c : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10842f485ea74723be20aa54634a9c62 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3dc3c8340cd4d90a5ebc3c7eb312939 : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41b12fae7705496394f36bc558e9427d : SUCCESS in 12m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/48a7451ecba14000a83d54de8286933e : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59f15c0ae0304449b27c3701e8148d32 : SUCCESS in 21m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d55d01db3ebd4a8daf219f24f7c1a8b0 : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6488b71d7c544cd9dbc824d0ea2defd : SUCCESS in 18m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b1118cc65bbb4ae6931a23a118a3e019 : SUCCESS in 18m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/249960cb3bb94694a498624ca14b7813 : SUCCESS in 24m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/903abad921964a6a85005c7e296291ed : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c80e0f9f544f457a8b25c1ddec83f9d8 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31e5ae0b314546d0bc4c4c43e85fb478 : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e05174fa404403ea0d10af9cff14141 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9a139980743495aa0800db87ea82965 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6bfb1ccc7ad0487a98f49b000ea22f65 : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ed0b65138aa4048841985b3470707f1 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ced2ae22c31c48008c5664e64e9b635a : FAILURE in 14m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b7576382b69454a820060cef1373962 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48d8fc416bb2494eb82567451c3a2149 : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ec1f0baba0e4f5588a9088009967a9c : SUCCESS in 59m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f82420e000154e8eb1d0ff0485e92be1 : SUCCESS in 56m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/694b42cbfe5042c58e7888dc40e571f0 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b98c206203b845f8a12007cb9453cbfe : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/201902094079492d8627f1d621fd0276 : SUCCESS in 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/211b333fdf39437aaf5aa73e74e3140a : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":23},{"id":"7a6f6fa2e2eb29bb082a7fb2cbe97f898ce92512","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-07 17:40:21.000000000","message":"Patch Set 23: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":23},{"id":"9587ee479c61fbe95e2199b1b68b92fb49ee97dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-13 23:07:27.000000000","message":"Uploaded patch set 24: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"ac62c2056a89372e4612760b16c6b9175db85ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:30:10.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81a87f6cdc4b4187b7c64d6804f52461\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3b16279d1e4e598abf2751eb5e2f07 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/c8e9369deeca4aa7ad6e5f4b2099b38d : SUCCESS in 1h 08m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce8e7a558f824663ab7fb7e970be09f1 : SUCCESS in 1h 00m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/538942818b804f8894a6206fb522caeb : SUCCESS in 1h 04m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c12bd19019944b4aa1ebc78f4d3a967 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6f0c98652f4485ba54074c82341c165 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1358c8e30ba7442fb50d6a388a1fd707 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf3e26c515c24ace9a887f831844fb1d : SUCCESS in 11m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24d5331b91634e6289cc8bb5c676df68 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c888956007e84324b2d0e90a60eb18e2 : SUCCESS in 18m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97b2c872d70e4475a093ae889581e2fa : SUCCESS in 18m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f34ff1ea21514b4eb4ff368113d6002a : SUCCESS in 18m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1f2abd2357534b9080b828aa2566f439 : SUCCESS in 18m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0c8e881382374ab8921fb61b4cc59d2e : SUCCESS in 22m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b26af8a71fe04a25a5bc7b842e25baab : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1e0219c6d90424b900f8993b1fb5a88 : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4de9aa626d2646628f183a6ea7724b32 : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71f670dd007a42b98600831d5a6e2588 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cab982f9e4f34ba4a96afb007d84bf82 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5b0d0bc8a70d473f9d37a0ab67a6e0dc : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1acac52c9b34e48859b840d711bb943 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e7a4683120240c5991d79cfddf5bc33 : FAILURE in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/884bbf04932a4b68b090d566dd066b69 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/47787c9e8823457099fa7352ddf61780 : SUCCESS in 54m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e648d93f07444f185d0924046c93a4e : SUCCESS in 40m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74560aa14e8540a3ade94bd10c53f60c : SUCCESS in 54m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a7d2b1f1492499a874a9db6cb96d6cd : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ed4ebcca8147aa8c3b9854a41c653f : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7fae8f0d3b5b4cf18bd17bad7c50cf02 : SUCCESS in 18m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a331016314f64377a26614174d8e56c2 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":24},{"id":"291e8d2598c0bcca9de99e90d872b4db08ea9ce6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:40:34.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/044323b92d164ce492eeba5cde0a3353\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7703216e3a1c4ba796895a8a93bda7bf : SUCCESS in 18m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b51c5da8f577422a8a959abbfb872d15 : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/691e29a7c60f4d988e7a7ee56d1de041 : SUCCESS in 23m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e0e2ce3f7ac4eb5a0a6b9230752da98 : TIMED_OUT in 1h 30m 51s","accounts_in_message":[],"_revision_number":24},{"id":"f6a38f71a18b4cb52bbdac84f7f512ab1b231e34","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-14 01:19:48.000000000","message":"Patch Set 24:\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"b18c00d8efd48f1f2113f42b6ab8d9907fc87db9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-14 19:04:44.000000000","message":"Patch Set 24: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":24},{"id":"32dfa770aeb0b9db24a9e8529d34bef5b10fb857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 19:05:33.000000000","message":"Patch Set 24: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":24},{"id":"46fe57fff8e38d299bb931d9eeb05fcd829db216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 20:29:11.000000000","message":"Patch Set 24: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49f5878cdb14853b15f19c337fda224\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/846899fd585446f28e8d6078000ecf59 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/fbd6cac2f8234bb8b9f786a0f9077cdb : SUCCESS in 1h 04m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d6e05d66d3d42ffad61d1fe05db1d86 : SUCCESS in 50m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c375b66b4d244978a174bf84b4a2564 : SUCCESS in 1h 07m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/037ae619c4954ccd887977d19b6e4eb3 : SUCCESS in 11m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fe21a83242934e20b19bfd0acd28a1c2 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a531ed2f29634d099d06e64c0b4d850b : SUCCESS in 20m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93c6fa970eb04090bab04095e072bd32 : SUCCESS in 23m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/71a353b49c5f4027a4d73e4b148ce977 : SUCCESS in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/760c136d98934a28946cada3a73b25fd : SUCCESS in 17m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a7c1a840a8f4f7b81fedab069afbdd6 : SUCCESS in 20m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/906da2248d3747e480cbc066915ae96a : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73d9ab1145d84ae3be8386335b2a20df : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c9c66f4747794b7693733889a8c7e57c : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8772f03bd96f4bc6a8060c8c4d3d5b32 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b5900e11fa0431ead282228a51c40c5 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d91ce164500c40bdb0ad8248ede195bc : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1088f28b51e4f418bfbd92f23f3147d : SUCCESS in 5m 54s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d790e561214649b69eaaada0dffbde06 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8bd8fb06a9e497495b1330e92a93141 : SUCCESS in 1h 08m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b522b916a6042868620f170a2072675 : SUCCESS in 43m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/390577a42f9f41e2871df5d5b13385fe : SUCCESS in 43m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/268637fc3e7840dda8329dc0e4a56eae : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75db26eae2e345edb59343f43f5cbcb0 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad9d7adff4154aa795fd50137407192c : SUCCESS in 19m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58f6d7cd51cc485490f56ced97a1f832 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":24},{"id":"fa6f20aa24d799bdca3f5f5187742984e04a00a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 20:29:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":24},{"id":"e0835a4f1117e7c2c084e28cea11b1734bc2a303","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 20:30:35.000000000","message":"Patch Set 24:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01533bddf1a45a8a489cee903654a36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bda8ca54994fcdb40b921bc9b09662 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d840e24f8bf84604b80fad658f7051e0 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ebad72c9be76451f8b34cfde21932325 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"3637b1abd9240f7feeebc7f299d42fe8815d7dcd","revisions":{"ac7492f664fa8c0948ee29999bfce09a2d3a85fe":{"kind":"REWORK","_number":1,"created":"2024-03-11 23:06:14.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/1"}}},"commit":{"parents":[{"commit":"5f753b64956a8bb7292ae034f44c674a9872fb99","subject":"grace period and x-open-expired for expired objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f753b64956a8bb7292ae034f44c674a9872fb99"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nupdated every time the byte size is updated so that the latest\nx-content-type (with the latest bytes) is used when the content\nreplicator runs to ensure eventual consistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac7492f664fa8c0948ee29999bfce09a2d3a85fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac7492f664fa8c0948ee29999bfce09a2d3a85fe"}]},"branch":"refs/heads/master"},"f8c8ee7f15cdbc00fdc46675d34c89182cd27541":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-13 21:36:13.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/96/912496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/2"}}},"commit":{"parents":[{"commit":"8e7d26f249249f324bf9953b577cd47a1bddf85c","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d26f249249f324bf9953b577cd47a1bddf85c"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-13 21:36:13.000000000","tz":0},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nupdated every time the byte size is updated so that the latest\nx-content-type (with the latest bytes) is used when the content\nreplicator runs to ensure eventual consistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c8ee7f15cdbc00fdc46675d34c89182cd27541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8c8ee7f15cdbc00fdc46675d34c89182cd27541"}]},"branch":"refs/heads/master","description":"Rebase"},"d910937495f86036824a6bf98ae4b3c259418d64":{"kind":"REWORK","_number":3,"created":"2024-03-14 00:31:40.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/3"}}},"commit":{"parents":[{"commit":"8e7d26f249249f324bf9953b577cd47a1bddf85c","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d26f249249f324bf9953b577cd47a1bddf85c"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-14 00:30:45.000000000","tz":-240},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d910937495f86036824a6bf98ae4b3c259418d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d910937495f86036824a6bf98ae4b3c259418d64"}]},"branch":"refs/heads/master"},"bc13de6c1ee658b2f31e5c86fa2c4cdfd25ec9e3":{"kind":"REWORK","_number":4,"created":"2024-03-18 22:57:47.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/4"}}},"commit":{"parents":[{"commit":"8e7d26f249249f324bf9953b577cd47a1bddf85c","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e7d26f249249f324bf9953b577cd47a1bddf85c"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-18 22:57:39.000000000","tz":-240},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc13de6c1ee658b2f31e5c86fa2c4cdfd25ec9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc13de6c1ee658b2f31e5c86fa2c4cdfd25ec9e3"}]},"branch":"refs/heads/master"},"0e8fef54cc78f5ecadafc845e708f5f94727fda8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-08 18:51:21.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/5"}}},"commit":{"parents":[{"commit":"d51168cc493a7326f5af30e15550efa272dd571e","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d51168cc493a7326f5af30e15550efa272dd571e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-08 18:51:21.000000000","tz":0},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e8fef54cc78f5ecadafc845e708f5f94727fda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e8fef54cc78f5ecadafc845e708f5f94727fda8"}]},"branch":"refs/heads/master","description":"Rebase"},"171a6672f5571f5ea9a9af7bb485a5d027b0c238":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-17 20:08:36.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/6"}}},"commit":{"parents":[{"commit":"d875bee2f4458f4e47652c147335c6611566a684","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d875bee2f4458f4e47652c147335c6611566a684"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-17 20:08:36.000000000","tz":0},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/171a6672f5571f5ea9a9af7bb485a5d027b0c238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/171a6672f5571f5ea9a9af7bb485a5d027b0c238"}]},"branch":"refs/heads/master","description":"Rebase"},"12ddcf2aaad10df71e13838794fbf0bb5587b1d6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-18 16:12:32.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/7"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-18 16:12:32.000000000","tz":0},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12ddcf2aaad10df71e13838794fbf0bb5587b1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12ddcf2aaad10df71e13838794fbf0bb5587b1d6"}]},"branch":"refs/heads/master","description":"Rebase"},"ba168d66877c5a4d53eb33b142e6596d55c8f444":{"kind":"REWORK","_number":8,"created":"2024-04-19 20:46:27.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/8"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 20:46:11.000000000","tz":-420},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba168d66877c5a4d53eb33b142e6596d55c8f444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba168d66877c5a4d53eb33b142e6596d55c8f444"}]},"branch":"refs/heads/master"},"3d9e14aa7b89b307eede9523ddc29951258fd1d2":{"kind":"REWORK","_number":9,"created":"2024-04-19 21:39:02.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/9"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 21:36:30.000000000","tz":-420},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d9e14aa7b89b307eede9523ddc29951258fd1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d9e14aa7b89b307eede9523ddc29951258fd1d2"}]},"branch":"refs/heads/master"},"c318cae3dcb5111745bb630b8584bcf50429c8ff":{"kind":"REWORK","_number":10,"created":"2024-04-22 20:49:50.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/10"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-22 20:49:38.000000000","tz":-420},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c318cae3dcb5111745bb630b8584bcf50429c8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c318cae3dcb5111745bb630b8584bcf50429c8ff"}]},"branch":"refs/heads/master"},"2f4e6bc410e93e7911d994f72af5a53139de3fe7":{"kind":"REWORK","_number":11,"created":"2024-04-22 23:03:44.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/96/912496/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/11"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-22 23:03:33.000000000","tz":-420},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f4e6bc410e93e7911d994f72af5a53139de3fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f4e6bc410e93e7911d994f72af5a53139de3fe7"}]},"branch":"refs/heads/master"},"7cd3efdc5c4a5ea28d7a7e88e67d4920590fa9c4":{"kind":"REWORK","_number":12,"created":"2024-04-23 03:12:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/12"}}},"commit":{"parents":[{"commit":"9c44f5b5da11e97e4996b6f34e23ba3d1f60990e","subject":"support x-open-expired header for expired objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c44f5b5da11e97e4996b6f34e23ba3d1f60990e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-23 03:05:00.000000000","tz":-300},"subject":"add bytes to expirey queue entry","message":"add bytes to expirey queue entry\n\nThe size in bytes of objects are stored in expirey queue entries\nunder the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cd3efdc5c4a5ea28d7a7e88e67d4920590fa9c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cd3efdc5c4a5ea28d7a7e88e67d4920590fa9c4"}]},"branch":"refs/heads/master"},"d01b968373a0f18bc4144b1e094e9511329bab50":{"kind":"REWORK","_number":13,"created":"2024-04-26 14:53:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/13"}}},"commit":{"parents":[{"commit":"b6c377e7e5b9980442d1c70618af7b7f0c0c66f1","subject":"Merge \"support x-open-expired header for expired objects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6c377e7e5b9980442d1c70618af7b7f0c0c66f1"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-26 14:45:29.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d01b968373a0f18bc4144b1e094e9511329bab50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d01b968373a0f18bc4144b1e094e9511329bab50"}]},"branch":"refs/heads/master"},"07dcab6a8c3f6e08fb811f1361d7d1442ada8263":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-05-02 01:19:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/96/912496/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/14"}}},"commit":{"parents":[{"commit":"0a3c46e6c4a56e425020ed37f582b2fb29508315","subject":"Merge \"Refactor utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a3c46e6c4a56e425020ed37f582b2fb29508315"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-02 01:19:29.000000000","tz":0},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07dcab6a8c3f6e08fb811f1361d7d1442ada8263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07dcab6a8c3f6e08fb811f1361d7d1442ada8263"}]},"branch":"refs/heads/master","description":"Rebase"},"cb776885bc74111941e2cfad78cf6bb3d43db850":{"kind":"REWORK","_number":15,"created":"2024-05-02 15:58:47.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/15"}}},"commit":{"parents":[{"commit":"0a3c46e6c4a56e425020ed37f582b2fb29508315","subject":"Merge \"Refactor utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a3c46e6c4a56e425020ed37f582b2fb29508315"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-02 15:58:09.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb776885bc74111941e2cfad78cf6bb3d43db850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb776885bc74111941e2cfad78cf6bb3d43db850"}]},"branch":"refs/heads/master"},"8381a309952be9d2da9861069f6744dcab228c13":{"kind":"REWORK","_number":16,"created":"2024-05-17 22:43:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/16"}}},"commit":{"parents":[{"commit":"0a3c46e6c4a56e425020ed37f582b2fb29508315","subject":"Merge \"Refactor utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a3c46e6c4a56e425020ed37f582b2fb29508315"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-17 22:42:38.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8381a309952be9d2da9861069f6744dcab228c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8381a309952be9d2da9861069f6744dcab228c13"}]},"branch":"refs/heads/master"},"6eac37ecdc6d2af9b2c5f63888051b220cb712c4":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-05-21 18:04:24.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/17"}}},"commit":{"parents":[{"commit":"337079f21f66c4f288c5ffa736a177f446472e5e","subject":"Merge \"replicator: Ensure handoffs can clear with large handoff_delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337079f21f66c4f288c5ffa736a177f446472e5e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-21 16:54:18.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6eac37ecdc6d2af9b2c5f63888051b220cb712c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6eac37ecdc6d2af9b2c5f63888051b220cb712c4"}]},"branch":"refs/heads/master"},"f8ba7f1aea58230fc729f9e18bb486645be3cba2":{"kind":"REWORK","_number":18,"created":"2024-05-25 13:36:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/18"}}},"commit":{"parents":[{"commit":"f3adce13753f9b32be60a68b49c79321cb552a61","subject":"expirer: bad config should not loop forever","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3adce13753f9b32be60a68b49c79321cb552a61"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-25 13:30:55.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8ba7f1aea58230fc729f9e18bb486645be3cba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8ba7f1aea58230fc729f9e18bb486645be3cba2"}]},"branch":"refs/heads/master"},"ad2ed9602607d5efe4fd34b0fcef5d23fc440c23":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-05-28 17:00:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/19"}}},"commit":{"parents":[{"commit":"82debd2cda0a3f2727150be37e001daee857d292","subject":"Merge \"docs: Add tempurl/formpost metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82debd2cda0a3f2727150be37e001daee857d292"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-28 17:00:22.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad2ed9602607d5efe4fd34b0fcef5d23fc440c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad2ed9602607d5efe4fd34b0fcef5d23fc440c23"}]},"branch":"refs/heads/master"},"5ad3a5a4fe193e8cd9b4e67bce377363e890098a":{"kind":"REWORK","_number":20,"created":"2024-05-31 20:10:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/20"}}},"commit":{"parents":[{"commit":"99c629edb813c7d9fcd736f5d34b3a47d224a457","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c629edb813c7d9fcd736f5d34b3a47d224a457"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-31 20:05:07.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nUpgradeImpact: During rolling upgrades you should expect expirer queue\nentries to continue lack swift_expirer_bytes\u003d annotations until ALL\nobject servers replicas have been upgraded to new code.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ad3a5a4fe193e8cd9b4e67bce377363e890098a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ad3a5a4fe193e8cd9b4e67bce377363e890098a"}]},"branch":"refs/heads/master"},"28441819065b59e4c89c62f10310f0679058282f":{"kind":"REWORK","_number":21,"created":"2024-06-03 22:24:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/21"}}},"commit":{"parents":[{"commit":"99c629edb813c7d9fcd736f5d34b3a47d224a457","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c629edb813c7d9fcd736f5d34b3a47d224a457"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-03 22:21:49.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nUpgradeImpact: During rolling upgrades you should expect expirer queue\nentries to continue lack swift_expirer_bytes\u003d annotations until ALL\nobject servers replicas have been upgraded to new code.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28441819065b59e4c89c62f10310f0679058282f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28441819065b59e4c89c62f10310f0679058282f"}]},"branch":"refs/heads/master"},"d4f329b19ae7fb819796f2e133a4afb32f52f907":{"kind":"REWORK","_number":22,"created":"2024-06-04 18:11:10.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/22"}}},"commit":{"parents":[{"commit":"a666010aae8aa9ba67341eeb5e69380fc20d02f5","subject":"Lazy import is not needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a666010aae8aa9ba67341eeb5e69380fc20d02f5"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 17:42:35.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nUpgradeImpact: During rolling upgrades you should expect expirer queue\nentries to continue lack swift_expirer_bytes\u003d annotations until ALL\nobject servers replicas have been upgraded to new code.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4f329b19ae7fb819796f2e133a4afb32f52f907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4f329b19ae7fb819796f2e133a4afb32f52f907"}]},"branch":"refs/heads/master"},"4e1c0e16a925bfaeb78cea9a34cf88feb7145caa":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2024-06-04 21:05:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/96/912496/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/23"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 21:04:02.000000000","tz":-420},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes of expiring objects are stored in expirey queue\nentries under the content_type field. The x-content-type-timestamp is\nprovided along with the x-content-type update so the container\nreplicator resolves the latest content-type and ensures eventual\nconsistency.\n\nUpgradeImpact: During rolling upgrades you should expect expirer queue\nentries to continue lack swift_expirer_bytes\u003d annotations until ALL\nobject servers replicas have been upgraded to new code.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e1c0e16a925bfaeb78cea9a34cf88feb7145caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e1c0e16a925bfaeb78cea9a34cf88feb7145caa"}]},"branch":"refs/heads/master"},"3637b1abd9240f7feeebc7f299d42fe8815d7dcd":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2024-06-13 23:07:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/912496/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/912496/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/912496/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/912496/24"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-11 23:02:10.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-13 20:47:51.000000000","tz":-300},"subject":"add bytes of expiring objects to queue entry","message":"add bytes of expiring objects to queue entry\n\nThe size in bytes from object metadata of expiring objects are stored in\nexpirey queue entries under the content_type field.\n\nThe x-content-type-timestamp take from object metadata is provided along\nwith the x-content-type update so the container replicator resolves the\nlatest content-type and ensures eventual consistency.\n\nUpgradeImpact: During rolling upgrades you should expect expirer queue\nentries to continue lacking swift_expirer_bytes\u003d annotations until ALL\nobject servers replicas have been upgraded to new code.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n\nChange-Id: Ie4b25f1bd16def4069878983049b83de06f68e54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3637b1abd9240f7feeebc7f299d42fe8815d7dcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3637b1abd9240f7feeebc7f299d42fe8815d7dcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
