)]}'
{"id":"openstack%2Fswift~912528","triplet_id":"openstack%2Fswift~unmaintained%2Fvictoria~If2a9200483bc0414ec261a485e369ba2249bf5a7","project":"openstack/swift","branch":"unmaintained/victoria","topic":"create-victoria","hashtags":[],"change_id":"If2a9200483bc0414ec261a485e369ba2249bf5a7","subject":"Update .gitreview for unmaintained/victoria","status":"MERGED","created":"2024-03-12 10:48:21.000000000","updated":"2024-06-27 19:09:52.000000000","submitted":"2024-06-27 19:07:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912528-create-victoria","meta_rev_id":"83cc3b10953c5c7b2933960502e8d0f065a39101","_number":912528,"virtual_id_number":912528,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 19:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-18 15:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-15 15:34:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-18 15:39:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 11:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-15 15:34:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-03-18 15:39:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1d5246b2a39695b61db50d2710d5990ad4cb1af7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-12 10:48:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e4af83bf2393231a4299a762252831833e0ac0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 11:47:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0ef85f7cec4954bc0c958d54ea79a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed5643d89004d65a37d1992ebbf07b8 : SUCCESS in 9m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d4abb8bae5241a580af3277bbf6fe58 : FAILURE in 3m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1728bbdd671441a8ac103bdbe94f654a : FAILURE in 4m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/20113d0a13c54fa8b115c4ac2eba14c1 : SUCCESS in 29m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e92c59b45104851b49b28ec11836371 : SUCCESS in 14m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a07319d2ee8e45969d6b315e49530012 : SUCCESS in 12m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e1131af73860472994b1809524cf1e49 : SUCCESS in 13m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1eb78efdb3f04d71821a49838af8ea02 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae62f2fab26e4a498229d590bef6d653 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a67edb7e92efac3a0e15f370dae57aa7fb162a06","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:34:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a9b0826edfc01c02a49e9c7e89672b59221fc4bc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-18 15:39:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dc77d1265354b763e89671c57a1e44c7f74742e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 17:29:51.000000000","message":"Patch Set 1: Verified-2\n\nUnable to freeze job graph: The nodeset \"centos-7\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"73709a1b02e2bda4699e70e0863919f9e3c75929","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 16:05:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7c60d8fd80bd734dfb74907f910882279167ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7555cbe7081d4e51b3f6fe3c5b5f9fc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf36106265344dc9df7b94df55a7b8d : SUCCESS in 7m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54fe40ef6ca147b9a8710747deb5feeb : FAILURE in 4m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8aaa7085856d42dabc951b34d1f11414 : FAILURE in 3m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b533d4df4aa24b09a6320b2d7c8daeb7 : SUCCESS in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98c75a7fcda5442bb6f092d313c736bc : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c6efd93bb114069816f367185b68453 : SUCCESS in 13m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c1adfb18b2a045a3a0490a51fd82aa2f : SUCCESS in 12m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d31c3edd2d425f9e3b635e16c23ca6 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939e55488bd0497096489dc7b1351e1b : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"118f6f3cd978509bbff8b47343893ab16edec08a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:59:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8c8d26c5d2fc549671c3c3434ad32f8ff9b0e74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:07:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10162b264724b5e92bb6fa4fd3b4bed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffdca9157be4f6483cf1017a28f6e83 : SUCCESS in 7m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dc2f27e2ccc0459ba9b3460a275c194e : POST_FAILURE in 4m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0aa70ae1500b4d32a004bdd48e075d50 : POST_FAILURE in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73502047b0604777b2000a88f683ec3b : SUCCESS in 14m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc782b0590bc46baa0c757761e6f0d92 : SUCCESS in 13m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8b60d85aa0cd4be18483d4203b8d6386 : SUCCESS in 15m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0d18c89f75634e3497ad673bdfcc3a3f : SUCCESS in 14m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a64889bd5cfb484bbce1b28acc23fffb : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c7bdfc8df1a4fe997740a559d187653 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f7b656cf25143959d4aa1fa2e3f2e4d : SUCCESS in 17m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/efa6456ab16e4d3ca18e32ae18ff5d19 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/38a1d7c4e0fe4bb59194dc7847bf65a0 : SUCCESS in 16m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5b78600602f496b8059d1b283acf8c1 : SUCCESS in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af2c184ae3a40bcaf959862d54cb3ff : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1f47fe812f0fbc24a480f41257aa822b2f35d315","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:07:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83cc3b10953c5c7b2933960502e8d0f065a39101","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6200f9f9c75443c5b11ee2af9920a833\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c27de8799a94dff8c30c3bca29cbaa2 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d642135eb8124279ac99c61ca091e928 : FAILURE in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b1c3959c7cc34412ba4059237d891fee : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43954bcdaeb1720b9a8b01993b6bb6dd8805d0ed","revisions":{"43954bcdaeb1720b9a8b01993b6bb6dd8805d0ed":{"kind":"REWORK","_number":1,"created":"2024-03-12 10:48:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/28/912528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/912528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/912528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/912528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/912528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/912528/1"}}},"commit":{"parents":[{"commit":"f80dce1f1921ec7812893bb22243b86aa7356f93","subject":"[stable-only] Cap virtualenv/setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f80dce1f1921ec7812893bb22243b86aa7356f93"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-12 10:48:19.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-12 10:48:19.000000000","tz":0},"subject":"Update .gitreview for unmaintained/victoria","message":"Update .gitreview for unmaintained/victoria\n\nChange-Id: If2a9200483bc0414ec261a485e369ba2249bf5a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43954bcdaeb1720b9a8b01993b6bb6dd8805d0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43954bcdaeb1720b9a8b01993b6bb6dd8805d0ed"}]},"branch":"refs/heads/unmaintained/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
