)]}'
{"id":"openstack%2Fswift~912540","triplet_id":"openstack%2Fswift~unmaintained%2Fwallaby~If20bb1675d0fbe7b20196d55b784c767e3f15682","project":"openstack/swift","branch":"unmaintained/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"If20bb1675d0fbe7b20196d55b784c767e3f15682","subject":"Update .gitreview for unmaintained/wallaby","status":"MERGED","created":"2024-03-12 10:49:00.000000000","updated":"2024-12-22 17:53:10.000000000","submitted":"2024-12-22 17:52:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912540-create-wallaby","meta_rev_id":"12aaf2d179a442b352af1a2cb65d4e6b7c4192d9","_number":912540,"virtual_id_number":912540,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-22 17:52:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-18 15:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-15 15:34:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-18 15:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 11:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-15 15:34:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-03-18 15:36:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d564225de101731f4d3633d350670d693417eff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-12 10:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"115532d3672beef47b4d733080852c4294ef4d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 11:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a2eac4469e14334b2c5621feda33fa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4265178da444454bc7e891d05ce8e64 : SUCCESS in 7m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/070b5f91ed8246649c26058d0cd54ea9 : FAILURE in 4m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/627802f84e8e4e02a2ba5645073d5b5e : FAILURE in 4m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ee1def9458d45f9980eef9c9bf591d7 : SUCCESS in 32m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/036a333f23d846aa923de69daf5d52e3 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27b9642889114fd6ba4b5e7b09ff415a : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc85aa38636477baf644cf593c1b489 : SUCCESS in 12m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c8f07b03921485fac1769019a871de5 : SUCCESS in 13m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a2ba391e5484a7bbc28a393ecf5cf60 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb0b72506284970a782ca86342f736f : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cbab36a6f93aa8cafccbd9920e4eec0f28c020be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:34:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6caa98f2cdff5e081fd82c9c732b337b7fa462b9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-18 15:36:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39224877744808a354fc7e9e66dbd09d5cf001ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 17:29:50.000000000","message":"Patch Set 1: Verified-2\n\nUnable to freeze job graph: The nodeset \"centos-7\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"2a7e9fb024e93d8b79caba1a2a9763eed0209868","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-30 23:45:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ca3758c3ca1f8516865aefcc603739c994654ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 23:46:05.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: The nodeset \"centos-8-stream\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"d3fd0f937df84dede464b7ee60ae52f6e82a0187","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 10:01:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"459b48c6d35a1a04ff5930b4edad494545ee139b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 10:01:42.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: The nodeset \"centos-8-stream\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"de450b062b1a35c6cb5d448be299e532cb62bc6d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-22 17:14:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c22ebb878434fe706d69d67c95f187e87c15957e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 17:28:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1923184eb1e4f988f2563771e736eb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2aaf0c329404df992808bec53a15c91 : SUCCESS in 6m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43f4b637495c43eb84d5cdc7de819f09 : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2ddb9fa6b0e47aeadef7d7f370e5df6 : FAILURE in 3m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33d6eb6c3a864442b3f7057b2c7a7d19 : SUCCESS in 12m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe390c67bcae44de8f5324230f4439c9 : SUCCESS in 12m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/86e617f04c934700903d6834d7264eef : SUCCESS in 11m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5c2d36962f044b4fa6dc5a35e7e95591 : SUCCESS in 11m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/acf62e617ce34e91aca2b4e338feead2 : SUCCESS in 8m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82d621a6ef4642fb8794085e0a922a2e : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/141bfeb2481248fc90297247752bb213 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"08eaa3176f06f6c500e5838982f613d12fcbec37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 17:28:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd55ccae364592bc2c0b4cea6dab128d2cb78043","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 17:52:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/814e053b603b4af8b8dc8314dd462fd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968cc6122bc04da4bc62a3873c6b87e2 : SUCCESS in 4m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e9da30e893ff4397ae82d00ae06082a8 : POST_FAILURE in 5m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4f8d98862a744411a3921fd6b21f0bc7 : POST_FAILURE in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7db40d05c91a42ffb11e110654293958 : SUCCESS in 7m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c057fc656c3b49e7940381ead8435a4f : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef9e917343104a378e92f7958d6c61a1 : SUCCESS in 7m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2a12f1eb2c64e34bc5d62e3fc364b27 : SUCCESS in 15m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c319267c0bf4dce94975fd149738faa : SUCCESS in 7m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e625215b31648a1b0ea91c728fa01bf : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8368731bdb2747edbcd5278b0887d091 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93d589558ba649b0a7a561e04406883b : SUCCESS in 14m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0910b311e8024e6b8e38270d79fdb653 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/27865bea621f41f890d1d6be5ccbee45 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/447fad5288e642df8d346799faa82ed8 : SUCCESS in 15m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cc035c70b8f4a91b97d59c7ca0c37a5 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faaaa83802a34c8f861cb56e566cc523 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0e186dad4c72babad3bf8f4ebaccec73a6674cca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 17:52:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"12aaf2d179a442b352af1a2cb65d4e6b7c4192d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-22 17:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edc969d3cc84db2a0d38827c40786f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9cd968e5a940668d4061fc1074dc2e : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6800e68ed10d4133b6a72f034d964422 : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0039888bcdbb4e06a6ff2f3c942e87fe : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0279d111c8ca2a3ac72b598e699035afd5a52596","revisions":{"0279d111c8ca2a3ac72b598e699035afd5a52596":{"kind":"REWORK","_number":1,"created":"2024-03-12 10:49:00.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/912540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/912540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/912540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/912540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/912540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/912540/1"}}},"commit":{"parents":[{"commit":"16dbcae9df9773f2b2bfea41fd4156e16aa14b37","subject":"Merge \"sharder: Prevent ValueError when no cleaving contexts\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16dbcae9df9773f2b2bfea41fd4156e16aa14b37"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-12 10:48:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-12 10:48:58.000000000","tz":0},"subject":"Update .gitreview for unmaintained/wallaby","message":"Update .gitreview for unmaintained/wallaby\n\nChange-Id: If20bb1675d0fbe7b20196d55b784c767e3f15682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0279d111c8ca2a3ac72b598e699035afd5a52596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0279d111c8ca2a3ac72b598e699035afd5a52596"}]},"branch":"refs/heads/unmaintained/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
