)]}'
{"id":"openstack%2Fswift~912595","triplet_id":"openstack%2Fswift~master~I7700aa3c93df311644655e7ebaf0b67aa692ee80","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I7700aa3c93df311644655e7ebaf0b67aa692ee80","subject":"CI: Bring unit test jobs in line with 2024.1 tested runtimes","status":"MERGED","created":"2024-03-12 16:20:22.000000000","updated":"2024-03-12 18:53:54.000000000","submitted":"2024-03-12 18:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912595","meta_rev_id":"e89c39133aa3f65a7e93fcd6b3ba131116e56097","_number":912595,"virtual_id_number":912595,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 18:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-12 17:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-12 17:16:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-12 17:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-12 17:34:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cd932d39dc45adf34f35dc15926edd5193249a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-12 16:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74b9633baf8eb7c2832c65f72be4111946aa1358","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-12 16:20:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f8a150fc5b37248c5dc72116170d21dd3721a30","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:06:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e053bf68a3a24f8f876b129841dd193d\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/09915d0de3f6430facaa7a1eb4d6c425 : SUCCESS in 21m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41f3e737e6844a0ca3846d0a89648d76 : SUCCESS in 19m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/52dfaafcab1643fc8ce86493ddb51016 : SUCCESS in 20m 11s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8c411a720b64551aa278fd902e8b930 : SUCCESS in 16m 35s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e94ea7b0bec43ee991322c775070746 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4606ef8da701449293a327cd49812093 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa31c1894c4241f18fce93e9f36acdcc : SUCCESS in 19m 34s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/04247b289f364794a9abd753ba810f42 : SUCCESS in 24m 30s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/22dbbb4aacef48ee8563cdf916cd8f07 : SUCCESS in 25m 10s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/974ef301c39f42efa11ea115973f22a8 : SUCCESS in 18m 45s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/57cd0e6be6c4426db279874eea9b225d : SUCCESS in 18m 51s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/47fa5e64333245bcac7b4faa06211cde : SUCCESS in 19m 32s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/2f593d59597a4772b54b9026b1e2b555 : SUCCESS in 21m 04s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/eecbe003bb0546cf9bc01f4fb58345eb : SUCCESS in 18m 40s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/abdc35a0537144d8ae69a10c38f8e53f : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6b777d6f835de56267185ae29ae6cccea90c3125","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-12 17:16:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94e3880c9eafb194fb3566a6b5508be7537bcfa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:34:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7043d95578841b691c3ecd021385778\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb77560b1264e239eec05433df8ca4e : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/48e88f82e5874617b77d21bfbc0ae29e : SUCCESS in 56m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/516de598a6924f6e9888668665fefb02 : SUCCESS in 51m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d29a21f28b8c40fcb6c4becf0a9d42dd : SUCCESS in 33m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fecf493b0a2414697d8ba77044b9fc5 : SUCCESS in 44m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a7dd3e19c6b453abc4d37c0fd3f97b1 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1013c47321224be693d47923ddf525d4 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/659200a885634a50907372fd834dde06 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/305a42dea14147b290572cf2d40af149 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9388a97351bd4f14a8dc39a1bb6f5c59 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36b718d3d2bd434aa191dcb2d2464f08 : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/235bd76912cd4d22aa6c69886aa686c9 : SUCCESS in 18m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad06072e63cd4bc887d5aa31a65987ce : SUCCESS in 16m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe6e051905c640f2839acf5ecc193893 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/484140eaacde47a0869a546e16fa84de : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08a103b6ccd346778a04b16815719f58 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e2af7ac2ded43a5bab83ce4cfdf3287 : SUCCESS in 20m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/301cecf3951b4953b0ae42f2c30cf78e : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/81424d9c4acf409e8296a892e96a5cef : SUCCESS in 14m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/affd68dc13a1482bbe2a131007a72e00 : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7f84eacd6694b4ba7ca6e1500e00d18 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9035b21e2098409687324b097c746ec6 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aaf3456502334d658c62a274e25eadf0 : SUCCESS in 8m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8b487074dfc4b36a96e6d2bb7bd85ad : SUCCESS in 1h 05m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7dbcadfa4dc41a3ba5db77a15c7c3d0 : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/322da5535436462fabd1e87a58398d40 : SUCCESS in 56m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/767b929bb3e54f1d92a8474abe7084a6 : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03d51485c074be697a42f0812e95701 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea4e13689d6e4df7ad29a799bb178afe : SUCCESS in 20m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa2314ec5626414abcc56307b1d6fbf4 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5b4fd5b0c5b36772a9625a3077519b8d3494174a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:34:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae89a4be0e25a41e99b88aa692fdeb6ebcd170c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:49:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf2eed086f34da3a67c595a14a023e3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51ab60884173499d84c1c82a9d2a3f35 : SUCCESS in 15m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7ef6de6497a413ea2d0d65f7abcf493 : SUCCESS in 23m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b065d151891141469a6a168575e03e01 : SUCCESS in 19m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/87e92c4170784c3da531f6553eb434f3 : FAILURE in 1h 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3afd6bc8aba0740483d7275e7db57f9d9f206691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 18:51:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5da475ed730465cb9d59ca372f85f50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be4754171754db182a1aa419934bae3 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e62c4f04d1d446d692966290a64bfa4b : SUCCESS in 57m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ed9be7b5fddb43f2adb42d0a4be7052b : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/710f5fd9a243487c8929266999afd991 : SUCCESS in 39m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ae1f6c6e29d49e8a4db7b2709d7a2a3 : SUCCESS in 54m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f61ab78438a047938bb1d1a17cd11ed9 : SUCCESS in 11m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cce6c6415aa64bea93395656d910a199 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7195103e22404008ab16918d504aa5f4 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c32c70b4f7e54ea482e15b6afe178212 : SUCCESS in 18m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/128514cfdc3a418ba6cfe5b5b0a712aa : SUCCESS in 17m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/da23f07828ec4eac9ed775b490a9a966 : SUCCESS in 20m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e18b846d310744cd8d7f40864bdf0018 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85b220a9668542d19e7e8d016377e908 : SUCCESS in 24m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99e3bb259e7440a2bc1f4feb6186b2d5 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/78a2e7591afa476b82854d050c46e2b6 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b377f1b2901407fb69b3ff72329b103 : SUCCESS in 23m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5320ad75462d4a8cbf6b45b76d121d5b : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b937f2d126d483db26e5c2a17af5d70 : SUCCESS in 8m 57s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/508768b7306c4cce98efac72dfe10d11 : SUCCESS in 7m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ac32efd7c734a478a50ece4a5c5ccbb : SUCCESS in 1h 16m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86315133efe249e9acea887c2aa3ec13 : SUCCESS in 53m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/071042af612144ca8f7a5c57687cca8f : SUCCESS in 52m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/056512bc59ca42e6b3cf570e33b0ba55 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91e34f1fb6c43e1b317f9d8ccce1b28 : SUCCESS in 5m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/754e780188e34e04922abb0e97bd37fc : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a3a1b08a5d55dd03de751eb1668f420af9c04fe9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 18:51:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e89c39133aa3f65a7e93fcd6b3ba131116e56097","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 18:53:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d3bdada2f742538377de502fd34eff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5ba40093424375a3ca43de1a687af8 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4b8694f24a3b4c54873a5095f3c8c867 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/02c4f869ed5545cdb8718b97132d433b : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9279e3d2c01fbd1d4cc9d95005aff5deeaa77c6d","revisions":{"9279e3d2c01fbd1d4cc9d95005aff5deeaa77c6d":{"kind":"REWORK","_number":1,"created":"2024-03-12 16:20:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/912595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/912595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/912595/1"}}},"commit":{"parents":[{"commit":"b06ffea9413168240d31110fcff0a5df2cf62f47","subject":"CI: Update rolling-upgrade jobs to point to unmaintained branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b06ffea9413168240d31110fcff0a5df2cf62f47"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-12 16:13:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-12 16:19:53.000000000","tz":-420},"subject":"CI: Bring unit test jobs in line with 2024.1 tested runtimes","message":"CI: Bring unit test jobs in line with 2024.1 tested runtimes\n\nThe last time I really looked at this was probably Yoga, when we were\ntargetting 3.6 through 3.9 (and left 3.7 and 3.8 as experimental jobs).\n\nNow, though, OpenStack is targetting 3.8 through 3.11; as before, we\ncan assume that if tests pass on those two versions, they should pass\non the versions in-between, too. (But still have them as experimental,\non-demand jobs).\n\nSee https://governance.openstack.org/tc/reference/runtimes/2024.1.html\n\nKeep 2.7 and 3.6 testing as our own self-imposed minimums.\n\nChange-Id: I7700aa3c93df311644655e7ebaf0b67aa692ee80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9279e3d2c01fbd1d4cc9d95005aff5deeaa77c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9279e3d2c01fbd1d4cc9d95005aff5deeaa77c6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
