)]}'
{"id":"openstack%2Fswift~912995","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1358bdc0c21d95fa879ee2fef58e72679f1068cf","project":"openstack/swift","branch":"feature/mpu","topic":"create-mpu","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-14 16:11:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1358bdc0c21d95fa879ee2fef58e72679f1068cf","subject":"Update TOX_CONSTRAINTS_FILE for feature/mpu","status":"ABANDONED","created":"2024-03-14 11:53:41.000000000","updated":"2024-03-14 16:11:07.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f42ef9cd5a85cd2a058b6fc50a7da7ea2b924c5","_number":912995,"virtual_id_number":912995,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-14 13:22:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 13:21:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 13:22:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 15:41:25.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-03-14 16:11:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"43a3cac9a4d0f51c8622cb0db35cc95940e3e6d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-14 11:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af206bcd342ffc7551dcbdc03aaaa8d159238121","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:21:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f864f446484d9db2be1291b0731e80\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f74f9730ec0e486cb285cd334be1b367 : SUCCESS in 16m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/187a928dce224a03ad236a693a10a4f2 : SUCCESS in 48m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7880f9574e145f5a84eae49045baa5f : SUCCESS in 20m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/15a571278b3443648b22f3dfa834533f : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c4e52d5ac01071a04c31bda038ace0bb4b8f67d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 13:22:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eac707b43ee4f2a9a0c6ab872e3dd59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1beadb15194fa8b3cf6a3364941b19 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93c5df0252614062ae12fd9f185614a9 : SUCCESS in 4m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/566bf579f39c4a049c34fcccb6d9cff7 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e866908a39ce4e80bdc4c066074c7239 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0967acd0ae474f5e9c6f8e9695ab02bd : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e5479ddce444776be69e976606e39ae : SUCCESS in 11m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4936c6fe867b4cd186ba4cc4886618e6 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44a832d6fd494babac9c0a146c12dcaa : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed2aeea64474f40a6ab42613c170eae : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e9b686a629b24e94901baefec991ebe8 : SUCCESS in 18m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0d7ee621eb9146cfa730bd1a7ec05c2f : SUCCESS in 14m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/877312755ec848c08f2ff3b727c41527 : SUCCESS in 18m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ed661e087e04e018554100e798ba0d7 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1af9f3c9b97b499e87d46535bbaded9d : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/910e0091422d420bb821033d5c03b916 : SUCCESS in 24m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e14aeb01c184fb3a82e0cfb06ce75c4 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/18abe4fb684e449ca0f7117c628c46a3 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f0a497bc240c4d7a9e495dd36f5e50cd : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/413c8176ce34470fa2b81b2f24ad7890 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/987c3a6b373b4eb48832634190ab3bb6 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7f2e4d5c15f44ce872af98a75c33fe0 : SUCCESS in 7m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3961e31e0d44abcb00369b606c48f53 : SUCCESS in 1h 06m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a995ab47717046a8942cc18b4af813f5 : SUCCESS in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bdbf26ab8e4367b2a9b3f6580e3755 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e30243f58fdc468ca76b9becf0efd862 : SUCCESS in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"070e772ddc6c11ed5ea81e001d3ec1cc42dd8aa8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-14 15:41:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05b869e9ee7f3e1142e0a619e351940e6fea327c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 16:11:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f42ef9cd5a85cd2a058b6fc50a7da7ea2b924c5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 16:11:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e2f2c7ff6c6113419ba0ebd98a63c31b0315120","revisions":{"3e2f2c7ff6c6113419ba0ebd98a63c31b0315120":{"kind":"REWORK","_number":1,"created":"2024-03-14 11:53:41.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/95/912995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/912995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/912995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/912995/1"}}},"commit":{"parents":[{"commit":"6706f79459d9fd2f054b57eb9a4c76736e9fac5b","subject":"Update .gitreview for feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6706f79459d9fd2f054b57eb9a4c76736e9fac5b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:53:40.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 11:53:40.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for feature/mpu","message":"Update TOX_CONSTRAINTS_FILE for feature/mpu\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as feature/mpu branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I1358bdc0c21d95fa879ee2fef58e72679f1068cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e2f2c7ff6c6113419ba0ebd98a63c31b0315120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e2f2c7ff6c6113419ba0ebd98a63c31b0315120"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
