)]}'
{"id":"openstack%2Fswift~913237","triplet_id":"openstack%2Fswift~feature%2Fmpu~I8d181d70b5bcdff0f8fc35e700789185d1a57fd2","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-14 19:02:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d181d70b5bcdff0f8fc35e700789185d1a57fd2","subject":"Merge \u0027remotes/origin/master\u0027 into feature/mpu","status":"MERGED","created":"2024-03-14 16:29:58.000000000","updated":"2024-03-14 19:05:08.000000000","submitted":"2024-03-14 19:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913237-merge-master","meta_rev_id":"83c49cebd04bf8d18cdffb20d3d112cbd76e8f48","_number":913237,"virtual_id_number":913237,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 19:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-14 17:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-14 17:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 16:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 17:06:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-03-14 17:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8be3803fb76d43b82386fe0871c1a1748e736ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-14 16:29:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2780893fad88f4ffa206aa1b8ee4a3f51b3ed7b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:50:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11edbcd92fd4f24908c1205b4d05b73\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6ce28ad4a424141b54d353dd561486b : SUCCESS in 16m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35f464e9c2bf4578957c0fc21a79e1ae : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebaf91c9c38f4c14b93321ee298e8248 : SUCCESS in 19m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/19800eede579462a98a1e5240b34f1da : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0f0526349f4e06ac82f5ba7ab66d1172deb3d196","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-14 17:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d98fece8f2abdefd1703388b7dd9817b7e971b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 17:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8399ac1bd634685a66caadbedd2f757\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc95f1c8e4e4d81a181230c5f78f07d : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/078bd26997e8421081f6ea8ebcc97910 : SUCCESS in 5m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e300737a61a4007b9efa8ec35e60769 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4bdda8cd407438a8cf31fd13c9f9c89 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41e8b19b88444267bee82085352e3902 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8e46a4f20f4418ba7d8167c4667519e : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33481de6c9e049aeadde586c336182cc : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d4eac62e1774929bedfe0a6ef8627c3 : SUCCESS in 20m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fda919195bd4cd8a758b609b51ab7c7 : SUCCESS in 17m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b45bd63d0003463faa023d2af7617a03 : SUCCESS in 19m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d29ecbe71ab84c1e96eee79cbc0c099c : SUCCESS in 26m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d94ab543d00c454cb78bec1db575e21a : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be0febad075348509c358959faffe085 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e60bd04c832e4362aa1e5c2eee5d3c1d : SUCCESS in 22m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/84c91b3ce0b347fd9723342da42be8e4 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/01ce586f07384f44ab1feb3d94539771 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9cfc7938ff5a40119030b1c7abdb7409 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b05cc2190c494a849754ef76017cd848 : SUCCESS in 9m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96d4dafbb5c7461da2766ad3f0cce939 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/534a87b105c243edbc6680647c46d7b5 : SUCCESS in 7m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3a99aed1e3c4ccea567ae9dcf626ed6 : SUCCESS in 1h 09m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b28c451e0974f87bb4f115ef5799c99 : SUCCESS in 24m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6a445769a34861949b46693b61103d : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69a4a09863624a9287e4913b85c493c1 : SUCCESS in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba861e5009edc34c580c67b4c98dd945aa2027ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 17:48:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b8d689de14dae9ae7ded7311c538db89726b74c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ca37f53982647e59d9a5a5d4dbddd18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfa72ec071c41a7b6f4b02b75ee7088 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80860fa307a14ced9ffd57bac2703570 : SUCCESS in 4m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5f03662aac0441caa955fb97c3895a68 : SUCCESS in 12m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b020884e48714f91bf9042f52f7a7dc4 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68bd696b95474f1cbcb135c3bb696ac7 : SUCCESS in 13m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd25eb8548040228f63ea43aff8c251 : SUCCESS in 28m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/590b58165d634ab8ad89f3a73bbb58a1 : SUCCESS in 17m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/00403ae307954f75818fd7c9216917a4 : SUCCESS in 22m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/943b2f6424524fe3afc7c86c5a3558a9 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0e374b1cfbd48c4acd71332dc8c7895 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/45560c088fd34cbeb5864e274b0d7387 : SUCCESS in 18m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c772052bf448468e9398020e586a5c45 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ffd58337e03421aa1ba1813ffd478c9 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d1cb509776641568fecb97603966441 : SUCCESS in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8cf6d719e0db4e269d641dba34dc670f : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfc4b3adc36648c8b71a1862472b6562 : SUCCESS in 7m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2b01a08c3764d918c2cb7c4e355a051 : SUCCESS in 1h 09m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a6aceb2cb574641bb38a50c547978fe : SUCCESS in 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2506468735f144719444b7a96b02c786 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d06f8ed527364f030b1395cee5880490ffd27cc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:02:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83c49cebd04bf8d18cdffb20d3d112cbd76e8f48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:05:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71bf1a961a444f69a6c8d776a1463461\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2e723eccd8493cb6ec2bae97f5df44 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5384f28ddb4b4cff88d3b149a8ed5cde : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/910d9ba567c04200a2ebc67320f77ec8 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ad650970b2af428d9a71eb3c3dd98b3e : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"774ab78455dd0856d3e0beb47c4b088512667dac","revisions":{"774ab78455dd0856d3e0beb47c4b088512667dac":{"kind":"REWORK","_number":1,"created":"2024-03-14 16:29:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/37/913237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/913237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/913237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/913237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/913237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/913237/1"}}},"commit":{"parents":[{"commit":"6706f79459d9fd2f054b57eb9a4c76736e9fac5b","subject":"Update .gitreview for feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6706f79459d9fd2f054b57eb9a4c76736e9fac5b"}]},{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-14 16:28:53.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-14 16:28:53.000000000","tz":0},"subject":"Merge \u0027remotes/origin/master\u0027 into feature/mpu","message":"Merge \u0027remotes/origin/master\u0027 into feature/mpu\n\nChange-Id: I8d181d70b5bcdff0f8fc35e700789185d1a57fd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/774ab78455dd0856d3e0beb47c4b088512667dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/774ab78455dd0856d3e0beb47c4b088512667dac"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
