)]}'
{"id":"openstack%2Fswift~913285","triplet_id":"openstack%2Fswift~master~I5ab762dfe0f85e346c4868ec4540884ba5f0a7f4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-03-20 03:48:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ab762dfe0f85e346c4868ec4540884ba5f0a7f4","subject":"tests: Update CORS geckodriver","status":"MERGED","created":"2024-03-15 03:28:59.000000000","updated":"2024-03-20 03:49:36.000000000","submitted":"2024-03-20 03:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"913285","meta_rev_id":"6dbfab56e48b8c509476880458ac81e64d994563","_number":913285,"virtual_id_number":913285,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-20 03:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-03-20 02:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-03-20 02:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 05:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-20 02:16:51.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"e6961096e56ff67dd52bbbd8bf2e29039b34115e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 03:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78a1ceca4288e0031b367e7c6ca20d74a5158ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 03:46:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ece95b2e1c1043517e631209ea391cd2a9e40325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:01:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a72ed9e129e4d48be7ff511a8a75cc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f186c35828d44a2b3dbc75ff0efa6d4 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5179a326a2a9416aa11dad2d503fbe7a : SUCCESS in 49m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/50a2ebbdcd0c4f8e8b9fe0e65c6ddcc9 : SUCCESS in 59m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2033113aaef643cda333f074b9bab94b : SUCCESS in 38m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d850efeddc04f82bf53fdb9bee689bc : SUCCESS in 58m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07465310840b475fa920edb748bb96be : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d92822a48f74767b35a74b4eb9ffc80 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e9e9194ce694a4fb48356b14aaa67a5 : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/136955327062414ca32bb8a2b0353a1f : SUCCESS in 11m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f424f2e75cd64877b1d581f3f1643f28 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef6ef4e7664943e9a4873a37539cf0d8 : SUCCESS in 19m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6878cbb83a3f4bb2a5d2239bc2c60d08 : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0ce6f18d73442d8bcec3575ec801ea : SUCCESS in 18m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb329afd97e845b59d5449359c79c43e : SUCCESS in 19m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bccb5db89c848ec8896974ab3b26860 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b187264b053434f92673c3d122c89f7 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1b420be1772749ed97324765a540fa5f : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a468aa21ff1e4c228d3eed7194b3bdd4 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/320ec6d8df814eee83500f8b9ea3a319 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/75e7ed6e2661461cb19cbb995d30f5bb : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/861f6cf3dc47419b8f9d051986227025 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/343be91c57b04b319048edd3d8399ada : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5f9a01ae0c8474292cf289d2eb4d920 : SUCCESS in 6m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa872a3598d843f0b25e60c7306a7f90 : SUCCESS in 1h 05m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48438d4fff9d483fafbd5c9a92db12f9 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee90397a4c0542d2b8a1693203d54437 : SUCCESS in 37m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00f5c59e6a7d4f97af918f5262abc45d : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9300dd8bb3b4aafb38b88b204ded464 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25235dc8ceb341dc820cc1aa15cb4b01 : SUCCESS in 19m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26f2bad168f54341b14df5f297cc87bd : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8e84709cd9c6d52d39566e3046141fd523e4cf66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 05:19:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d37825686214b76a969020c0e65cf0c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9252f12719d4f33bba6a07791fc454f : SUCCESS in 17m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/16b1f60cf26c4f1190c24571559b561f : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/508d1bf97fe04daba9637210681f3427 : SUCCESS in 22m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e200527fd2654559b900f8bb308d1122 : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4adee96f956156a67f998a7ba4103bc74bdb705b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:23:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2e13397a84d0382e1507a7e25169940b01e9a01e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-20 02:16:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b8f7ce0e1e60a3c360e7e8f39ce4f86644c862c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 02:17:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a71f3798ae3b4384c4cb9149bd60bfc2bed4cbae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 03:48:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e1c0dd8a20844afad1fbfe727b82757\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64c8d2e10e147438395669e60cb6256 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/64232cd575194061bfb4365e35b5f862 : SUCCESS in 1h 01m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/937c14421b4a43dcafa2232d76337d37 : SUCCESS in 49m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2c9182bb8ed43229e23569aee9f492e : SUCCESS in 47m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a52d90e36bda48278565eed586e3f9c6 : SUCCESS in 58m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8d7114ad21cb46c0b341ebf45820b77c : SUCCESS in 10m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0ef0c2d1ecc44112accedab3245518d3 : SUCCESS in 9m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8344fc458b5440ec9a5f5bd7f593c3e9 : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6bd3229aaa841d58118590e4a495865 : SUCCESS in 19m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78971549d6234d048bc0fa4c906724a0 : SUCCESS in 19m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5a21a916e1ae4afe9c5d2e17506b1e9b : SUCCESS in 19m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8b5b045acec4fcb8f28f31f34b385cb : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a76452e70284be9aab071c97ab99a18 : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70c27fc4ca3c44dba6d88ed1bb350e4a : SUCCESS in 19m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c52f7c28f2f04d10a4ff127f018dcc82 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f5796b7fc1a429281416529f0607bd9 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7429a7f9080b4418b09468093c7b8778 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d0dbe43aee74abaaaea031da546b6d3 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/360637144aec4db0b7c041f34ea7422f : SUCCESS in 8m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b0ce2eb219243f78030939e23fb1c31 : SUCCESS in 1h 08m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6861ba163aae481e8eaea9a39c77552e : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/862ae7b77aae4df688d9e5a16191c2e1 : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/355b8dca20124801853885e7f226a3ce : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d13b9e500434594afd5474f2e9de511 : SUCCESS in 4m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c3871c72a5c4c23a14131a4f508f8eb : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1174a2d4e594bd7799f03823082b6a0e047f4ba0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 03:48:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6dbfab56e48b8c509476880458ac81e64d994563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 03:49:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b7ae4429834c6b9a356c988f0a4499\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5380032389e543abaf8c9070fd173056 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/98713582a13a4d478fca4060460dc324 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6c536b6b2b564565bca2e2088c01c564 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af15ad53fb7818cdf14c92523fd6799bfabc12ec","revisions":{"d1696f55c2791692a23389e1b3e028e8141482b8":{"kind":"REWORK","_number":1,"created":"2024-03-15 03:28:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/913285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/913285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/913285/1"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:25:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:27:55.000000000","tz":-420},"subject":"tests: Update CORS geckodriver","message":"tests: Update CORS geckodriver\n\nChange-Id: I5ab762dfe0f85e346c4868ec4540884ba5f0a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1696f55c2791692a23389e1b3e028e8141482b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1696f55c2791692a23389e1b3e028e8141482b8"}]},"branch":"refs/heads/master"},"af15ad53fb7818cdf14c92523fd6799bfabc12ec":{"kind":"REWORK","_number":2,"created":"2024-03-15 03:46:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/913285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/913285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/913285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/913285/2"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:25:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 03:45:47.000000000","tz":-420},"subject":"tests: Update CORS geckodriver","message":"tests: Update CORS geckodriver\n\nChange-Id: I5ab762dfe0f85e346c4868ec4540884ba5f0a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af15ad53fb7818cdf14c92523fd6799bfabc12ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af15ad53fb7818cdf14c92523fd6799bfabc12ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
