)]}'
{"id":"openstack%2Fswift~913411","triplet_id":"openstack%2Fswift~master~Ief0771dcb17c4ed3d239d60872454e2d35936e39","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ief0771dcb17c4ed3d239d60872454e2d35936e39","subject":"CI: Add rolling-upgrade job for Caracal","status":"MERGED","created":"2024-03-15 15:15:12.000000000","updated":"2024-05-02 16:34:26.000000000","submitted":"2024-05-02 16:31:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"913411","meta_rev_id":"4d660e660297541e5aba84a2dd90c993bd61c351","_number":913411,"virtual_id_number":913411,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 16:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-02 15:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-02 15:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 15:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 16:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e6378281a490b7740112ccd18ee657929b0f3fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:15:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bfff53b9b80cd2491435c453703004e56ea5894","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:15:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8030acc3e688baaa0800ed3c79fff60897092d88","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 15:16:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca245810d1891d57b79f57e24af5663318366697","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 15:40:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ab81ac89f72495ea54a613c79895d07\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28767841714546a6815f6f90eb5747a7 : SUCCESS in 19m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b17b2228b16042eabbb1a62c9ab965d0 : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ced2121e1f784e1ebc573f28bbd543b3 : SUCCESS in 23m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/01cc62c0aee948e292c2c7fcb09ad0ea : SUCCESS in 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a4eb74ca00defcc9aa4c7849352ba573f6f55a68","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 16:20:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6f961c6ddb4bf6a6e9113372c2ba7b\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5d513320d2540ed81b6345d1946c436 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/442fde65137d4788abaee5a8a25e6d57 : SUCCESS in 16m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00cdeee645484ef298140bf268d5afb3 : SUCCESS in 18m 52s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/725acb63d32e41aeb94c85aaebace33a : SUCCESS in 22m 14s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9212c9a80d694c80b5b8fdb2931bf5e7 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8421e58b3fea44ecb29ae9ad93f413aa : SUCCESS in 16m 27s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0431238a96554912906f2f98510c7346 : SUCCESS in 17m 31s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/cb96644d6dd447e889b8254b76ff7cef : SUCCESS in 20m 17s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/b9988b83633e46f398b2c1817fd3ea7d : SUCCESS in 25m 19s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/da254b6ae09444829663c3ad216d0465 : SUCCESS in 21m 03s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/35ad289ac1e3420c9c9a700c0f02dc37 : SUCCESS in 18m 14s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/078bac31999c4796bf9e8f057397db87 : SUCCESS in 16m 55s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/182c9556cc3241a09352a3abcaf206e2 : SUCCESS in 31m 41s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/c4d22c754c6d435a9a1c0ce2de3de8b8 : SUCCESS in 18m 56s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/3e2e8a174cba415fb063db45ad48bc1f : SUCCESS in 19m 59s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/577c434c08e44ff49d89024ea0e4239d : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b11831778d77333f821ee6807d0a1880e1c36fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 16:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c1dc2144f4431988d16a3d0fd005a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c230a38a694d44708e52ab3b6bad19c8 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/c63f9bdd32874451b1f40b761fd30e00 : SUCCESS in 1h 01m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a6789feb34114caebd8cee5f8808863e : SUCCESS in 1h 01m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9513e2d1a7b24371a302667e36457fbf : SUCCESS in 1h 01m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db780a74bbfb4433b4386570738caba0 : SUCCESS in 1h 02m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/167e7a90117d4b23a4d67d048be4a2f4 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4ff298358ae4d2c818a677cd98040e7 : SUCCESS in 25m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8090f90827654e928c8613d59d780260 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3604ebd296ed4e36bbb8e00b73787585 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7dde976b6334bc595dee5c4a028835b : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4b3bd76ead64ff5861ec89e2fdd27a8 : SUCCESS in 20m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5638a5e1e4034ff995a47031096255b0 : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9248a5112eac4b309fe57cae5d083237 : SUCCESS in 17m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/563c18493753414ea604ed944654a894 : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64e846b78a4445e7a68ec7d11ff087e0 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d074f95e3ac421cb0b8d15da7f146fb : SUCCESS in 17m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca69e02e64c14f1688c3655e9ad8c544 : SUCCESS in 18m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2a06bf39b1c4fdb993e2143d8a3f7dd : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8645f3439ff14d70a998ce3e8c35b287 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d69baa95106428f960a6a4f2159c361 : SUCCESS in 19m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0377b6569204483299902c03b7016f97 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1765ac7bc39a47258dcd99e988415587 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecfed4e6a027462bae433054608cb34b : SUCCESS in 7m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4284888c8d24422843a985715e7e646 : SUCCESS in 1h 00m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5efea78ef3943e4aa8b5ec36ca7307c : SUCCESS in 58m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f3382cb45dc4704bad08100758071bf : SUCCESS in 51m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49503e39f9648739a5c872ad997a863 : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd522b4345248fdb47522b9097e436f : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87fa6771ca91496e9075d0b5cc878809 : SUCCESS in 17m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b78b9d7d0b8c4c4c93664b3a2ae61848 : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1513ccedc38401e03fb3b6aaecc8911ba7152d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 05:10:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f2a25a26d0a3c80b07cdf1e3f44853c461e5c9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 05:11:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cea1d64b51dd7256d14c9efead76c17bdbdf96f5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 05:52:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04745b42fbc543ab84508346e77e1acb\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e960b9aa05f42cd94e860d48cf5f153 : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/380013e83f514380b305adb83deb3c2e : SUCCESS in 20m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cba64179fa6a41c0b59cfa6135922165 : SUCCESS in 17m 10s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/f40808436a2e4904a3c828a1c6f7bb17 : SUCCESS in 23m 03s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/df024eb59dda41a98f9d0a0b1195edf8 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/12cdc898ae69414dad530c6e4bdffc76 : SUCCESS in 14m 16s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e60b86c5a064b218d800289e95ded40 : SUCCESS in 14m 24s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/3a2b80b168b6478f8d2599362835520f : SUCCESS in 16m 46s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/84998713718a48a8bc0d136e29e01f9a : SUCCESS in 24m 00s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/421b470df3024f909dd6fcb033a5a8ce : SUCCESS in 20m 44s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/53a61f0db1f84ff8b1443e6f98fb2760 : SUCCESS in 19m 18s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/3634dedbdebf4ec8bbc92030ad8f82c8 : SUCCESS in 17m 12s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/cd7b9dd5d1b2431c8fcdb4fd94e7aa92 : SUCCESS in 21m 21s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/44f3f2bb1344449cb010be10b061e107 : SUCCESS in 19m 23s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/c63c449891af4582b1ae5db96f78c785 : SUCCESS in 16m 08s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/d57681b21a304d258adad30ab4066069 : SUCCESS in 25m 03s","accounts_in_message":[],"_revision_number":2},{"id":"33d577f3bb0578b701be4b740e9bfe2af673a566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 06:22:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ab03a5c45443f1afa26201f613a74f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d68281987924de48ebeaf38eb0c1fa2 : SUCCESS in 18m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf933c4c4311422f8929b6674c8cce1c : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c130d42ed4844637a225676eb4c1fa0f : SUCCESS in 1h 09m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab26eb45fa404aeea80342fb0ebf1710 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":2},{"id":"025cfc22cb599ee7a8813297582d325dbd46f489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 06:25:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48c8a451b1574404810584003b096c22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e749b4599943408937a82e1e68b093 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/ce0d60cd6c5c45e5bd17745b9d091750 : SUCCESS in 44m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b8429ca4f5547238f74eac1c745554c : SUCCESS in 38m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2788f1630326486f8c6dbf5e7ed47207 : SUCCESS in 48m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3df88f438dc847a295ff4a3c208cd4c6 : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9991e964403e45c78545baa3c4c394d0 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29f899753d1a4f5ead426e373d1b04b5 : SUCCESS in 19m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13325c613f0541e08de32a97a1bc147c : SUCCESS in 10m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59f7340b45854edfac9e2cd80aa0da1e : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2aede1566b94d6da1467dab7007f965 : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f98415e865e4d86aeabd1294717af07 : SUCCESS in 18m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ea3cad94c82d492aa79ebb0346140198 : SUCCESS in 18m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f46fd3853a334568b3f5bd908e3d043a : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/736cb8ea0963426fae565dadff190a70 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c3db11a927a45cda81c8048ae567bb9 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/963b4ff7058142fda266c7444b1e9b89 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/517169d26edd4e20a2ab3c2237f2eabd : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e668baee758f408ea645d3291825e042 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09f59e8ad11348e79c64a0e7acd2c84c : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6edbb62601e4ba696431f93d608139a : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d3b97b62f004607990d5d6bd3afeddb : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c72735b41b1a4982acc5551f5e5729dc : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c89405b30fbe43369af8504373ac7950 : SUCCESS in 57m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05a09d007a2e411a84a5295125e2a23b : SUCCESS in 39m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b80bb4e3c904431b2f6ccefe0f0d31c : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33a2dbf9fa414d80bc6121c15b9f4c04 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ae538f21a24bf09d505e0dc61294f4 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f11349e95514821a61e1de3f4b76064 : SUCCESS in 19m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8476784a62ac47fbbc97a3beec3c9542 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8df2a3651d5e148fbcee6f0dc06830aaa9605cae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 15:06:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b95ded04e4385da0c8bd927a188e11a253c92421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:06:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b267ca9096e2fc9c92d9411178eb53fc31cdbf0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:31:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f504f6522d7c47948513436c46cd7ce8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8034dc00883e41598648249d5133bcd5 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/780033a6a68141adbde499eb1168cc9f : SUCCESS in 58m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5856adf62b0549928a3b0f1264ed1f44 : SUCCESS in 47m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfe74eaf7b454e9ea2fde9f03baa3b93 : SUCCESS in 1h 04m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ac2a5a0e941544d3911047e849059b82 : SUCCESS in 10m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ae0c814313a1430699e1222feb1c452e : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89534f6ca1c6490090c7349eb9169262 : SUCCESS in 19m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24f25f356b384e178fd0e93aeeef5eac : SUCCESS in 16m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d4d272abda4172b0f0e095d88f04b1 : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cf7ea5d5a94a4e8e9215cb4e8162437a : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fddcbf0aa24f46b6a41b303d6d44bab5 : SUCCESS in 21m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35f4735907d947f7acafc6a9b072c99f : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b19a73436ba4082af94af39c83212a2 : SUCCESS in 22m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16d8d1cc980e49d2bb29aea41ff66448 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/021603be86d54477bf7498e32262f812 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6cddc091abdb43a79082aa422fe1f261 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a772e1c1ffcf423586c60852a266d4d7 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/018346999c224687b434f65738af8ea8 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39268aa4a60e4719806c6ef8a0718303 : SUCCESS in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a2a1aae1de54f3ba241952b57728bfb : SUCCESS in 45m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f81cff977d74c2f824f1c1d4aa41186 : SUCCESS in 36m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cd5c63d60f4446bab354e3777d0d964 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef93cdcd11da444a90ee130feeec3efb : SUCCESS in 4m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc43c350d4a94d37808187a89ed14dcc : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5a2ba383141c5f67ef162cc1631ba31845febc8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:31:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4d660e660297541e5aba84a2dd90c993bd61c351","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:34:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff7f50635e54d28bf45e408d82f64ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6962e993e10b47c4a20134dbff092b8e : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3ad8fdd94a1e4b43969dac8af4a10d63 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1c156fb2de8c4179a1d329ceb20f7482 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3eab3facbb3229afb62cb722714f1fef1833b46","revisions":{"a28fb91be84c654165059641600ed29b61620552":{"kind":"REWORK","_number":1,"created":"2024-03-15 15:15:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/913411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/913411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/913411/1"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 15:15:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 15:15:04.000000000","tz":-420},"subject":"CI: Add rolling-upgrade job for Caracal","message":"CI: Add rolling-upgrade job for Caracal\n\nChange-Id: Ief0771dcb17c4ed3d239d60872454e2d35936e39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a28fb91be84c654165059641600ed29b61620552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a28fb91be84c654165059641600ed29b61620552"}]},"branch":"refs/heads/master"},"e3eab3facbb3229afb62cb722714f1fef1833b46":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-02 05:10:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/913411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/913411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/913411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/913411/2"}}},"commit":{"parents":[{"commit":"b4dddb740639401f8166ee167b4173fc51ed03a0","subject":"tests: Use @with_tempdir more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4dddb740639401f8166ee167b4173fc51ed03a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 15:15:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-02 05:10:38.000000000","tz":-420},"subject":"CI: Add rolling-upgrade job for Caracal","message":"CI: Add rolling-upgrade job for Caracal\n\nChange-Id: Ief0771dcb17c4ed3d239d60872454e2d35936e39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3eab3facbb3229afb62cb722714f1fef1833b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3eab3facbb3229afb62cb722714f1fef1833b46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
