)]}'
{"id":"openstack%2Fswift~913425","triplet_id":"openstack%2Fswift~master~I0509adccdbb7c416382556c642113cb32fc82bb9","project":"openstack/swift","branch":"master","topic":"p-coop-token-alt","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-23 05:05:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0509adccdbb7c416382556c642113cb32fc82bb9","subject":"WIP proxy: use cooperative tokens - alternative","status":"ABANDONED","created":"2024-03-15 16:34:56.000000000","updated":"2025-09-23 05:05:51.000000000","total_comment_count":6,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"933df9add43febc4616a3c013701ae66f0a60c35","_number":913425,"virtual_id_number":913425,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-15 17:48:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2024-03-15 16:46:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 17:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-18 05:23:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"923429b2d3f47df8da233c97d72eed128ca358ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-15 16:34:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29d30db8b54ea2d8abbb609a60abadc61a309764","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-15 16:38:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"429f632a720fe4a4ac2a28e2c3b8186407f84357","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-15 16:46:03.000000000","message":"Patch Set 1: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6185ddf789e97611e316549a0fc033033f4ee2d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 17:48:42.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9100e3a41d6b421aabfd63aacd9c31b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8c75453f15422ca68f056bd73507fb : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/68e33edc2db4467789a890c672a52f05 : SUCCESS in 56m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/27807935f6024bceb71812a1b7a1c4a7 : SUCCESS in 1h 07m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c4addb5dae8b4c31bbe20fe90aa9c4b0 : FAILURE in 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/687392db466848939390fa35f5f9edb1 : POST_FAILURE in 26m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/964ed172378348cea9b9e90b3b9e0043 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef51211dbde94c2c9b621592b139bf09 : SUCCESS in 22m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa615f9871d64a86be49881601471e40 : SUCCESS in 23m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d04d1c84d4548f6a9464ef978bfb6af : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d2ecd49a09e40f5b4c5a314f11eba89 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6211cb69386841e08dc902fd133f3fa1 : FAILURE in 21m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b19e39b41e0942ab926b74a9cb195fac : FAILURE in 18m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6836b08361b46c28e811c2121362417 : FAILURE in 18m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/94b67ba925364af6b2384e121162935f : FAILURE in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e11b4136e4a47338912bb9a47bbc2bf : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac4bdf82549f43858889d6917b68538f : SUCCESS in 20m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/213e0517a2e14835bc2827ff9805a5ea : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/70c279ab131348608bd4eb9302e169e9 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8e1ed57f4d8940fca5596ce9be542a54 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99d802cfa00d481fb3bd8399a7358bfe : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19a09531e3ac4778a39070caab8da384 : RETRY_LIMIT in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8498c85d3083426988565dc7f668feef : RETRY_LIMIT in 4m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee014390b6e44da590ce6843b67a8250 : RETRY_LIMIT in 5m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f016eda14f64e1890f14543b7f4c104 : RETRY_LIMIT in 6m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/090ccdaffb4e41fa84a976542743e269 : RETRY_LIMIT in 12m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59e7285543d945709fdfc3c4cc605857 : RETRY_LIMIT in 11m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ec8481b912b49668b4b88bf54a3a285 : FAILURE in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d26c1692ab84746ba2cab635c23152b : FAILURE in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2532d94e69445698db890eb0b4c9c2f : SUCCESS in 20m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4390d3c364ce47b99fccbdc7f9894282 : FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":1},{"id":"626d181bdcc7b1321c0c04a7a8dba00e762c0f0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 18:07:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d71e01d355d4c93adfe221f4ca99790\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e31ac8fcf03a469da766a5da511ea3d1 : FAILURE in 1h 02m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d24dfceed0a43a380e11e2d8107468a : FAILURE in 1h 03m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/60d176e7789c419c8ea9cdb6cc70ae48 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cdc77ec6c5945f6b6a341e0e47fd998 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7d6ef2fe4fc827a3c681dd73e73afc63a9b6233c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-03-18 05:23:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"933df9add43febc4616a3c013701ae66f0a60c35","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-09-23 05:05:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db4b830044aa7980a889af58bbb8cb03e1b55026","revisions":{"db4b830044aa7980a889af58bbb8cb03e1b55026":{"kind":"REWORK","_number":1,"created":"2024-03-15 16:34:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/913425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/913425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/913425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/913425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/913425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/913425/1"}}},"commit":{"parents":[{"commit":"cd5ae367d72004f62f552a44076b040bf2219967","subject":"common: add memcached based cooperative token mechanism.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd5ae367d72004f62f552a44076b040bf2219967"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-14 05:12:22.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-15 15:53:55.000000000","tz":0},"subject":"WIP proxy: use cooperative tokens - alternative","message":"WIP proxy: use cooperative tokens - alternative\n\nChange-Id: I0509adccdbb7c416382556c642113cb32fc82bb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db4b830044aa7980a889af58bbb8cb03e1b55026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db4b830044aa7980a889af58bbb8cb03e1b55026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
