)]}'
{"id":"openstack%2Fswift~913452","triplet_id":"openstack%2Fswift~master~Ibcd7ff6f184f935819b54f6062f04244a8c0075e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2026-03-03 18:05:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibcd7ff6f184f935819b54f6062f04244a8c0075e","subject":"Move statsd client into it\u0027s own module - testing part","status":"ABANDONED","created":"2024-03-15 22:16:16.000000000","updated":"2026-03-03 18:05:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"303f0d00735e1a5d50209876be8714140ba86d42","_number":913452,"virtual_id_number":913452,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-15 23:27:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-15 22:55:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-15 23:01:50.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-03-15 23:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5daa9501e0bf0558ac5f7bbfa363b21c2f3ab0b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-15 22:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"879117d33cd4bd03a092c56e249a530d31b9d431","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 22:55:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f39341a4eef48ce9369b9782032287a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58c724466ef441acb8c157778f02f789 : FAILURE in 16m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/babcdf8179654bb387f5ecaf5237a739 : FAILURE in 18m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e985f03a20646769955f808e6a5f303 : SUCCESS in 37m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dba583c25bf4b21bd8756af9d13066a : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8cf8ea351c02a5da33029f7dbaeb685ffa7904d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 23:27:19.000000000","message":"Patch Set 1: Verified-1\n\n(31 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/158862c857124decba15082a0e1b8fc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f3e3c21b0c4f7f9373c0bf311b4ee8 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/c0ca0fea09db44f4bcd225103579f414 : SUCCESS in 1h 04m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ff6f156e908f4b2b9cd229c90fbc5bb0 : SUCCESS in 1h 05m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a099c6aef994d789277e0d88861cd59 : FAILURE in 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4160021ee0cf43a6851fbea252490249 : POST_FAILURE in 17m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6462f05074a4c1e97e9023fa69a3de8 : SUCCESS in 24m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0953aa4d7fb49148131dc4ca3c95664 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83064e0895e545cd9a136bd9c4107b52 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/343909906f934fbf8a3fb2730469734f : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c18dcd7bbe4b45ba884a5f784d0cf8f2 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bcafece8d093444caf96118b51ecd9d4 : FAILURE in 5m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73b06fb912ca45ec8c0663f6ad13cfe2 : FAILURE in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/61204b634922401293bc3dca90640e94 : FAILURE in 18m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8a7af40d2fc14c78a809e88b8804d5ac : FAILURE in 19m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c71c71eb7d6945f98ae7677aa77350bd : FAILURE in 3m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f73d8073f3cc4541b2433431bc804f99 : FAILURE in 4m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca6826a11adb4b128ba1889fb697d5a9 : FAILURE in 3m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16b78f9c1d7a4ec59b96976667736064 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/99c3828d2f7843159ebd4595dac7c63a : SUCCESS in 23m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2567ab9b3a7443ff8debb8d756e37c99 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70f414ee0b3e446b931b386fcaa235b7 : RETRY_LIMIT in 4m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db686b2a3f334633806be7dacd43cfe1 : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f4b7b4cd90846a7898f823a714631cd : RETRY_LIMIT in 4m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6afadf83008f43789445923827cc167f : RETRY_LIMIT in 3m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8650e3171a04c6e8e4dd60297a043c6 : RETRY_LIMIT in 19m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf9d6d1dfe43445fa2f4a9842e30fef8 : RETRY_LIMIT in 8m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e30995c465647a1ac964f2003c0a1d9 : FAILURE in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d16e443bf15d4382beac8f7bc700d851 : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c6a69b4d151458780ddab7894a30f6f : SUCCESS in 20m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d18fcb321554ca68a372037a708ec06 : FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":1},{"id":"303f0d00735e1a5d50209876be8714140ba86d42","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-03 18:05:01.000000000","message":"Abandoned\n\nthis looks stale","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e801264cc376ed41e67f3ba4689ecd3534687213","revisions":{"e801264cc376ed41e67f3ba4689ecd3534687213":{"kind":"REWORK","_number":1,"created":"2024-03-15 22:16:16.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/52/913452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/913452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/913452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/913452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/913452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/913452/1"}}},"commit":{"parents":[{"commit":"627448362abc1f83d9385165de232f47545f18e7","subject":"Merge \"CI: Remove centos-7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/627448362abc1f83d9385165de232f47545f18e7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:24:25.000000000","tz":-480},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-15 22:12:30.000000000","tz":-420},"subject":"Move statsd client into it\u0027s own module - testing part","message":"Move statsd client into it\u0027s own module - testing part\n\nChange-Id: I41b9b228f76ab06dd70f63cc2c09497efe373b48\n\npushing changes\n\nChange-Id: Iee84d3b1c149517708f0e8a1b85eed32ec18e5ed\n\nchecking for 2 failing tests\n\nChange-Id: I949d1093e0e9ed0bf910129f874e32858503e06f\n\npushing changes with one failing test\n\nChange-Id: Ibcd7ff6f184f935819b54f6062f04244a8c0075e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e801264cc376ed41e67f3ba4689ecd3534687213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e801264cc376ed41e67f3ba4689ecd3534687213"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
