)]}'
{"id":"openstack%2Fswift~913723","triplet_id":"openstack%2Fswift~master~I34adb8141cc9e62d17a27f01c63f40d1dd25991c","project":"openstack/swift","branch":"master","topic":"s3api-unicode","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-03-25 21:40:58.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-25 21:40:58.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-25 20:06:18.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-03-25 21:40:58.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-03-25 21:40:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34adb8141cc9e62d17a27f01c63f40d1dd25991c","subject":"s3api: Fix handling of non-ascii access keys","status":"MERGED","created":"2024-03-19 22:38:15.000000000","updated":"2024-03-26 19:51:42.000000000","submitted":"2024-03-25 21:40:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"913723-s3api-unicode","meta_rev_id":"26761bd7e1d5e6948fef7523b026a86a702b1763","_number":913723,"virtual_id_number":913723,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-25 21:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-03-25 20:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-03-25 20:06:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-19 23:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-19 23:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-20 00:22:03.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2024-03-20 02:13:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-03-20 14:00:55.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-03-20 14:43:54.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-03-25 20:06:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"1a0a85eaf4465fd6e1f169621e982077f06e4b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-19 22:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40bd11dd02b527e230ddd706fe1715fe03da5f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 23:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bab2f3e89e94d5697ff375f2fa3fdf4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63f660a1cf65462487bf7b40eff04889 : SUCCESS in 15m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc6945d6d90c409bbaeae48057862e7b : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b37f6521768f4c51b770f48b842fcb74 : SUCCESS in 18m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1dca2983977c4837be7c07e833aec56f : SUCCESS in 22m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d08c5340b4e6dc4a53a17a74581f8270c2ef7ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 23:50:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699cd35f99454274bee43c1aeeecefa5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d265c97d894e1fa7eab8bac9c68c54 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0ef7db1a14c24033b5b90f5eca11ce10 : SUCCESS in 1h 00m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca58b773085448fe943e83939dfef979 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e926c62c8479421ba25a0a4a122372b8 : SUCCESS in 35m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d450ce453b74492d937a550742d34bd5 : SUCCESS in 51m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/329e239910194e8aa0f5456b1c1e946c : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8de82ffa8328442f96b21622ced5e495 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df6d589af1684dacba14098f76e9f4eb : SUCCESS in 27m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26f1d89c412c4cb495298d622b191682 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e59f823f881f4623a3009d52ae28fa6d : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3befa5ec182940dd9f6e2c9cacf839e8 : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90fdb14c142b4fdead5a06125e28bc71 : SUCCESS in 23m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f16d2b17dd124d6f81115d539b54f55c : SUCCESS in 16m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/94f52e0bffbf4674a30cbec01b56ae57 : SUCCESS in 24m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a7ef0eb5b35450ca011ccc170367f2f : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f8d4acae20e47138df3f869c34784df : SUCCESS in 20m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1d49771b6564251a8e89cbfeac780c7 : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e9cfc263c79457ea702da11369002d8 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/650d1ff0ae01481bbebb7ac7e9499b10 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6fd340d6866d413698ad9e387c33cbc4 : SUCCESS in 23m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5d1e809fed4490590a590946c651d6d : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/60c3b53d7c744bf58d7b07c512614867 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d84a9bae3854a26bceec07053d5203b : SUCCESS in 8m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3eff5722710432e8ca03df84bbb1f64 : SUCCESS in 56m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3dd9311e124d4a3b990ecdf42b0a62cc : SUCCESS in 54m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffb447c15fa94a12a0b53855e0d213e2 : SUCCESS in 57m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9de0622226764b2c86e1a47782275eb4 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2091018d55c42c09697aa71c3048320 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/426bc8a540554c3e9d96f64174e41c78 : SUCCESS in 20m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c01c15496a764d2ab7c180e404fd4d3b : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7969d2c3b7aad34ee42f05b63e2b96fb463ab341","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-20 00:22:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5ca95d56b7d7feb3e1a6c307422985fec0b0db9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-20 02:13:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03df37ab64408d0fce13247412aafbe8ef2ad3dc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-20 14:00:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbd92cb7805e7c0d0e7a9ccd08205abe68c01432","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-20 14:31:12.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"da50e150e2b8ad72521c9bb788b3f62a7759b27b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 14:43:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9975a5680a3683e86144013887ebdbcd95085e7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-20 16:11:29.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"af1a095ec4c98c8212458be38e10637f0f558690","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-20 16:22:16.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2313569bed17628a0006f1fdcbda340c73ef7f89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-22 14:07:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce176b22207d1a49d44245cdcc9dfecc0782cbe0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-22 14:15:34.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b7ed4541f714f64c80ef6aaf66d7fc3ef0140f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:23:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/142b1db52f9843e19f47e8cc53a35d9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0508679f191495585b58f02255d0a32 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/ad845f50b0dc4a09aef39d7123bd148b : SUCCESS in 56m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2648a5eac15440f69959a4739cbb6ff8 : SUCCESS in 55m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/227843b3492d4c2c816672299aa840ba : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c3a2cd159c48f2948ec1c44c5ba087 : SUCCESS in 1h 06m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdbf2a61f75f42d7972754f24141ea7b : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f38b9b793ebc4a9a94e0e7a8155dd53a : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56024d6fd0cc4b089019c3dc8e201ef3 : SUCCESS in 25m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c228b4aa43cd4971a33723de5bd4e802 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6bf4ee2aceb94749a34d43064146e554 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab5fed69876b46fb8f04c59990a14e76 : SUCCESS in 24m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df78f562732544aaadaf6eda0bd62d18 : SUCCESS in 23m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3aad0397008d46ca963efc1cf763fb49 : SUCCESS in 21m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ba2afc099dbb4530a671a275b3133b3d : SUCCESS in 22m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6b04f3b27384ad9babb4054d1bd02ec : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af2896523abb4d429b57147c45f3b891 : SUCCESS in 25m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e057a482795c42978a692c72a2f0c840 : SUCCESS in 22m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44f04feb98094617a8cbba93a771169c : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8d49f7eeddfb4751839507e181da6470 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eeecd5528ae344ac962da885fa809563 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/122d1767c60849339af9cf35f669d44a : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ef98fe61348462fa606236730ac5963 : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f80b99875e234135aa417ed7f0bbfd4b : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4e817df942b4ced82d57d0df4ae6a20 : SUCCESS in 1h 10m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5795df8288ee4bb3bed2fa79731e6e75 : SUCCESS in 54m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a900a1ee27a84f388027b4d606c5fb99 : SUCCESS in 48m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2eb4a94de3dd48dfb8b69c97a37e96f1 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc061bd765c34c488d2dd45ae30697f2 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffaf12cde56744ac9d5e20100b9a5e7c : FAILURE in 47m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad1b000c32b64417b5f869ba37c84b43 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9960380072e143659552c506ae1ae9b638e04937","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:40:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0586b4cd08e4b0d8b37884b7b42060f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22789bee4bcb4e97bee072aeda2d9ba3 : SUCCESS in 20m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4566ba5dfc3641a9b19fa3debcadb35c : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcf8bb19fc984473be122ce12d0a42dd : SUCCESS in 1h 30m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e51b879cecb04f7d9bcf8bbd806c587b : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9026504a0b59aa7689ee605f398037450d149b2e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 15:58:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"327166c63e17fec127a11d58b953fd81a5f2ca00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 17:02:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0ec546210c7f3ae04274aca244efb7b2db6a5b15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:54:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b881ebee50d4af7969550d4a8ea4c6d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9f72b4538464e9b847becce69f3d792 : SUCCESS in 16m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e1689c08a8249dca44a204f601ab507 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/157ba216d4d642858cc0d8bc1a78daae : SUCCESS in 50m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2e213a4c16e4b5f84ca5452d7680850 : SUCCESS in 22m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d5050b800c6e020f8f72d2712468cc5b783c5818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:13:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61cc1e5518b4a5494d990633574d6d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfcabfb1a11a43a5b2f7fdc8e8b75f10 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/7d60aee088ca474daab55ecb0717e5eb : SUCCESS in 1h 02m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e39fd28aaed4fb7ae53fc18ab6f4b8a : SUCCESS in 48m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16f62f22c8dd425b8c431db6ad841949 : SUCCESS in 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c9dccbc8654b7fb0b8f33cb9068c2b : SUCCESS in 58m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4936b948c2b1482f87bfa6a81f3f1b9e : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5655f54af1d943a8a0f8454df48fa259 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/043b187b7a5e4ebfb893797ef822597d : SUCCESS in 18m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7eaa10abfc614fd7996eff6bfd6f5b2b : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7864f2d645d241959edc0bc5a4e9c516 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e158773cb0b6453c80f58a9ec7e82ce6 : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58f279881cb44b3d953394fa7d1b2976 : SUCCESS in 13m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb76cfd3ad7f458a95341a90ff35f047 : SUCCESS in 13m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6f3f738e573e4b7cb6cc5eca7c1016bc : SUCCESS in 23m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3735c1149ea4265846bef4fe7244fad : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33da0a641eff4d6bbc52ce0046664400 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8b52759520b40758c67d741cbf0c0a0 : SUCCESS in 19m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3da6c1cdac83409882c0bf22a3d84892 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8e589ab12ae54c2bbe2c0c46a110c938 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80342b2941c74008b870135a5c54ab9d : SUCCESS in 13m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ef7e554df5f4ebc8e60739830142965 : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a7a252b52ec4091b976ca429f4638ba : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5609fc7f9b284a88912874eff80837ac : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f90db35967234aabbe0782b2e68b97f4 : SUCCESS in 45m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebcac9310a1f4bedab6be75c80fd5bb7 : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07bda0bff6f7488bb4943bc2bdc55047 : SUCCESS in 56m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/830bf3781de1467d95480826cae32ef1 : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9586601927e48ac931774f4986622fa : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/670f09f392514966b18acdf222fdc5bd : SUCCESS in 18m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3a7708c42cc4b28945c89a0ba2315d7 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":3},{"id":"908d40308d527a742933dedf0b1ace4760749320","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-25 20:06:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b9040ec58df9ebb4e99cb09ea2825573697e8dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 20:06:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0a18cf4504bfc220f12425a9e3903c55dc3f2311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 21:40:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173b26e6995849d396b4461f9e802881\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03223bde0bf94504a7d0f065f9f18fd4 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/f6c945f9a14c4f658014f9d247f925d9 : SUCCESS in 54m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/62555feca4644ba7bee2c41d029af65b : SUCCESS in 47m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edf3fa0b4b11402490228cdd5e936626 : SUCCESS in 55m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bedb7fa5ff2e457597c6332ce6637293 : SUCCESS in 44m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ae8c2f4ca356477d8d429a47f58a0eb5 : SUCCESS in 12m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9f224fdf84ba46e88dc8350803626dd0 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/352fbfc5d36f422082bd67366c4f750d : SUCCESS in 23m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c088d5f29c4652aac979177495a9c5 : SUCCESS in 21m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/80c7844cfce343709b141cfa43850750 : SUCCESS in 17m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b0e46067777045469609d30928e67b4e : SUCCESS in 24m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77d3a06039024f05902c5c3558cedbf5 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70772a6dce5d4f8096c1e7186b803ad9 : SUCCESS in 24m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27f234bdcefb41cb8783a158326b9511 : SUCCESS in 19m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/82540830d67740608985fb3a65e4b055 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75beef571a024fcc9b06de326d7f6d66 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/749a2fed699b4303b18b1ab668c45843 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5f0879106c7400585ffb629e11c8e15 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67ecb64cb71e4906a34fd2114cf8d5b6 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3b57357b7e64826b7c22e4bc6e9d71e : SUCCESS in 56m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db9fe98b8afa4e5da0506abbfcf076e8 : SUCCESS in 52m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46b7b64049c641e88deef6ef2c7f1ed5 : SUCCESS in 46m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28102e887a164df091beb4ddc48052b7 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759ea2d4503b4c1aa4d8365bafb16b71 : SUCCESS in 3m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f573b63fe6e2493997fe0bd1048826bc : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8f5d003f5add369b83b6bc709d9f0f8fa7002d5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 21:40:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4c6e4b1a9da985f07bbdb81033321f9a5fa1b69a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 21:42:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfb7ae8e975437c8e7621c18aaf1542\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0411ab735a584a4bb6fa8b04c229c8ab : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bc33a3b945dc4928983543270b9940cb : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ffebc3143da341d3b071d6b4074ee6d4 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"26761bd7e1d5e6948fef7523b026a86a702b1763","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-26 19:51:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8424b02290c75a7e1eb2e36296b41926f041249a","revisions":{"b37c858f5f8011145c85ff185483024e68a89715":{"kind":"REWORK","_number":1,"created":"2024-03-19 22:38:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/913723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/913723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/913723/1"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-19 22:36:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-19 22:37:08.000000000","tz":-420},"subject":"s3api: Fix handling of non-ascii access keys","message":"s3api: Fix handling of non-ascii access keys\n\nWe stuff the access key into the request path until we get back a\nmore-authoritative account name from auth. But it needs to be a WSGI\nstring when we do!\n\nChange-Id: I34adb8141cc9e62d17a27f01c63f40d1dd25991c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37c858f5f8011145c85ff185483024e68a89715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37c858f5f8011145c85ff185483024e68a89715"}]},"branch":"refs/heads/master"},"fdaabf9a3d77a2772f48944555bb6f2d3bb0bd88":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-03-22 14:07:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/23/913723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/913723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/913723/2"}}},"commit":{"parents":[{"commit":"fd3997f027eeadc9a7783a1cc26372e28e644ff8","subject":"Merge \"tests: Update CORS geckodriver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd3997f027eeadc9a7783a1cc26372e28e644ff8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-19 22:36:26.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-22 13:57:25.000000000","tz":-300},"subject":"s3api: Fix handling of non-ascii access keys","message":"s3api: Fix handling of non-ascii access keys\n\nWe stuff the access key into the request path until we get back a\nmore-authoritative account name from auth. But it needs to be a WSGI\nstring when we do!\n\nCloses-Bug: #2058748\nChange-Id: I34adb8141cc9e62d17a27f01c63f40d1dd25991c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdaabf9a3d77a2772f48944555bb6f2d3bb0bd88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdaabf9a3d77a2772f48944555bb6f2d3bb0bd88"}]},"branch":"refs/heads/master"},"8424b02290c75a7e1eb2e36296b41926f041249a":{"kind":"REWORK","_number":3,"created":"2024-03-22 17:02:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/913723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/913723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/913723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/913723/3"}}},"commit":{"parents":[{"commit":"537b6885e37cb1830992f8586400fe6f8d7f2345","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/537b6885e37cb1830992f8586400fe6f8d7f2345"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-19 22:36:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 17:02:39.000000000","tz":-420},"subject":"s3api: Fix handling of non-ascii access keys","message":"s3api: Fix handling of non-ascii access keys\n\nWe stuff the access key into the request path until we get back a\nmore-authoritative account name from auth. But it needs to be a WSGI\nstring when we do!\n\nCloses-Bug: #2058748\nChange-Id: I34adb8141cc9e62d17a27f01c63f40d1dd25991c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8424b02290c75a7e1eb2e36296b41926f041249a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8424b02290c75a7e1eb2e36296b41926f041249a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
