)]}'
{"id":"openstack%2Fswift~913756","triplet_id":"openstack%2Fswift~feature%2Fmpu~I3f78e9bfaf2eadc77114ec3c21c7f15fc3ff99b9","project":"openstack/swift","branch":"feature/mpu","topic":"p-rename","hashtags":[],"change_id":"I3f78e9bfaf2eadc77114ec3c21c7f15fc3ff99b9","subject":"rename probe test_mixed_policy_mpu.py to test_mpu.py","status":"MERGED","created":"2024-03-20 11:52:47.000000000","updated":"2024-03-20 15:44:21.000000000","submitted":"2024-03-20 15:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913756-p-rename","meta_rev_id":"670f13f609454670d54d9d935150de3fa38ab358","_number":913756,"virtual_id_number":913756,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-20 15:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-03-20 14:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-03-20 14:16:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 13:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 13:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6805d413e25c0204acccbb2b05d69fe91b8afa17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 11:52:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"946eb7f364fdfbda641e44abb8709e47a3828b73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 13:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb30b27eb36945fcb09d5d73c2632029\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98c994ce805c41cdab573034a5bfb136 : SUCCESS in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3f4915eee1e4955b178b98235fcf8b1 : SUCCESS in 1h 17m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/613f21050bdb43fc842c2de0aff15f5f : SUCCESS in 16m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/742d543245eb4467a8caaf0f24f02ecd : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d3441994d244fbf99fac42af32feae34c4b0696e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 13:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ef663cab8bc413c8788d7d3d0ec426a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082d551643f646f6a280c4c32b3bfe7f : SUCCESS in 8m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/460abe7223d14e2aae42d7c34b7b26d9 : SUCCESS in 1h 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19dbcb5d98ee46b0958b1d0a8ec56a29 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6b47d751f829fbf1bb5009b56d255d4bd031fbc9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 14:16:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a165a40c3cc34ec2f1d18e97f49705028fd9c1ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 14:16:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2fb6871aabceef6c76cd936e4ec39050cb53e6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:43:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d30f382990f47bd9d1855adc471b875\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757e9c4738d54abcb97207eba6d26114 : SUCCESS in 8m 17s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b915e098c6047ad8f7099f15ea53e27 : SUCCESS in 29m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95adde8bf7c541d1b48684b5f8c56502 : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60190ac35f884c46b5a17959d4c69852 : SUCCESS in 15m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d841122ece5c46d68d555ae4a2e5dd53 : SUCCESS in 29m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63a2b233f8b14479972aad516921e123 : SUCCESS in 21m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b6c5b4fa4fa4ce1b8ff1c8d46792b90 : SUCCESS in 29m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aaf6d8d7b78e426aabc1d9f9c41ea653 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d3f0beb12c0143d6966ee732d067fc53 : SUCCESS in 27m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/392c17d868a7457fa2dab430009af2dd : SUCCESS in 17m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/69eeffc1853f4d01841cdd552e99ff1c : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e64aaff159141c4b345a1478dba6cc7 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/effa4a8dec2e4ff4a77bc7c93c230456 : SUCCESS in 7m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fec52a637d5d4fed8596f4efc459839a : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70106ba575e4ecb97403c7d94f75145 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7b5cce867f5744423ad3ec834995c27efdf86c1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"670f13f609454670d54d9d935150de3fa38ab358","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 15:44:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/884673eabd8946948461153d62769b43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360e99bf35154edfb225a2e3917b3f03 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6b89fb2e720c4fe79ceb9504aee37c8b : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/598b34d7af80493995deb4d36890a971 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec9bf688a552db6455e78f3347524f8dc91fd072","revisions":{"ec9bf688a552db6455e78f3347524f8dc91fd072":{"kind":"REWORK","_number":1,"created":"2024-03-20 11:52:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/913756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/913756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/913756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/913756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/913756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/913756/1"}}},"commit":{"parents":[{"commit":"256938544919924323e77a962bfc8830fa98ad89","subject":"Add doc re working with feature branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/256938544919924323e77a962bfc8830fa98ad89"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 11:51:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 11:51:59.000000000","tz":0},"subject":"rename probe test_mixed_policy_mpu.py to test_mpu.py","message":"rename probe test_mixed_policy_mpu.py to test_mpu.py\n\nChange-Id: I3f78e9bfaf2eadc77114ec3c21c7f15fc3ff99b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec9bf688a552db6455e78f3347524f8dc91fd072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec9bf688a552db6455e78f3347524f8dc91fd072"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
