)]}'
{"id":"openstack%2Fswift~913831","triplet_id":"openstack%2Fswift~feature%2Fmpu~I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-22 08:07:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-04-22 09:16:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7","subject":"object-server: include deleted metadata in PUT/DELETE response","status":"MERGED","created":"2024-03-20 18:39:53.000000000","updated":"2024-04-22 09:17:06.000000000","submitted":"2024-04-22 09:16:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"913831-p-mpu-mw","meta_rev_id":"aa49f2dd5a57a0e02f6c7aac69be6b66152efe6b","_number":913831,"virtual_id_number":913831,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-22 09:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-22 08:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-19 14:57:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-22 08:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 19:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 19:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 20:40:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"6a9d36f019a9ea77e7c7b43f3cfc0aa6f3ebe72c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 18:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae394ec1c664100a24ae18890cddd8fb41d7d15c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:03:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fcd5b2fbc34db4a9d1745b207de3d1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2159a50ba2f44ad1a457c54bda7b2ed0 : SUCCESS in 18m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ce5b073afaf498b9f8f278f32dee720 : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6517ea417d9549208f234f2afab4c933 : SUCCESS in 21m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e474b2d1ea9a4692a0072d62bcd26456 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5eafdc44ebac1df29e52c7e67c126d56db93b6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:41:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dfec309bddd4dc7994db55a39a085ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4eb8e3d4f9486683e9a1dcb125ae8c : SUCCESS in 8m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5768dff16387404c84bd81e1f271d786 : SUCCESS in 22m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b8e542f8b5d49fb922274d0562acf4e : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/605e95c99066493d8720ed1a73225e70 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d20f70c759d9446b83fda65687aa5030 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/683b37cd7ee14facb7bcc5d8543673d5 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e1964647f8f43189f063d2f5b0b1000 : FAILURE in 4m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd106beba7a7432cbbaac75252a85212 : SUCCESS in 22m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1e5294038f4179a0f58e1251c2c572 : SUCCESS in 18m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4a3df4f64a4d41698f62bc6891441e7d : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/396b6496a0284c5e9466789ae694a598 : SUCCESS in 22m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/372a3f10d9ee4994a7d4b53c695613fc : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2aeadd557aec405aaf7c7bc3ff5e2836 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e5971697cc24b9d984c9f9c0e372eb9 : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e399961a23e4cba9e7a14d5e94e8bf6 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ac02586de944a9ba0d614f9db9384ca : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0183ca541e76487889da6d17a6f56ae6 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1dafa4a1e7674b359f05de65041b33c4 : FAILURE in 5m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa23dfee13444417a4448a6f17fedebd : SUCCESS in 7m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d0c497d9f2d4e4bbd5957f4c5dcbece : SUCCESS in 59m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34faddf1dee244bbb741f959b6080ecc : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460bc11e5d93452e993992231cc1a2b5 : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5531132a0e4b4b4088966727689fc17d : SUCCESS in 23m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/request_helpers.py","accounts_in_message":[],"_revision_number":1},{"id":"1f14d402befd5a845d87ae4161a9e6f536dcd453","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 15:29:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f1039aa9df3551604ee992201c881cd1b08ad68c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:32:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ec10c9e19624882b2f54b6cb3adcaa3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e01886023dc54461ae0d781ff7cfc050 : SUCCESS in 46m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9efb9f01f29742a98c752bccae65e722 : SUCCESS in 25m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0f51bf24a474c878811c8a682529916 : SUCCESS in 1h 00m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fd75d8108224874b23542b825af6334 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c30b05d39aa75d40b7af77a6f854620ece3e13ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:44:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae33a04378294028bdf9cd4c406ee368\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1291f0ce66e4263b8b3fa875f891a4a : SUCCESS in 7m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6e677411aed4a48b8b51a11b5817bf1 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95ffb9b52bba482c928398b6f9982f26 : SUCCESS in 22m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d930557132e54f5792a17788d6785fda : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71b7d5d6270f4942a414c26e763d3d7e : SUCCESS in 13m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6cc5543ff7ee473aa8c1a4675e1e0da7 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85757979ce22485d8d4dd8235e84f28d : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99dd35080dda4249a95b0ec122ef5690 : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/49ad18ec9cd84f078623641bbac5e544 : SUCCESS in 23m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cb6ef5eb9589421182d7a14c0eeb6eb7 : SUCCESS in 13m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e8fedf3c2c449d89d3421b8c58dcdf7 : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f024e84960441f3a65bcce22da552d7 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e477fce9d6124697a6dd61941fd5fd86 : SUCCESS in 19m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32be9a1d4b694ba4b00d0ca62bb526bb : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a052c6df395844c9a5bcd0ba37ea5695 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d2ec8e04ba0d4ffb89b2bc227a213fff : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/947658a63ee845c5b593d124a25009f2 : SUCCESS in 9m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4469ca7feda44d0eb513f9f00f7f518c : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/660e6202680c4edfbe62fa19760bc998 : SUCCESS in 7m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a60f1d18e31848b29a098ac1466ced98 : SUCCESS in 1h 08m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afbec641fb6c4b35a9e31132dd489f87 : SUCCESS in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43aa04975c942e48a9828096bd16ff4 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/227b17b8add14488bebe1805705ffabf : SUCCESS in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cd88ed48b7eafa2a1034c405fd99900b15c8e28","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 20:40:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a573a770fcaf983899ce4c8bc59b0141d824c316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:17:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d9c8aa2a58070207a8834e85c6d5b01374c5dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:40:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"421da8fb3471e1ab2cd37cac219521ec75918f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:44:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82e67a757d54380b6a0ae2fe3858589\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7176b2a197b471597e3da3c4ed74695 : SUCCESS in 9m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed6412637b48402592bbaa6b295faa00 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee21acf031514faa93aff440b0e5b6bd : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d82664e1200c4d2886c457fd1aac82e9 : SUCCESS in 29m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d4ffb77273f447698c43e955b255a26 : SUCCESS in 10m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c87687b71fc343e29133b76cc8cdaa66 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37c07f1d0aaa4061a942586eedf67598 : SUCCESS in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/487d79b4d31f41b3832d3d36f2093ca3 : SUCCESS in 16m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/99016bd1f6d24cc6bcc73f1db76eaa7f : SUCCESS in 17m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/77f9d7d04f0c482eb53764dc0b9cc061 : SUCCESS in 17m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be19572da4f94ae589e9a71fce39f936 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/93675510e417436d88d7f8f0e7dfbea5 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17d92831798d45c0b00c8a5780f31ec4 : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cadd3c4219a14ff3827fa9d75108b375 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/45265739d2f64fc19fccfdc28a38ccb3 : SUCCESS in 23m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7136f3a1a39d4ae081bd88bd53754ea4 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8882eb65a85040dca1176d2906519d2f : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2929e7c9cc0471a98ef908c49ef0a6c : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/320f7e534d6545e18e6fd73d42cbd2b2 : SUCCESS in 7m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b30825a6cff4dd78f6acc13e06f24d8 : SUCCESS in 1h 00m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0cb68773c354db4b33499bcaa875fc0 : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbf912288994f63a45aee85d21c8be9 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55df0b34ec8546c99c5adf2b868fac28 : SUCCESS in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8102d53e6fedf0514f731f4fe6ad212b054fb385","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:48:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26528302b6af4b9aa4886ce1578af3ed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b871dad344354bfb9cb41a3bce55da9d : SUCCESS in 54m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/742640b479a34a9095fde1e2a0cc632e : SUCCESS in 1h 05m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c7c433e9fd405c9832de9561fc8ee4 : SUCCESS in 57m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3a661f9d87544f99852ab71de43cbdb : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":4},{"id":"997e6764fd8459b36f3b8c0b1d8f4ed2fc1abe06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:30:31.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"acc4d2a6e335eab2bb5c34a893eeaa157bf0178e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 09:45:09.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/469534a27f8740acbaa4ad1568d0096d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13548f2a093848d9988c52a4bd7a5526 : SUCCESS in 56m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2125dde524f34b78b779d86cce7fd849 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/135dab5e93a1472ca20635ff27662791 : FAILURE in 1h 09m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5f222b80db84361aed7fd093bec61af : SUCCESS in 21m 20s","accounts_in_message":[],"_revision_number":5},{"id":"4222a1aa5b80a6969148b340c99e73e8fc89d65f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:06:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1062a38d9a34e2d8702d6d9c62e0107\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96f84f4eaf244bfa5b762c57dbec112 : FAILURE in 5m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0e889e7770d4ad68f38ccaf3e4fab00 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2aec8d25de4d4ac48bf0987fad520728 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c834269b8bc4069a9f18288c46e3014 : SUCCESS in 22m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a95fdbc1ea324baa82e2827b07370e8c : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cbef1e97aa6241ee807cab632df8bf1d : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9668cbb8e03a44808d108f173aad7d58 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c652529c56c416cb60640cff118a90f : SUCCESS in 12m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/afabee8c16cb4ba381f0047f75fcbeae : SUCCESS in 17m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb9e56d2599f4721bbbd86108847fee1 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/283de4d548fd4a0f8bec70df7d29bf08 : SUCCESS in 27m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ad1e700f23d488483a509f59249d75d : SUCCESS in 22m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d6057e7487f419abfe6c42267f55d81 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad39815f3298430a888d43b3f7da065d : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b35006a655854b50981e49a2011f5f7d : SUCCESS in 16m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7729fa8c46e64801a4f8483807f0c691 : SUCCESS in 13m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a322320b0e0e49538e9a5f861a7be862 : SUCCESS in 14m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e40cad7784274b9d80e5031acb6315ec : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd12ba75e1b34031bd4ede1981bd1d6c : SUCCESS in 7m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcae11b1bcf64b80a92232f3f48cc269 : SUCCESS in 1h 02m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ec7a63f813742e88d55954cbb3c03c8 : SUCCESS in 20m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b71ed487e44921a9bab7b0257b31b3 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3080fe35afbf4706bce456223be5671f : SUCCESS in 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c9c222deaf429f257b31b41725fdec49269ace94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 14:53:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f55b7f2cb95cf3690cefe9ec7e3bc1c7ceb5cb5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:22:32.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aec1907bacb4344ba9e56db6e722591\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f34177956add4dd6b39fb6f3af36f367 : SUCCESS in 53m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecc9f870c2d24cc09597fd8a20abf530 : SUCCESS in 22m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/331574b101d04c969136253ed00e6384 : POST_FAILURE in 1h 14m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f58ea233c6e64413ae984d5b1964ceec : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":6},{"id":"16901dce0b8bd28450cf59176cbb3c968f492d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:26:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454e3572ebf8416c92a3b8208050155c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee16251b9eb412b9703fa96be3a8255 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f1bef58dcb84464b2a5d3db7f6708ff : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8df06200de84e23a8c200df50d83124 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a055d20d7c3a4864a457d4d0368fffbf : SUCCESS in 23m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c8743da677f418eb8c4c8a5da34c00c : SUCCESS in 14m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/330370accf5842078325f96d74e4aa8b : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abe85c93036b43b2b836ddfd8f2b4b4e : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/10ebb0d83d3c4dc4b3c8c008f496ab63 : SUCCESS in 15m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b99b548f4dc14f68b078c13ba0f87214 : SUCCESS in 16m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c84efc53a2014f79aae3ac7be942f250 : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43a7659b1053468e90b6e237c407c977 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/faa3c6d3daf44f379307f26b5f73c111 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa42119e321642029e27d8b7d7bfb803 : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e58b4a49e66846cd82fb671b8e7a4dcf : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e4c88712df742998b849687b6174679 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d4214342796418cb92b9bf49644ec36 : SUCCESS in 18m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/291ccc1ff7c34a70b750623b95fef64c : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fc7d33a83734fc1b54ec9cbbd50f1e8 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3171061112954ef6862261ef22186125 : SUCCESS in 8m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb63019ccc44401ea7549ebf1f746b2e : SUCCESS in 1h 04m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5b046dc7554463b21524664970dfa7 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934c322c222b4323b2a8460d0f0b7a59 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/009a2d0969954b35a33c54b6bdb95f88 : SUCCESS in 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6a8244bd903101c9937e3be07bfac78843d362cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-08 17:14:45.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"a035d5352e4b3162c15d5d7812872d0158de4b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:14:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b792a15be8b4ffc8e04cffe28bb3e8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ac81775d7445f1986ead7d04fe5176 : SUCCESS in 7m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae2c77496b3942d980d5b541f404de30 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5062396037d44f885c61b8abf964e65 : SUCCESS in 22m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52cb58fe7ab047f7b367d6ef3d788756 : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca346639d8444827b482f327750778c2 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/465424a1723a49ce96e769be835a853e : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12ea8e9fd0be442691fcbfa0ca3c3353 : SUCCESS in 20m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d10875ed70f47afbfd0ef117b00b241 : SUCCESS in 19m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8bce844ee75c4d78b3fdd20d4fab3211 : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0f453c03c5144160ad75be2853fe6ca4 : SUCCESS in 18m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db490e1e89c84337bc01d8b3158eb010 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb970184f0024a258f128117de81a3d6 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f613676f1cae4467b38099a09a6d8d80 : SUCCESS in 19m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47d89e583069403496d7d67b6d626549 : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3338fd1ea8eb43e49360e97b6211b791 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fdd35362b7a7482b92b80a2063f15895 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d30fd6f83cb479e925ea0b9f792f087 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc57c3485592479d8c0aa556095b2d8e : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/65aea944bf4c4c5d962bad71d6faeb86 : SUCCESS in 8m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6cb3ef5a9b04e46919d394a524d54aa : SUCCESS in 52m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9a3f560a2d24a9c9db4127bec31d83c : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b79d5122024305b2276e169ac3c6d7 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20e2006638124fc9ba1a976bab38eb1f : SUCCESS in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6850aa0d7ba6c7909fe13db25f793f1d7d7d6eef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:47:54.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e3d41599134443fbc6f73168502b7c3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6870d52ca354e0fbe0f1fb4e89bbc08 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ce82407473c438097c1eeb5fbc125fc : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/740c15f6c1834138bf9554b525660031 : SUCCESS in 1h 29m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/70faa7aabb7246b1b8dc50ce59697d41 : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":7},{"id":"4248b2bd4ff80bece98b26ea9152cc66590e0d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-11 14:01:58.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d9b08b043fe44fb326a0fc15f8baeb3a905ca195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:13:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59d22c62c369420e924a9184bd12e4b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4a5b3b4cbe46f9ac139e76b9cc6841 : SUCCESS in 8m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/300e71571fa84e94814a54abd106430d : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/238a9f9c317d4ff380b331f17c5c1723 : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b99b8fd8eda453984463b640ffb9c32 : SUCCESS in 21m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b355d3803eb4373a53578f1de91e5cb : SUCCESS in 11m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3d1bd13e7dff42e582c6f870cf7c1a5c : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66e74c00cc6348c986032d825cd00eaf : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c80b52e3a64bf6a1ce9cefdeae999d : SUCCESS in 14m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/717e07bc8cbc48aaa43d374d846d94f3 : SUCCESS in 19m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c62f586d583b4568ae139236c0446243 : SUCCESS in 17m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5b61b97e9254fa78d853a5d7b563b64 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59963e92f547467693c17c1d0300a140 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87ac608929484e3e8b0fc03d9d190062 : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f18e7d05be4846c59bcb84c588326d9f : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aebf7f39599e49839ed86ffd272430b0 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e57a5473c2884b098bfd1af140585400 : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d1996e36a33413b8aa05e48ba79dd43 : SUCCESS in 9m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84dc58cdb46d43ccb4dd1920f9eda528 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e605d2f4698846fa815a6aa1dbf59d20 : SUCCESS in 7m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/95dea6aea1e24d279526af03f2c4f207 : SUCCESS in 1h 05m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27dc74fc3edb417db364e74c38bf96d2 : SUCCESS in 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908319d8645345f28ac138d01a9a8e8d : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38b18d9e98d24605b2ce18511fe62eb4 : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"49389eb642cbc9874bd7db3de6c09e7845d63649","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:28:28.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96f2180800df48f88b3a2a9221d97e20\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c35ecb9e1f454644b5c80ac55f10c910 : SUCCESS in 1h 23m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7a2e09f55144c7ebfd003abe489bd87 : SUCCESS in 22m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf2ac429ae8c42b68fb42466b68777e1 : FAILURE in 1h 14m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a95a1986cc684c2aba5e04a198871bbf : SUCCESS in 28m 22s","accounts_in_message":[],"_revision_number":8},{"id":"be0cd5b9695168b5c68250b7d0f3f07d081abefe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:19:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"35409b521d2724956a935ebe0786e143c65359c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:18:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62627690b4a44f5cba763d15ae523bb8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40c9eb3bf6d3457d98ecad993a7580e6 : SUCCESS in 21m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4982b9084304f2796e873e349c6ff3b : SUCCESS in 44m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/58c5a37f0a7945b39f0eb3a38a3ea11b : SUCCESS in 26m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/47f5fe04fb5f4d08856ddb27498814e6 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":9},{"id":"9f727632200acb11230e400b093d61ab29371cee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:43:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2c7a68eb86647818337e8b2cbf278cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7c50e3e2794c67b1f76f381b7f7aef : SUCCESS in 10m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9b2724de73d43289f2684578620935e : SUCCESS in 22m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56dd6a86a4e84e7c9cda75bd960d2564 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42c24a9d5e024d1c95fd1df251116e6a : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec016478e776436c8bc5e085d8f6608b : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5722fb7ecf54101a6a0e084a661d7e5 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8fb21123201f4e5caca3e8c410b494f4 : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54daab97238749bb806934e970be6e96 : SUCCESS in 18m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/83ee1a9235b34fa196c8eedd2eefdff0 : SUCCESS in 14m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f4eb00fa0d974f70872d74fdce2edf33 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d867d9a2a5bc4f57bb3628d246394d9a : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/370c3b0f8e794d729ad27c61265b9b2a : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8166ceefa0e64b71adde10d7fc81bc3d : SUCCESS in 17m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d27f8ef5d3345db97a4e1a0d8e67f10 : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/635d7e13231742f689a9e51cbc35db7a : SUCCESS in 13m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2f5cd366bb384a9d85658b8f4079c816 : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43134504c4a74133994bbb051170de65 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5978d2641a74a36afe51ba512b423b0 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/637d7420ba97467d803a1b807b8dbdf8 : SUCCESS in 7m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7218a069f5bc4837ab84e19cbdb0b765 : SUCCESS in 1h 10m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c04352c590bd4029ab444a808cfbb342 : SUCCESS in 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b298a91b7394a2887cd956b7ae5a83e : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77d9788e1b1d478993037b30a34bfc2e : SUCCESS in 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d96afe7f14c8dd0e2b65e9db75a6e573500c601f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 14:57:56.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ca1406cdef06640efe461d671731714666ecfec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:17:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd695950b42a45048f0b9a6107c031fa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96591789984148fe8c0e7ff1cbe95a88 : SUCCESS in 1h 02m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/872dc42ce8ff497482e7be6971c7c0c7 : SUCCESS in 22m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb3eedd0c2fe4041906a84ee9ff152cd : SUCCESS in 1h 02m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1491265a12df4fdeb51f7736204cfdd4 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":10},{"id":"db29cceacc36df9c7aeb2c304134969d0955bda6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:21:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4c54d84d4df4f06871f9cb8976d6ab9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d507e841f02d439dad9eee093aa84d2b : SUCCESS in 10m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b842efe3f6db4080a89aaef2a1760e14 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fea98f7cacfb432c915c4bc4d4b57544 : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3262242c5a4b49b680693ea5df1ea287 : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3dae1b02c2524334a11476ab292233fd : SUCCESS in 11m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09469621db0142e3b716070e65126ebc : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55fd3d6456d747468aa6c9dfbd3b5e0f : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ebd69ef475a433fbf6b33fa61500922 : SUCCESS in 23m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/56b047ee6b4c4d32b7004930d1388015 : SUCCESS in 18m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a2b2631c50374830883fd7f5ad30acd7 : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e1e0145a5684a6e983c03f6c44a13b4 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/acdb82c793904d7eb32e733ee4249f83 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86d3dc4262f449eeb848912e02655817 : SUCCESS in 19m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ebb1b64b47c4d67b7e3bb80802a719e : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbfdec709e3f4f7b88579ba34a5256f6 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8bd9efe83d6e427e98092fd27cf8b16d : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92ab5a760ef6411280ec66bbac985fcc : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a171e490d784750bce2e4008fb69859 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9f62841aef94170a9ea86eac66cbe9c : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6faf69564fca4b3e9328f3b35fe9cea7 : SUCCESS in 55m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cec020cc19d54532bcde74a7960b258b : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c23edd61e24907ba4a776b35a92485 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b578f8ce12b94803a208401f0cb76dae : SUCCESS in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b166213e751ab1479cdfce4c3c189161fbce9bfd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 08:07:12.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"52e6ebeef575b6048f84c621ee4d894ea94bd75a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 08:07:36.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"51087c6c1405a75656a0a2497337828fe304f414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 09:16:07.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de03dab701aa48edb90c1b17120f73b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b29d6093f1947928dc86bec28f3a138 : SUCCESS in 7m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ada5dfaa59874c76bfb19d21d010f450 : SUCCESS in 10m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/010e400533f140ae91c27dfccab6f4a8 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5aefe7221b948f6a37a83fefec62c8e : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a531567d81f24922800280cd67780924 : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/93b5dadbcfed418cb63bce68a25d2a6a : SUCCESS in 16m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/779d046707bf4e1baa3652b7482ba9af : SUCCESS in 18m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de0525c4b4f54dee956212bb8b61d2d0 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b43e7c507e942ba9d03ae7524e6a261 : SUCCESS in 18m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a0772eb5b934baaba71f9d1dd2331c3 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd1cf6a9191c43c3932889be38251108 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b15d129559543b5b5a33f94e29f650a : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d0a17f6a0d5a4663b30954a5c8b38724 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6536173ddbb4c2ca49b39f399da974b : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/42297b8e28484b49b38ad59c15863924 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cf9d4f8c35540bc8d5160fc30802820 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a401f4f4ad74cb6bdd5d66b3c31a559 : SUCCESS in 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59da5bf981c14c328204dcfad38687ff : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":10},{"id":"87d6f7c8b57e554643bb3faef93078d161b87854","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 09:16:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"aa49f2dd5a57a0e02f6c7aac69be6b66152efe6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 09:17:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50c63ea037414f5a8ac7d7a3c2e40808\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cb1e12fa524b159b97b9009dfaa397 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9b94f10a5b5b4a6cb58f8aa6f0e2eb18 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/075ddd590ac84e1d961f5c8b40ed2fa2 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"5e2ab1cd5a529590b810e52d340f9deeec858458","revisions":{"f352de5d2f28116f8284a603bdc39225830548b9":{"kind":"REWORK","_number":1,"created":"2024-03-20 18:39:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/1"}}},"commit":{"parents":[{"commit":"16160b8cbc38cbcf2e9051f15ca7b3d022bf8337","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16160b8cbc38cbcf2e9051f15ca7b3d022bf8337"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 14:53:38.000000000","tz":0},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f352de5d2f28116f8284a603bdc39225830548b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f352de5d2f28116f8284a603bdc39225830548b9"}]},"branch":"refs/heads/feature/mpu"},"f6f6ba6215cbdf3836604407cbca267bd330f9f5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-21 15:29:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/2"}}},"commit":{"parents":[{"commit":"f60824717718578e072f51560656b8fae00d6868","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f60824717718578e072f51560656b8fae00d6868"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 15:02:47.000000000","tz":0},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f6ba6215cbdf3836604407cbca267bd330f9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f6ba6215cbdf3836604407cbca267bd330f9f5"}]},"branch":"refs/heads/feature/mpu"},"7589992571384cc59d486b30bdd4d2e041e6b7ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-22 15:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/3"}}},"commit":{"parents":[{"commit":"f4b6cbb78075eb5e191ab519011c235f5de004aa","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4b6cbb78075eb5e191ab519011c235f5de004aa"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:16:47.000000000","tz":0},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7589992571384cc59d486b30bdd4d2e041e6b7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7589992571384cc59d486b30bdd4d2e041e6b7ad"}]},"branch":"refs/heads/feature/mpu"},"07009350a44abf245ebc3343db94ac46f14441f4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-22 15:40:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/4"}}},"commit":{"parents":[{"commit":"3d7a6d0b0558a13892383e6288bbb1eb5846ecf0","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7a6d0b0558a13892383e6288bbb1eb5846ecf0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:40:33.000000000","tz":0},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07009350a44abf245ebc3343db94ac46f14441f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07009350a44abf245ebc3343db94ac46f14441f4"}]},"branch":"refs/heads/feature/mpu"},"803248d08ea296dde9859c45e28b440dee4c9f9b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-03 08:30:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/5"}}},"commit":{"parents":[{"commit":"f96bcde6251815a04d5e4d04f61a8bc7f493eb57","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f96bcde6251815a04d5e4d04f61a8bc7f493eb57"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:41:03.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/803248d08ea296dde9859c45e28b440dee4c9f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/803248d08ea296dde9859c45e28b440dee4c9f9b"}]},"branch":"refs/heads/feature/mpu"},"6e91e46f489385c72dd885a583cb16d8ec3b11bd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-04 14:53:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/6"}}},"commit":{"parents":[{"commit":"e2cc09fec856219c8a295b9d1f5e0b905fb5162a","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2cc09fec856219c8a295b9d1f5e0b905fb5162a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-04 14:44:29.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e91e46f489385c72dd885a583cb16d8ec3b11bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e91e46f489385c72dd885a583cb16d8ec3b11bd"}]},"branch":"refs/heads/feature/mpu"},"e34cc70ad223c4164d33fd952a285a1861149a08":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-08 17:14:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/7"}}},"commit":{"parents":[{"commit":"5ab917a6e517117d1594a08732a2cfcae3b52c18","subject":"MPU middleware: support list MPU parts and list MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ab917a6e517117d1594a08732a2cfcae3b52c18"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-08 17:04:25.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e34cc70ad223c4164d33fd952a285a1861149a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e34cc70ad223c4164d33fd952a285a1861149a08"}]},"branch":"refs/heads/feature/mpu"},"e1bb0dbaa1b825d141792a7d707fc1bb994a491b":{"kind":"NO_CHANGE","_number":8,"created":"2024-04-11 14:01:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/8"}}},"commit":{"parents":[{"commit":"bf04be7c145f53edda9efb83ceb451805c205cc6","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf04be7c145f53edda9efb83ceb451805c205cc6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:56:54.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1bb0dbaa1b825d141792a7d707fc1bb994a491b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1bb0dbaa1b825d141792a7d707fc1bb994a491b"}]},"branch":"refs/heads/feature/mpu"},"c33c4a479fbaf2fb22d797ecb1c87f0993c5ba87":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-04-12 14:19:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/9"}}},"commit":{"parents":[{"commit":"53cb2d2457802812b13704df2622d1968b6c42fb","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53cb2d2457802812b13704df2622d1968b6c42fb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-12 14:13:59.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c33c4a479fbaf2fb22d797ecb1c87f0993c5ba87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c33c4a479fbaf2fb22d797ecb1c87f0993c5ba87"}]},"branch":"refs/heads/feature/mpu"},"5e2ab1cd5a529590b810e52d340f9deeec858458":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-04-19 14:57:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/913831/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/913831/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/913831/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/913831/10"}}},"commit":{"parents":[{"commit":"6841b627c85e37703e3b4d5cf6392b251edb6020","subject":"mpu: un-skip MPU probe tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6841b627c85e37703e3b4d5cf6392b251edb6020"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 15:18:00.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-19 14:43:07.000000000","tz":60},"subject":"object-server: include deleted metadata in PUT/DELETE response","message":"object-server: include deleted metadata in PUT/DELETE response\n\nChange-Id: I7e3b471a8c762db25cffc3c9e52047e83fa7d9f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e2ab1cd5a529590b810e52d340f9deeec858458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e2ab1cd5a529590b810e52d340f9deeec858458"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
