)]}'
{"id":"openstack%2Fswift~913832","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-22 10:48:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2","subject":"proxy: Add list of backend response data to request environ","status":"MERGED","created":"2024-03-20 18:39:53.000000000","updated":"2024-04-22 12:17:26.000000000","submitted":"2024-04-22 12:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913832-p-mpu-mw","meta_rev_id":"9c599ac694ae3510be26996c45405771579d94dd","_number":913832,"virtual_id_number":913832,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-22 12:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-22 10:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-22 10:48:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 19:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 19:44:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 20:42:10.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"2b26e210e52af31c54f578e64302ebccf6ab224c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 18:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0f2246efab7a4a472788bb143a91d887dbcba23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:30:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee7c8a5d5a374437918c9a1e16003974\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad38e85f0a6443639008963ba2386ef0 : SUCCESS in 41m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ee34b83cb9d4db08e011c7a91efdee7 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/54415af81afa42bfac868edad6cce79e : SUCCESS in 48m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1f9f06304e342d187bf6d655bbb6204 : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"08faee45f3792890c94ce9ee11bebbd66d2e65a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:44:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b7c2fe31efe449fa201127e936afc69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5618643f8d86434ba7e6b3a668a79c39 : SUCCESS in 7m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6315c45f6aad4148bae04ffd51ba71ab : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78ea15aa4a624461bbebef34ef923fd8 : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99219da1717c46d1a1cc10fcf637804a : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab8f4ce101b24117981fece208d9abc0 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0aa5efb7b1b740c1b3de9dab961065ff : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b4277aa2d7a14f558f6f54437cc44afd : FAILURE in 5m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3d50b2e484948b486765a176331f993 : SUCCESS in 21m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79afa0917d0b446ababa300c237b1ec2 : SUCCESS in 16m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0b095dda2ec64718a5a6714c3723b2cf : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f15da7f84764d84a439ec7bda973cde : SUCCESS in 22m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a43679e69084403da7df44f2e7d08976 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d62b211e2894f478dfb0d5bbecf01fa : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6eeee1b962d84375a507d1788d640309 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b7333e3aa124c9b9a6262b6b11e07a0 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4850c2c399124555ab1af682426f76e7 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3804b12aa754466888d0c81cb8ea3dc8 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e60b840e126f44519fa54494d527cb1d : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/51d14a415bc7466e95a5607e057f0c54 : SUCCESS in 7m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f3ec2d788b648e683b282e171c44302 : SUCCESS in 57m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ff6031b937749d394e5eec7a3140404 : FAILURE in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b0d8d20269445f9c7105385b6a5760 : FAILURE in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b386af220fb4088b4560ffb81b21bf3 : SUCCESS in 20m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/request_helpers.py","accounts_in_message":[],"_revision_number":1},{"id":"e05104022d6d0e76420a4b140c3ee7b1fc8b3cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 15:29:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"98f16f23af7dd72f7a4ff321b4384e6dce268aca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:52:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b8e1d948a747fca7076bf357850abc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68356ef21ac6468aac3e7c85a20e3599 : SUCCESS in 1h 19m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5bd03e1ef064b1090794163ab5a0809 : SUCCESS in 26m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf0c9adfa77a488989b5ce5e73adbc4a : SUCCESS in 56m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8823a814efc4397b773b31d64623c72 : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1c69865f7a75dc7572e22e5e975f033699416056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:56:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3de5006594f4f14bedaa36f8c4ffcaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb20568a87cf46ef997a9d01e58cbabf : POST_FAILURE in 27m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb9c023054d0455ea85b0c6e15394509 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17b95baf6d964520bdc0cc9a8e39202b : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/236fa784ebdd4299acefb2320b62310b : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9e7145f910764148ae1c5b4737053f73 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6aa67f209e54dc6bc6aa54409f9bba6 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/586f966ce3d5437fa852ecd0ddbbf5c9 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/459b48fffdd74adfb8c7cc8b6f3ac9a8 : SUCCESS in 19m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/21a86dc5d19344799068a93095a73ad6 : SUCCESS in 19m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f999bd6167df41fdaa7912c16b5b5f2d : SUCCESS in 13m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3acc40b132214b459ed00f9bf8b10d2e : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a5f4395c24f4cbea3bdd12267165e2c : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf4633f94d4543bcb40fa5e9358fd251 : SUCCESS in 21m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a837abf27364ff7b9fd6925a6710eb9 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b9d4293d79c464dbc9d6694eb8d5d26 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2acaff5cd2424e7689d0a33e74e362e7 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2055e24240b74467bd7c707fbb1cdc6f : SUCCESS in 9m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7429573c4e4f43948303693b9b0673fe : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb1dc7ce4f2d4451aecae30d29c7c389 : SUCCESS in 7m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/db70fd0c1ec142179d8b560ac8e6f86d : SUCCESS in 1h 19m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8348881b0a964aaa8c478607e1ad82f4 : SUCCESS in 22m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42309822c82249fba6b9a0fe5168c944 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45c0e4fd8ea34368b3bfd3cdf5d4d491 : SUCCESS in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4cac3a3acfe545828a95257dce4bad23c45633b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 20:42:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8921a1372a7daa21e7e961ee8cebb5eadc4d1125","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 21:39:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc840c71839544ae8f8f2c84685889d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a349226b82f448d6920dc650aa9a5889 : SUCCESS in 16m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4abc14a9400e48e1b29aad75da19ea37 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/05ff3d37b274474aaf21545d068ffe6f : SUCCESS in 55m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/70267e3e1b9742aa88a7f0b549819d91 : SUCCESS in 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"58474afe295860b074aea66c1ee2ec633fd7fe1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 21:54:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05271ea5bf2b440b89b2907f6c800e4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f0eb73aaede4c82a934b8b6eadb9323 : SUCCESS in 6m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/540a1a6ec92c484d908638a37b493de8 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02497acbc8734552b0d2fede7fe0809b : FAILURE in 21m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/483c193a5d5143578715841abcd7ce9b : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c362d2650f474e6d828b639e112453c7 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6d4338838be4731914a9c10765b5404 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b93781f5bab4f9fbd6e6ac9ca01f85a : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0b37aac74b64014b7c8595fd623b429 : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d35ddaeeb21b4a3186ffbee85e7e2537 : SUCCESS in 16m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d036eb3849284efdaf57c3085a40ae65 : SUCCESS in 19m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4971a52e42054c92b0cb0a99c13bb444 : SUCCESS in 21m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c68ba148eb04785a0963813d3c07fa9 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb452a4e25324dae87fe9fe7d4c336be : SUCCESS in 19m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b19d18448edf40f8bc1c7b5bdb791256 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f62dad1546845cba18b0f86d02ecbf6 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d5b9c48f111401b9c1607dc3dbaf2ae : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/716e691ee85541b58f9e6e90bc729986 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d348bd57a2a4aafa594bbc29bd9354c : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6663d1a032e649a08a851a17390dc170 : SUCCESS in 6m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f70284225f614106b3809c2622a59a05 : SUCCESS in 1h 07m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a4d8fe2c1e348af810d5462e92aac81 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110ff78d90e9451c90a5e563ad01cf70 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/73dbeb06793d42bdaab71852a04777a6 : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f955546dcb5b951153a38ea0af1783a3f5da375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:17:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8a73740231029c4d64bc90986d5f53b164d1cb48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:40:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5cb09ac0e4d37f0fc5c97fff9c82d151ebc5059d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:06:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41f29af32b2a4e43b6782eaa6f8f5a3f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57de2aab40d1471b9023f24e9ebe1b7e : SUCCESS in 19m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df4220747e454091ac1a0875e91ba719 : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7024111ea5644011bdb769d019e36189 : SUCCESS in 23m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8158d2c9df72428d8667061087e1d86a : SUCCESS in 22m 35s","accounts_in_message":[],"_revision_number":4},{"id":"50e2f85e48618d346556033f3579185e0d4c812d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:50:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c709e31a466243ccb52c25bfb63677b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7afffbff7174151b7882f728636ff83 : SUCCESS in 9m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/798dd4f6d3994923a4f62563a91167a7 : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad5f786844694dad9c1c7cd8ebffe378 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e66152109c740c3927376502ccd85ad : SUCCESS in 17m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/039bb7c6b9cc497490578ce6d4677333 : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4a4317517940455b8a0a798b819a9512 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9f5a969369204a208d3ea0e5d37d5424 : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9f9b67aee2b4c659a5ee37b6bdf580b : SUCCESS in 21m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0c42284094504c57827c671a54e13bc7 : SUCCESS in 16m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4732a0da2ced41bdac6364de7a001437 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa2d8b64ec384f7ebd464574399d0d06 : SUCCESS in 21m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3dcb39d6abee427fab4454b019915322 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d03d8113c684cf283e11a5fb61cf1ae : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/979407dc5b7647ccac34c74db0b9cc82 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5086340c46447d38525701ddc63a5e8 : SUCCESS in 23m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ec599b00e94f44beb1e01fe8d392284d : SUCCESS in 17m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e005501c4b841b2afe515ad0a788b23 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df4772f3469947d2875f219d9bfb0c2a : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b335dc4cd5ca400c8d975aa7058257ed : SUCCESS in 6m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e79f424fe55a4ec5a16e532e651e48dc : SUCCESS in 1h 03m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f767ba9c7aa2456799ea7f5f51491830 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7845996f994a4b15ba0cc63079f774d8 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cde29a9ac18d4a049c2e171424623130 : SUCCESS in 19m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"370e56af93ce008067b361c4adc1265cae5f6fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:30:31.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e8237a905e2e2a9257177fd18bf9654a0460c5ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:07:02.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca7697a4f4a430fae26fc58c1c9a2a4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/207066e969124361bc0368991d8a216f : SUCCESS in 1h 25m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/728e8fd5b013480badf1e052c4042973 : SUCCESS in 53m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8b4085f2d6d4f35a38cddc639d38619 : SUCCESS in 21m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a58c3f504bf4664842d836efeed4c0d : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":5},{"id":"739d21aae9e52689d93943861125d3cc5e7001cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:32:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc166a9ae1f84f009e47c5ef1a3035f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d706937d59c644059da04cae0ea8a165 : FAILURE in 5m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88241d73e8084c86981df9fdece7d7f5 : SUCCESS in 27m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f08da92c0d04cc8804a92b5bac72eb5 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6193d731ce2948889609182e0166cad1 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/138aa66fb8c3456c910a0b38f36cc396 : SUCCESS in 10m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c42962760bac44e6b348fbf67d598fc1 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d9aaf367b40f4e70afd5748d38ba118a : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2044ceb6173841889733f4b89c99a054 : SUCCESS in 19m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e7fc8544819a4d2db8453310aebf01e6 : SUCCESS in 17m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f60a36c0bf294367b308d67f02454ded : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/649fa00310ab42d6887f8d9deb91fbba : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07d7b7e101094426acdc10347e635343 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eda32b0fc4e140dfbbf440313004ece1 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9aad6cf9c62440dd9adc1e9d7ed40c01 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7771070e11649bebf64a711ebc209b0 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f277c257faf545f881c5e549a2728229 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/615747e319274c8aaaf6acf23491c03f : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1410c4eed2340ce8474989342353372 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e60fe2d03414707905f935770de6128 : SUCCESS in 7m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cc67e0c401c40309ae33a67ea8326cf : SUCCESS in 1h 07m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aa4954a8767461dadc6f114a9786c92 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df045c69a3d0468f878f0d404add0e99 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dcd9ff76def4008afb4c6e60c3bd110 : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"867d1df292fa2726c4ede559087bf29e7c9c2f6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 14:53:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3edcf013695ed5564502303b8a548cb065253727","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 15:57:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68867a4dca84d148c3fcfa7aae3984d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc2892f266ff47a18141726ee5b57e19 : SUCCESS in 48m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21985fd14f1e46d8958aacbd3c9b369e : SUCCESS in 22m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/61f78af4271f4e13b9a43b584ceebe1d : SUCCESS in 48m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c6ca8fc216f4b12a2661a7d951ddd89 : SUCCESS in 21m 40s","accounts_in_message":[],"_revision_number":6},{"id":"612241ab3914fb8d3e09bbefc5182711ae550fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:33:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae0fc6f8fa0e4eb98a4ca0196cea6154\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965e1e0c0e5f4e708d1003a7cbf64b07 : SUCCESS in 8m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0013a0b0881348c491e7e83ca5f94bca : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82ae663584bc4492b2854e23307146d9 : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd6afe9e51e3450599d6a7a0cc16a327 : SUCCESS in 17m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27def7722ff344dda272a003ff5180ee : SUCCESS in 14m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0482ed55861f4f55a6d2a612ec7e8634 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bc087082c554f1eaeb6fe2fdfa938b5 : SUCCESS in 15m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/715720cf05814259818857ab577cdc4f : SUCCESS in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ed7c1b37bded4cf29ca0cd5c7ef77301 : SUCCESS in 19m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f6c2af5b2d4e483a9c00081433bb7904 : SUCCESS in 17m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c37b65bbe28a4b94bd83fea27da3ed54 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d31d74f2b3f344c98295fd5f2ec800d5 : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/48eaa40c53ef4713b58e8c5afbc713b4 : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b0dcc5f0ae764d53b56bcfc763a3bc43 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f85fe5766486435384f1b98b5501588f : SUCCESS in 17m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4123b4f5bb9e487c89dd99cf235a5580 : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba78400031f34f80a8caaa69680468b2 : SUCCESS in 9m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ddf3cf936d71401cb4d830df88854e92 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5841f423e6ae41078cd2edbd9fb95da6 : SUCCESS in 7m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9e9a552809946c79f6b13643cc9e7d6 : SUCCESS in 1h 00m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7a59ce4edf84b59b899ef6c60c3aebb : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2546e923434a18b6eab949b3312e29 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6714f571b7d84d5491ffe6f16d87a41e : SUCCESS in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6e76410a30e39c07e58e7a1aa992e105fd126fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-08 17:14:45.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f531ccb6210fe4d03871e195447d372931733718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:27:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0965068c65964748ba508ec8cf5ae491\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401df69ef7ba444aa062f0cde326da52 : SUCCESS in 7m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb7ca64fffc74271bd9721d22d4fe52f : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8903a23cc9884bb9a90563e1f751063c : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a44fbaafd31048709fd9fd002ca34fcb : SUCCESS in 18m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05250e473d734975bcf058e9b398f99e : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/835eda4555fa44159f63fe0158b8e00e : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78f0bb59514e49f69d296b515c46ec28 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86802cbbe3e44970a71142a219420bb6 : SUCCESS in 19m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/98ff9f7d0a6644988264a5c2a4e1e327 : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/affa93e8bc684fc5a4a104a78dff22ab : SUCCESS in 19m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ebbb3fd906ba42e5a8879bdde421981b : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d98ccdb6ca54e2c8e617974f7204238 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4ddf7b0ccd5457394c49cc7c61d570e : SUCCESS in 20m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ebdc183539274f65b0c2b60f499d739b : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92ab0c7b1f5a464daa7214eda088cbd4 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37d6505bd3ce4496aa5d4ca7f6b0ee38 : SUCCESS in 19m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9cd4591bee84ce69af0df3bf3a177f5 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44a17935fd21491894c16afa796b90ee : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b50eed7ef0c44629ad7254a6d62badb7 : SUCCESS in 7m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c212718142f244428392d585164c296b : SUCCESS in 1h 05m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3efd81fbc254546aeb0e1787f7fceba : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b76db9399447028984c46a0a2d749b : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ce5f66b0e024bfb89691ded273c06ab : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"38c2619e296a328e4fc91ec675ddd79adf9ab145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:37:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5768a24bef4a238d7c7e7a70e68442\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71d7e95f8c0843809220e55f860ce5e6 : SUCCESS in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd392f72089d41f9933e279854a43687 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e607401f9c346208c06b9bb8d3a2c58 : SUCCESS in 1h 20m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/95bd72832802470aacb06f08daeeb2f3 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":7},{"id":"565f9f371926e882b9092f3e6f1cdcb7bbb501e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-11 14:01:58.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"96b43dcaae3cc865ac0e5a43808b0123854e0467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:08:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cebddedc0cf475a95e78773659e8273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab04ab71c68a4682a4325898c052a9fc : SUCCESS in 9m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/587c57ba797d4db2b63207c594364848 : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/710c77e7050c4fdd8e549465d715dd5c : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41ae0211b7424539ba7040cc9dde5482 : SUCCESS in 22m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67fa9a2329bb4c47b95ea26fba04dd87 : SUCCESS in 12m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5f0e10522ff45629dff4370eb0be0f3 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c920b847b67422f9e3f4688943d000e : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14887bd414ee4d59885beb9947918b16 : SUCCESS in 15m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91b8e139b2f54eec964d4de1b57f2474 : SUCCESS in 19m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3f2923a7b875420c8d1c00053a6f72c1 : SUCCESS in 18m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3679b6394044b25b175a86a67229ee4 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d3acb2fe7da947329586c72ef3b4a127 : SUCCESS in 19m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8efc81a853664baa89ba13a0963204d1 : SUCCESS in 18m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1cc15b2b13746739f22bd857e6e8b80 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9f8ab431e2644c17b1f3914345f89690 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ba70e299f30c402e842c5af674af1bb7 : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62790f7bbcf445cca17513dbff2ed07a : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6304a3791ea4005808b5613bd673622 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa145803767e4360abc8597efa7510c1 : SUCCESS in 7m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/920dd023df0d4fdca4ef64347cd7d86c : SUCCESS in 1h 04m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f9b24a25d041f0b4b5eca93e6ff0c1 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/accc671a64e34310a03e335e5e2ab261 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2163aee75e204213aef5bb463eacfdab : SUCCESS in 20m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3e413604c804859700135e716e2f2b79fe6e2bcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:14:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa26df8d380f4f098c69eefbeb8163f9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/faaa0da731524ed7b2a94e0021585070 : SUCCESS in 1h 07m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc0e60948b4e407bb76a2951259af592 : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e4d87d0fc93496197485bd0e9d3cba7 : SUCCESS in 1h 10m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/50c3aef6cc4641d586474a52a95a6fc8 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":8},{"id":"ca7a7484eff6deb154daed8214282984a0261677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:19:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"f9e421cb4b9752f61e85bc424692b161801bd5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:40:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/725eb30c693a4802a1344a27dc1e57e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6fe1eeb4e743f98b1ddb2019f79ee5 : SUCCESS in 9m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/284780540d044508996c7e42f0d36acb : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e16b23dcb3c24314a2f5fec1411697e5 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f9b28ac9af044f7a36452ad36f111c5 : SUCCESS in 24m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b4e48439e934b6893a283f15f29e3cb : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c24aa5c709041d190b9e83f250e7a14 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9352cdbcf5334cb888b68dd3e56d4be7 : SUCCESS in 19m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d75d7346ef98428cb6bbadba0e8ebe6a : SUCCESS in 18m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8c1b05c8bc4033b757a96b1775ebe3 : SUCCESS in 16m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bc92b3724e9941fd8a94705d2330c0e5 : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cade6d0618654569965143ca92c3d100 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e9b499ed09c40a682ba2e6396323876 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae25019e7b8644569c7181ad204200ce : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/915f436ec8724b7b8502b422d0ce34bc : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/781e2551473741308debe9f02b73f18f : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6253747538884b6080bc91937d20871b : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eac28863ad8345828d18756ada368ef8 : SUCCESS in 14m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e72e9e9d614043649a780946591a2afd : SUCCESS in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d9590e80a1e478d8d8c52ac2d0caa88 : SUCCESS in 6m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebc6f806106e408e8097e1e0d51258a1 : SUCCESS in 1h 08m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c176fb0694040a2be6bbeb798f52a67 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731677b6e69a423e85fd253b4b111ddc : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8a007cc68a54a2697659279511f6749 : SUCCESS in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"96833fecb0a84447b6537d1b957618d45d8da4e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:45:35.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d105f089552a4edc8e340cc4065a900a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94cfcc4258a44b3d8caf2e1f684b9fff : SUCCESS in 20m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4074263c77e04d25af596572979a30a0 : SUCCESS in 1h 23m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4879893c3a634cb3b647f3eae162193c : SUCCESS in 27m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad0181762b6a44d9916302bf3451aa6a : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":9},{"id":"bab454bb326c99bbde38a725db6e77fd90a75ddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 14:57:56.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e744f00c8c132b81e91c7067054d6be5d4f6d2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:11:17.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41f61c25b81248138cb5f70bf5a6c9ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863c3b1f8a93486fab764a495120ad22 : SUCCESS in 9m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49494e9cb90d40b88041f928f54a778c : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d1bd7af0f554ccc9f95ec93a601fa11 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffa538f869344fc98dd5cfae9394abba : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2ac16cdda7f4524a9078ed472b912f1 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e21221d650b64589985d4f2863bccc08 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/984598858f4d47a7b7a431cab702cb55 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c09c30537d2d41c699bbe0f2d5275752 : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8a707d01f8bb41afadf249f95466c26b : SUCCESS in 16m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/59b2fe38d4d043019b5290263dfdb0c3 : SUCCESS in 20m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/941da210260240c4b21920eec9170f7c : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e8787c201bc49b0a5f83552afe7e1c6 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c671d7d8519447529d0a2d5622f5ef68 : SUCCESS in 20m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8af0280902844ea1bc5db7e9b5097880 : SUCCESS in 20m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d875ac42887542d7a72a1466e04a76f4 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b30761ba98e84be0bc77ecca5fc89e81 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/963cd5ef2ee749fb801908a5f8709846 : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51397c78cedf424b81d83745816e27b1 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59ed72a15f18494ea60a8230a047a750 : SUCCESS in 8m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/29d59d815c08400ba0889a3c4b81f685 : SUCCESS in 1h 02m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14dc8378ef5040faad78d41ba1bf2d1f : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b26f837a9524f4d9f4a588e16a52fdd : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88a16beea2214e5f9a2e74c5747233e2 : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"714d66a55e85379699d271c8ad6b0bea1f3c16b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:44:25.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/961251878aea4f65b068ca1b65269a86\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01e08312c1914211b7e28ea8ce14720d : SUCCESS in 24m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba7598dcb692462e805a0498d03333c3 : SUCCESS in 22m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/597e625aef984a87b21edb71b1d73dfd : SUCCESS in 59m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9db0a308ff452095b84f158ecb9063 : TIMED_OUT in 1h 31m 53s","accounts_in_message":[],"_revision_number":10},{"id":"8d4a534407988e32432178ca9fa6eb1ac2d3f214","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 10:48:16.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"8426074ec91e5919b1508ec50da98068a9f48f52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 10:48:43.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"3dd87f4387345324f4b266505947230b340b6ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 12:15:53.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/724ac2b65b5b4bef9c58c0088e4b1125\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6db2945efc498490e778c87b45edcb : SUCCESS in 7m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3260efca252401eb5720ec1dd0f8f10 : SUCCESS in 11m 56s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c2d3ed01d1ad4c94a25dd678e35678cc : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a6a544c2c3344ed90f3567571452f84 : SUCCESS in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1af9214125549e28ef04be90741ae72 : SUCCESS in 18m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60cd3c0d7aa9469f9e7b411963424d82 : SUCCESS in 17m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0e61bcc9be8f4b508317e3c9129b2bc8 : SUCCESS in 18m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7deb7d31c1b648d09325021a4bd015f7 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d91edf681715496e80881ae3b175d60b : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b182fce147048ac9fdc6858a5b66b48 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/40e3dc3de4ba4e5fa2c7f0601e426280 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/422d79ab2f6749a08f7245b60242cd11 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ac80b1a24f248f8b09825fbca08c94e : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20676ce49519442f80305acaf2eeac14 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e98eef57d47e4179a618fb29fa64cfe5 : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67f2c9280ad64adea9bd1d93c77f37ce : SUCCESS in 59m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c29068f04e1f45db805047c9dec6e30c : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbf424764774d5aa9bcf2926572e8c2 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":10},{"id":"bda71418f8b0f83fff1538ee3c60ec051729a69b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 12:15:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"9c599ac694ae3510be26996c45405771579d94dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 12:17:26.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13abe7daf3364b4dac80265fd3f44c6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73769b674e474d5daf5b2973e2f92150 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1a8c4033c07d4138aa3a109d90b244d3 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5cb3f00a62e84f7c9641c970cbe1829e : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b8db8afa5216ee69e6ddebcecd5e747fb6143a08","revisions":{"a9d82983337f4acf4a69995a4b3715bc7d0071b7":{"kind":"REWORK","_number":1,"created":"2024-03-20 18:39:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/1"}}},"commit":{"parents":[{"commit":"f352de5d2f28116f8284a603bdc39225830548b9","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f352de5d2f28116f8284a603bdc39225830548b9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 17:21:40.000000000","tz":0},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9d82983337f4acf4a69995a4b3715bc7d0071b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9d82983337f4acf4a69995a4b3715bc7d0071b7"}]},"branch":"refs/heads/feature/mpu"},"4fa6921171eb2ad2446327ebb2365682815e304f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-21 15:29:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/2"}}},"commit":{"parents":[{"commit":"f6f6ba6215cbdf3836604407cbca267bd330f9f5","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6f6ba6215cbdf3836604407cbca267bd330f9f5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 15:02:47.000000000","tz":0},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fa6921171eb2ad2446327ebb2365682815e304f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fa6921171eb2ad2446327ebb2365682815e304f"}]},"branch":"refs/heads/feature/mpu"},"67acfaf190249b8f0c4b83f9ce9fbcd6e7b9a9a8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-22 15:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/3"}}},"commit":{"parents":[{"commit":"7589992571384cc59d486b30bdd4d2e041e6b7ad","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7589992571384cc59d486b30bdd4d2e041e6b7ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:16:47.000000000","tz":0},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67acfaf190249b8f0c4b83f9ce9fbcd6e7b9a9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67acfaf190249b8f0c4b83f9ce9fbcd6e7b9a9a8"}]},"branch":"refs/heads/feature/mpu"},"0a2026a7fb1ba75607c8cc5af27901e6a97ec963":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-22 15:40:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/4"}}},"commit":{"parents":[{"commit":"07009350a44abf245ebc3343db94ac46f14441f4","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07009350a44abf245ebc3343db94ac46f14441f4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:40:33.000000000","tz":0},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a2026a7fb1ba75607c8cc5af27901e6a97ec963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a2026a7fb1ba75607c8cc5af27901e6a97ec963"}]},"branch":"refs/heads/feature/mpu"},"15d180e2ef192b980eb5a505516dd3be92bb3d89":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-03 08:30:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/5"}}},"commit":{"parents":[{"commit":"803248d08ea296dde9859c45e28b440dee4c9f9b","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/803248d08ea296dde9859c45e28b440dee4c9f9b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:41:03.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15d180e2ef192b980eb5a505516dd3be92bb3d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15d180e2ef192b980eb5a505516dd3be92bb3d89"}]},"branch":"refs/heads/feature/mpu"},"73616d4faf562936c9955f671e200dad9ca198da":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-04 14:53:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/6"}}},"commit":{"parents":[{"commit":"6e91e46f489385c72dd885a583cb16d8ec3b11bd","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e91e46f489385c72dd885a583cb16d8ec3b11bd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-04 14:44:29.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73616d4faf562936c9955f671e200dad9ca198da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73616d4faf562936c9955f671e200dad9ca198da"}]},"branch":"refs/heads/feature/mpu"},"a7b204cf971212f9b5790ccc8e4256a0a7e2c820":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-08 17:14:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/7"}}},"commit":{"parents":[{"commit":"e34cc70ad223c4164d33fd952a285a1861149a08","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e34cc70ad223c4164d33fd952a285a1861149a08"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-08 17:04:25.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b204cf971212f9b5790ccc8e4256a0a7e2c820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b204cf971212f9b5790ccc8e4256a0a7e2c820"}]},"branch":"refs/heads/feature/mpu"},"ea699f975cca1ac8ca2b708179013ff17d9d49c8":{"kind":"NO_CHANGE","_number":8,"created":"2024-04-11 14:01:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/8"}}},"commit":{"parents":[{"commit":"e1bb0dbaa1b825d141792a7d707fc1bb994a491b","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1bb0dbaa1b825d141792a7d707fc1bb994a491b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:56:54.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea699f975cca1ac8ca2b708179013ff17d9d49c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea699f975cca1ac8ca2b708179013ff17d9d49c8"}]},"branch":"refs/heads/feature/mpu"},"cb4ad09f62a37c5843645fdedcd04e9d78fbdc72":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-04-12 14:19:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/9"}}},"commit":{"parents":[{"commit":"c33c4a479fbaf2fb22d797ecb1c87f0993c5ba87","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c33c4a479fbaf2fb22d797ecb1c87f0993c5ba87"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-12 14:13:59.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb4ad09f62a37c5843645fdedcd04e9d78fbdc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb4ad09f62a37c5843645fdedcd04e9d78fbdc72"}]},"branch":"refs/heads/feature/mpu"},"b8db8afa5216ee69e6ddebcecd5e747fb6143a08":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-04-19 14:57:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/913832/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/913832/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/913832/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/913832/10"}}},"commit":{"parents":[{"commit":"5e2ab1cd5a529590b810e52d340f9deeec858458","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e2ab1cd5a529590b810e52d340f9deeec858458"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-06 14:52:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-19 14:43:07.000000000","tz":60},"subject":"proxy: Add list of backend response data to request environ","message":"proxy: Add list of backend response data to request environ\n\nMiddleware may want to inspect the attributes of backend responses,\nfor example to find interesting headers that backends have\nreturned. When there is a multiplicity of backend responses, the\nheaders from all of them cannot be added to the single proxy\nresponse. Besides, it is not always appropriate for backend headers to\nbe added to the proxy response. This patch therefore adds a list of\nbackend response attributes to the request environ under the key\n\u0027swift.backend_responses\u0027.\n\nThis patch refactors the proxy controllers to replace the multitude of\nlists holding statuses, reasons, bodies, headers with a single list of\nResponseData that encapsulates the attributes of each backend\nresponse. This single list to Controller.best_response(), where it is\nadded to the request environ.\n\nDrive-by: Stop passing server_type to best_response because it is\nalready an attribute of Controller. Always include the request method\nwhen logging an error if quorum is not met for a backend response\nstatus.\n\nChange-Id: Ibd70d9b631075a23bff258c4f6f55ca4df59c0c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8db8afa5216ee69e6ddebcecd5e747fb6143a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8db8afa5216ee69e6ddebcecd5e747fb6143a08"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
