)]}'
{"id":"openstack%2Fswift~913833","triplet_id":"openstack%2Fswift~feature%2Fmpu~I84604a7ea049850ad4d0f0cea2096ab8a98dfb4a","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-03-22 17:59:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84604a7ea049850ad4d0f0cea2096ab8a98dfb4a","subject":"FakeSwift: support environ_updates","status":"MERGED","created":"2024-03-20 18:39:53.000000000","updated":"2024-03-22 18:00:50.000000000","submitted":"2024-03-22 17:59:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913833-p-mpu-mw","meta_rev_id":"26afd3437372ff6620a700a279228f90d86acfce","_number":913833,"virtual_id_number":913833,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 17:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-22 15:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-22 15:38:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 19:16:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-22 15:38:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3b7bd3412340f20c23610b9a823ad53626d0cae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 18:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9141b687eba0d7654ca4f077857dd53232b2fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:16:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcf9c00ee98a4b548c69e7ec72aa17a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074cab97752d408499364d0cdd0b0639 : SUCCESS in 8m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f7866a36d7f44e9a3a46f7373c6a8d6 : SUCCESS in 23m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/245b75f57ed149339e2fc1d21de6551a : SUCCESS in 23m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65cd8ca3f27a4cbd9d7090403a0b7632 : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f5e1257ec9d40739386ba648277d7d4 : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/679ab382c0b6413a82a11830dbeac471 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73ee9d94ff38416f9c2e6f87185aabd5 : FAILURE in 4m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efc850bce0ea4b13bf50fbcc4d21ee39 : FAILURE in 26m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/77b85df3e8294b34a1435941111adc22 : FAILURE in 17m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/265f8a6727b14bddb647d00830044377 : FAILURE in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35e746bba82c4ad08dd9f181b67c69c9 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32faa7eb1dc14638931f4643e0526b3f : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d685d79c5d54814ac2620ab8da3f94a : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4470b757c66a47518491537a81bcd58c : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2f77f26911ae46c6848856b2cdc52af0 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b9875e9b5ef41f38d584dc8f32b80be : SUCCESS in 25m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f2ca399ad064417a6f1cb8baf049b62 : FAILURE in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83c5f3ff2ac4b87b1e654c3f33140a6 : FAILURE in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ba8c19c62294cad9ea6e53c592be3f5 : SUCCESS in 20m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/request_helpers.py","accounts_in_message":[],"_revision_number":1},{"id":"d6d1dc1036bc3f42b83fc278d0a4272c7970bc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 20:00:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7feec76b644939914906188ebaf2c8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/144ff5ae15fa4ba998ef195a3ad2012b : FAILURE in 40m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c27f0312236b4d2ba0a97180982cf0d5 : FAILURE in 19m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fde2d4dbf65341b18e11db03a4f2c47f : SUCCESS in 21m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7b56aa1ba494b2a970527667510bfb5 : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c73e05a538419d1de873847d165a7fb5d3952f3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 15:29:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9279b85fac903056a4b46a62691726e85c1f3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:18:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48966b5a3aa547dfa74ab72c3c4caa81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4dd2f1bc00b45979855370d90fe7a8a : SUCCESS in 10m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6d762edc71747a18b82d6617a6f751b : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e480252015aa4f8a89c0adac064984b6 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d75696d878843e5ba260060ca3e88be : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bfc31bf0eaa4f11baf4a575aaf70062 : SUCCESS in 11m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cbd96619c10f4e6d9f0667ac61dcc15d : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38544a03104f4a75b2090464ac19b68b : SUCCESS in 21m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e44eaf1905e4f56926971c06543d896 : SUCCESS in 19m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b40e909569ff4e3c8233a81ccadc4343 : SUCCESS in 18m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b8124266a80d44ebb6ad9611ab9f6553 : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6db1a2a14506431285bcf189bd2b389a : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5fd1e9a1e7f2402f9ad6c51cccfac083 : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c80a88f7965240fd913337a125f368f6 : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d438d3584152435aab18c24a30591794 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/61352455708644cfbaa76a758577f1a5 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f3bdc2ae8f1e482cb5affa040eb72ab9 : SUCCESS in 16m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5061a9868c8435ca6f1b68f8aba5d78 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177e604bc1a24f6a9b9aef2f19f0face : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6155be258c6d46c39915ccac68ebaf5e : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7c88439090ac41073c13b516ad20b555e80321d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:41:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3875c6975e4226bd180b062dd843f8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a551ca4fcc046abbd8d8cf1c079e482 : SUCCESS in 41m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8ad5de3c4284643bb98eae045e3b552 : SUCCESS in 29m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cf909f2c05b4b16b14e2017fd79df61 : SUCCESS in 1h 10m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27d07f5ef5514cf7a0113d60d728d6c9 : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8ee7e1f2211682f3b418865b5434c351e5840e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:17:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4dbca56c19a31009323fcc302fbc006bd7c47e5e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 15:38:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b26f128e3d1c19e603d5b22ded2c29edc0533309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:03:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98d74a0288145e2a053af76f4238c2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc013ecea264ec78e2416fa88b9241d : SUCCESS in 8m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b43d7866384f40ac6d254b0441c474 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7b86a4ecf39452fa463164a89b7a747 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fab1dbdaacf242879e5b0be873e05b0a : SUCCESS in 28m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3acd39f98e1940f3bf9e7ffb3752b887 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e093314113b543b697176c1770c191d6 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eaa98998623144d5be8fc8c2b1a9ce5d : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a72f579721841c7a63644b94678f181 : SUCCESS in 22m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c500710ad3f44daa3d240a8c285a6e1 : SUCCESS in 23m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6e778f7965e84556a78e4991a984dd8b : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39fff9991374493b8cc5998ca2927497 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6901968f60b74274936005b8c6aeb1d2 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/637eb5c147f24bb5baaa16802cd0fb57 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b29fa2db19643bb87fa4d86a3de150c : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1461e28ed0cb49c0b43df4d4a95424a7 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1f44881d0ec4984a23bd8da92f60d34 : SUCCESS in 17m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52db5b8620c84b059c5d67e543ce39bf : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01304d78798f463285aaa9823242633e : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/688cfdadce6e43ab96955c41efb31e5f : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"232a5648fb5735a2cfa26f1c58365eb8cc9c6632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:03:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"caa54574319b5adbaa4aa6f4a58b05bb79dc1a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:50:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7dd7e4ab4a4f2a886a3ce117466d38\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/078031c567a44088868e1022cf7d389f : SUCCESS in 49m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bcae98ab39341c79f8997c5b8697a38 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/760dfc86323141a59d66769f9f9a527c : FAILURE in 1h 07m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b901878542c84de69b5019709d440d80 : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":3},{"id":"b22b0b4f8f6e38e8249bc90683d0049efab11cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:59:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/063d9d5cf8634b55870206b9edeb1113\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43bbef4c55a4e70a5fdeb5c08556a36 : SUCCESS in 8m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3474fddcb53472ca71e63e52533ca01 : SUCCESS in 11m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d2ea7e432ad94d4181c4418b5ffe5bb3 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d805c619de6548019dd5c07a2f39b14d : SUCCESS in 23m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61f8a180580347709482338c59aa73b6 : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/95047b3712a74ef18b6b045327e45101 : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/399526a57e6d4b6ca523dbe60da0cae4 : SUCCESS in 20m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14d93b0fb54e4d93b833f07e10146e07 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0121b0d2813b4994b1d7698eba90a748 : SUCCESS in 21m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15b7f02b670644c69808400c7c19cefc : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3aba3ef69b45421db194eff85b8db056 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4c17caff02ad418facc2a199d135fd3f : SUCCESS in 18m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/23c826b507fd48aea5eb914814d4d9b8 : SUCCESS in 20m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8655a678e394902bcb9fbbb40f42979 : SUCCESS in 8m 35s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/366a44b2e22846448ceb8bd4a5492d8b : SUCCESS in 7m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4eedf3fdce84b3590fca306e4fd2423 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076dcb1deef247e09da7aeaf62982baf : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"e3be55dff1bf0738d33a958960d9768d4e155728","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:59:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"26afd3437372ff6620a700a279228f90d86acfce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:00:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e3e2d30847a4dd59480e4550719e7b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bae259d6b494eee9d932467db401c9e : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/496bf97d284d42f6a5c418ba3df23a1a : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/af7124f479cd449985459f32b31f5577 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ce9accfbfb087ae3703964b77dddcbb2f8dbbc4","revisions":{"53c468ca553553d4aae19c9bfc4e5e524111ce6e":{"kind":"REWORK","_number":1,"created":"2024-03-20 18:39:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/913833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/913833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/913833/1"}}},"commit":{"parents":[{"commit":"a9d82983337f4acf4a69995a4b3715bc7d0071b7","subject":"proxy: Add list of backend response data to request environ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9d82983337f4acf4a69995a4b3715bc7d0071b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-08 15:27:35.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 17:21:48.000000000","tz":0},"subject":"FakeSwift: support environ_updates","message":"FakeSwift: support environ_updates\n\nRefactor FakeSwift to encapsulate response specs.\n\nAllow request environ to be modified via environ_updates attribute of\na ResponseSpec.\n\nChange-Id: I84604a7ea049850ad4d0f0cea2096ab8a98dfb4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53c468ca553553d4aae19c9bfc4e5e524111ce6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53c468ca553553d4aae19c9bfc4e5e524111ce6e"}]},"branch":"refs/heads/feature/mpu"},"409325f2e0a97e478007a7169aec369616846780":{"kind":"REWORK","_number":2,"created":"2024-03-21 15:29:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/913833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/913833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/913833/2"}}},"commit":{"parents":[{"commit":"4fa6921171eb2ad2446327ebb2365682815e304f","subject":"proxy: Add list of backend response data to request environ","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fa6921171eb2ad2446327ebb2365682815e304f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-08 15:27:35.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 15:11:50.000000000","tz":0},"subject":"FakeSwift: support environ_updates","message":"FakeSwift: support environ_updates\n\nRefactor FakeSwift to encapsulate response specs.\n\nAllow request environ to be modified via environ_updates attribute of\na ResponseSpec.\n\nChange-Id: I84604a7ea049850ad4d0f0cea2096ab8a98dfb4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/409325f2e0a97e478007a7169aec369616846780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/409325f2e0a97e478007a7169aec369616846780"}]},"branch":"refs/heads/feature/mpu"},"5ce9accfbfb087ae3703964b77dddcbb2f8dbbc4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-22 15:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/33/913833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/913833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/913833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/913833/3"}}},"commit":{"parents":[{"commit":"6e8d6fa6bfb5707ae71415820873834666e6c231","subject":"Merge \"rename probe test_mixed_policy_mpu.py to test_mpu.py\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e8d6fa6bfb5707ae71415820873834666e6c231"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-08 15:27:35.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:15:58.000000000","tz":0},"subject":"FakeSwift: support environ_updates","message":"FakeSwift: support environ_updates\n\nRefactor FakeSwift to encapsulate response specs.\n\nAllow request environ to be modified via environ_updates attribute of\na ResponseSpec.\n\nChange-Id: I84604a7ea049850ad4d0f0cea2096ab8a98dfb4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ce9accfbfb087ae3703964b77dddcbb2f8dbbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ce9accfbfb087ae3703964b77dddcbb2f8dbbc4"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
