)]}'
{"id":"openstack%2Fswift~913835","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ib735b660ce3622ffcd74269c2dd9aa42fdb31711","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-12 14:25:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib735b660ce3622ffcd74269c2dd9aa42fdb31711","subject":"add first cut MPU async cleanup auditor","status":"ABANDONED","created":"2024-03-20 18:39:53.000000000","updated":"2024-04-12 14:25:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"59b7ee388a5e8dfca7f28daac5f77779377e8c6d","_number":913835,"virtual_id_number":913835,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-04 16:29:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 19:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 19:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 16:31:11.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"ac4ed7e75a4832e6f386695aed925a3a30b9325e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 18:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d45f7a4642a500597c0a7f34e199734d59758d50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:41:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5597653e324bd182ae364c7ae19311\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c7b652859e5414aae3566877e4347e7 : FAILURE in 59m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb7d8fa1f78b4396a95387b1fce29cf4 : FAILURE in 18m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/31370c59558040a1819f1c9d59d85b0d : SUCCESS in 42m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b6007e0c49a423091c10e24431563a0 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9dc7c560f1ea7569ac7bafc7f1ba87176b7c3dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:51:27.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/464a88099ee44bd38ac93040235c56e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e41d5f9e8e4cd9bc0ff8b2b0926862 : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b815c3ed4b8b4d4b84ab8478ae77169d : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e688e615bddc4255854cf181b911163a : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb1aa23ccad943e095517b11670ee8e7 : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15d0afda5c754095a7b47d4e819ba14e : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b368c453e5ab490cb3769192277b67d7 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68f31cce54ab4693abf05da2951ed031 : FAILURE in 4m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57d09f081f4d45b8911effc065414f08 : FAILURE in 19m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bcb711499abe442fa898905dd61e4b2f : FAILURE in 17m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/496e0c83e39c4de3ba73cf5665ee6024 : FAILURE in 18m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d60ae923aaa4d1396894ee512e8c667 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df3445cb7f384183afa0838aa5c9da11 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84c891b0805a40e697fb07babdbae8f2 : SUCCESS in 18m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b80b8db0cf82458b85e32a0d355c5666 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/921c1e172e1043ecaf3ee02e5abd400d : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b6d3d1429f842e4841bf194773d7a8d : SUCCESS in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1c4d9289db741b9b8b81efef6e1a276 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac69c8ba25e5425aafa3138157933479 : SUCCESS in 13m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d510f775b2534ff3a84edec0d0ae36b3 : SUCCESS in 7m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/12a5903f55064409997340b1240fcea6 : SUCCESS in 1h 04m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1d4ee6625224e148d82eddd628902ce : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df659d6002f4e29a7cb633d60e1370f : FAILURE in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76999c94ddd84790b3d73e391f23b2e2 : SUCCESS in 18m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/request_helpers.py","accounts_in_message":[],"_revision_number":1},{"id":"01f7e6fb5a49707b369d234a2f3e0f75659c20f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 15:29:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4d9ac905f907a69379e97a0a604e002e5cb9181","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:50:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde924829ce546d2a01b72a774d6c16d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a420b49c9ea4e2e842298f2da705a2b : FAILURE in 1h 18m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/486e173267d94ca79c667e7f1af84c92 : FAILURE in 21m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d73652fdcca4709acdbb45305b3b42d : SUCCESS in 59m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/86a37b049f6043bbad094ee53f8e91bd : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5634ef48c9084dc69602f0659c5444477a48d66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 17:05:12.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a6d07789b9142d392c96f8f41e624fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1df4151d6c94453a9bf82761c4115971 : SUCCESS in 8m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af357a8f0c014b7fa56d7e07f1e2b776 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc5c77e02cc04b02a1e8cd1193ef2518 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3289260c14442e7b77246bd52e60f43 : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b623a7b732f245bbb57124700deec8a7 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f14200ae91164a88a9e10c1b441cf748 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8e2113904fa47b1935740230ed3f77c : FAILURE in 4m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06c9f34d1aa04e5e8383e861c1453208 : FAILURE in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c21a82f8270744fa985cb44684db8906 : FAILURE in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/67854964203c4ae7ac6e2c55131edd09 : FAILURE in 16m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/225d466450e44ad5bfc96fd0f1101056 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29c769798ed44c26b1bac715ee737f41 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95dfd33a21d749f0ba7aaef1f0108794 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87be450deb81403a92a0b3bb5cd84b3f : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b36bfd3973b644a5ad132d4f4cd7e3b4 : SUCCESS in 14m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dae25cd9cca043cb9176919b66168f1d : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d69ac19334c47f3b0b163f892e05406 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/347376a0e9c8417ab546a1dede3f7f35 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aaae7cd06ee748b3921766fd7afd0d05 : SUCCESS in 7m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46fdf55b66824293879f1da3dd5f092a : SUCCESS in 1h 04m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b007c888bf47afbb117a142681bfb4 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042dfe22a0f949f08dc15c4c34d924b2 : FAILURE in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e56f399cfee744d9a52663a43abfeffe : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"313fa6975f71d8fedffeb2d8d29889eec1db24fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:17:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"694b570d1624abf07a3da612de6008881c999cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:40:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6f149f19ca4750a9038544cdfc157732d821fe01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:41:30.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0bf3f5e9d4466787b97865f2016e83\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14a9de7bd27149d4a7ede8dce9af4fc9 : FAILURE in 20m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c1e6950c69f4b84899eff089f554938 : FAILURE in 48m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4263e57a960b4519a09cf897660472b8 : SUCCESS in 20m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c65cd81b259f472c8d95a47a3ff62827 : SUCCESS in 22m 18s","accounts_in_message":[],"_revision_number":4},{"id":"512cce7d2b3489b61a2278718b802e9275563967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:46:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd84b24649d14ddda14e0128dcd5f07d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68759cb1366345adb259fb2460d80c81 : SUCCESS in 8m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bb93b7d7aa14b99842522c5efe587c8 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/168f484102534fc190e58ef36b56d143 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51deca30db5b4c8b8f74b2c628868cf9 : SUCCESS in 18m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e885b50e96044bfb9fa511793b17b549 : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c842e528ba4047d9a5859f0cf4483a23 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f0f58c0322d4d1b852954cc31918045 : FAILURE in 4m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d628490bb2c42cc8a8a7a4273aaef18 : FAILURE in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e167bc42c0134635905c8b8f7f42bb63 : FAILURE in 16m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/22d68a851aec4f9b960c4beb2ffa6da4 : FAILURE in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bee7033d07fa4a239ddea527c53ef5ca : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73b6d4dd7b6243928db7350e9e35ec27 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/625ed78c7c14433a8d4c43883c019784 : SUCCESS in 27m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d418b2bc9c442f2b484e1e37054b9ed : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c5383b14746482d8c816e9f0baf3346 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4dedc9c1561b4f40aaa9b58d2083329f : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/346c3b00c35242179f7cb7cc16b9c8f8 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26288a6b07124a819506592ddc689688 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46a27ca72d144c708ccdd99351ec1726 : SUCCESS in 6m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/98442492543b4b69842dde1808a2e81c : FAILURE in 55m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/696ff88b062047fdb2c7f1d07a048f53 : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103b513bc7e24fe6acc08f7faf065d9f : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ac178e136054196889aa008dcdafa13 : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aee0619563eeb9c33a47de954209820a98150606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:30:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"405bf58c64321be5dab6a16a877980a173a5fe5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:07:18.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df71acb49d5b40d7aad4836de9c4f453\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/836d7ab1eb0d496ca03ba697532955e1 : SUCCESS in 18m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb05f95458c24fd6a67423652f41cf8e : SUCCESS in 1h 15m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7eb4f124349b40df846738385ce1a31f : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/72afdb90fce9456b8816a63b45cb6b03 : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":5},{"id":"5abc7f4d1ffc7bcdd336d85ef92a88b8cc16ba4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 11:05:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efeb225023f04084bd0983db35da0bdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ac831a197148ebb951a5871bcdb7e0 : FAILURE in 4m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4f99283706941e1add714f081ab4fbd : TIMED_OUT in 1h 31m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d14733ec8284d3d82820cb598b213fd : TIMED_OUT in 1h 31m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3047ef01cbd74ee4be1ba53165fe4534 : TIMED_OUT in 1h 31m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44816114490e47f999a678b634170df5 : SUCCESS in 11m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12b3e04621cf4a64ac6cb65422bf7cb4 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ccaa290ca1c429d95c9b60bf63f3730 : FAILURE in 4m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3af3cf5299cb48a089ead575f7655d2f : SUCCESS in 19m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca81f2db4f142b59d8f17f1710de8e8 : SUCCESS in 17m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f42153a94ae546d38cd7aade58183cd5 : SUCCESS in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0b774a169414ed39f86e1e7dcc76c61 : TIMED_OUT in 1h 31m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb3370faa8594fc7bb8c4c426dc1c59e : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90152d0444784588b67c8d971e9cf291 : TIMED_OUT in 1h 31m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/976688a43c2d4ff1aab39e7971fb3fd0 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/945ab64b9ff244c18aa2d004e604fbdd : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f93bbb5853242dd9c571a765c3e5c6d : TIMED_OUT in 1h 31m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31d7d9e9b20e4f2eaef1eda25e99aa79 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a49efafa7934ad0bc6499140763775c : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1d3ab1f634ac4554884dcdb217aa4d7c : SUCCESS in 7m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e0a7d12e71d41c89a0cf7580d9418a0 : SUCCESS in 1h 11m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/470cf42dba2741d09b9462f91e4f1289 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56813a2a0c44cc1be3350a52bfd7172 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6643e069620343908fd9c3508dc419ad : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3a773582647c3c3c8a2716160195e240e849f8d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 14:53:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ae4f9686c9d2a4281e85e48bca010a4147d64be8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:07:52.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb29f340dca2448fa42e3c78948756fb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e69c91415c0a4338ab8f9698f2f0f856 : SUCCESS in 54m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c420c235d9f444c8626116d5d4266d0 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ad07cef08a54afead964ebfdb84554d : FAILURE in 59m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/662b8f3b31c6421282763337832e9ac1 : FAILURE in 22m 52s","accounts_in_message":[],"_revision_number":6},{"id":"6711c10f609b97c9d5fa103b5b3ea76e9693947b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:29:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d47ace772db54981871e002247620fbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5270a1617a04682882870d8ed4f97a3 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fedf3be9989840d09c47d407fe7bf16a : FAILURE in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d14c5fdac1344d6a74275f74772e2a4 : FAILURE in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0e6e275536d424786e48250711eec6d : FAILURE in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b8f428cef0949c4a9f1f9cd1f72fad0 : SUCCESS in 12m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f7deb0889ec14c4d881eb4e30ed13d23 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9246d0fe475e4d1ab76037964a9aa769 : FAILURE in 4m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e74c7ad6b651490b9cba32721f61b364 : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df71865bc3b04a1e9435d5348608a51e : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6995c8e4af6c4a30a195a81ba1a32eaa : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be2f50d42c9448bc920c945783bc3c9e : FAILURE in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/353852600e8e479b89a5c3aacdd1f39e : FAILURE in 21m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc27d5d9f5d94bd0a735d69f0f8558c2 : FAILURE in 16m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2623f5d9e97747a98fb2f2f7e55cc8f7 : FAILURE in 17m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5bf29377d6fb4c028bf2d3ae89707ee3 : FAILURE in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/70ff7c51ce874c8cbcfd07a535ae5ffe : FAILURE in 19m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbd62bbe88154ee89539d236047782bb : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad52a2eb6c21461297caa44d5c455c19 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab438b00d26b4ee7b57cf2e358bdedd4 : SUCCESS in 6m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ea5b60a182e40a493009ed3d8112f38 : SUCCESS in 1h 07m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1615aa5c20c946449619c89217dbde48 : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8db9530e8ba4db49ee1a6f59e93ca50 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9d7ccfca518489092478b3e8fb8ed1c : SUCCESS in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"59b7ee388a5e8dfca7f28daac5f77779377e8c6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:25:43.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eef39261b0742794e961f41040edf03e4f4ea2cc","revisions":{"dd2c7abef6457a39405015e44466a4cb55708f89":{"kind":"REWORK","_number":1,"created":"2024-03-20 18:39:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/1"}}},"commit":{"parents":[{"commit":"c4d3c57739e59fad0838da712d0a8c886115e3d6","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4d3c57739e59fad0838da712d0a8c886115e3d6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:37:01.000000000","tz":0},"subject":"WIP add MPU cleanup auditor","message":"WIP add MPU cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd2c7abef6457a39405015e44466a4cb55708f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd2c7abef6457a39405015e44466a4cb55708f89"}]},"branch":"refs/heads/feature/mpu"},"12b2c6856bd52246e999d55cbb72d0feb103ed4c":{"kind":"REWORK","_number":2,"created":"2024-03-21 15:29:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/2"}}},"commit":{"parents":[{"commit":"14796e3fbde73a723024758eef75d920373526a8","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14796e3fbde73a723024758eef75d920373526a8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 15:27:48.000000000","tz":0},"subject":"WIP add MPU cleanup auditor","message":"WIP add MPU cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12b2c6856bd52246e999d55cbb72d0feb103ed4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12b2c6856bd52246e999d55cbb72d0feb103ed4c"}]},"branch":"refs/heads/feature/mpu"},"137a7c28cb38a60ad247e47147a736f4d2039cdb":{"kind":"REWORK","_number":3,"created":"2024-03-22 15:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/3"}}},"commit":{"parents":[{"commit":"c68bcf620fd796e71587f12963915ea8b61cd7e6","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c68bcf620fd796e71587f12963915ea8b61cd7e6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:16:47.000000000","tz":0},"subject":"add first cut MPU cleanup auditor","message":"add first cut MPU cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137a7c28cb38a60ad247e47147a736f4d2039cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137a7c28cb38a60ad247e47147a736f4d2039cdb"}]},"branch":"refs/heads/feature/mpu"},"6839b1fbff43ae19a629552a878c81b980655c23":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-22 15:40:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/4"}}},"commit":{"parents":[{"commit":"619a41078a80abb44da8c51b9d066f39ca546737","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/619a41078a80abb44da8c51b9d066f39ca546737"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:40:33.000000000","tz":0},"subject":"add first cut MPU cleanup auditor","message":"add first cut MPU cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6839b1fbff43ae19a629552a878c81b980655c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6839b1fbff43ae19a629552a878c81b980655c23"}]},"branch":"refs/heads/feature/mpu"},"01993b2f92e82295d674c9eb9ebd444c982a5aa0":{"kind":"REWORK","_number":5,"created":"2024-04-03 08:30:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/5"}}},"commit":{"parents":[{"commit":"28c4cc6a7dd1114134c7a12b423bab7f7c4b6a07","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c4cc6a7dd1114134c7a12b423bab7f7c4b6a07"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 16:34:08.000000000","tz":60},"subject":"add first cut MPU cleanup auditor","message":"add first cut MPU cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01993b2f92e82295d674c9eb9ebd444c982a5aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01993b2f92e82295d674c9eb9ebd444c982a5aa0"}]},"branch":"refs/heads/feature/mpu"},"eef39261b0742794e961f41040edf03e4f4ea2cc":{"kind":"REWORK","_number":6,"created":"2024-04-04 14:53:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/913835/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/913835/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/913835/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/913835/6"}}},"commit":{"parents":[{"commit":"8f01b01c81c29f07f0caf540677b6871dbb1a469","subject":"Add first cut MPU MW async cleanup markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f01b01c81c29f07f0caf540677b6871dbb1a469"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:31:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-04 14:44:29.000000000","tz":60},"subject":"add first cut MPU async cleanup auditor","message":"add first cut MPU async cleanup auditor\n\nChange-Id: Ib735b660ce3622ffcd74269c2dd9aa42fdb31711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef39261b0742794e961f41040edf03e4f4ea2cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eef39261b0742794e961f41040edf03e4f4ea2cc"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
