)]}'
{"id":"openstack%2Fswift~913836","triplet_id":"openstack%2Fswift~feature%2Fmpu~Iaab46cc4b222579e30f92bddc7346ffa01e7c173","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-04 15:00:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaab46cc4b222579e30f92bddc7346ffa01e7c173","subject":"WIP s3 mpu async cleanup","status":"ABANDONED","created":"2024-03-20 18:39:53.000000000","updated":"2024-04-04 15:00:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58fcd836c9c0311baf5c68a18e4d6acec016ba50","_number":913836,"virtual_id_number":913836,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-03 11:55:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 19:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 19:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-21 16:30:57.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"2c415c1d95829a09c9274d3da79780632a728eff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-20 18:39:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb467ddaa6bbb23b477cf4a02f074a43799fa7ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:36:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80bc5aa9cb6044d2ab57ab967337f9c1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e2d1839d68a452aaacbe3590cb0ee3d : FAILURE in 51m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8af9cdd3d2fc4f34bcfd519827a98bd5 : FAILURE in 20m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f59618041ebd4996872ce021cbfb88d0 : SUCCESS in 54m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e2734f5cf54450ab3918acf6a7b7661 : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c99645920f4155c49f1cbdfab7bbc419f0099d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 19:53:53.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dff1eb0f18d24631ab835b59c9da88a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbb0a5e086247aca806f252276d8dfa : SUCCESS in 9m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be8a1231199449f782a49eadac3742c3 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fc1517f842a4f25abbdf50c28b836ef : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/540825736ac74d678a395dd34e1e4db0 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14fbbf314beb448986dd9cbe7189dd80 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc1f6b073b094eddb356f5b38a7dfa89 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b50f5de6db849508b412b263d45ef5b : FAILURE in 5m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5429c1931786429b8240e6f24dc98e5c : FAILURE in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42ab78748e5f4cedbd482ebdb51f6c06 : FAILURE in 15m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a614f462d0314cc6a5e8945c3238df4c : FAILURE in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a1ca6fac3764de897a04fed42ccda73 : FAILURE in 3m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae7f53c2db5d4e6c9771c7bd9c0cdf45 : FAILURE in 3m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fae15a05b8ff4d929671fc54dea439a3 : FAILURE in 3m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16e48bdc9dfb4b3285579981f5f4dc83 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4f45985ca2594e7ab20a43b05212367c : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48c6376e557944a68b8ad14892386e00 : SUCCESS in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d835e5a5a8e64dddb7ae5da0a468702a : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6aa6e470c032468590f3ebf9e246a477 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b32032f700d54759a561d6e3ad188696 : SUCCESS in 7m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/956301d17aa7455fa1b60025e1224c7b : SUCCESS in 1h 04m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee307c5e516d4d02b04006d72225652c : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ae2acc739a4ee2ba2cd4e69a5e7e6f : FAILURE in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a40f4520b874c05ac969d68fe2f40a8 : SUCCESS in 20m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/request_helpers.py\n  Comments left for invalid file swift/container/mpu_auditor.py\n  Comments left for invalid file test/unit/container/test_mpu_auditor.py","accounts_in_message":[],"_revision_number":1},{"id":"f2f033878ca84cabd2c9daf5819e27437ba9ed36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-21 15:29:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"50f54a5e06e93ab438dbe0115f11cbed5c808fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 16:40:47.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac23f0ca16c44daad89fef424a572b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4d9de34fa1462d83941a520e6fd17d : SUCCESS in 7m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b30624b3840a457c985337bea97415b3 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d82e7c8c481466d95c58d2b12fd397c : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9aab88d145724a7d8f4beb67a80106e7 : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9b73b7795544414a75c71794a7b0f6d : SUCCESS in 11m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5148358ddb2148e7bad3215cec115753 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e53085cf65342aeb7e668cf00ec5730 : FAILURE in 4m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6467357e76cb40839b83184c1b79566f : FAILURE in 18m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/982ff582d85246659500d5fb4fcdf61b : FAILURE in 18m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8fc32ff8ebd040c3ba744f8f1c7d1e04 : FAILURE in 17m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9324d68d90ac440cb2728c412c4f3d08 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/685add61d2d04ad9b62c688d7fc31e3e : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa89e176f35f4c46ad419a756bd0cfb0 : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1205bce7eee9425abcc82632951b0843 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd67a38793774fd3a411708c4f795325 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/21b701c0ca51486185de650d4716a02e : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/649a031c3c2c41c9b6584122edece39b : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e04d4b2641440f6852576eb2a5991b3 : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/381be269ef764800a3cec3f72818b101 : SUCCESS in 7m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7c2cf32575c4a06ae02add5cb370290 : SUCCESS in 1h 01m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4acd3a0ed6cd437291cf82346fdbc874 : FAILURE in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749554bbe0af48bba9c62d7d390c8eb9 : FAILURE in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/497c0656dde04c07ae28035850c308da : SUCCESS in 17m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/container/mpu_auditor.py\n  Comments left for invalid file test/unit/container/test_mpu_auditor.py","accounts_in_message":[],"_revision_number":2},{"id":"3526b11999a4c847d974b09a802fa58a8748f1c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 17:20:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c718f0296a646a5be99cdfa21dc9314\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2a764a6405948d9b2c968aa03364f60 : FAILURE in 1h 10m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9bd39847bee4d8a9a44dded74d6b95f : FAILURE in 16m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5be20b236bd456aadf37e65481e697c : SUCCESS in 20m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cbaa269f7b543919be9234ce473f95a : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":2},{"id":"04b74e22b697aebd88dd1f7f94366064368fe435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:17:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6375d26b3df6397a6e9a133c037bbe5b1863b6ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:26:15.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"5268ce37ae2891e4c64699e413dd5a7ece087b77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-22 15:40:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"001d7a919299ff0849ad49d85d5cad836ff19b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:54:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fc75ed3531149f6b7a1df7012894c10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25de64802872439eb5cb50ecb55cb48e : SUCCESS in 7m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f008503ec0ea4efb9c66f2f91f3f338e : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c01fe934a6d442cf89f995bba091669c : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74698db8c0ee4d86991fda8a5b2b039e : SUCCESS in 28m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59c1467bd62140de8d46d939cabe2988 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17b3cb7714cf410c886ddbfcaa6766dd : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5570742ca63948b89ea3e7fb13e25ef4 : FAILURE in 5m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/414b0a96e74f472d820c60f97b8a9e36 : FAILURE in 19m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2a9ae4e555d8466e9c75693136e94cc5 : FAILURE in 18m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e1615f87d44a432eba603e29138f8587 : FAILURE in 17m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1304f616030b495897582410cb68fad7 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b42d3d2f8b364f6480554c35e12c0a7d : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c375c8233ab0442aa0475852c93c6571 : SUCCESS in 20m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2d28db89fcb547889a274618991629f4 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4668250e7e5a4a6da18b5550d57a3a0a : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/503e3dffb0ea4ddbae7278de30ea3c04 : SUCCESS in 17m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7eabe5eb894745f6a79c643b48d61700 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ecad5dbe3bcd4ebbb7b1cc6544e4f4bb : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/abbfb890a53e4f19b6771f273bba593d : SUCCESS in 7m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/09f1409fda164110b8b41b2481e9e9a9 : SUCCESS in 1h 04m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5edcfc153d954cf0bf5e5e4af2c9ba09 : FAILURE in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366a04a1a543430ba1d405944c96f118 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2bd2d75a2eb340358d8477c5eb7f2d84 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e4a4d41a77380247b8f4e5a9870570d86a69720f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:00:41.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f6a0ce93274352bbd6889e9ed60e75\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/534a1aafbe004be58dd15dbdd2cc19a7 : FAILURE in 18m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5bb9584fd624341b8a2530c854efeed : FAILURE in 19m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e6f398931114fbb872539448eb13385 : SUCCESS in 50m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cedcf02941c40d4a4d1ec863ed25dee : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":4},{"id":"75bc00791f6bd810e50b82e53387f8db7e3e32d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:30:31.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"026adb9a826e35e64e3b286480e7dc35f6232afd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:27:30.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8008e66d7273481cab56380612cfa238\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7719aeeeee044c5b80d0558e6c264b3 : FAILURE in 54m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/685e74f9fa1a4b2f8d6ef6f029e59f49 : FAILURE in 18m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f4a12bdcfc7464483581363b3f153dd : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b6bfe26412d4b36918f2845eb563f63 : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":5},{"id":"a17df0406c96c451690fa0484ca111fe37482377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 11:55:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb997987305040a5a2c9b7f89c45cb5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d2312fd981438b9d20f36123f5ed51 : FAILURE in 5m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/933f3ce7e4a94e609f027293296f29cd : TIMED_OUT in 1h 31m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/134e019e62b84e9b8db7df2a07dec8f8 : TIMED_OUT in 1h 31m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b405e6516aee4f7980cf75cf1501a958 : TIMED_OUT in 1h 31m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24e03357a9e4476ca987519ac5ff4580 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c7f7c1e1ccf4020a726e9a1307d0edf : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e2d49d34c5640c09e1c4da4c3596075 : FAILURE in 5m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cbcba7b311ba4908827dd77a7ed76e61 : FAILURE in 18m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4a66d88384674e82be8325461eb710fd : FAILURE in 16m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8aa682e453b24af4a3073ad90804efaf : FAILURE in 18m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a6d79610ba6940d8aa2727296b5dd065 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b43effa424984ff7b3a2a512ded5ad6f : TIMED_OUT in 1h 31m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/436089b3064547ae9eae17e091408265 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1bb0fa2568444e59827e9db16962e193 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/927f9c7109a64920a10bcb16674c47b0 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca6b235181874ac1a0f7f157caa1d3d5 : TIMED_OUT in 1h 31m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f83496444e04f7390e73cda4fb39961 : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed18bb6b35e74215b38d7c8be260a426 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2068e164eb44da4834295c8fbf79237 : SUCCESS in 7m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9dfdd184cbb4a5485187c70f96344d5 : SUCCESS in 52m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50faffe981bb4f5da0f1cd235f778855 : FAILURE in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9867e69e3814fa9868a6aeef343538b : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1add29abf9b4613be3450b9da525f0b : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58fcd836c9c0311baf5c68a18e4d6acec016ba50","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 15:00:35.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/913835/5?usp\u003drelated-change","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7cd6d91d7ea9bafd5e08e8066ff2fa4943a0383a","revisions":{"d2a10a3757d77772ecad266a3a45cfa121c3eb6a":{"kind":"REWORK","_number":1,"created":"2024-03-20 18:39:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/913836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/913836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/913836/1"}}},"commit":{"parents":[{"commit":"dd2c7abef6457a39405015e44466a4cb55708f89","subject":"WIP add MPU cleanup auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd2c7abef6457a39405015e44466a4cb55708f89"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:30:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:37:05.000000000","tz":0},"subject":"WIP s3 mpu async cleanup","message":"WIP s3 mpu async cleanup\n\nChange-Id: Iaab46cc4b222579e30f92bddc7346ffa01e7c173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a10a3757d77772ecad266a3a45cfa121c3eb6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a10a3757d77772ecad266a3a45cfa121c3eb6a"}]},"branch":"refs/heads/feature/mpu"},"43d317c1c4e65e3fa45027bff34806cca8efdd52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-21 15:29:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/913836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/913836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/913836/2"}}},"commit":{"parents":[{"commit":"12b2c6856bd52246e999d55cbb72d0feb103ed4c","subject":"WIP add MPU cleanup auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12b2c6856bd52246e999d55cbb72d0feb103ed4c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:30:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-21 15:27:48.000000000","tz":0},"subject":"WIP s3 mpu async cleanup","message":"WIP s3 mpu async cleanup\n\nChange-Id: Iaab46cc4b222579e30f92bddc7346ffa01e7c173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43d317c1c4e65e3fa45027bff34806cca8efdd52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43d317c1c4e65e3fa45027bff34806cca8efdd52"}]},"branch":"refs/heads/feature/mpu"},"fe997d9df34cceab607890208abd20b51f24cf5d":{"kind":"REWORK","_number":3,"created":"2024-03-22 15:17:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/913836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/913836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/913836/3"}}},"commit":{"parents":[{"commit":"137a7c28cb38a60ad247e47147a736f4d2039cdb","subject":"add first cut MPU cleanup auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137a7c28cb38a60ad247e47147a736f4d2039cdb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:30:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:16:47.000000000","tz":0},"subject":"WIP s3 mpu async cleanup","message":"WIP s3 mpu async cleanup\n\nApply async auditor cleanup to legacy SLO MPUs.\n\nChange-Id: Iaab46cc4b222579e30f92bddc7346ffa01e7c173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe997d9df34cceab607890208abd20b51f24cf5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe997d9df34cceab607890208abd20b51f24cf5d"}]},"branch":"refs/heads/feature/mpu"},"16a727c27ae232da2a318c900a3af2de3e965ba5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-22 15:40:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/913836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/913836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/913836/4"}}},"commit":{"parents":[{"commit":"6839b1fbff43ae19a629552a878c81b980655c23","subject":"add first cut MPU cleanup auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6839b1fbff43ae19a629552a878c81b980655c23"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:30:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-22 15:40:33.000000000","tz":0},"subject":"WIP s3 mpu async cleanup","message":"WIP s3 mpu async cleanup\n\nApply async auditor cleanup to legacy SLO MPUs.\n\nChange-Id: Iaab46cc4b222579e30f92bddc7346ffa01e7c173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a727c27ae232da2a318c900a3af2de3e965ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16a727c27ae232da2a318c900a3af2de3e965ba5"}]},"branch":"refs/heads/feature/mpu"},"7cd6d91d7ea9bafd5e08e8066ff2fa4943a0383a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-03 08:30:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/913836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/913836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/913836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/913836/5"}}},"commit":{"parents":[{"commit":"01993b2f92e82295d674c9eb9ebd444c982a5aa0","subject":"add first cut MPU cleanup auditor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01993b2f92e82295d674c9eb9ebd444c982a5aa0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-20 18:30:42.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 18:09:31.000000000","tz":60},"subject":"WIP s3 mpu async cleanup","message":"WIP s3 mpu async cleanup\n\nApply async auditor cleanup to legacy SLO MPUs.\n\nChange-Id: Iaab46cc4b222579e30f92bddc7346ffa01e7c173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cd6d91d7ea9bafd5e08e8066ff2fa4943a0383a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cd6d91d7ea9bafd5e08e8066ff2fa4943a0383a"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
