)]}'
{"id":"openstack%2Fswift~913841","triplet_id":"openstack%2Fswift~master~I1dfc0ee1d9e70cb0600557cde980bd5880bd40b3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-03-25 06:45:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1dfc0ee1d9e70cb0600557cde980bd5880bd40b3","subject":"recon-cron: Tolerate missing directories","status":"MERGED","created":"2024-03-20 19:20:50.000000000","updated":"2024-03-25 06:47:02.000000000","submitted":"2024-03-25 06:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913841","meta_rev_id":"8d808c0b9392654c1f9103fa69ed332e76766ce2","_number":913841,"virtual_id_number":913841,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-25 06:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-03-25 05:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-03-25 05:40:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-20 20:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-20 20:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-25 05:40:40.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"762f71402a4cbeb364a01c230eb5ad89c1d6b013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-20 19:20:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cf077883818d8c5f50333b06fef9baa48fa84ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 20:17:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca36f89dfe4d4ed799f9d2ebb6ab4f2c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2a10668263e481ab89965c3d80efb8d : SUCCESS in 16m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8233f1adfe74b189d162125031bcb14 : SUCCESS in 32m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d9511f8063046a79c46328636d7ff9c : SUCCESS in 17m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/471613ad75604d0e88b2a8f1ccffcbf6 : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e323cb12fb25340e596da17ea65e626cabb2d6b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 20:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70039263d8da44c6870d88e4cd8c5e75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1bb8d39232446b4ab6ca92b36790366 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6de08ffc337543c0a541169b471ba609 : SUCCESS in 47m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e274583b658142b09b4be9e4b53faae4 : SUCCESS in 45m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6efc0c5405c4f61832536a71302f1e1 : SUCCESS in 55m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4548c880dd5e48fbbe0f183646595ae6 : SUCCESS in 52m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fcb24c9751b423dbfee1cb1e4e1c7bd : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66c39a5afeb14c10b326ffc02ed257af : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b49a548942f48baa23ab73c7034a888 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a745e812ad640b78d8a3ec3df03e18e : SUCCESS in 12m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0412526f26ae4c9198a75ebea7cf7798 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d797264b8d754bb38ab91921d9aeb02c : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab38d61bd93c4a968638cc4e68cb902f : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1acfada659874649acb3de2272c9462f : SUCCESS in 17m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b478081c177437c83cdac75971db5a4 : SUCCESS in 18m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a7e98ec3c4a48fb8941ad06cb1f80a6 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff52758206c043da9b72341fe650bac8 : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/347209a614484c2294f9bc0916e2cdda : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d522f9144d745a29ddfb31da3d5c49f : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12e0b3c2bcc449bf9a3f14e1a687e72e : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/990314a1d7c2493c91bd98b569e37964 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/94041ae9fd9e4a75957f20657a053abf : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/159dd1c1f73a416eb9745f5edf2f47b6 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2caf46a9909a4ba3a03f2257a3f892eb : SUCCESS in 7m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6de73cf5d122450b9a4617c8f116ee97 : SUCCESS in 48m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2320eb97ee524c57bc1801e7bc0f206a : SUCCESS in 37m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4bd8fb5aea245b1b68e24b8b772181d : SUCCESS in 52m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e03b4c517f454b17add533079cb70118 : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84493c1ca0e544c1b4e18a8238d35b0d : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0f17da0548d45008767532cdafc4b0f : SUCCESS in 18m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59fd8dc517ea401b8c0933049b1a4e91 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e85ec43aa5ce046183306227a5c9a287426648c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 21:11:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d5ee1da9a8a817a47529bdc3463af99781f985c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 22:08:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed606de35abf40bf9f81e2fcfe12664b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11f9e4d7fbad4414a982737cb9088736 : SUCCESS in 15m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/992f962a04ff44d48754607ebcec6581 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebafbe4a15d74c638ef2d5cf1a80dc65 : SUCCESS in 56m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/539adff017294f3a80a10710b9323d5b : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1d0fe5c4ea907f25029b392962123fd7dbc45469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 22:30:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf08aabb7f794aedaea6b4c108c11cd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374e2c2eec614f359cd76d1e2b8bc978 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/5b9d9607cefc44659c7f3424124af8a2 : SUCCESS in 1h 09m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/157a7fddaa9048b2a1eabcd054bed681 : SUCCESS in 1h 12m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a370aa228964bce926453d097423279 : SUCCESS in 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2fb9610908e493f9e755bb4094bae5b : SUCCESS in 55m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fe9aeee64af47dda4e49a532fe7a264 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eb212c9c94842048ecfaa7de8d14356 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dcec4d85ddd40b9be55d9bfaad1b830 : SUCCESS in 19m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4985a5efa6645ee8357fc21b31febbf : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d33a124865d45eeaf5f4a3b31ed6dfb : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9665a17ef48449588ab02cdb6697fdf0 : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bab16fe881e94d4586a6d470ee1821c2 : SUCCESS in 16m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/db5aa3e107ac4b6b9dd51cca2a73e173 : SUCCESS in 15m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bfbbbf3cf2284923bba7a17825a0b67d : SUCCESS in 12m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46fe1cd369f64601a2d247407b0d484a : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3bb2d1bfdcc4b2d913ac1d819d46c13 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a886d4b8c9cc4c80a2bc9ce9a558f5cf : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2ba7c6b39c9a41e5a9d80cebff401596 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6ca2262418549f2b9d4bbea29439ec6 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/447806f9cb5a4348b645f0acaf58285c : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53e908655088430fbbfe530530c71530 : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9e0514a492d415cbbbb226917829575 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9781a44f51f4a5f875672d2b59e1776 : SUCCESS in 9m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f349d49ca854aeeb956a2e6d7669113 : SUCCESS in 48m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d874a708ff0f4ae89a8f54a7f2bd8032 : SUCCESS in 41m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac3d6ba9a23e482184ef2aa4d33e46a4 : SUCCESS in 34m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04dcabdbfe27418a84c0c5a7f5a485da : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0216f82e019c4568adbfe58c81fb6425 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/becabbae624548128c932fae1af39cca : SUCCESS in 20m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c11f524c7443cdaa7238b240ec9fa8 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":2},{"id":"dfa95e8e3145cf26b754258608abe5fc5b6310ab","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-03-25 05:40:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94c6382d83bc8da9fc866253e790bf0dc0de69e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 05:41:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5385b557e7a17887caf05c8b6ff1ac11b9429eb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 06:45:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb8ff09489b489a80dcac83d1e500ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb643f3cba74d0a8d37ba29a28d4f35 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/971bf82c344645ed817a2ab38e933e2a : SUCCESS in 1h 03m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ad8b93b94eca48278c762c5c54e5fc3e : SUCCESS in 54m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9c2d2965619463384ae4f2bd45e96cd : SUCCESS in 41m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d121eab3007444f88a7fe3cd2fa57d8b : SUCCESS in 51m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/adbc913beb3f4a8caaac4dc9b0e7dc4a : SUCCESS in 12m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dc06eab129f5496f8a64a0f796f5ed34 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ecde3ffbbdec4f248b850322113704b8 : SUCCESS in 22m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad5c623fb71e4dd28f29ff00f7208aae : SUCCESS in 17m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19f18f5b47bf4726b88302f41fc856cb : SUCCESS in 19m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d630f4fbbedd4eb99ed69d331b2117b0 : SUCCESS in 17m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c63a2c83c1bf4e25b706e51cc88d2efe : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3cfe4b060f54d63be5c10d3e3e6eb77 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b73bdfdca1147189606e2265474887d : SUCCESS in 18m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/475c99e8cfa74f96a0010b2172c93b5e : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/893a9f4365b24f53817d7feac6b7ccea : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/afec8968dea948e9ad72aafb10ac8995 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05286062eeef42f9b2b859c661d154d7 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2c2a56a99aee45528b4e6d016d35c5fb : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/30aa30cb180a431eb8f26b31b4de2549 : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fae64278865f4dab903605f376ad9bcf : SUCCESS in 48m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac7fa8ec4eb8414a97634e633dbe0351 : SUCCESS in 54m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc59fcab3ad7402bbd37bfecaade3eb9 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f71723146a347b59517ed856aebbf5c : SUCCESS in 3m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70c7a422fd6a4e48a1dcc562791218a3 : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":2},{"id":"28d0870de585c437c0dc850273d0361937335155","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 06:45:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8d808c0b9392654c1f9103fa69ed332e76766ce2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 06:47:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f651e3bcd1064bfd85d934ce6a33bdd0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c043e5d02cb34ababb749ca065ba10cc : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2abf78fbc82b41b79f78540b97b05164 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/20093131e77b4452b8f27c19fc4b8a25 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f31b6f7353f6f8e91a452bf3586032b94b7898c8","revisions":{"e24d8d9a38a3f3b420c27712fc1cc9a5db0ea153":{"kind":"REWORK","_number":1,"created":"2024-03-20 19:20:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/913841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/913841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/913841/1"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-20 19:18:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-20 19:18:10.000000000","tz":-420},"subject":"recon-cron: Tolerate missing directories","message":"recon-cron: Tolerate missing directories\n\nAny of these directories may get unlinked between when we saw them in\ntheir parent\u0027s directory listing and when we go to descend.\n\nChange-Id: I1dfc0ee1d9e70cb0600557cde980bd5880bd40b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24d8d9a38a3f3b420c27712fc1cc9a5db0ea153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24d8d9a38a3f3b420c27712fc1cc9a5db0ea153"}]},"branch":"refs/heads/master"},"f31b6f7353f6f8e91a452bf3586032b94b7898c8":{"kind":"REWORK","_number":2,"created":"2024-03-21 21:11:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/913841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/913841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/913841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/913841/2"}}},"commit":{"parents":[{"commit":"b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7","subject":"Merge \"s3api test for zero byte mpu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6dc24dbc0eca46141544dcdaabe7cdc281c9ce7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-20 19:18:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 21:10:14.000000000","tz":-420},"subject":"recon-cron: Tolerate missing directories","message":"recon-cron: Tolerate missing directories\n\nAny of these directories may get unlinked between when we saw them in\ntheir parent\u0027s directory listing and when we go to descend.\n\nChange-Id: I1dfc0ee1d9e70cb0600557cde980bd5880bd40b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f31b6f7353f6f8e91a452bf3586032b94b7898c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f31b6f7353f6f8e91a452bf3586032b94b7898c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
