)]}'
{"id":"openstack%2Fswift~913859","triplet_id":"openstack%2Fswift~master~I2f6760f0d40ced849f60906ebd6cfdd478ccd345","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I2f6760f0d40ced849f60906ebd6cfdd478ccd345","subject":"tests: Validate some assumptions in test_signals","status":"ABANDONED","created":"2024-03-21 05:46:48.000000000","updated":"2024-09-04 16:52:53.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"131b58d790467e020d18aaf83dc7669fccbea342","_number":913859,"virtual_id_number":913859,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-04 00:44:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-21 06:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdd8200bbf9a0108084cac4e0b8a63b7f0f92319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 05:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"814d1442bfd83332770f0db710400d720fa2f347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 06:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecaad2712e0246a19e70ad802bc11728\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055482caa4534cfb9ff30cbe4030645d : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22444c139ca74502bc4e51ce9af3cc1f : SUCCESS in 54m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af4ddae42f141c8981ef49a1dacd1db : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"03fb762de0cecaa723371c5a5ded57e5a66c44ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 07:06:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2ffcbabc5e4d9ab3045b7fd09d06fc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a77b542500df4a8eac09e6d9f02d66fb : SUCCESS in 1h 17m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11b9245b45a1418596f6d4a365c2de68 : SUCCESS in 1h 07m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bcd37ae8ff14207ae3234960df54c34 : SUCCESS in 20m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/108028e0a00c46089ecb1b83c43c9f77 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3caa8a779a40cf6fd3fc9a58d62a0ddd6ea40787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 22:49:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a0f065efe6b473267931edfd5024dcbaed42f95b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 23:47:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe03548d53454a578d6ab3ed586e3d4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7b631781bb846b49c9f53674a17a151 : SUCCESS in 17m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b9d9c0a284640e0835c12988487c14b : SUCCESS in 55m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/698e021b15104ec6abd4fa82e7ec825e : SUCCESS in 21m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac291ba7319d4c428ee6c68a397f6688 : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":2},{"id":"425cc39c8111d7f3fabc3db00094f535dbf24ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 23:53:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d86e6bb2dc2947b99abcaa69e62fce18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8bd050e75341178f2d8785135311b7 : SUCCESS in 9m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5b4cc0d420c4a158f24b53246a14557 : SUCCESS in 58m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be77cd5b06434923bac7e85e8202b683 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"cb2adcceda5fa83626a97f8bdc768538d2175859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-01 20:21:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":3},{"id":"8acd0fa3f9f3cdc0e974204e886b7ff1b6d9dbf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:29:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba357dae69f4947b6efef3398df0f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d52a25e5352481e8ceb412134aea544 : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5ea36a791304ce6966f264e5d028c2a : SUCCESS in 59m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451c73b660a64baab2a70739e3843156 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8fd9a129ed64f750bf93d9b0937d58b873ff24e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:55:13.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/288e05c289394943bb1b4e7770e6db13\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5219d6821ab4703ac844e8c74dc9d79 : SUCCESS in 18m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d90e632c699a4f68a3983a0e349c5c87 : SUCCESS in 20m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1a02325eccf4e5c9acbedf0914d10a0 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8fe1f9bc4d34601a84f02a9e4855a60 : SUCCESS in 23m 42s","accounts_in_message":[],"_revision_number":3},{"id":"86c33f422cbe5037732a36ca65de2afd474cc29b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-03 23:43:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f3e5c8237350b40435aa005bc67446043a626adc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-03 23:44:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"97098a10eaf8c6f354ebfe7addfe4de612ee483a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 00:44:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86442bc3affd4c5b9bcf929649628517\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df18733fc814bab9df54d590aba9aaf : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36961f6068c14c6598cdc11b512391f4 : SUCCESS in 57m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b0e38a1a964635be5cfe12d78b7b37 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a653e24a528338ff97c208e71c5a2e437f8ffd5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 01:16:48.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461c62c9143f42518afaecd1b393498b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffe53bd341544e299ceea456e3d151fb : SUCCESS in 21m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/024cee771a0e417396205c04680933c0 : FAILURE in 1h 00m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5221f67b47ef4a3f92893efb948e8e8a : SUCCESS in 19m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d3499614e59462388fa59df731f1248 : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":4},{"id":"131b58d790467e020d18aaf83dc7669fccbea342","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 16:52:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"aa924756257c49118c36aebee080f88acdfab5aa","revisions":{"83b8ed4a723c37a817b308acd721b8db5bb95f6d":{"kind":"REWORK","_number":1,"created":"2024-03-21 05:46:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/913859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/913859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/913859/1"}}},"commit":{"parents":[{"commit":"190243a66bf3b8b0df6662b7421a1c6829bdf688","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/190243a66bf3b8b0df6662b7421a1c6829bdf688"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:39:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:42:35.000000000","tz":-420},"subject":"tests: Validate some assumptions in test_signals","message":"tests: Validate some assumptions in test_signals\n\nChange-Id: I2f6760f0d40ced849f60906ebd6cfdd478ccd345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83b8ed4a723c37a817b308acd721b8db5bb95f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83b8ed4a723c37a817b308acd721b8db5bb95f6d"}]},"branch":"refs/heads/master"},"42807a07eddd026924ae6e6a45b9b77cd5ec5ae1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-22 22:49:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/913859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/913859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/913859/2"}}},"commit":{"parents":[{"commit":"1d91d59514a5c662ae443455a82837a388679479","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d91d59514a5c662ae443455a82837a388679479"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:39:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 18:59:34.000000000","tz":-420},"subject":"tests: Validate some assumptions in test_signals","message":"tests: Validate some assumptions in test_signals\n\nChange-Id: I2f6760f0d40ced849f60906ebd6cfdd478ccd345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42807a07eddd026924ae6e6a45b9b77cd5ec5ae1"}]},"branch":"refs/heads/master"},"993b2bda8e4b15b89795014306b04b0d3ac3f925":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-01 20:21:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/59/913859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/913859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/913859/3"}}},"commit":{"parents":[{"commit":"7ed859bdfcb0809e323ddef281f572eec5aa6d5d","subject":"utils: Add crc32c function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ed859bdfcb0809e323ddef281f572eec5aa6d5d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:39:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-01 20:21:43.000000000","tz":0},"subject":"tests: Validate some assumptions in test_signals","message":"tests: Validate some assumptions in test_signals\n\nChange-Id: I2f6760f0d40ced849f60906ebd6cfdd478ccd345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/993b2bda8e4b15b89795014306b04b0d3ac3f925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/993b2bda8e4b15b89795014306b04b0d3ac3f925"}]},"branch":"refs/heads/master","description":"Rebase"},"aa924756257c49118c36aebee080f88acdfab5aa":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-03 23:43:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/913859/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/913859/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/913859/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/913859/4"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:39:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-03 23:26:55.000000000","tz":-420},"subject":"tests: Validate some assumptions in test_signals","message":"tests: Validate some assumptions in test_signals\n\nChange-Id: I2f6760f0d40ced849f60906ebd6cfdd478ccd345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa924756257c49118c36aebee080f88acdfab5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa924756257c49118c36aebee080f88acdfab5aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
