)]}'
{"id":"openstack%2Fswift~913977","triplet_id":"openstack%2Fswift~master~I9b97e8d03e263ab054091b5901a1118d7ae0b892","project":"openstack/swift","branch":"master","topic":"s3api-unicode","hashtags":[],"change_id":"I9b97e8d03e263ab054091b5901a1118d7ae0b892","subject":"s3api: replace_s3api_access_key_in_path_with_account","status":"NEW","created":"2024-03-22 14:07:38.000000000","updated":"2024-03-22 15:38:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4c632f76bfc6c46f1c59e79229e93489070f542","_number":913977,"virtual_id_number":913977,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-22 15:38:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 15:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 15:38:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a90a3035d84fdc4bd21ade0e6c45e10c29cdb3d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-22 14:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81f5cf9f3b88fd2f401403fc047f2b4a736b1254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:29:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92f829db20644ed9bfcb86c4e65b6cee\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9c1bb873df84d6b937bccecea6bb687 : SUCCESS in 1h 18m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d703ffa4307740dea0d0d861c217e4a3 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cfbf77e5e6e45a286f6e8e32a78e0ce : SUCCESS in 1h 19m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/09d6b72541814eeb9a85b5be5167ebb6 : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a4c632f76bfc6c46f1c59e79229e93489070f542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:38:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4bafd3ca4014fdfae9aa022758b44d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9166e93349ee4e7abb67fd7b902690fa : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6933d5ee5c454690bdfe965da1db4bd7 : SUCCESS in 1h 08m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/db55365dcdbc4faeb98d0ce2b394fda1 : SUCCESS in 48m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/404abd9772b64c0e80e8839aafb32c55 : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c88308beef94689852c68f9afe25062 : SUCCESS in 48m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92482b716a31443a9d66a40f8b7b7fd6 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c4f5dd126a34d2c93c6704a46516a25 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d4c1af43cee4dc786ad4b7538d1d2ba : SUCCESS in 26m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4221d2aeaf3450aa5ff0a344e68741f : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/188b690b57304f5a88b8895cda64409a : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b3f827a18a8463ab58cc085288d0655 : SUCCESS in 18m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f25dc3a5d18140ebbca1675c4bda3e09 : SUCCESS in 21m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00714bd8167b434f9d0def9b5d002b80 : SUCCESS in 17m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/13462b2bd14147cd96d47958bd32ec46 : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f39c3c902bf04fe6881da17b5952541d : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6ff937223d9444cb4a2be1e5d066e5d : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b51f2619788474cad4b27844cac19e7 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e1b630f348a4e79b67afbb531a1e778 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/20b2f6f3f3cf40a6aa7c773d1295c4c3 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca465efb18bc415f947c09274edc4c83 : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c34fa44d9394e6e8a6d2b362ccceee6 : SUCCESS in 8m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a9057a20a4d94446abc4ab765390daec : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17dd65ae7bb04865b8080e44501e2dcd : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45b1059a52dd4d729cfc941f60fea7ac : SUCCESS in 1h 03m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07f101a976e840c4a24df12dc509f672 : SUCCESS in 48m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9f5e2cd5fc4470582f27a76604f5338 : SUCCESS in 55m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d1de6d3bb8449d89598ef3ffe78cb07 : SUCCESS in 24m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8315bd848749849197577a5aed77f1 : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4aebcce98964d3b8b5c64aee85ece4f : SUCCESS in 19m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16d60ac7b6f0459e9886966209809712 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15ca6cd2e665f29d67b94bac71a026c65796881a","revisions":{"15ca6cd2e665f29d67b94bac71a026c65796881a":{"kind":"REWORK","_number":1,"created":"2024-03-22 14:07:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/77/913977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/913977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/913977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/913977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/913977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/913977/1"}}},"commit":{"parents":[{"commit":"fdaabf9a3d77a2772f48944555bb6f2d3bb0bd88","subject":"s3api: Fix handling of non-ascii access keys","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdaabf9a3d77a2772f48944555bb6f2d3bb0bd88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-20 23:21:43.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-22 14:02:35.000000000","tz":-300},"subject":"s3api: replace_s3api_access_key_in_path_with_account","message":"s3api: replace_s3api_access_key_in_path_with_account\n\nIntroduce a new helper function in the s3api module to help middlewares\nmanage the s3api authentication protocol.\n\nAdd tests to exercise the s3api authentication protocol against bytes on\nthe wire and the strings in a stub auth middleware user database.\n\nDrive-by: makes an already working swift tempauth user like:\n\n    user_test☃_tester☃ \u003d testing☃ .admin\n\n... work with an aws-cli config like (again):\n\n    aws_access_key_id \u003d test☃:tester☃\n    aws_secret_access_key \u003d testing☃\n\n... instead of returning 401s in the authorize callback because the path\nsubstitution had missed.\n\nChange-Id: I9b97e8d03e263ab054091b5901a1118d7ae0b892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15ca6cd2e665f29d67b94bac71a026c65796881a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15ca6cd2e665f29d67b94bac71a026c65796881a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
