)]}'
{"id":"openstack%2Fswift~913978","triplet_id":"openstack%2Fswift~master~Ia47f2ab4f92479756a12dda40b08189410e0da64","project":"openstack/swift","branch":"master","topic":"s3api-unicode","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-03-22 15:28:26.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia47f2ab4f92479756a12dda40b08189410e0da64","subject":"wip: fix s3request.S3Request.blank","status":"NEW","created":"2024-03-22 14:07:38.000000000","updated":"2024-03-22 15:28:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d77e886405e5a5c2ad3a06dbfa4f43cb132865ce","_number":913978,"virtual_id_number":913978,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-22 15:28:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 15:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 15:28:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d5a0cf528df5b55824c1d0f9b9d902cc4cd7f4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-22 14:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70c1657f07e91491338e1eb0b08aacd667bc31a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:24:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92cdb9dc80dc49ecaedb87c41969d75b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d19b9ab076144a48ec9618561036665 : FAILURE in 1h 14m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/635e38e66ebd4350b798d2009f241292 : FAILURE in 24m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f640bee9d06f4eb799227b0312589ba4 : SUCCESS in 1h 03m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f117a893c57473298f5ff3e9c255322 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d77e886405e5a5c2ad3a06dbfa4f43cb132865ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:28:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8091430ed5a041ab8124dc0a1ee86451\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1d2b721c69450e89f1061c6d6dac55 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/9aebffbab33d488e943668eb5f3d5200 : SUCCESS in 1h 11m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a1463a947d24c82a79025e480595290 : SUCCESS in 54m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01d54fbe7b9641a59a41959a94d1b39e : SUCCESS in 46m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba9f3877550244b2add21a1d01af2831 : SUCCESS in 1h 04m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23e8fdd8293f48e59059c14f02a37713 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cf6c63aa83a4bc08e10b46a2b106924 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c64d0e3caca94a2293e9aee7253e5604 : SUCCESS in 22m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a10f06fe89ea45fbab5a864b5b0c50c6 : SUCCESS in 10m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14824b18fbc441778870458a9c2a10b9 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc3d92c297064bfd90411550799b49fa : FAILURE in 28m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35acdf42ae7e49a595d2a0f04fbb9315 : FAILURE in 28m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00877b1b8aeb4054b5cee8a3740a7d1f : FAILURE in 32m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f18304f4bd2f43239a3d8d2acd671722 : FAILURE in 22m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b13a42f731e74a358cfd772eb97fbe76 : FAILURE in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5fae084dfb884883bf69d8610b97706d : FAILURE in 24m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af549c0908c04fbcab16c17033600d35 : FAILURE in 23m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f3915cc292b04286a2a74de419f75982 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6594b03091ca49149faa12102e4815b3 : SUCCESS in 22m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/febbecb0f34e4a509cfa5cf620d1c7bc : SUCCESS in 22m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a668b8d2531843c98d7958d8d802ce52 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a373ab5f0bcc4f1d96121dab21e1fdca : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38cbe7505fe4412da3f1cc1f02851dfb : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f89947cc2145ba93944576402d19dd : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82cec417e3474cf89ffa3b1e477e58b0 : SUCCESS in 48m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c501c9cfc084447087f5bd1d488e3607 : SUCCESS in 48m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc45ff4e9d544569bd5cee6d4d5e45be : FAILURE in 34m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6441068924047b5b1d96b5312fe90eb : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9be7550e66444cc3818d5dec0266b2cb : SUCCESS in 19m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c14ac093f1124f83934fb0691e5885d8 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88e5eaf606bfe4c556acc1816feac2a5da5c4663","revisions":{"88e5eaf606bfe4c556acc1816feac2a5da5c4663":{"kind":"REWORK","_number":1,"created":"2024-03-22 14:07:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/78/913978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/913978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/913978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/913978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/913978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/913978/1"}}},"commit":{"parents":[{"commit":"15ca6cd2e665f29d67b94bac71a026c65796881a","subject":"s3api: replace_s3api_access_key_in_path_with_account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15ca6cd2e665f29d67b94bac71a026c65796881a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-22 06:28:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-22 14:05:55.000000000","tz":-300},"subject":"wip: fix s3request.S3Request.blank","message":"wip: fix s3request.S3Request.blank\n\nand start on a few other wsgi-str related fix-ups that sort of jumped\nout at me...\n\nChange-Id: Ia47f2ab4f92479756a12dda40b08189410e0da64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88e5eaf606bfe4c556acc1816feac2a5da5c4663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88e5eaf606bfe4c556acc1816feac2a5da5c4663"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
