)]}'
{"id":"openstack%2Fswift~914022","triplet_id":"openstack%2Fswift~feature%2Fmpu~I7d73b24620e208d1f8ae61fa87b341c8fecbf2e3","project":"openstack/swift","branch":"feature/mpu","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-22 12:48:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-04-22 15:49:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d73b24620e208d1f8ae61fa87b341c8fecbf2e3","subject":"Put a few more links in docs","status":"MERGED","created":"2024-03-22 17:48:55.000000000","updated":"2024-04-22 15:51:31.000000000","submitted":"2024-04-22 15:49:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914022","meta_rev_id":"ae26355b70ab2442b19bfaaba68c87256586983a","_number":914022,"virtual_id_number":914022,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-22 15:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-22 10:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-22 10:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 18:48:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 18:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-12 14:28:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"bb5267e66e23661015e5843ac43b44bbf608e728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 17:48:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbb0929d8ab49a3e891b3292b686ca0539a65818","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:48:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b444f3f6165474fa5e3571dd830eff5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2018b3842171405ca3694926aef95a97 : SUCCESS in 17m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e4dd3db02eb47318827bfdd1e5e5573 : SUCCESS in 57m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b67331b01210466bb9b0093e02ba15f7 : SUCCESS in 20m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/90aa1410182b40c89cdc9305b9011764 : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8d7ed944374fe6db200d7261e26864b2074b77e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:59:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5529e4dd37044a696ae40bd97458b38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdaaf5952644196ba5aaf07b930a238 : SUCCESS in 8m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/647305a0fd6249029ea366f2939a3db2 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97b535ee226c4e15b7c8e457fa59236d : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e92948edd6441e5bd9d9b57a578080d : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3eefd295206407da2768444e79e475a : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91a83eac737945da95a361f059641bda : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bfc14df066624498a5414ef44222a115 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/673b01f91ce04d209817e84a855aab04 : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85cc9e0ffe7f418b823494748f4798e8 : SUCCESS in 16m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2faf340a444647529e23c93a3722b79a : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c3c7abf2b55842edaa458601301b1a1a : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/850ac8b400e84bf1b2bc1f4f7e2ee55b : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23ae1b589320497596b33e10cb3a235f : SUCCESS in 15m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb378e940a7b44d68d03ba2326b4ff2e : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8cde173572964bd9811af9153e69c08e : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da696134e9dd4ace8ab9d8df07a5a65a : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78db35679317433a9c88326cdcedf986 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff767a7bba8b4e4eb01f51c161e79bb9 : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3fe0badf7d740408891135ef708f5d1 : SUCCESS in 7m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/09acbf05091a4014a26d78cbe5600eb0 : SUCCESS in 59m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e77ce4487d64835a31a8693a4d00f48 : SUCCESS in 21m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e4a0a73fdad419abf512c472679a527 : SUCCESS in 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3df8cff480c9c884cd4b255a49eb847bb8365b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:28:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd5782228b2ca5d46366d96255a6f734f86fd697","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:28:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96bc3281c2b5f37bc636c99b9bbcfc03185b2d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:43:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21217bb6b9c94742840f248f88f80b8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7552df9a18a4f17a9e072ad84c86f58 : SUCCESS in 10m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/204f4f4148ee45ca977851ea59220d94 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2758a9eedbfe4425809fd4824a62c1a7 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/348499da4d0f4733bd7a2a5a6639a913 : SUCCESS in 24m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7b6bdd2106647a2a71134efd4c99df8 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/11419c0fb2cd4a8dadb7faac38f72340 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10f2391669514e518eb982ea2c628075 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2768efaeb95f4a708a696a98f1a7a207 : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8580ca811c6342719fff2bf0641847af : SUCCESS in 22m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2426b9f1467048d88488322b80d86ce3 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a10e8becddb244658496327318d75aa5 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a86c691841b429284617f6ec729d878 : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51ce5d242fcf4982a9aad2e3276e3d54 : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ab21446823e4078898e1273992b0d84 : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ef3f90c8463e485c826cc3c906c733f1 : SUCCESS in 23m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b20b2bf3d152418fb0fca23a2cb597ca : SUCCESS in 18m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e4aa087d4e349a88b826541759f3bdc : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa8d37b43adf4a938d7d47328af82947 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a810e3483f248c29c4dcd398944ae2f : SUCCESS in 7m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3eb65a72eaea4c7aa226e4d8eb7ccbf9 : SUCCESS in 1h 08m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/439f2b07443e49c086e7b102c7ebeb7b : SUCCESS in 18m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ef5db2b52474ca09b479b0d0441d544 : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eaf1c0cfed47159c9b4be827f5109442e311471a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:52:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff47e51023c494c9be1d151711008c7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e57c852336b2424aaa221e4b3793c2d6 : SUCCESS in 1h 22m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f6cea5f5ce44ac68e2d7d651c51b556 : SUCCESS in 26m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10de13999eee4b85855c55c645fe91eb : SUCCESS in 36m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8e25f13116542f49329ed2d664bfdc9 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":2},{"id":"bb9b91ca0608448187258a38a5e710dca68190c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 10:50:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"73c8cd3861dfc95a6cdc5c892dbf5704c4e9007d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 10:50:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ffc0476ff00e9eac43ec8f979e81903222479ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 11:55:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a54457c097b4a4c8bb5fb3b564b5958\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd21b2dbd59e47179a15e03acaeded83 : SUCCESS in 9m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c774b1fa2f9b4ed499e67944ab2bfca1 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e86af63b6c134cb3b02fd1405e4698a9 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1f01e48363a46c98b3aeb2e44bcd4e1 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34ad73dbc5e74e64a04e3e6b1d645b14 : SUCCESS in 12m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aed0a38c250d457ea64604d37f9631af : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e0606afc2fb4471bc148924651e094d : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97f08ae6349a498d91ca09532f46e5a9 : SUCCESS in 18m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fd51167f2edb41bbab18ce9744ae4780 : SUCCESS in 17m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a496bea81dc34cf88064bde181a3e818 : FAILURE in 22m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/230eb1bf6a854cc4b9522418884eee91 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73e6e0280eef40e782e1c4709422b956 : SUCCESS in 18m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae3b7a78f63a4bfaa76ef469c9eb60d3 : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1bede892a4264a9da21346203cad5258 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b257dd36e0c4a708c808772d79ca231 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/caac73ad87c94878830322e9a29e67fc : SUCCESS in 17m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e69f1ffbd9c54134872354cb36c74992 : SUCCESS in 5m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58d2b1593200403e9a52cb0476709bdf : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e7d8dc078ae4ce2ad630ba00c0cac56 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac863854e7bd460b8006e440e6b9a097 : SUCCESS in 59m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e00cb7061bc04d5a93bf749b658371ad : SUCCESS in 27m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/983dadbca6984b51b13462ab1d72d059 : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb27fcf3d10894366ba994045c963e93e15b7692","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 12:17:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7fb3b52fa1740608ce2f79e64b4445f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af94173e7539427dafc825bae8f3824d : SUCCESS in 16m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45abe1de2068434eb965b3f3a9ca14ae : SUCCESS in 58m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe70bc1de945445cbd769b72adc468d9 : SUCCESS in 1h 24m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/415a9a801339463ca6fc13e3f6dd556c : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":3},{"id":"558924fe0fde3f41780042b2bd8e055ee6eea3a9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 12:48:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8011292d697cc47d4dc8737357db1f49931b21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:47:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1376718324842faa4e7fe11196909cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7b0007cd6a43c791122bd34b856528 : SUCCESS in 7m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/412c30dc24714e6a83d2302ddab51bb9 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe179cfc885a43c78c4baa44e9d353d8 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99565147932a49148215da2608cce980 : SUCCESS in 24m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db53791cb66447b6a30c8c591621c11b : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/98380e7b7de3431b8c01782b3760f89f : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c7abc49b5ad4b268b34a6226fa188d7 : SUCCESS in 18m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6de405e8f1e24ab798c28b059dba5698 : SUCCESS in 22m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0453c67a914b4cd5aa1ca6c490a76b12 : SUCCESS in 17m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f426198de28644c3acec4f568bf9c4f8 : SUCCESS in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/897ff200c694459fa0a24a2aaad93477 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e4a1c6dc244481ea6e2945decc5f0b0 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72bef9bada684c9a809a20401b14bdbc : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e04292f1606427390ce158d7184f955 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f5d0fd8556ea41a2a236cda9a42e9b5a : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/04fdad662377439e9ae6b444cbf20780 : SUCCESS in 14m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/518667814cfe444dbd9789b68cfd5f76 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/53f17ac6617f477a897ee47066a24460 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efda533532454bcba98be3bb7ee16bc4 : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc1d1fe5846c41618efe40dbc07b8c1f : SUCCESS in 57m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56de78dbfc6e437aa544a154c93146f8 : SUCCESS in 18m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb7f67dcadc14605978a7edacf98ef18 : SUCCESS in 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2bda5dde7b768974e49f8dae9abec43a65e5d5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:47:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"09f859cd5066535e30a9ccb91eac2c1353900297","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:59:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d4fee6379c4f5dbc0bcafafa51e119\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/928327ec41a244cbaa8b41c8fd6d9687 : SUCCESS in 15m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5e4ffe32dab426da7529d6c3ea05734 : SUCCESS in 47m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a659614b878346d18b14a347c1c54c8b : SUCCESS in 19m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a0d99e82f6d42369cfa82aa77b4d400 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":3},{"id":"7eac4865bbfb2700e954955ca930c3844234d49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:49:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be63d6a8ddc544019c152ec7064eeab3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045561b8203a452a901394a94a01a9b1 : SUCCESS in 7m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d7cae0eeb2b8467894181cb2bd3fc573 : SUCCESS in 10m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/abc43528ab7446e5b5c53194fb243603 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca68df1361714d32923b97a994136d4c : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae7f14eb1f704e88aa3da34cf1c28476 : SUCCESS in 18m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab2f54db5aa44d039867a94a9458b515 : SUCCESS in 17m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5f8dbce1ba92431b8c1cb8787254c52f : SUCCESS in 17m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed6e7b65e27446d2a5c1e87a97fbdbf2 : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92b434be15db4c5aa810b4afadf0f2a0 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7caca46f0d245b29582ff6716a3b79d : SUCCESS in 23m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0870b369828443cf99e3abea3b904c8f : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddf98b14e549473d8356e0b2f0d9b777 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9946d214cb1246deb9a453bd57c39af5 : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bff7ef7e21b04300bd935f7f30c9cfd7 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6fe5d003b238402a9d15deafb917ba4f : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9a3fc571ae34bd2919fbf2639704253 : SUCCESS in 58m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81a36cdc49e34b5e8244d58a7f8c69e0 : SUCCESS in 23m 13s","accounts_in_message":[],"_revision_number":3},{"id":"24217707d12150214a0418315d1ba83b2785ec5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:49:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ae26355b70ab2442b19bfaaba68c87256586983a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 15:51:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbcdc5a801644a5a872b39a1c13efd24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e595152ce3e8490cb6c6b9f4192bc437 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4a8644647f1648f6895f7fc9263b74d8 : SUCCESS in 1m 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f158d63334a449cfb1a7b8a3d470e337 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3fbc8bf04a2a0f4fda9da4b629504af8ee6be98","revisions":{"a64026a26bb9d56ebc9b5c219e66ceabfbc1dd67":{"kind":"REWORK","_number":1,"created":"2024-03-22 17:48:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/914022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/914022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/914022/1"}}},"commit":{"parents":[{"commit":"3d7a6d0b0558a13892383e6288bbb1eb5846ecf0","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7a6d0b0558a13892383e6288bbb1eb5846ecf0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 17:47:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 17:47:44.000000000","tz":-420},"subject":"Put a few more links in docs","message":"Put a few more links in docs\n\nWe probably want to re-think the whole overview_large_objects doc,\nbut at least now it shows up in the branch diff to remind us.\n\nChange-Id: I7d73b24620e208d1f8ae61fa87b341c8fecbf2e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a64026a26bb9d56ebc9b5c219e66ceabfbc1dd67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a64026a26bb9d56ebc9b5c219e66ceabfbc1dd67"}]},"branch":"refs/heads/feature/mpu"},"1086c01727b5cfc2391fa19afaa5e7feab7ee0db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-12 14:28:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/914022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/914022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/914022/2"}}},"commit":{"parents":[{"commit":"53cb2d2457802812b13704df2622d1968b6c42fb","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53cb2d2457802812b13704df2622d1968b6c42fb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 17:47:41.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-12 14:27:56.000000000","tz":60},"subject":"Put a few more links in docs","message":"Put a few more links in docs\n\nWe probably want to re-think the whole overview_large_objects doc,\nbut at least now it shows up in the branch diff to remind us.\n\nChange-Id: I7d73b24620e208d1f8ae61fa87b341c8fecbf2e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1086c01727b5cfc2391fa19afaa5e7feab7ee0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1086c01727b5cfc2391fa19afaa5e7feab7ee0db"}]},"branch":"refs/heads/feature/mpu"},"a3fbc8bf04a2a0f4fda9da4b629504af8ee6be98":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-22 10:50:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/914022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/914022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/914022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/914022/3"}}},"commit":{"parents":[{"commit":"5e2ab1cd5a529590b810e52d340f9deeec858458","subject":"object-server: include deleted metadata in PUT/DELETE response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e2ab1cd5a529590b810e52d340f9deeec858458"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 17:47:41.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 10:50:17.000000000","tz":60},"subject":"Put a few more links in docs","message":"Put a few more links in docs\n\nWe probably want to re-think the whole overview_large_objects doc,\nbut at least now it shows up in the branch diff to remind us.\n\nChange-Id: I7d73b24620e208d1f8ae61fa87b341c8fecbf2e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3fbc8bf04a2a0f4fda9da4b629504af8ee6be98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3fbc8bf04a2a0f4fda9da4b629504af8ee6be98"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
