)]}'
{"id":"openstack%2Fswift~914029","triplet_id":"openstack%2Fswift~master~Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics-patch17","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-01 15:02:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-01 19:49:34.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-05-02 00:27:02.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-05-02 00:27:02.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-01 21:53:31.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a","subject":"Refactor utils","status":"MERGED","created":"2024-03-22 21:04:33.000000000","updated":"2024-05-02 00:28:10.000000000","submitted":"2024-05-02 00:27:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":208,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"914029-tmp_labeled-metrics-patch17","meta_rev_id":"966ea4739933c1885e19f789d8290507c12963d2","_number":914029,"virtual_id_number":914029,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 00:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-05-01 16:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-05-01 19:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-30 20:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2024-05-01 19:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 21:24:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 22:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-25 14:34:01.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-03-25 15:59:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-04-03 11:28:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-04-10 17:59:29.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-04-25 14:10:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-04-29 20:16:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9cf1378e7b5a29c7ed08cce93b55c5950eaa7fdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-22 21:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf8d8eee680fa3049233e1bea7bc92a73ae5484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 21:24:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0e78b48d5c4ce7b93534d0f6b82548\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a623b10865c242189e4fad4d386d844f : SUCCESS in 17m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e743e02d5727472c8bd89f3675363726 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/757cb6a916a44f918cac5c8826f89268 : SUCCESS in 18m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f0b6262f9b84957b6778ae175fba5c0 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"989dfee6fa80e9cd2312cbc88dcf520bb10740ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 22:12:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/204bbbd9efc844d19147281e6133a438\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333fe4d56ac449319b9200a08d2713ea : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/45c22229b563486e98983f3e43fc0e68 : SUCCESS in 39m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b5a8ccbf22b475a806ad97ad21af241 : SUCCESS in 39m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c09babafd608449baf6e25248c506ee7 : FAILURE in 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca6ce4ade6014bb4867ae2b39899617e : POST_FAILURE in 28m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bb8bb339e8640f5b60d462649843adc : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e83fcf89eac5420caee6e3e5b14fdfa9 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/113ced19d53d4e41bb72a912d2d8dbd5 : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/031cf8d458f24ff9bb9d10bad1f3ce3b : SUCCESS in 10m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b07991ef9134905aeccb9f8afa0df74 : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e05c0e01d0f47849d9c47773565186c : SUCCESS in 19m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4044f76f2524e368b01a81440502047 : SUCCESS in 16m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3ca5bce1c55049d8805d07008d4d199c : SUCCESS in 16m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/24501a797a9d4cf888e83b1a8e0334dc : SUCCESS in 16m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f56fc1bbed5471ca404438721c988cd : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28355fd756b242bba14dea22a73d8825 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe00d9dec1e74f3b8dacc24780bb49af : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6dbd1f7390cd45838880a9dbb3b5d2b5 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c51453049c04a4280d5a6a509d72a26 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8fa72d1f483947ccb833e39efacce4b4 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c58e11fee99a463f997157185b7c8f63 : RETRY_LIMIT in 3m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/402a1454167347a79f21141fa80c043e : RETRY_LIMIT in 4m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c194ffee6cca4f7cbea242453fc4bc7b : RETRY_LIMIT in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3c3849f3f05459794336a7934679ad5 : RETRY_LIMIT in 4m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fe461d677114fc58192477cc5ae464d : RETRY_LIMIT in 11m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21a446b5b395414ab9d471d40017a478 : RETRY_LIMIT in 8m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15655b7932b5403f9acf015b70723db0 : SUCCESS in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25f30d4702934293b6f21ce1eb2ebfa8 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bcd2f58002d44a629a6a7e73f80dee43 : SUCCESS in 20m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/905fcf9901db4098a7acbeff34a0fbe3 : FAILURE in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3d3320f16fb9b8b6bbb1adcc0756b87c1ff25c83","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-25 14:34:01.000000000","message":"Patch Set 1: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cf0e0f37a2d530e5d10b22ac6ef1fa9efe58f830","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 15:59:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad3ef840b2573932838f3269eac98d8a5b4a2731","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-26 18:01:07.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7a153f9372fe83a77c9f5cf7678776668053373a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-26 23:32:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eda1482085aa339b454fc6dcdf2485120da8c8b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 00:51:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e0e6c36459447283da8f7e5475cb8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0aecd2f4e947c58c761cf5b6723850 : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b20f5a855f764eb6887b8cc49befd5ab : SUCCESS in 1h 11m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8390ab05187a4795bf2d4a87765d7836 : SUCCESS in 39m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51b99884484540c480c84e3fda72db5f : FAILURE in 14m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87914462b7714e64aebec79c5576ddeb : POST_FAILURE in 32m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8532d60829d94b81af9c116ad8ca7650 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6d541b6b44c4b3eb645ecfc1305d315 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccdcecf66c9342d0bf5af53551b1d1b3 : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b96f30a537848c98cde17f2a0718fb5 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c43e7caa00a4f6fa8911e471d8a0947 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fe95c48d41834b52aa68b941d0db20c1 : SUCCESS in 19m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f34062bd0a2640ffb4a28b564ad99de2 : SUCCESS in 17m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3cd766d8266a4270a12161be1d72c1fe : SUCCESS in 15m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5b8c10fa0f7947d8a0674d36352d087c : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6684a047a77241ed95ee99570d11030b : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe824367cce54931b118bac9910a9afd : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ab48896e8624bfb949b44e5b60a8209 : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50c331d222e1429bbf06ddea70e14dd0 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/815fc492286a4c1c85ed249af86db0c8 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c562962cd0574b1686812ab1fe05f4db : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8e99b70bd4e4989963a540aa38fd795 : RETRY_LIMIT in 4m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e054f13c430a4528bf81518af8dce1d2 : RETRY_LIMIT in 3m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38fda5c1861a4f6786d8297230d62945 : RETRY_LIMIT in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f98d303448dc410487d917269434f91e : RETRY_LIMIT in 4m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b2aa384bc30479cbeae7b75486b4f9a : RETRY_LIMIT in 8m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6e75eca1ffe4040a823f0880f4b07cd : RETRY_LIMIT in 8m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4716fba788184a8ba4d1610aa8836f0d : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b8951bec38431381708134ef850568 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e0714e341b545bc9b6771f37af19df9 : SUCCESS in 19m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78458565a49f4aab95aee0f652e24b1a : FAILURE in 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e668a0607a8ab0dafb453dc440f3b030834918bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 01:05:40.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef06f4e558b46b99e28baef1f6c990c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6783a202404c403bbbfd252eb0488557 : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2452d256a23e4728857d4d307f994fdb : SUCCESS in 57m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/315c12a468264d18b62abcc84874e4d7 : SUCCESS in 18m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5eec0c4a14c4e3093d0954d5e56cdac : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"666f9d6e406a136ab772ea36f528cce2da3352a1","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-03-27 19:08:51.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"afc0d44cd716bebf1b07eceab656eda53bba8d19","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-27 19:11:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"289c723dce2b98be284dbceea3ac0f313788a021","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-28 00:24:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d845e97549c4cf336d1b6d0ac0ff1f9b6480fe7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-01 20:25:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d991594720c4da957665c522f53714eea9db2817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-02 17:38:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8f199cf48ed0bdc5fa6c4d68dd8856946646ca54","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-02 17:43:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a2e1ba0f0c209fbaef32d60bbf0369b9816578a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-02 17:58:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b0cf45e611135d8dc167c790c1cf2156a3242dc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-02 18:54:42.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":4},{"id":"80b453516f6cf02edec0fa487f47d73d98ab91b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 19:14:51.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ded2747772264766aa69004771826bbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c807e1c35094c4fa7a742d8daef42cb : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b045160776d74ea88e3ed12f64d57b12 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/151cf9a349594d47b55aa2353898cf03 : SUCCESS in 53m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4402f8a7a664b1a98306ff595581944 : SUCCESS in 1h 01m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccaaa21262a9499b996e50c2c2fe6999 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9362f3a34723473cb4b23f9332f8d2eb : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/864f0a84e3334b11958b0428faf52a6d : SUCCESS in 22m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db3e69d47d2a49b68802acfe3c651e17 : SUCCESS in 11m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/533b76adce134f0f8fdc538cede99bd7 : SUCCESS in 11m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52fb8b0cbfb7420ba5d5d2e87acb61df : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23b7fee927a7437fb058627601af8575 : SUCCESS in 21m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/28b3272a04be4ca39328a743f71860b0 : SUCCESS in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/283118ec5272415c93ca393b074f0a3c : SUCCESS in 19m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff31447f743846f8bd1efeefbd14c00d : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4610a8bbb42a4d6fa48fe3b2742eeb54 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/26c1f0844eed488cb4ee1b720b98e4fa : SUCCESS in 22m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/353a0e70fc594cce93bc6d1c289b5411 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6fd3615032f74071a4c7f1384f4ebcb4 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6404a22cb7e419e94b51f39c0e47c1f : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d6084a1e45145e9bfbcd2eaaea9732d : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d374a4a830a54055a83e528f207e6705 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57c2ea9a05664d2c8ab20d9bb8e165f9 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3483baa1cdf84756a57c0fc28318e56a : FAILURE in 57m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95d664fd321242ba8210fb53f72b9b3a : SUCCESS in 44m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0383364e495b48a693f4f92e5a98daf2 : SUCCESS in 42m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/453ca8a8174441499cac7e5ffca5cfcb : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6f2567fe77476ba16835ba2ec4ca90 : FAILURE in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb1295cd9bd94a5f8fdf38e0fcbd6dcb : SUCCESS in 19m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07321671514f44c8b64085e3fcc9b7d8 : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":4},{"id":"bb622372b1e03143933f14b82a5df86920efda14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 19:27:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77817d1c9b042109749a662b9c26c55\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4c7cb4383f34924a59fc6b4f9bafe3c : SUCCESS in 16m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e08878ae91c4a4e8606bbd4c8971df9 : SUCCESS in 44m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9599987fa1494360ab9a070e2d3414cf : SUCCESS in 18m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bba7c8e14cd641f48e73a92a07bc25be : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":4},{"id":"83af21036c554d2fc0b64053d590988d14fe59f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-02 21:40:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3ac7edb1606c7586d47064333dce8b406bc95242","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 22:28:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdd88555c8d4f85aee0666fe7b992bf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfaf87defef84be8b2a1de683535d438 : SUCCESS in 13m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a373b2b29df145da93c1be0a1ab597ec : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a6d8fe14d3c470caeb78009a391b634 : SUCCESS in 46m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb056c5fe7874442a196077b880a8ff2 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":5},{"id":"646f443b049ecd40602a9de09470ed51e9827a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 22:45:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33a9311eb37547db82d6c00a3140367f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eee35162817462f965496cb0e401f10 : SUCCESS in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/71011e3c449c4d1f9ea38dcb07bed2c7 : SUCCESS in 44m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35305536f7514d409667b337614a7117 : SUCCESS in 35m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d3c4cfb22d4e1ab4f1312a7df004fe : SUCCESS in 42m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4726e2629304404fb262e1a9bf926258 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07dde15f19364e228266e670f091fa12 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f865d8a60b5d4a36803c26bf0b1d82a1 : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20093072a2614b6f9837295b836242fd : SUCCESS in 14m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14645342a5cd45a49701ea6579683a41 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8e32519fe3e4dec8a3e140f2dd6ee2b : SUCCESS in 19m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ddc0846f1ad4917b1ed64c01e856a87 : SUCCESS in 17m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/589c8f14b0d84129b20daf40c1dd33e3 : SUCCESS in 20m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d73407e6b82541389f3d9950d9d0b526 : SUCCESS in 18m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b18b308e63d648dfa4351ba210bcf35a : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/acf60e61afe14c42b2361a156d9c223a : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f4fc62f48394b04a547905bf80b8f87 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bd0c8bf7df9c4923a4ee577d3a532476 : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/01f8afdaf5334e3e9814daf88fee589b : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/993f7829a9354e65ba44c6c41bfdd289 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0fb33fb0c7b4f27890d762622f0bfd8 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fbfa5e9e5bd44ea9b03a08757b09bbe : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6a96607b43f49218b772735cf907109 : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5da888590bd14da28b353911f09a48d4 : FAILURE in 59m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7940af9dad3240e1b2467e35cb34b200 : SUCCESS in 40m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e02fb21472c400983f4fda1df0eaac0 : SUCCESS in 44m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bfe142de12949aca91c09f44e6a2ec6 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afbdb0b75c294439a26743f6552604cd : SUCCESS in 3m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76a89175d7c24dd280b57b47f2232b51 : SUCCESS in 20m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07d5146ed117425d8fa98be62ca5c11a : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":5},{"id":"524c519dcfeba3496fa07d5109f0ed12c536e9d5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 11:28:23.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1fcc46da9ce743220706f1f1db3358f5ec062769","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-03 14:46:45.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"199d31c06a561a9873c5c6f7fe206b909d4d62c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-04 03:23:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a78bf61cf68f5db1ee5deba3f5b826fb5f8cf43b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 04:22:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1694d55301548c0a29b413cafff46ab\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bfe7c21259940d589d93d8c71166d8c : SUCCESS in 15m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a176ce7b56ce4e5c83c3035b6057cc9d : SUCCESS in 49m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0586818f94949eca66fe31db12848f4 : SUCCESS in 28m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/89857dcc06ff4546b7060e70ce22ecbc : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":6},{"id":"2f27775cc9bdf3fe26d7ec9611e0b274e5a7fdb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 04:35:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a8ded9ca5284839bc7628563b3d532a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1149b9ab27d45f283ccbc1d71360efe : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/a3db7b65c23c4202b6f5b3750d93f1d7 : SUCCESS in 57m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8f5736fb57e4bb3b3f9c7f9e5485079 : SUCCESS in 39m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e91b1b7300024206aaeba4065e052926 : SUCCESS in 58m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d69521d0e84341188af66666d1c374e9 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc9b4d1a701444aaa7b0757024e19fdd : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13cd890234aa4487b949f59703202e30 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f88e980ca7ae4790931f4fe8733db33a : SUCCESS in 12m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05664f2012234600b2465c992f980c8e : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5fa460791fc24936a44f13d1d82770eb : SUCCESS in 21m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/573635f867214b1eb1933867702a4a5f : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6d64ace2601d4d829d0735d24fdda22c : SUCCESS in 17m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3858eb537d494ed4a6a156c05f48ad2e : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f21bc4f393649e99ed23e4909c1ca2a : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddf40c1829db40b9820a58ecc4a8e029 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/db7aab929f864953818b0ecfb01805c7 : SUCCESS in 14m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/700d0985df8848dfb7b331ea7a95c4c7 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7750e3ba5a344c39aa10532c34f75796 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/79f0190333ba48d6b5a52fbf92b49ea7 : SUCCESS in 17m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbb411514d444989ac4d9536b7527287 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aff22bc34eba4975a46781bdcf6c9621 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/997f69e797ea4b9ca183ac70a7432f3d : SUCCESS in 10m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48b49404956c41eb99793467ba00b48e : FAILURE in 55m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6123909fb8974441aa474d3428cd0039 : SUCCESS in 54m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae2089fe878a4e828b4d30c3241d286d : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d3e3f5096840b581ab08b33798e1b3 : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa5c17353cb47d3b84b8206f5e9eb12 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/488333803d714df6b609144f070591f9 : SUCCESS in 19m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f31ac93eecd24c6ea8434e3ea8708e2a : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":6},{"id":"86962ff07395fde539de7a15dfbedcf1d41ff36b","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-05 17:12:36.000000000","message":"Patch Set 6:\n\n(18 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c64df26b784885a10d5eb63c6da0e5fe194e7780","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-05 18:41:42.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b4d5304e059eeadb27188f1291b9d74b191d1e2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-05 19:34:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"88660bcf41d0899e134ae0522e2264ad17deb6e9","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-05 19:41:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8950807a36452822b7d828691970d43c4b87d528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-07 05:21:27.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"71c7f5494a1efa8f016f3722ebe3a6ce4644ab6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 16:39:29.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd84316407d54b97a3296dd361db4a02\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/500ab36a0f744245a52592fcdf264045 : FAILURE in 37m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87b7fa3b261a445c97ad9c2ac3babd14 : FAILURE in 15m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5942a7dbe0614ae6a6271c429c3e1ba9 : SUCCESS in 51m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbd73477d03f49459e577b54e59e203a : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7c90a94a12c7622d406b8fd0418ce9642d092d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 16:52:51.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0226ba12b06e413296f3dc6313eb7134\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235b2a9b384846708d659bb5714d910d : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/72809483623c4f67859b2a1d95726780 : SUCCESS in 1h 08m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2458182224e34a7bad2d0a637987a3aa : SUCCESS in 1h 02m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/399a3ce6a7ee4f489086277e6f203623 : SUCCESS in 57m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b860990c6e9c49eab73880304c4a20c9 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65101bf337b14df7bf9b7a9db27f1946 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69051bc98acc4b189e3efa204b824624 : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64b9cd5adc454db4819f3b2f0c7f7143 : SUCCESS in 41m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6232332c6ca47969ed33e1d226ebdc2 : SUCCESS in 41m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/999f58c4db304bd7acefeddf6e622213 : FAILURE in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ca8d5b88c6044638deee4ef2e324cc0 : FAILURE in 13m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68d2b087c05e474f9e293b5b8413062f : FAILURE in 17m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f7231cfda4a94d25b667cc75b3a0f80d : FAILURE in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1fd57992fc3b4b9e884c0b298113d67d : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0b7499641fa43d3bdcba948f85e8522 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51aaaf9a7a934821a8b1acc0580559c2 : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6980a5f146a648b9aa479210ed930305 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/771d2fa6e48046db9c1068420a3fb4d6 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1945c4fdcd5f41ab8adfd794256dd5ea : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c78b2836e3974eefaa7b3c72493db1cb : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec1c279dcfe04c4aba72b2b4c0d219f1 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/56635061d9284525897f48c717c3f6bf : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50b867e759574aeca522b4e8ac2868cb : FAILURE in 1h 03m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b86b95ceaa84d80969f49ba540e89d4 : SUCCESS in 39m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4323682a2b84573a9db725f94710e61 : SUCCESS in 47m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc4faae169144a6bcc1b34fbcb683c7 : FAILURE in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a765e8faa4c044209dbc8026f49f7e17 : FAILURE in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f40f0ea3ee742dd986f3341e47b6b02 : SUCCESS in 21m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e10326bb8a9c42abb9822b320ae2c87f : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":8},{"id":"0d2dc404ec79960a807322d4b93caff0be2a0e83","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-04-10 17:59:29.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2a425d065805a57e668cf8a095677c91c0c427ba","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-11 17:33:27.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"98f2b844d30cf5b74b1ae33351bb1dc569df8167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-11 19:41:20.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ca1b8cc2795a94c9d3eceb7e2e83f1e208d0ab18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-11 21:08:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d478c03bb866199772345b5c65fd11fbe4cae5ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 21:15:20.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5516ebb35cf44685ae6ae109d378341a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4551957096bc427e8083ea96e13d9127 : FAILURE in 4m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cacd46d98464b359db28db2bda089f6 : FAILURE in 5m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c6f47153cfa4c5d8549bacc1c36fae1 : FAILURE in 4m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/64a4fba0b3114f5e95289dbe048231a4 : FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":10},{"id":"d048c5559cee454ab8a746aab4e381c3f75dce37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 22:17:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9993e074b17044868f4787fb16b21207\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f1ff4daef5b94eaa8f8ec8ddbf50dde3 : FAILURE in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e5438f365048079a312e3eab94e60e : FAILURE in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b043792f5754bb587491ad0d270dcdc : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9b71f3b23f5f46148ecdc82f60cffad3 : SUCCESS in 53m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d8bfd44a7854da2be20008ce1809fb1 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f001a37aa240a19d573d094735cadc : SUCCESS in 1h 02m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8439489c72546358c3a70f611804ffb : FAILURE in 5m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2250f2f4f904f6b9f575b868fb08fc9 : FAILURE in 4m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6570781d49244f587afd9bc398debce : FAILURE in 7m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff984e155ee147409890e050bfccc0f2 : SUCCESS in 11m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d7c4f0d3225465cbf707448eb52a84b : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2d26bdcf26c4eeb90f60faedb2edf64 : FAILURE in 3m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd058b47c25441aba55bb7615d43b01e : FAILURE in 2m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f9b3ad999d848ea8db9a71bf9b7d3f4 : FAILURE in 2m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/70cf16caa00c4f66971f7d5714b7b86d : FAILURE in 3m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7514a5f441fa408ea20ceee4bd835ed0 : FAILURE in 2m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7bb0f7683fb144629942a6968342c4e3 : FAILURE in 2m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d33eb3cccba44288fd1edbbc11b5e1b : FAILURE in 2m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/663a5c20c44446b892b107893c053130 : FAILURE in 3m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f2e91c23db274b00afdc91570382a45d : FAILURE in 2m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/de52521951ca44e285172573ce34477e : FAILURE in 2m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f389ba1f3804486385483d36182e6b94 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5dbbf7cf527542f58b7ad9c36cdf406b : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a334a6606264baeb22a6bbcd35f7a1b : FAILURE in 5m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d0f2a316dd24691a9d12c0cc8c81690 : FAILURE in 50m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f70a1da6c54494999d8006be2f86189 : FAILURE in 22m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7be0b20ac95345dd956ce623b060363d : FAILURE in 20m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21eca793bfae4e8b95e84a3ba069bf59 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a16c1e27b64b8d9624bf6d9fec53ca : FAILURE in 3m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e6e54b5a52045e3b544ae9697b29bb4 : SUCCESS in 23m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c799ec4f4ed4b17b0fb0aa205695d38 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":10},{"id":"0a6285e9c606a40065991b29027e3baf92bd8ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-12 13:31:52.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"0fd4944916f1fedb371cb9d3ff6ca240545f28c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 13:37:37.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed9704b2e8f47319e0782f444d7ef0e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d321b3cea499473491daa97a26646bdd : FAILURE in 3m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c790fa885f94c05bf21a69070fe706d : FAILURE in 3m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2e33b0fc7964120a831b3934505016a : FAILURE in 2m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/42f5a8ca04dd45c5aa2ec41c07450d9e : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":11},{"id":"2f01d785cbe5972b4e99ef7ca9308d7df01c6395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 14:37:01.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/970976bf25524e8592eca53a8c02bcb8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b9f04b2cd79e45b6b6eff852496d7513 : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc500563c367447d817fbd47d1f7ec0f : FAILURE in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ecabc29ddc2470b81615f34b3d2f9bb : SUCCESS in 6m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/397cf213099b4c04b098b31bb4ab667c : SUCCESS in 55m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08b7946eaf2046fcb89783f479739ba8 : SUCCESS in 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a376a1610936469b8539c5a347c27571 : SUCCESS in 55m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/637eec76fbc94c38885413dbeac12e0c : FAILURE in 4m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/408e723658f74816b96c570a40115b7d : FAILURE in 6m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db1e8ff1f75f4d3fa9e8aef635ca7c54 : FAILURE in 4m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/833cb0d6246c4d86810f0e0c5eff0d11 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1cfa02abddf747eeac2d89566f890b60 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c54e2c06e784e15b7a0715e9328ce90 : FAILURE in 2m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6250cd2411044780ba14c2f18b3b9ffe : FAILURE in 2m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9f9ba1adf2b461894f70dcdbb570257 : FAILURE in 2m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/655faa903b5b4ec499e43081cef0b4b8 : FAILURE in 2m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b06766da172465c851e31a7bc0986d7 : FAILURE in 2m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef64d56f449a4f418b07189fee138535 : FAILURE in 2m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/715f81b37d734e928502940222698774 : FAILURE in 2m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b6892ec6ee144d39fea0d21b8daa79c : FAILURE in 2m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f96d28a6883340188ee5dda6a666f3ec : FAILURE in 2m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f80e4df821854a29a020073382fdb2b3 : FAILURE in 2m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d131beebbf114a988f85866f49959f84 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df4009f5b4c04025b5aaec61950907e7 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6cdd71e0ea96403981bf55ef1284b21f : FAILURE in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0af54a91268c4a20b6c0a9bb4fd58ab9 : FAILURE in 59m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3070219789244bc1b6c7a0535ad47b84 : FAILURE in 22m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6ff7f1f972841b882307d3595c55b86 : FAILURE in 13m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6de6ba9a8e8427b880964c57109be7e : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dfb177cd4bb4cc2b61965fb05fc5406 : FAILURE in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44b63ac6bf704372a48182e8bb0c9c5b : SUCCESS in 26m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1e44975390c48ca80d99ba6e27e7f1a : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":11},{"id":"40dbf650051312b4362217c2eb0c280f0a38cfb8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-13 14:11:58.000000000","message":"Patch Set 11: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":11},{"id":"805cae5b9d6f9f64178eb09cb0ab171456572a16","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-15 13:05:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"69c0760c5e225b5f348485e4b0ee3e5a7dd85736","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-18 16:55:49.000000000","message":"Patch Set 11:\n\n(19 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ff4b40200dcb2101c7255f28993d6b22db849eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-18 16:56:09.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"ad026c9f3b18af307d73e9e8ba85cc024154e16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-18 17:34:03.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2c3b1a0adcc63b807880426fc92a18626fb3266e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 18:28:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2203bb63b71f4100aa04ec70f41128be\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8eee785264214f3390da92b0f7760979 : SUCCESS in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2a8ec8d4b6f4b688330796fba36d33f : SUCCESS in 53m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f8c847016504a429589fa45b6338274 : SUCCESS in 21m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5a068516234499db19a4c7f003ba5d6 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":13},{"id":"c226db2216cdfb95aedda2ca6613315631aacc57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-18 18:37:01.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"cace7d1b1815c46b636d4bf3b9abcd9719dcb9b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 19:38:01.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f9d09b7187844748c709983db8c954a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/671287776457438493585b1ea521dbba : SUCCESS in 18m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6da7a933dec4d2e93c57b9379f92faf : SUCCESS in 59m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/163d287c9e70420e80531fbbb754436d : SUCCESS in 19m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/58db032be3eb4aaea6edf69cf634d8d4 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":14},{"id":"e3834e571dc5a31e4561222b80ea5d9ce9135b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 19:53:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cee7cb7783494009a69ec38faaf9deeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1699803d6248e0a211864dadc4cb17 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2732651d6a9e4a8c950616b0e340b1b0 : SUCCESS in 46m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1e0a6055ab04ab59d2430ed67f7e26c : SUCCESS in 41m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51c26cd339114b0fa8ac3c4d2aaa4b1f : SUCCESS in 58m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abb8f83f762e4a21a0e303af5a441871 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edf969e82b8b4f9198ce4e83e7daab0d : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee8af4beb4144543b7eb85cdcd9b09a3 : SUCCESS in 16m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a528736fc3104f7d8b55f547116c0a29 : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2861664ed8c24bd29c772ad77272282f : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f184e14e90b44ba8e7c84535e3b147f : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a077d6a66ffb4d9bb4363731282fd766 : SUCCESS in 12m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91b10415d89a4627a3839146fbdb0918 : SUCCESS in 17m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/63cedb585c864a91a7b2c720700c5fcc : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2deb982da7c54d648db2345087544379 : SUCCESS in 19m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff29fcacbf1f40e4aad398ef2a45229d : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6e2d66fb9e146599643d1ef333f8b2c : SUCCESS in 21m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d09d9b4e2ca448cabbcc18ce7a1e613a : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6de1ac87897b4d5bb69844aef528402c : SUCCESS in 18m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/856c2fa50497483d8342faed89b402ab : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75a5cc967acb436580c9a0c9924eacb8 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a34d42657c5c4efaaa4e2a5f1ec75c0e : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d8bca5d800b464fb8e5fd3d5aff5797 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1fc6ea93c76467799b60a5caae448ba : FAILURE in 1h 10m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b945722896644281b4bb62ffff017576 : SUCCESS in 49m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43257d0760fa405fbc5b3266128d86f2 : SUCCESS in 54m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1044ba90f7254cf091bb2c64e306fe18 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a896e1c9a9924923a89d3344beb29eb4 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24e78851b9be4efaa8ed728086b34244 : SUCCESS in 19m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44d9989384214c798c24a003a54bd23f : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":14},{"id":"8841a7344e2591cf1f6bcb16b5a3124debb22cb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 13:50:22.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7c9cd58e26a5014271dccd5cadf256c9b76cf21d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 14:10:56.000000000","message":"Patch Set 14: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":14},{"id":"6f51cfaac29447d117b3b42c79b975252441569d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:15:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91f6d33d25244d4b45538bc21280f6c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55120e9010794b50a299e2308cd1042b : SUCCESS in 16m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce90332a22d541bb9d2da3c339478eed : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2255ad70b75d45c2897de944dd47cbb1 : SUCCESS in 22m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae239a18fdf54cdb93314254d7ff0db0 : SUCCESS in 23m 09s","accounts_in_message":[],"_revision_number":15},{"id":"59bf9dd856432a3d866308c236e8fd7429361f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:07:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27d8e2246e104b23b0bb949a561fde05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9f3282143d4bc68f5180c997db0434 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ab7efef9fea149e9bdcb406855fa91f4 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4eff8d51cee240c1b4fa2a63a7e03193 : SUCCESS in 59m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33b1c34395d4141ab4d9004e62108e3 : SUCCESS in 1h 04m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61c502a087844b678031fa056fa6058c : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24631a2637164bdf857ac106bf465a38 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1561355bae2243c4b6dcd873f1338b7f : SUCCESS in 22m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08eade7bfe304a1aa727c1e1b63bbd8b : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1be17dc1336498f976f33ea66976e38 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e05f0d88e47b4164b1679ecdb68ca066 : SUCCESS in 20m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/caa8ddaad95245bc8c976f729411c183 : SUCCESS in 17m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/391394a1d77c436aa122b0ddef78d013 : SUCCESS in 17m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3866005eef7b4b9e88b896ce338d52ac : SUCCESS in 16m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/970143af59634cb79368e2d43f603f68 : SUCCESS in 19m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/950ec05cd3e44f299f2fe695cca34c58 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bbda2c708244456fa70982228dc8238f : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6c49401126224cf3b620d491ddae0603 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b5c438dabff54f6bb98b94e248540052 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2eec1f0bb0d34f058cca4be64b77bf59 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f57a564a03a9448e913a1d313b5325e6 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7deb3f3b6bd453cb705667cdc8455d3 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3fdc49997204b2c8e81db024e48a44b : SUCCESS in 7m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/75cd02bf91cd4cdf80b2732eb6b168ff : SUCCESS in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4012fc3b878b4041b8c919851b7003c8 : SUCCESS in 54m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2de04efce3a64b98a3a1884e07b41d7a : SUCCESS in 54m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e27a45e2dc5042dc9c69b19e7b22bb0f : SUCCESS in 23m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7229ac347f5a45d2910b14a466b15cd5 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c06d01fa143468f9b9d7905ddd389ef : SUCCESS in 22m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74978763817d404a87ed22d082e8759d : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":15},{"id":"d3df0a570585b19aaa8fcb1a431ec1df021f10d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-25 20:33:51.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"743e6958c13ce36ff012e9502fd4da305fda38a8","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-25 23:48:49.000000000","message":"Patch Set 15:\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cde7115c8be3b18a7580f41fa983dad520b99289","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-26 00:12:12.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4416a1c8338b1425c0edb85e67a58539b4b4c69a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-26 16:10:42.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dba196f6d6f5effd65e46201dfa30b0676253ecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-26 17:51:49.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"f2e6d44954b0f8f8dcf99df92f93c0948c216495","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-26 17:52:19.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"ededa96ba7948d1bef34d634ba97e5805758ca44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:12:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80c9617bef054d25a4a650abed4c54ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ac524209cc42d48110911426d6e486 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/660af880f75447219d6a127d7cd33dac : SUCCESS in 1h 09m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/586d9922f3584545a0f9585edf7923b8 : SUCCESS in 46m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23363ad013dc40139a5fe455b20f1fd8 : FAILURE in 58m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb670c586f684a299b6c24109f9876c2 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec659e21fc2243b9a007a1e14c9dbe6d : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10a26829205a44f094718e46dece8a20 : SUCCESS in 22m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19502280a39d4b7f9a18cddad2dc4973 : SUCCESS in 11m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4875ccecf84440c9a7509b10a42c4da5 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e4339a06625c4b2eba5486ba2f058950 : SUCCESS in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f54c3f021d0e47d99da542e973b6b512 : SUCCESS in 18m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b30453384c1546908bf9809d49e2de01 : SUCCESS in 15m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8dab58b8bb7f4615bd95290a1b7dd8b7 : SUCCESS in 16m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6002aebe34c4235a3ca31a2724bd318 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aeb0498174e04316a4f5d7d16ef42b1a : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65e877a7760b4ea1a77df830ab2cbba3 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/57a09fe005af4357973d667aac1894ab : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/011d88ff0bbb4e9c8b55b3195a0304c7 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e86ece9ce9e84fe8844f7473980666e3 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b22951965c0b407ab2a777f4c53812f5 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/413d19c0091f4dcfb03d8f186dd9cb2c : FAILURE in 20m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d833de9068145d1b89ea41f1effe3e4 : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/51b9cc185a5c47c1a8ce0136661cb582 : SUCCESS in 59m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/71effba80b7144bea4c02ee47973f4bf : SUCCESS in 55m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4be273a5ae54965b7043a38bc5e70be : SUCCESS in 50m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b8a8b9e30724546a08297989a8ff445 : SUCCESS in 21m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88023cc6db4414582c414137e25b550 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8d2bb1f5d5344e6929180efa6edf6ee : SUCCESS in 21m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d889166f8d8e4bd79a7a62561ba0df78 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":16},{"id":"44c092658626be3830f1c49063dc4436390c9554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:26:56.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c966971811654db0ad231c06b5dfbb58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e659afb9ba7448a594c17f3d34cd1ee7 : SUCCESS in 50m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba82c587bf5c4850a9b81fd3c16bce1f : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fb3a7cbffae4348a0441f7c274f7e54 : SUCCESS in 22m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cab3c49fd2d84140b80bc5d59ba92c79 : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":16},{"id":"9fd8e6c18e742f6cd6e22b0073bd1f2d4efe5d1f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 17:14:49.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1adb8eec0fd10f51c407d2b95899b99a9a8ff7e2","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-29 18:18:27.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f094216e4dbd3493c69a2ce521b6357c44e6cf09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-29 18:20:45.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"4e9569b999b57a9b0c463fcf64962262f180fd20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 19:32:16.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888c9b92ad1e495c9de86afa957f124a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383afd841ad540288cb4324bceeaca5a : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/30d569810c4e45159b164428f6f09791 : SUCCESS in 1h 05m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33c8465e2f3f44a88039a24f8fd5ad16 : SUCCESS in 53m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28e1d00141be43b2b8beb1156ae7af0a : SUCCESS in 42m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2483f40299ed42c69f7b00f250a5d358 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/def3e268aab749d688a40046d36025cc : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3b93a88fc424389b70fa280274b4c79 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9d268fafaa840b28edd3fb55e375390 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a2b3385212144e69a903c8e1f5be7e4 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8921bf90d724c0584e54e887d4a3c29 : SUCCESS in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3bad1e155e54494a9ea65c073c6c3cbe : SUCCESS in 17m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c59f2059c104e65ae9e378552401c23 : SUCCESS in 17m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/48379968d9a54218aae34e9ceff70315 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f60b26c8a404bb9863e0ccf161adbb5 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c1e49d53b654ef78a7624589259d7f2 : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/39e5bfd407de4c76aeeff6233efca0f8 : SUCCESS in 19m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe3a59ee1d0c4cae9f638c59f6a6f22c : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f8e5b683ebe44589b2f3a1b73764a004 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50ace94e0d0e411f8643314bebcbe593 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db1b1ea73d4544f98e2f9d4eb8fafb90 : SUCCESS in 15m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/809ea37974c649849485651fa4418870 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6485594e1fc940a69456ae0251d1395c : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bde162f3755347a8b80d24a97c5a152f : SUCCESS in 1h 00m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d285f880768a42f18be8884557163bd3 : SUCCESS in 59m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4947e9d425241d08b844c0cb0e32bbc : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb50b565ceb64b25819992989fac2e4e : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be48a0716c1149a980cb39e506ae25cb : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef104012939041618d48d5a1eb0542d2 : SUCCESS in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/863ab4be11d44352a831fe73af8fb4fc : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":17},{"id":"cc50b5e6c4c02c61611c5e06deb0c2bacdb3a09b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 19:35:55.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6db00fb612eb438eaa95a9d69f1d5fa3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/680580894f4b4054a9a7f10d5054644f : SUCCESS in 14m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15d21043c1634b02ab7151d4e857be5b : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff002ec3caf04c9497ea58b07d42b2a3 : FAILURE in 49m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/60723528063a45a8a88e71bdcb91ebb0 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":17},{"id":"fea1e8daddf19f2c71683953efbc31b42ec498ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 20:16:32.000000000","message":"Patch Set 17: Code-Review-1\n\n(31 comments)","accounts_in_message":[],"_revision_number":17},{"id":"a2f3272e1ebd012fd347b2dac6bb812298eccbc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-04-29 21:21:32.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"1fd55b0b34d46eaa705a7a75d21048fa9e96716f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 21:29:58.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"0c1e6eadacdc889d3ee952027eba2d8cd4c04f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:42:21.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a09295fd8a44412993edc052daae960\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff78f934dc7046aaa4ee86f085b40ce3 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c126849f73e5444083b72492ca4cebad : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbcc9a0e012b44bd83f6dc62972d92b6 : SUCCESS in 47m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5516c1ed4d784334ab26505ce8c522eb : SUCCESS in 46m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53a2b388e88349b3bc2a6ad2cd7f78b0 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4168567c9b2448988e35146efba6e67e : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/019729cad38e4d05a8960242b7fd2ba7 : SUCCESS in 25m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c584335242247b386ed011286b838ce : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb02b481b5b44518807658881f0e757a : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2846d7bdb6dd4582bb59aa2f1fd12407 : FAILURE in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/38eb5ae6efc047fa96a952764f207a66 : FAILURE in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f0ee734dfaa4925aa4506000546f805 : FAILURE in 13m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f0a59af37efc489ab3627a918d775e15 : FAILURE in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/563585098e8d407a9fdbf81d0aa2f508 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed5616cfb9e74cffb3f9112db887774f : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47cf160bba9246119fec573ac944911d : SUCCESS in 18m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de21ae1f4b2b4cad9e7513f1a4d8a4be : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3441a373fdad484ba0b2457b62b5d6c9 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d41df92a4bdc470095475794c4906da9 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55449865cbfc40f285bf33f5bf806b1b : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c4a79520a9d4445be726eeaf43bcc2c : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/966c5da54d3843d0b78f3914e2f13bf6 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4794ac1f73a64bd88a9c63b6d8ef4736 : FAILURE in 59m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c001c40859a942eeab88157340c52eab : SUCCESS in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9087b4d950a49c8bcf9ec00c45af3be : SUCCESS in 52m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8533512962e6401796c79b43a10c3bf6 : FAILURE in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37b02e5d3dc4d6bae2360e26199274f : FAILURE in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d2fc3b184f54b4587c7d31110b4d246 : SUCCESS in 20m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acb6407ba6b24a10b56b377bbabdb9bf : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":18},{"id":"bc461ebf674fcad9b117b6dc02521e036c8692f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:54:38.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ad68f0932e476687e436ab9e253095\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/929ab149a4ca4fccb83941809b7eb659 : FAILURE in 52m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efdd5021fc554e61975a58bc1c4e73fb : FAILURE in 17m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6da4769c251c45469c8b83483390b36c : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2502cbae78d646cbbd9cbbf4fbc41505 : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":18},{"id":"7a0991dfe29ae0e386dd2c60a379ddd617e49541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-30 20:24:45.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"e8d59508d52298c887870794852d0269f1480efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 20:27:47.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"73e825f22e81d924c52a847e9423ff6fa3c5a72f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 20:28:37.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"39fd71d98d53bebefcf22b6f1035a16067c248f7","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-30 20:38:35.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"029ec49aede224d4e9657948d23b45eab7c8ae0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 21:23:21.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d7c3fc0b6734e0f9887ecb9ca270978\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8de0becf86554d029ffbdeba6b3c1b87 : SUCCESS in 17m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e90c284959e45ad8648125762857b5c : SUCCESS in 53m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74720327699c43b8abcde818a052fd90 : SUCCESS in 19m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c2b7fef629b4e2284bc6759652791aa : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":20},{"id":"ae9bc3f389cf32088eead2c046a4773c75d5b66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 22:29:39.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d723a688d4d547cb8d93bbe16ec3ec78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003c59575850471d9c09d123428becda : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/8cf5c7b004124e12ac002e120c283e45 : SUCCESS in 55m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c2e0dbb0d8b4785b0b71a846f50e34f : SUCCESS in 49m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/777e27754f49487f9e8e9e6d00c170f4 : SUCCESS in 1h 00m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6903f3a825a94f4688d24fea4b76b9a1 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e995fde31a0743fba938e2ed02ca4995 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c4fc15a48994ed5b248df6a4588842b : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45b11385f41b48c4b41a64e0bb23b253 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1c2c45b56ce48728ac6764cd13a4005 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da74232f377a47f4bc643504dc161531 : SUCCESS in 22m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f456f8c0d894c6991c8968d647004cd : SUCCESS in 12m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1601eafbffe0453e84cd61f5bcdc98e9 : SUCCESS in 14m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/749e8ae7e6ab4fe9958f6ea85d605c31 : SUCCESS in 14m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30c5ac5ac68347c98a82769508e874ba : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5fa856afa9de40e29ee03901b1fc8bbb : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4507a63d8a8b4ef1aa5034f4321b6361 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4037014adcdb4b7282974ce56c30c15f : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/29eb46d2c3c64d86b59d611ef78bad4c : SUCCESS in 31m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb6bb9c3937d49e489c16d559976ae80 : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9a829990cb44214a8d7208ba39fc471 : SUCCESS in 9m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/21f7b8c9596a4f658381bae422ddcc70 : SUCCESS in 30m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/68e23b819f3a445c9f47543226ed5440 : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/849cff3cb4b446f7946eb9c946877731 : TIMED_OUT in 2h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b23b046f8d7e4904b2fdafb1b1b88489 : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a79345fcf3649e0a72ab43f1c3a08c9 : SUCCESS in 49m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59acdcf4fc694f98af055efa36003c69 : SUCCESS in 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0496404a24049a1be5e41dd8d17db25 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41d864029238425581be2bea299856aa : SUCCESS in 24m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16735fa3bf034ee2bf236c53cfa44d8d : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":20},{"id":"138b0b5ab03e3798a78211ae5f36cbaeba6c5eb8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 23:00:58.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"17c49aad1c13ea333b481f174971b3e35dc31750","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 23:50:42.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c133bea45c3248bea771208d23bb9a38\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe263420bbd944f5878ef18b7c900b30 : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebcf32f22c154e41b140c2275b083ecf : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b991c74cc44e4a02a8e857ff94316fd8 : SUCCESS in 48m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/281f08a56cb3426996c01c39d4b697e5 : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":20},{"id":"f5dfea4014e0a376625aae378c2107b8133cffb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 01:07:45.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/032924d3653a41a2a914aad6cf0871b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a61b23ca0b404c82e33fd1f61fc8d8 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/cbd209a4b1514eaf9c423a3a467a5be2 : SUCCESS in 1h 01m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3084ad6d552b475b83a6e28ac8bc379b : SUCCESS in 52m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1bd08b2d9dd4f5b980790db91c74342 : SUCCESS in 1h 07m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dac1fe4e1e94404993c8f4b226d12f31 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ef604c075864eab8faecbfe20ed9170 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d55bdedd74d4f119ec7549d7d6a0195 : SUCCESS in 19m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ba31fd30e6f4e2389e2210f698a869f : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bbd14b0d136a4e6aa508dd24b1f3c148 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b0eaec568f8491e89b10502e925e6c4 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7266df4f1356458eb2b5d534d5c8d262 : SUCCESS in 18m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8260904ef235445fa6f6bece9ba5d753 : SUCCESS in 23m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/90251f1eb91a4185b4aa79744270e72e : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4293d3be89ed44a1876b113cef501fa2 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05fcf53c433749ba8b13186662c60544 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c27954a8038c4c2aaf8d3e33020c45db : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/58b9a5696c6d4422b254252db71f5deb : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dac062d6ae88460e80938d3c1233b404 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3fa7222f69464c9ba7fc0ae1854cddc4 : SUCCESS in 19m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97409696913640cdbbd95aa5ea39347a : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dca2a24f9264adcb0cee37470c25a20 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad91cf30a6ff49c1af80c75ddab345fb : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8c3a7069da04602ae6375f5c2e93ef6 : TIMED_OUT in 2h 00m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1429c10599a8428a94123836230ba698 : SUCCESS in 35m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5a928da9f7f4be5b07efdce32cb3176 : SUCCESS in 36m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b00cc7083b4b4f8087d7c0c2e1430872 : SUCCESS in 24m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9bccce8bde4960b750d1d2c4f9ee1b : SUCCESS in 3m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fe3c9282d8e40b0968721ea6405e480 : SUCCESS in 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6abab7f0842c4ce38740f402b5625ef2 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":20},{"id":"36dc8a8a533f3809d4ea2948c3b954f9ae04769a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-01 15:02:21.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"4f53455e767eb8bce7f5569c81d10c82c5cd16fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 15:54:29.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e48975d0e24f3db9b7184af0b15b42\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03b2a19829494cd0803bcbc3221c8e76 : SUCCESS in 49m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd3f09688213438c86403d70f5f53300 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbc16c5013994ca4888b0a75d61acd13 : SUCCESS in 50m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ddfa018d67d4c1cbb32da5c1421deaf : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":20},{"id":"e73f41cc593095aa650a103db4bf8fd57e0ab348","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-01 16:22:30.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"c1f34fbd8c5b7a2f550cb3f62f3cd8c686bb0d31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 16:40:40.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1111ea80480a4082ad9d8f728ac29954\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f86c3dff6f941a69a02dba10e36d482 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/b328df74fe334915b841f1bb4f80628b : SUCCESS in 46m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6529e04b21514908a49c5633f59630fb : SUCCESS in 36m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a21f0134dc094f299cd0d7549dda0e55 : SUCCESS in 1h 02m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fa80249c86946b991596d58445f38ae : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6bd25a4a262437d893ec29b02c7d800 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/915a357d6fab42cf84b2fe3b937d73fb : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fea0e2da017a467088c8cf8c28629087 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f98b04575bb4acbb620d6716e8e8eb9 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62534838aee841ca972f339297cb6b29 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cbd71e6d623419ca9aa665c77fdcdad : SUCCESS in 19m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fc7e67d7a3cf401eb2128a7e3ee1434f : SUCCESS in 17m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe3cd55f01c94816bab3d386f6e4b2cc : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb7fc98a9ff147f496010a711b4d8fad : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8013dbb91daf47a992b85468927e588a : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31f6f2e6af4445fcb54423adfda18131 : SUCCESS in 17m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b296cea1f33a46b3a6857858c6bc4e10 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83a69ae520d74d1398e9099c2bf45bb3 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b005251e87534936bfda8128bc02a25d : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0972d2be52e24beebf352da864ae5a77 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e30b6831ed2a41b6a7ae8811d90ee129 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b8bccd202574bd4b0286cc78278f94d : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b3f376ff27d41d3911c379a79c4d56a : SUCCESS in 57m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d13eb112a4f454b854188896d5e163b : SUCCESS in 50m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b57467eba6154ec4a180bffc6b35b876 : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/186ddd98515c4f588f8c7e9d7c7b4a4b : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5295791ff90f4a199920f9d0479fc05b : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d12103e94dd542c0984704a3828d5543 : SUCCESS in 18m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d2e34f130164f74bb4051905f961608 : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":20},{"id":"5b025d5d6b9a3984e1776a6775ab4d76b380d8bd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-01 19:49:34.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"7e7ae9c0faf5fda93bdbfd163afb2755d7777c36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 19:50:18.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"adceece34562968c3753550c34329e20aebd17ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 21:49:47.000000000","message":"Patch Set 20: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9023abc92da54df9a1cd32408fab5bbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9c3c6c921f4f56ba61477bad83722e : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/e7a5d888f92f42b5a12ab0e71bb9d8d2 : SUCCESS in 43m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01a4a1d9b7624caf83da42ba07a858fd : SUCCESS in 36m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8af5d382f04a464eb65e5b8ee90d4b6a : SUCCESS in 44m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/597fb3a6f7254c1b860cdc191b763685 : SUCCESS in 12m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f909cce690744a688bb0bfe14b6347b8 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2db030c0c7a4f5095a12bf3541cde28 : SUCCESS in 23m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/449b22a7dfc34dd58ac51f3128661abd : SUCCESS in 20m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f7b17cffd41f4221947720f86f3f5457 : SUCCESS in 17m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/046e84a9e9694356a23c3b74eddf2f68 : SUCCESS in 19m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a22095121d544209739e568ceec37bc : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/791e223a63d14dbbbc09ce04fd90c81e : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2b29847e82e4216ad62b7072ad72362 : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/977859f1c7e74ba291d3a120a98c6c27 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f47d18f537a841838f958e1691584b0c : SUCCESS in 16m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cdfe3ce3d0ed4442928dd5d9b4208340 : SUCCESS in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c019559294674e49b842b4a2120a01e3 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/796c5152631640a0900410b08376bf0e : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/33701ef1782c43dfbef271275a373835 : FAILURE in 1h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c40d857411f4f9a80c0d70414296730 : SUCCESS in 40m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64d266cdc99c467ba7ea5e49febf30db : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf47714a93a4e9ca1b274283c1e6db1 : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7cfc9e4371b4772a9395639db824886 : SUCCESS in 4m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2401f345b6840948575d1d5815b3997 : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":20},{"id":"a6d9c079590080d9dbf966dfea5fec5e3e21fff5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-01 21:53:31.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"326031b39c6a590c8a76c75f3323e1627cb3244c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 23:05:23.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ca7ba88d4b46f0a116e98cd98b2958\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42be4f5f78424d3f9548111e23315d24 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/63e8c14bdfaa4cf58ba4b74fe84035a3 : SUCCESS in 1h 01m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17b48b5787274b0681653736a3de5da7 : SUCCESS in 49m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec9a14c8a52c4b25821aefe1b5096ad7 : SUCCESS in 52m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85c0bcb1b6bb4026955707ddecccd0b5 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22b1fc00773641188eb077fb491ea403 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5acefdc276ed4bc4bb607af1bb310460 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec010b333663428792e059054ba6fc7f : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/28d95d1ad6494c0eb184ee3ed7468f44 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2401840d0c1d449d809ce973d4217dd4 : SUCCESS in 19m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf59bc81cc3146239b26331538112747 : SUCCESS in 19m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df8f386136b44331bc834cb5f275eac1 : SUCCESS in 17m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/da409d0445c4417cb8faba7bd6abc802 : SUCCESS in 11m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bea7596c6de640e883c618126d20112e : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7dd0194eee274700984c9bd5eea037b4 : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f87c32c2eba541dfa30444dff89171bc : SUCCESS in 22m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cdaa7670646640248f181c0e2d5f3cd4 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b953c481356f48d48279df85f22ff910 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6a6296a0c294020943066be54163667 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1bf98196c3a3431a87324f8eb2a4cff5 : SUCCESS in 5m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/868816e2d7f44a18b454650f3309fca3 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f960d706dbb4a00a4f17203bc40d756 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48770f2808814f1fb885b2f0e30b3223 : SUCCESS in 47m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c15891276d848968a549dd8cf8b7af0 : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62679b3e4bce4a6aa9b2bc209eff4c34 : SUCCESS in 48m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e5e2ade1f324e42aee5c7ea65cd49ec : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c607ad246c954162b5165aa3df95b927 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d221e79896de4f688e3dc7a8dd20acf7 : SUCCESS in 19m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/158a4dd9241449e8a39bf57a89d2db4e : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":20},{"id":"d55b6fbac6e3fe5fc6aa8703862159d9e2d3ba7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 23:05:55.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"1c68297d9f5652a36281a66096ccf5242973f0b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 23:25:51.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1429abb29604a378937ae0afac43ea5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a94572f06f844a39d9b49f52ae358ea : TIMED_OUT in 1h 30m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e779934fc5a44b3fab5950e51e3f1360 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/87dd4a5111e14e3ab42fc657cbf01fba : SUCCESS in 25m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b14202f531964876aa979a135f2c10f1 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":20},{"id":"530ee2b15dd38b779f107a4fd4b175c3214d328a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 00:27:02.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/022864a4fc194aa2925ed3ba35f28ba6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f902cdbecd3342de876f89629df989d0 : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/7aa8376494c7416d962be6c2ada33b71 : SUCCESS in 1h 03m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e6715319fc1452b834e049b8f54110f : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abaeaf30850e432eacdea29f972ded0f : SUCCESS in 1h 04m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9b9a0429588d48d4b63358d0f2228583 : SUCCESS in 11m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4d15d1dfeb704968a2d00fb0b4b53f28 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52a98f009e824559a82d53ea05969600 : SUCCESS in 20m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a251e34288e1440ab075bbc97b2b63d2 : SUCCESS in 17m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6cd0c6d7f6a743cfaca63d7363d92c61 : SUCCESS in 17m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dbeb55985ab74637ae7b2362764bdf45 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a57644a8962e469ab7a893e08d7a2826 : SUCCESS in 19m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20062281530048cf867cf8173d06f344 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4371a7c4855a4b00a561fb35f5782b35 : SUCCESS in 16m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1309f1fec154baa92be164a56097b60 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/22a5b61afd7c4631ab8454a84fcb8dd2 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef8e34336a9b4a7a857a19ada13288bd : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a469a4cffb24b98a6e3fa3c48fb1820 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b5e6bfd33cf49f2a5f0c1925eb8594a : SUCCESS in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ba1e9d6ccf54fe08fed7d20b6fdb3e9 : SUCCESS in 50m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdea86c261e04c72aa01c468439c6df3 : SUCCESS in 33m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c580e55cfd4549b89ed614df089b4787 : SUCCESS in 37m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a004ce5d55524f5aa54354e4e416c8db : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dcd3ddf1bbc4661bd42a2a3cd9892ea : SUCCESS in 6m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2a0804071b4e35b7ef0e336fc6cb0f : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":20},{"id":"421822ca1753c6100dc3296b54dc386bf2d21e00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 00:27:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"966ea4739933c1885e19f789d8290507c12963d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 00:28:10.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c699d5e8aebe4432bc0926e0fe39a179\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a4e17eb8144da787a6d9e051c618c8 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/213b9297973d42e59b4dc2da7c0dc3bd : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cbd2bcf6b81548658e621e52e2fa1d3f : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"bc3a59bdd36fbd90554e73d419d8767a3845f9af","revisions":{"f3085c50a57988ad3a7d719dea1eaef518c03521":{"kind":"REWORK","_number":1,"created":"2024-03-22 21:04:33.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/1"}}},"commit":{"parents":[{"commit":"27ef11ea14240d8611dd1cbadaf8412727c85f49","subject":"test: implement cache expiration time in MockMemcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27ef11ea14240d8611dd1cbadaf8412727c85f49"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-07 00:24:25.000000000","tz":-480},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-22 21:04:14.000000000","tz":-420},"subject":"Move statsd client into it\u0027s own module Move all logging functions into their own module","message":"Move statsd client into it\u0027s own module\nMove all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3085c50a57988ad3a7d719dea1eaef518c03521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3085c50a57988ad3a7d719dea1eaef518c03521"}]},"branch":"refs/heads/master"},"22e60512cdbe1fd6d033cc01153f44a8a27e2630":{"kind":"REWORK","_number":2,"created":"2024-03-26 23:32:44.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/2"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-26 22:52:31.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-26 23:32:14.000000000","tz":-420},"subject":"Move statsd client into it\u0027s own module Move all logging functions into their own module","message":"Move statsd client into it\u0027s own module\nMove all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e60512cdbe1fd6d033cc01153f44a8a27e2630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22e60512cdbe1fd6d033cc01153f44a8a27e2630"}]},"branch":"refs/heads/master"},"cdd6157afc0816ef7922131cc8a91cda0985092f":{"kind":"REWORK","_number":3,"created":"2024-04-02 17:38:59.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/3"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-26 22:52:31.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 17:33:56.000000000","tz":-420},"subject":"Move statsd client into it\u0027s own module Move all logging functions into their own module","message":"Move statsd client into it\u0027s own module\nMove all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdd6157afc0816ef7922131cc8a91cda0985092f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cdd6157afc0816ef7922131cc8a91cda0985092f"}]},"branch":"refs/heads/master"},"f1bd573c3a833cd11b5e10826f982a48986240a2":{"kind":"REWORK","_number":4,"created":"2024-04-02 17:58:13.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/4"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-03-26 22:52:31.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 17:52:50.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1bd573c3a833cd11b5e10826f982a48986240a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1bd573c3a833cd11b5e10826f982a48986240a2"}]},"branch":"refs/heads/master"},"2a50f0dd6f37b3d6e0a20d1c3d46b65b85810e28":{"kind":"REWORK","_number":5,"created":"2024-04-02 21:40:22.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/5"}}},"commit":{"parents":[{"commit":"5e13de119d3039fb097e3089edeab5e4a3ce1ac1","subject":"Move statsd tesing in its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e13de119d3039fb097e3089edeab5e4a3ce1ac1"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a50f0dd6f37b3d6e0a20d1c3d46b65b85810e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a50f0dd6f37b3d6e0a20d1c3d46b65b85810e28"}]},"branch":"refs/heads/master"},"c6431d3af91c251aae6bdabd7f121ae9fcc700a1":{"kind":"REWORK","_number":6,"created":"2024-04-04 03:23:09.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/6"}}},"commit":{"parents":[{"commit":"7d6614bd60517e81a8e7a09c05774aa357df30f9","subject":"Move statsd tesing in its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d6614bd60517e81a8e7a09c05774aa357df30f9"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-04 03:22:51.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6431d3af91c251aae6bdabd7f121ae9fcc700a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6431d3af91c251aae6bdabd7f121ae9fcc700a1"}]},"branch":"refs/heads/master"},"28c2ad3416039c827f3b70a8dd4ee4d53b85bb96":{"kind":"REWORK","_number":7,"created":"2024-04-05 19:34:43.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/7"}}},"commit":{"parents":[{"commit":"5e13de119d3039fb097e3089edeab5e4a3ce1ac1","subject":"Move statsd tesing in its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e13de119d3039fb097e3089edeab5e4a3ce1ac1"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-05 19:29:52.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c2ad3416039c827f3b70a8dd4ee4d53b85bb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28c2ad3416039c827f3b70a8dd4ee4d53b85bb96"}]},"branch":"refs/heads/master"},"b3c421acbe7e1db5f57adfec855a4a77a25cd775":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-04-07 05:21:27.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/8"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-07 05:20:58.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3c421acbe7e1db5f57adfec855a4a77a25cd775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3c421acbe7e1db5f57adfec855a4a77a25cd775"}]},"branch":"refs/heads/master"},"37cdcef79c1087f2534c285bab33af37e66f60b7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-04-11 19:41:20.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/9"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-11 19:40:31.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37cdcef79c1087f2534c285bab33af37e66f60b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37cdcef79c1087f2534c285bab33af37e66f60b7"}]},"branch":"refs/heads/master"},"c422a4cac473fb2b2acb946e3f16faa8955dd2b2":{"kind":"REWORK","_number":10,"created":"2024-04-11 21:08:24.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/10"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-11 21:08:01.000000000","tz":-420},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c422a4cac473fb2b2acb946e3f16faa8955dd2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c422a4cac473fb2b2acb946e3f16faa8955dd2b2"}]},"branch":"refs/heads/master"},"d998bc81d45bfda7baaa85e5a6915b7b4d660dde":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2024-04-12 13:31:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/29/914029/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/11"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-12 13:31:52.000000000","tz":0},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d998bc81d45bfda7baaa85e5a6915b7b4d660dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d998bc81d45bfda7baaa85e5a6915b7b4d660dde"}]},"branch":"refs/heads/master","description":"Edit commit message"},"17fcd4c99f01725afbb47d8afed0bbcd3e6a006f":{"kind":"REWORK","_number":12,"created":"2024-04-18 16:56:09.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/12"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-18 16:54:32.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17fcd4c99f01725afbb47d8afed0bbcd3e6a006f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17fcd4c99f01725afbb47d8afed0bbcd3e6a006f"}]},"branch":"refs/heads/master"},"b6b165bb110144b15ab772f17b2bec169f27db96":{"kind":"REWORK","_number":13,"created":"2024-04-18 17:34:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/13"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-18 17:33:44.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6b165bb110144b15ab772f17b2bec169f27db96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6b165bb110144b15ab772f17b2bec169f27db96"}]},"branch":"refs/heads/master"},"89b027d821f79a3d1cd886e0503091f9996511ff":{"kind":"REWORK","_number":14,"created":"2024-04-18 18:37:01.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/14"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-18 18:36:44.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89b027d821f79a3d1cd886e0503091f9996511ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89b027d821f79a3d1cd886e0503091f9996511ff"}]},"branch":"refs/heads/master"},"0f79f21164f5b6db907b9f95a905a7d73a1fd301":{"kind":"REWORK","_number":15,"created":"2024-04-25 13:50:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/914029/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/15"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 13:11:01.000000000","tz":60},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f79f21164f5b6db907b9f95a905a7d73a1fd301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f79f21164f5b6db907b9f95a905a7d73a1fd301"}]},"branch":"refs/heads/master"},"5b49293c5267c0bf4e3e0aff1f932007c58468b0":{"kind":"REWORK","_number":16,"created":"2024-04-26 17:51:49.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/16"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-26 17:51:05.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b49293c5267c0bf4e3e0aff1f932007c58468b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b49293c5267c0bf4e3e0aff1f932007c58468b0"}]},"branch":"refs/heads/master"},"e38bfedd4d1851f444a7665cb6a1042f5f69b281":{"kind":"REWORK","_number":17,"created":"2024-04-29 18:20:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/17"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-29 18:20:24.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e38bfedd4d1851f444a7665cb6a1042f5f69b281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e38bfedd4d1851f444a7665cb6a1042f5f69b281"}]},"branch":"refs/heads/master"},"0bebadf20ceef1db47c36cd127099e21d0997fbb":{"kind":"REWORK","_number":18,"created":"2024-04-29 21:21:32.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/29/914029/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/18"}}},"commit":{"parents":[{"commit":"aa8604645c3245e182f0e00e5bc689247af24e1a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa8604645c3245e182f0e00e5bc689247af24e1a"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-04-29 19:59:09.000000000","tz":-240},"subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","message":"Refactor utils\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bebadf20ceef1db47c36cd127099e21d0997fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bebadf20ceef1db47c36cd127099e21d0997fbb"}]},"branch":"refs/heads/master"},"19f70e476850ec1245d6e59bb6201a86eb3f148e":{"kind":"REWORK","_number":19,"created":"2024-04-30 20:24:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/29/914029/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/19"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-30 20:24:24.000000000","tz":-420},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19f70e476850ec1245d6e59bb6201a86eb3f148e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19f70e476850ec1245d6e59bb6201a86eb3f148e"}]},"branch":"refs/heads/master"},"bc3a59bdd36fbd90554e73d419d8767a3845f9af":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2024-04-30 20:27:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/29/914029/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/914029/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/914029/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/914029/20"}}},"commit":{"parents":[{"commit":"6387949a540e871e19062de8131503a8ac2ceef7","subject":"lower-constraints: update to mock 3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6387949a540e871e19062de8131503a8ac2ceef7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-02 21:33:29.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-30 20:27:47.000000000","tz":0},"subject":"Refactor utils","message":"Refactor utils\n\n- Move statsd client into it\u0027s own module\n- Move all logging functions into their own module\n- Move all config functions into their own module\n- Move all helper functions into their own module\n\nPartial-Bug: #2015274\nChange-Id: Ic4b5005e3efffa8dba17d91a41e46d5c68533f9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc3a59bdd36fbd90554e73d419d8767a3845f9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc3a59bdd36fbd90554e73d419d8767a3845f9af"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
