)]}'
{"id":"openstack%2Fswift~914714","triplet_id":"openstack%2Fswift~master~Ibae08b198e685bcdb11621c4777d1a39e2c7a076","project":"openstack/swift","branch":"master","topic":"random-expirer","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-16 19:03:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibae08b198e685bcdb11621c4777d1a39e2c7a076","subject":"sq: make test better","status":"ABANDONED","created":"2024-03-28 20:41:08.000000000","updated":"2024-04-16 19:03:07.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4392e61ffcc685be4d12c0e1be53f893479aa58d","_number":914714,"virtual_id_number":914714,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-09 17:55:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 21:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-28 21:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-09 18:30:13.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"620c3bde842014c562561b03abfeac22beabdd90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-28 20:41:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8b0f8abdba8abe8ba373118c595bac8d997b562","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:20:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3737df794ee8470e87e6374579172081\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dd6bed345994e019d72c73b5f540cd1 : SUCCESS in 37m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ced6081c685740de945ff8532dccb1a0 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5dd68109cb5547fda6df39069a856b21 : SUCCESS in 36m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73e762f584dc4815b42fea77cbed2326 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9234d89f604dd60956d11c3bf403c3cb29db2850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:31:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eff1d62e9c72456195b03b5023ccb301\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c271890db540493393df9250aa5acf84 : SUCCESS in 8m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0611ef5621e4370b176fbd733e4dffc : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eca769c52a084ef6b6a1a36437adcb0d : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5746ea2b6a304f14a1e07e6490fcbd31 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58b3a4d579da44ddbb9cb1f382c0bcbe : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7dc8e2c5d87455b81e6e42926733fa9 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c75ba7d0c3fd4c3f8725908ac7e13137 : SUCCESS in 19m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d03ad921580a4ea98d26ba2f21110344 : SUCCESS in 18m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bec3caf9805d4bfa8d50ea4504e13703 : SUCCESS in 16m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/80a35238c5cf444e923dfc7e739f263f : SUCCESS in 20m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e02549c93a449b1b2a2c1ccf448b221 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49c9fd3fd24640b9bde71a7c042f57bc : SUCCESS in 22m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a339e38529934f22884a3257b1c1cf16 : SUCCESS in 23m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43eb4394bb554010b9db5ab7833380ba : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cfc10d9090b14ef7af4a8cb6554d6fe9 : SUCCESS in 20m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e8e1c9f8eee4883aa66d6e42d323787 : SUCCESS in 21m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3660e9e5c06d4cf79915b3dd6bf45034 : SUCCESS in 40m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84c1609b5a194badb751541d7c00b576 : SUCCESS in 49m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d77f6f3ba03c43d39bb47a066ad2d6af : FAILURE in 22m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd50f5959bc94f5d9001183e10d74660 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/112300f4437047b28cb676c9ad29ff90 : SUCCESS in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32f87759cb15a1f7ec82e4002bb7617c3fe64905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-28 23:12:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"112cf610ed3e4307efda3d1be9527e42031e9b20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 00:11:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861ce11bd1af4638abc122d7de7e75f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f3400f2d334460695275885b6a33eac : SUCCESS in 49m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/249dd616bc6a417891f35563635298f7 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1292013bd3f846ad89ad31c09caffe2a : SUCCESS in 57m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a68335cdce946ad918d7d02a16193e2 : SUCCESS in 20m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3dc22458a629b838e4c54853b6ba9551d02dd87d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 00:11:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcfedac2c64d4a1090cbe110fc460cd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e985bb087d4eceb0424b932dddaaf1 : SUCCESS in 8m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43454229550548e1bd0a071bcd68500f : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9f9447e880d4cdfa0cee9cf5ad6e170 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f7af75302074a27a97aa1fea29cc124 : SUCCESS in 24m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c66ce7686e04db7956dab6f2ba300ff : SUCCESS in 10m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20919b0fc05340dea27e66b42edaeac2 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aeb9fe82a16f4c73a2d9aa83b9ea5dc1 : SUCCESS in 18m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea2240eeaef459ba4b98511bf9794ab : SUCCESS in 25m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/520a64f0326a40879940e67bc458e98b : SUCCESS in 13m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9b1faae92f4a4ec28e116644871766da : SUCCESS in 18m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d248bd664cb94c4b855669af2ba95ac5 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5210cf2983f4ac19b3410e86fa10951 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd7f7f83c83f468396a1a480a8043bba : SUCCESS in 20m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9458173bf9684d0fab4a90d5ed067392 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d27b95c817014d6a8a2ea1d6ba67294a : SUCCESS in 16m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7ade2ab92f34427971ebbdc4e29ad73 : SUCCESS in 24m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ecd8087be264716a0606649d73e53d0 : SUCCESS in 50m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79d1bbe276c24842ace8086d7c9e38c6 : SUCCESS in 51m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eb75b42736b4b2e919e92fdadc695a7 : FAILURE in 28m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ed355f0f7f435c831664cb8807259c : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32e2b010957b403fa652592fa0440c60 : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45119c3f42aa4c7801a5daffc6181ddf1a175bbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-03 14:13:05.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"49306e6c80a1826a7ebedef03cbc70b2f95fc1dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 15:28:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d847872b5534bdbb79cee913c0437d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40e0dd7b643f4ad9a0302ca0dfb8b62e : SUCCESS in 1h 11m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f7ab1ac23804e3090987a98bcd112a3 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b3017c4e20847d187ac0a21738276b5 : SUCCESS in 23m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/103a4b4f36354d239d79ae90adf0b4d6 : SUCCESS in 1h 05m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d3e12e7c64aa4350c4d5ce444cf74244cc1448af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 17:12:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47820ddf8f20438686ad7fb34a3bbacd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e886131bfd44a58b1179ac976632fcf : SUCCESS in 7m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8aa9264b5e74f668d5c845c1ffedc59 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5eb4920b8334c58af12713145161fd6 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53aaef1ff4104a01a1a42e15369fcf52 : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13bdf8b22f2d4d248b2a369437117e45 : SUCCESS in 1h 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ea8860d879384addb0e3b909d4f5857b : SUCCESS in 1h 11m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d420df04d58844798fb92903529acaeb : SUCCESS in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/589b7db783344f2bba9a96624345b609 : SUCCESS in 25m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1435f84ff3764b50b165c404e78e1a0f : SUCCESS in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb0c9238f0484d1dbd06a8f18a1db6df : SUCCESS in 18m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/793929ed2f02406b89ff6ee427305445 : SUCCESS in 26m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b8a4c20e707547849a9f92e0171d3aab : SUCCESS in 20m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eabe5ae6a6874085b79ef6e49d4ecabe : SUCCESS in 17m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a14b8612ebbd4553849e8e8783a16871 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2a6e392b77343a78e83c4aba9e80c1f : SUCCESS in 17m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d23e3f6f13404707a4d7a50ced464dea : SUCCESS in 18m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9bd745a5bfb347daa3ff0d5ce21e1298 : SUCCESS in 1h 01m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c43d6e884b314f8887b022c074980eff : SUCCESS in 54m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60a5a6a092b84d79990d15ab37d296c1 : FAILURE in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10278d3164e47848dc4ee09e7431984 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c421ae3fff354d9daf7de5e33846f314 : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2b4f8feb0f57a9bb4aece1584b2d7206efac717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-04 16:20:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7d1a0b3b8e6d3459c637dfea5185e64bb6d97063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 17:17:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd858cff3f34f0a809d26cd17e01e11\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d66162595214a6aa22c7791619f3568 : SUCCESS in 19m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4250e49f9ca40ff99e4966f03b47951 : SUCCESS in 44m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59e0f5c1302f49e1af1b6f60865a6576 : SUCCESS in 19m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/92244402c60d4037a4b7fad5455e8544 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":4},{"id":"bff4eaf5e0baa797d9530a94549d28946a16b4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 17:20:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d3bb620d27242de94866967d1565b54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3c58c310f043ae8db3337409e50529 : SUCCESS in 10m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41f013ab0d9d41d0b709fd0e3c9acd56 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40f5c61bc71445ffbe0d84b7c5df6f9a : SUCCESS in 38m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a800cfaa52647448de4314a4f839a37 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f3e474e45d946f086471664c908be89 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bef4c7e219c541f68d2969d2dc90abe4 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e538ff3e91545f5bce09b0b8f777144 : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb12b753e4b4f80bdac2bcaeb902cef : SUCCESS in 18m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8264226cbac94b44af22b17203144feb : SUCCESS in 17m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d1d84dff70e5494a9fa9f66e9d888e1f : SUCCESS in 17m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f03de5ec7ca41d0ba7a26810e87d87e : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb6a19f24cbf4224b56950f011aa0671 : SUCCESS in 20m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/913a25ab10cb4c1da4f1f8f951cd626c : SUCCESS in 20m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/079bb05a14ca4383be64dc7268ad25d7 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f82ee5a991340d4a6c2cb62d91ae30d : SUCCESS in 17m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1ccd33b16d144b38fb319d5fa1974d5 : SUCCESS in 16m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03b4de85b47d49528e53301f3f0d0612 : SUCCESS in 48m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dd1abaffbb44cf0894e29d4129be672 : SUCCESS in 49m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/738398a535a04b9ca085696ec639fbc0 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd7c1e73fa94c76aafa39032162d7ac : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa7b330603764c01bdc0dcaa68b804e0 : SUCCESS in 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb11317e96c4fdcbbcc288f2200655674d225d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-05 18:29:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fc23645c7c0b229a97f35a25add1554436fcddaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 19:47:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7abf5e86084367a3eca426ae0a4310\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cba97dfb62349a49743441dec62b63c : SUCCESS in 10m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35f82e16f34642a38455035828a7aa1d : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d7d0ed5ff384c28a6e6c68afda6785c : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5adce2c101e14d78a406cf3a1c95cd0b : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08851a1de27b43558e357551825b177e : SUCCESS in 23m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ed073e6cf5ed44d0a74b1a1e77cfa7ec : SUCCESS in 23m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25802b9ddcab477292572101895429b9 : SUCCESS in 22m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8530cefea604fd1b500e53e4b8586e0 : SUCCESS in 20m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e2965fc30011488e9837edd76bfe212c : SUCCESS in 17m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4d5aa8bf02ff418d9eb20194281e9bbb : SUCCESS in 17m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8491712e14c14a3089e0a1c69127b157 : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a3dacf3d1d341b4b88ed811bc2c4ff3 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03b684403fdb45a793c367f37abf34d7 : SUCCESS in 21m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/978442d7cc2c4094bb3daf5cccf64838 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a21f1a1112e54e209b6a8923f76d4cd8 : SUCCESS in 12m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1ab4cc982944bf7a645cef3a951943f : SUCCESS in 21m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e62b7e4a53074fd6a8e4ac179afa91cb : SUCCESS in 1h 00m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4a6ddafb09141fc8e23781252213b6f : SUCCESS in 55m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8020848c7d6e4c69a4a05ece80b8dba0 : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020808e2c2934502b50b74cc61a6a6f8 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0061e7af78a54e698fa1e96df31c1fe2 : SUCCESS in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31e532b0b864d4201c1b1edab0b7e15d0e1c857d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 19:51:44.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed563cb188af49dea1559efcc1b073ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1319a64e1b034b69a21c975dd94161ed : SUCCESS in 1h 03m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25c68eab7ae847c9b9f5902b62881577 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/35cc15d6ec8c46e8a95c4cd57041420f : FAILURE in 1h 20m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/636be83ccfaa4b00b93861ef65aca203 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":5},{"id":"0628d35a2490d27a5691753959e2ae46ae895fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-09 17:00:53.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"63e1e5313e5d2542306bcb6c7d338f97be32596d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 17:55:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2395ba53d34b3cb22615f430609424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0524beac957a47548c86469221acd48a : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f59e6f15e9064354ad48ae8453660e72 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35fc09899d7649cba2d5c39173a04863 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cf2affbbca4415d8f7829dd4c55d7d6 : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e229990c09ee4de7b22fd007f01ea7f7 : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26573afefc18432593b7370ad927634d : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5ac02b2d16943ab8f8b72e885edc064 : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2b7fb1ba66c455c9f460107b78a08fe : SUCCESS in 19m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3e54305f66e4a50890f05a7dbca6f4a : SUCCESS in 26m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f10e8577be994d428b68934257ae4dd0 : SUCCESS in 25m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b5de1c87dbd4381baf55ba0d2a50b10 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/beb9ee63dac74d5f9d9df376c0b81ee4 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2621a677f4074f77a80d7c18936ef504 : SUCCESS in 15m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb9631d0ade8402f8e5c64fcb81f75d7 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa8f411e912c46d19583c4e1b4139e8a : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0d5986855fa24d92bde8cfa4e9d170d1 : SUCCESS in 17m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/365693bfebab45adbe85f6dfdee55892 : SUCCESS in 35m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b6e527c85f84651b8a645c348afbd43 : SUCCESS in 52m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7068ab3ac414a068fa240407e6e7ac8 : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462ce3ccb6b94ac199166df2f4fb5cdd : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49973ae540fe48819fbe23f887886760 : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c6618233a5ff77d52a46d9c77b92581b48c45086","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-09 18:30:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8d99cca8b3614019eff9bdf87ef25241977508ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 18:34:15.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c1298ebe864c3984a8a553c4a01fb4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad1fd3e46ee34984a6ffb66368acb409 : SUCCESS in 18m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9007484dfc6940ed9bb5946a390be677 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cc943dfe83e497787f7d71404b322ba : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/34630fb57a164859a38a13f6381d6a1a : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":6},{"id":"4392e61ffcc685be4d12c0e1be53f893479aa58d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-16 19:03:07.000000000","message":"Abandoned\n\nsquashed 914713: expirer: new options to control task iteration | https://review.opendev.org/c/openstack/swift/+/914713","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"239ae4ee57207cd840d94ca24a3dca7621897a06","revisions":{"ed81ee4fa30a50b298f9c77bf6fd1333a6510f09":{"kind":"REWORK","_number":1,"created":"2024-03-28 20:41:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/914714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/1"}}},"commit":{"parents":[{"commit":"0cf4fc9ee2ffb9cbc1b6061b497f6aa1d569d2e5","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cf4fc9ee2ffb9cbc1b6061b497f6aa1d569d2e5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:34:46.000000000","tz":-300},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed81ee4fa30a50b298f9c77bf6fd1333a6510f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed81ee4fa30a50b298f9c77bf6fd1333a6510f09"}]},"branch":"refs/heads/master"},"9ccdf7a6d6d480aea4b9e5db922d8ca182fa71d6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-28 23:12:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/914714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/2"}}},"commit":{"parents":[{"commit":"e4e5917364a1266b7fad3adcb7c24b5c8ea06fd7","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4e5917364a1266b7fad3adcb7c24b5c8ea06fd7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 23:07:19.000000000","tz":-300},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ccdf7a6d6d480aea4b9e5db922d8ca182fa71d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ccdf7a6d6d480aea4b9e5db922d8ca182fa71d6"}]},"branch":"refs/heads/master"},"821647dd8106b848df7d429660cf598f02319e49":{"kind":"REWORK","_number":3,"created":"2024-04-03 14:13:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/914714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/3"}}},"commit":{"parents":[{"commit":"5c245ad040c91e5899080f729ddde3755c38583d","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c245ad040c91e5899080f729ddde3755c38583d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-03 14:12:47.000000000","tz":-300},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/821647dd8106b848df7d429660cf598f02319e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/821647dd8106b848df7d429660cf598f02319e49"}]},"branch":"refs/heads/master"},"642bc69077fe45e85e3a0f91bf4e9237acfbc975":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-04 16:20:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/914714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/4"}}},"commit":{"parents":[{"commit":"7302e8e929942dfd61300ae3c0526771d0959936","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7302e8e929942dfd61300ae3c0526771d0959936"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-04 16:20:11.000000000","tz":0},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/642bc69077fe45e85e3a0f91bf4e9237acfbc975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/642bc69077fe45e85e3a0f91bf4e9237acfbc975"}]},"branch":"refs/heads/master","description":"Rebase"},"864ff79b068f8dc48637515c9333af5109cc80c9":{"kind":"REWORK","_number":5,"created":"2024-04-05 18:29:24.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/914714/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/5"}}},"commit":{"parents":[{"commit":"64aff91a2d9fd7e4ad399880557b41732133056d","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64aff91a2d9fd7e4ad399880557b41732133056d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-05 18:24:21.000000000","tz":-300},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/864ff79b068f8dc48637515c9333af5109cc80c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/864ff79b068f8dc48637515c9333af5109cc80c9"}]},"branch":"refs/heads/master"},"239ae4ee57207cd840d94ca24a3dca7621897a06":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-09 17:00:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/14/914714/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/914714/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/914714/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/914714/6"}}},"commit":{"parents":[{"commit":"4107f58699688085fc521b06a476969d834e3d0b","subject":"expirer: randomize task_container iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4107f58699688085fc521b06a476969d834e3d0b"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-03-28 20:31:03.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-09 17:00:24.000000000","tz":60},"subject":"sq: make test better","message":"sq: make test better\n\nChange-Id: Ibae08b198e685bcdb11621c4777d1a39e2c7a076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/239ae4ee57207cd840d94ca24a3dca7621897a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/239ae4ee57207cd840d94ca24a3dca7621897a06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
