)]}'
{"id":"openstack%2Fswift~914768","triplet_id":"openstack%2Fswift~master~Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059","project":"openstack/swift","branch":"master","topic":"open-expire","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-03 08:41:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-04-08 18:59:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059","subject":"expirer: account and container level delay_reaping","status":"ABANDONED","created":"2024-03-29 19:58:45.000000000","updated":"2024-04-08 18:59:43.000000000","total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca43ed08a3916d93e6034067c2288907002cf636","_number":914768,"virtual_id_number":914768,"owner":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-02 21:50:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-03 08:41:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-29 21:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-29 21:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 09:45:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"1a4e8ae9e5c1a25b205c36449df53729c1f57125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-03-29 19:58:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbb5eb5cce3c84fe9127281d96924ccc9a7ea8e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 21:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c9efbb3e264cee9b69569eae919b7a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e090b8e65e704495a21e30e3d6689ba4 : SUCCESS in 1h 23m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8f0b1fc7ea2499fb2144a06b508d01c : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d19da328d4064ffc8bd3c1393899588e : SUCCESS in 37m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4163b6f40db4617b99b89813d7c784d : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bb8c59341e42c66aadf0910cb856e4efb31e26a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-29 21:25:31.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ccc2cced21e4336b4075e891f6dd8d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/109b796fed1d4acead005a9271da28ee : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/93f2aad7444e48dc8e24e9bb6b7e0605 : SUCCESS in 1h 16m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0878ac22d2864977a715e8de095c8db9 : SUCCESS in 34m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87e4307ddfd0426f87f236b87e598c78 : SUCCESS in 55m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68df1c4daca44466b5ccbce7f9d65b54 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef71bdd0642140c39c8b8dee755ba85b : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8656a50dacbd458386006021755cdd68 : SUCCESS in 24m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28b89890ad6f4d21bb9b96e8875d5fe6 : SUCCESS in 11m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69adc8057409421797fd5904f8ff6dd0 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3417ed7b14e34f8ea2f0b69f936b22f7 : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4e2a01316234da282fbe11f86d9c9cf : SUCCESS in 17m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c551fc02d0c4e1ea8e0baf853325be8 : SUCCESS in 22m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/edbdaae14e274146b12bb78f90f698b1 : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0699f09be1594283bba27c09b42cf2fb : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b01e74104314cb58e93d6ddb45a6a45 : SUCCESS in 23m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1aa7d50c85544f7299fa4d8b1760bea4 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a32c9c914c5e4bac8c29f7095143192f : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/783265d478784974bfdd6abe19205fda : SUCCESS in 14m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac3bcb8189d94f78bd729ae07b2e0d19 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f9ad9328ce94cbca1bd412e423b3a35 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/321a894d3e5e4b89ac096e66d23b8f21 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/885b08dba7804407beb18b19b6bfb9d7 : SUCCESS in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/925039abb0b1483a948cb7878b6c6b20 : SUCCESS in 53m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21bf6f4fc5ca42f088b1028806524ca5 : SUCCESS in 55m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b864b38efb374438ac2114cbeb253930 : SUCCESS in 38m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f83cd9e9796e4ec79d837c511a5ac043 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6621b9aadd4ed999bc0b576d3a4281 : FAILURE in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cac68dbb04dd4323b7202d4b55ad3545 : SUCCESS in 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e12c734a13941578c55522200da7f57 : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e951669a594c7adfa5f2ecac4d6bd9da07eda0e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-01 16:49:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fcaefa8544e830cc32a07e3afe1a75671e25a164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 18:08:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9901192e7c35486fb8a8dabbcfc7f3e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727dffd983144abbb1d76db2bd62a4fa : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f946f1ff25cc4c97a381b4ecc6a14cee : SUCCESS in 46m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/92cbb07ecbb04cbf977eb50c0a858d8b : SUCCESS in 58m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c8fc109950441a6be8c68d6e8f2cf40 : SUCCESS in 47m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36f66c37c7a14b78be7f7c56aa9f67af : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/051c19cd0c5945f783e724bbdbb775c2 : SUCCESS in 27m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8c9e08c73404907a44fe238669ae0bb : SUCCESS in 24m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/250e3c1c1d3e42648b6a029c0dde14a9 : SUCCESS in 13m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3a5c42987bf481fb36ef4cbb5714f5b : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9c44df9157546edb094929f1eb5f406 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de2c40c49a284ae29fca2dd794eff02b : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4cbcb73adb864ad09258a87013bdc3b7 : SUCCESS in 12m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/72ba48e271d94d5aacb6f182cf57ef0a : SUCCESS in 17m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cbaeaa06125147c08f6b5b9dd2811884 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25b36bf6747e42a5a2ebbf25a4a3d45e : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b19ad73c8eb341cb96108e3bdf8d26e7 : SUCCESS in 19m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e9a3061f244a48b9bf17f050bfa4f1f0 : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/99ec5d87a6654facad356aa28576662d : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a79248032bed4317870a3225a5eee584 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffcd7561216e41c2b2e193b39ed9ae60 : FAILURE in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/935f283441b14338ac124c126144bf3b : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f1828359baa45edb2f2eb8ed828f866 : SUCCESS in 7m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/508bdc64cad041bdb21a08edd6b11270 : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80593d09828f489aa42adf81ea02595e : SUCCESS in 53m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cc6199fa97c47a09a3f8832621f6a6b : SUCCESS in 42m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12dac82c5acf47648fd77d3a58907071 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d3d4a3df28405799a8ba9485b75a02 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f879af4455654935a52dd445d09bee70 : FAILURE in 37m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb618ecb0aea49fb96e2de65038bd55b : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":2},{"id":"bb3f5e8c5ae8c83ca329b0e2521906ce3de5ef0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 18:13:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6825de2d410471e8aa6863664632607\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bc12a1358424bb290b0755b57647112 : FAILURE in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec8763b96bee42a089964170a6458f04 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a9aeea358b24a8c9f43ea0b96a2472d : SUCCESS in 1h 22m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8132eda33224f82b5a914400e9dbc6d : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9d3272d78a1856e3a02d0058473304ee4d3c461c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-02 03:05:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e325f8002df27019154922f74981650e78f2db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:25:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd8181e555ef4905aea33b2495533472\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447c6930fe434173b7d8f9611df2595c : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/473f620be2704b2fba2ef27254b1b2fe : SUCCESS in 54m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d830e6f2edbc497aaa9e16585533bff5 : SUCCESS in 55m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1922bf17d9ab43a0b774d39786987039 : SUCCESS in 59m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f372229fb048cea9448a8890f5bdb5 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9253846354e4c2ca974863254094577 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35e27be92cb6412a88463a9f460d6168 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73d01acaf5eb4862a4007360532b5a28 : SUCCESS in 12m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c581ff59cacc4a78b6958b478fed0363 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b83a640056c347c19e018bcb57317322 : SUCCESS in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dff55c2c24544bc999f7494ea1195653 : SUCCESS in 16m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f0817d845dc44688aa371682586f5388 : SUCCESS in 16m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d3bd382d90e540e2abb41d968a337b07 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed3fe227297c49b791ac0efeae917fbf : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/569f3668de694aec88dd0540d3c63eb6 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a17972fb7aec4dc38b9abffd81d818d7 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b712fd6b4e854e0f8b480db029b4cbb1 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e29ce73f020a491483601f06591fbf43 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b96be04a524f4f46ab51c09af950c6c0 : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e02a897575c34b6292c5978aede0b718 : SUCCESS in 9m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7dccfc07e2e49dcb79627f998edbc30 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee899a3b5d0848f680503a938aef5db3 : SUCCESS in 6m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/18cffa1c5de041f5a89992359de3318f : SUCCESS in 49m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/04478b2848e74971a6a59232e48b3af0 : SUCCESS in 36m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14baad160c534662876f8e36d5ebff2e : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e72abd0c54d4fd6830b60b80491b62e : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0bb2d7f2c6442fca0dd94d826b09b8b : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bdebb7d39e2430da1954b10180bd6f2 : SUCCESS in 20m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d05403f6a99c4881a2e123350eb12526 : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e89b78f0f5028bcbadd261621cd0d5a556fb7d01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:41:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50bedef6cefb4bb5977e9dcee7780e10\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd251ee037354b2fa667ddfa75a5692c : SUCCESS in 1h 27m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e0c02ea37b947a280494d28de84e2ec : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8dec78dc8364774adb0140c3bb2a90d : SUCCESS in 1h 13m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcd31d946feb415ea1b26be04ece3253 : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f61f59c640fffe3b87a2d9479cf1335c5f24c6c1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-02 09:45:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fa248a2c2345979f9ef69fb7fdd2bbc69f5b872d","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-02 20:33:42.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bd897c2db218138476cfb0c9ecc917e945bb2d22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-02 20:33:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ddfb85d0db11b01d89142af36de92c86f77c7a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 21:50:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c69f4e41a442e894ebe97666c3d76c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a002c1d003c8470d8c7616b6c6b7e7a7 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e995cfd3af5346de9b0ec7656b3951da : SUCCESS in 48m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/644c71b36fb640ccbc7b35ddd7c37e39 : SUCCESS in 35m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa8265ec783848e7afc68cd9aa02f6a9 : SUCCESS in 39m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e818717e2a674ed79ba9a927c12e79e8 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed6dbfab342d4279b20500f259009a81 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e26be96f99a456fa69ff3c0f3a6d4fe : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6dd27972e0ac48adb233145ffef7631c : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7915d7e9380b42cb8c7dec5f27024277 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2168f01d0ea74bb181f24e2c2d9508ab : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/413e9072a5844785b5df7644667f6134 : SUCCESS in 19m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/edef1fc7e4e44d63be64a3a1ddacd66e : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/49185055e3714f80ac289209ccd36ac0 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f1ec4e9af8b4178b1cf104cf5b4a513 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5dbdaaab4464ae995f385fe00fc7d36 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c10722a6b5a54705bb27d57f2f7dcdd1 : SUCCESS in 18m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d62e967f32114c06b44ff0d47f943525 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e91ea4b738bb4103b6c14703408461c7 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6dee406076c49daa7e43510af605cef : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5550f9112da44c6880291c8308fa6862 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e67ba7e20a044b486fae94341421890 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97b8e7e4a5084dcc9d0975c14f5684bd : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1551dc74990a4ecab21e39d11982d014 : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59654b5cc76a43d5bb142f1641d0a926 : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ce2dfe240504ac0b6baacd328b62361 : SUCCESS in 47m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1aa539888d040faa5b5b2dc5d3a8ba3 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642a25cf989e42179fd78949d050a948 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9de826a9e7ca4d738d9829371e22132c : SUCCESS in 16m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/150f138a09044ac2b77f1993747ccdf6 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":4},{"id":"29f08200e1ab830b9666f5ff67f7df43c5de028c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 21:57:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86cd69a8bd32440ab7a217e9901e7412\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9aa0cf7169bb43cba689956f92d0408d : SUCCESS in 16m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ef89c1e50e649f7b34e4913b73d2744 : SUCCESS in 1h 08m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccc822525c5d4d7a8d843c05a74f982e : SUCCESS in 19m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9516353d3a804f56af986205b53967fb : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":4},{"id":"45db39b3a3e5ba732c7b0e7167ef90eb74c3c923","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:41:24.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca43ed08a3916d93e6034067c2288907002cf636","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-08 18:59:43.000000000","message":"Abandoned\n\nsquashed with grace_period (parent)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"419a0ef7ba1599154852586463070729b96319af","revisions":{"a31b5b1f9164cffd583b7f86f406185f3ad6f63f":{"kind":"REWORK","_number":1,"created":"2024-03-29 19:58:45.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/68/914768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/914768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/914768/1"}}},"commit":{"parents":[{"commit":"7bffb150fd2281e125cdb5ac99a627aa035a5ead","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-29 19:54:00.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-29 19:54:00.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using account and container level delay_reaping values. The\ndelay_reaping value of accounts and containers in seconds is\nconfigured in the object server config. The object expirer\nreferences these configured values to only reap objects from\nspecified accounts and containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nChange-Id: Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a31b5b1f9164cffd583b7f86f406185f3ad6f63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a31b5b1f9164cffd583b7f86f406185f3ad6f63f"}]},"branch":"refs/heads/master"},"43bcfd5a7065fc772a1faf377b1e938740be842e":{"kind":"REWORK","_number":2,"created":"2024-04-01 16:49:33.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/68/914768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/914768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/914768/2"}}},"commit":{"parents":[{"commit":"7bffb150fd2281e125cdb5ac99a627aa035a5ead","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-29 19:54:00.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-01 16:49:22.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using account and container level delay_reaping values. The\ndelay_reaping value of accounts and containers in seconds is\nconfigured in the object server config. The object expirer\nreferences these configured values to only reap objects from\nspecified accounts and containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nChange-Id: Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43bcfd5a7065fc772a1faf377b1e938740be842e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43bcfd5a7065fc772a1faf377b1e938740be842e"}]},"branch":"refs/heads/master"},"14ca15d5eee0f04e8abaa8c62ae9bb2360eb1899":{"kind":"REWORK","_number":3,"created":"2024-04-02 03:05:51.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/68/914768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/914768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/914768/3"}}},"commit":{"parents":[{"commit":"7bffb150fd2281e125cdb5ac99a627aa035a5ead","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-29 19:54:00.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-02 02:52:56.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using account and container level delay_reaping values. The\ndelay_reaping value of accounts and containers in seconds is\nconfigured in the object server config. The object expirer\nreferences these configured values to only reap objects from\nspecified accounts and containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nChange-Id: Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14ca15d5eee0f04e8abaa8c62ae9bb2360eb1899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14ca15d5eee0f04e8abaa8c62ae9bb2360eb1899"}]},"branch":"refs/heads/master"},"419a0ef7ba1599154852586463070729b96319af":{"kind":"REWORK","_number":4,"created":"2024-04-02 20:33:49.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/68/914768/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/914768/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/914768/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/914768/4"}}},"commit":{"parents":[{"commit":"7bffb150fd2281e125cdb5ac99a627aa035a5ead","subject":"expirer: per account and container grace period","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bffb150fd2281e125cdb5ac99a627aa035a5ead"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-03-29 19:54:00.000000000","tz":-240},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-02 20:33:10.000000000","tz":-240},"subject":"expirer: account and container level delay_reaping","message":"expirer: account and container level delay_reaping\n\nThe object expirer can be configured to allow access to objects\nin specified accounts and containers even after their expiration\ntime using account and container level delay_reaping values. The\ndelay_reaping value of accounts and containers in seconds is\nconfigured in the object server config. The object expirer\nreferences these configured values to only reap objects from\nspecified accounts and containers after their corresponding delays.\n\nThe goal of the delay_reaping feature is to prevent accidental or\npremature data loss if an object marked for deletion with the\n\u0027x-delete-at\u0027 feature should not be reaped immediately, for\nwhatever reason.\n\nConfiguring the delay_reaping value at a granular account and\ncontainer level is beneficial for being able to keep storage\ncapacity consumption in control while maintaining a desired\ndata recovery window.\n\nThis patch also adds a sample configuration, documentation, and\ntests for bad configurations and grace period functionality.\n\nChange-Id: Ie1fe4a2e821de7966e77e0acb8b641ed81ea4059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/419a0ef7ba1599154852586463070729b96319af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/419a0ef7ba1599154852586463070729b96319af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
