)]}'
{"id":"openstack%2Fswift~914828","triplet_id":"openstack%2Fswift~master~I4b5b12a3b0288b696a39903264741bc862a94ad7","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics","attention_set":{},"removed_from_attention_set":{"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-05-04 00:29:34.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-05-03 21:44:44.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-03 22:09:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I4b5b12a3b0288b696a39903264741bc862a94ad7","subject":"Move statsd testing to its own module","status":"MERGED","created":"2024-04-01 19:34:40.000000000","updated":"2024-05-04 00:31:02.000000000","submitted":"2024-05-04 00:29:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914828-tmp_labeled-metrics","meta_rev_id":"58df4d5049db1aff6b1761a2bb616b3ef461c341","_number":914828,"virtual_id_number":914828,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-04 00:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-03 22:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-03 22:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-01 20:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-03 14:48:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-05-03 21:48:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"fb81b0596b5417c11f52c5f20d8b902484cfc5c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-01 19:34:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d893fd337e3954997b62d244279b9445b724c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 20:46:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ddf03b15f14a9d9760abec56dfd1e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37331045ff92465daa393073bf45ed60 : SUCCESS in 8m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22812144d4294c15b3e99ec05baf1f12 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b1c4b15a1434969bf3e7d78f3f4ba15 : SUCCESS in 21m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbcd8d2e71e44e618a5f7fe799197efd : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa414759837a4fc293774ab600abf50c : SUCCESS in 12m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f3bc00b2feab4c3daa595b6aba0008ba : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/686ad6710afc417097d96a7f31f29052 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ca6b1d4338148768d3a25aaca4ff681 : SUCCESS in 18m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b7833c9942714fc293deb75937ff1a15 : SUCCESS in 14m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e90467f8a49e4daeaac6b50a161005ec : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/713cda4482e547d38b49df8acb5786a5 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/639c4fd4ce2b4a6d808084da254cbe1c : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eac86468146d44a4a519bd51831e78d1 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f0ce18d9d4d54bca801d49c2741562ac : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8558bbaabc64417fa0d32a9458ec9ecd : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/23586128385b4b378b6480902e7b310e : SUCCESS in 16m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75d0ce73d239430698ce33baa2d5778a : SUCCESS in 34m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed667a1161a640709c880c9f1eb1a001 : SUCCESS in 53m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ec4e04474854bb39e8f92a7d1553211 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a5685f8caa04996856fd79712f40c6c : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/912d134494fc408ca5db4be87ce80e2e : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ecb0b4fb098cfcea00a14d436aaa58334fec35d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:06:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc6db409d0045eeb3b788019f9b76fc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb860bcec93e4f49ac916a200089a554 : SUCCESS in 1h 30m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2d18ec5323d4d79b20d610d326051c8 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e632b6943114cdf91f91901feda9679 : SUCCESS in 1h 11m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/04bc4554efd04488b17922ab7202dcbe : SUCCESS in 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5d2285d2b92087e63c63e7e710243137da58da85","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-03 14:48:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e365904472ea5e780c6e5649a8227a2c10db6e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-04 03:23:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c147fcb1086847cf671a439fab4022f6a68e31b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 04:21:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c943755ab5049e9b0d203882e55c74a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d65f05c10449168e682284f7cf6db4 : SUCCESS in 8m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/988ba418659c47529fc0f198ee036bdf : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28e535a93c5a4a6d8541e73b10ec445e : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2464f1766ee40ec9ba661ea5668e7c9 : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d517e836efad4948b6170e02d29272c9 : SUCCESS in 12m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fbff451dcd5f42d2809821942d945c2f : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/14beae3e6d23463888525819e5673bbf : SUCCESS in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0bf006d5d13411081a5bab93112eee2 : SUCCESS in 17m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/62ab724658394c35801ee12a5506cecc : SUCCESS in 19m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c6bb55126b4a4be7902a5d90cc6a747a : SUCCESS in 18m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/65f5ccc1282f4953a8bfa3b5e749812a : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/91119bd578574671a6cb0871ee4ae86d : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fec6fd3c0b8f4ef78b2b052d8f1a7f2f : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2139615952c54d2ea752d14089c78f4a : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/286ce30ecf3c4a0db94199c8568f3c2f : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/154f48479b6748a1af82fc717f2515a4 : SUCCESS in 16m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c82f5a61c564442cb3481bd0440e2528 : SUCCESS in 44m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f823ae614c8c4ad3ad04a125818217d3 : SUCCESS in 39m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2216c47944ab45e79de4bb84329f01fa : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb783e11c3e4b8bac9fd710e3fad895 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f569783407504162804415911de847ec : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d74511876bda31cbcb4c9c094b7d520c23313ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 04:36:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c954e1894a6f410a9a8be0a3bb4e0081\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ac0f0e0c70b4790a067253ecda1fe88 : SUCCESS in 52m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c22d0297540f4abf972d6dc0bf9a67fc : SUCCESS in 58m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eaeda2d7e2e94843b739115cae04aa3c : SUCCESS in 21m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da93ecb05b2b4cf6aa3b970fa2952422 : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"208cbb56ee5c95df2f700170a3f299aa43bd6ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-07 05:21:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"667f04fc2f85873f12cdb9dcde23e78a939bbd24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 16:41:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5dfbee681ab4de88fad082fffd15c2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9029cbbffcdb4c86bbb9e5d27eb3a77a : SUCCESS in 8m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f74bcef497d4491f9395c434221148eb : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e97fe0dfe32469fb77c3cd496be025e : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29f2393ea1e440dfb5db3e47a10fcb2f : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73f39dee9d3746fb810b4d582f324d43 : SUCCESS in 40m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f44167faa377418896374b92810462f0 : SUCCESS in 40m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9156195669e455dbcea98b4b0a0a943 : FAILURE in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4de45331b49946b9beeaab27ea1370ed : FAILURE in 19m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c83b550c20d04da58680ae9310855f5b : FAILURE in 13m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/92fbf397ac4d49d59897e79b1745f3ce : FAILURE in 16m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a639f5918e8a4928bed6d41ce8daa7f5 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75b92a729f38482cb392c111dc4840e9 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3324c0fdc2f648fea2a6d77a1734e354 : SUCCESS in 15m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc4c9933d78c4bb4a872051aba6f1dbd : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/66df20edfa684888a862442e3c674c28 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99052a07d3e44745885ac228f9d02405 : SUCCESS in 17m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/81886c360729416caf6453d619943771 : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9d999a22afe4d10b40894daf1b054d8 : SUCCESS in 56m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9b3bda5ba0649d6a253ab7c8cef5ea2 : FAILURE in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d25353c5bb0489dbc1f6765ae678bcb : FAILURE in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e6c3f0c69bc4df7986be2f3bd02c87d : SUCCESS in 19m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/utils/__init__.py","accounts_in_message":[],"_revision_number":3},{"id":"9acb7547899d9dc07e479a68cb42e229881dd4f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-07 17:19:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0fb51dc34744a46a4b0ff007c8f2df8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f16a7b41fc0e42c8a60578658dfa4312 : FAILURE in 59m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f179f5a0097946b9aa953c9022918429 : FAILURE in 15m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/880b68c69bb1458b80d8a36560ed9a9e : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9438087ca70b4c339883101532ed2029 : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3b97fe177625c2b7ac5c71d8fac3e14e0e5fc509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-11 19:41:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e6487dd227595abe6daac7ab5372dad90f64c430","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-04-11 19:46:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"105513f6658e4aca016534501b5b550d3016c515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 20:32:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6844eebb9eef4179a6ded92aa9917f7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac1265b5d584b90a26e505ba3cae756 : SUCCESS in 8m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a54cc767cf7241629dcafa079f744ea0 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4a4ef240453445f831c3124a36ed464 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52767436027f4b4f9dbb4477d708aa4e : SUCCESS in 20m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/50d3bb06a35947bfa1c4bee3d759be1f : SUCCESS in 11m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c38c1394d9fc4ddb934647404902d083 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c84fc347e2b438d813fcc992a653c1a : FAILURE in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/baca8a69c7454021af2908d87a7351e5 : SUCCESS in 16m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c2bcb5d3b3c34ea4abbd743ff97c4c7e : SUCCESS in 17m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6ca32104c1ac4fbdb17667bcea407dc6 : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf8ac853f20540f5a306ce3d1b1feadf : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1e448e9548f840c9afa7b8a9a2faeb5e : SUCCESS in 20m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/939d2ad7b504429fba8d0264ffd4a51c : SUCCESS in 19m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/80b8e06e43134f17a375483aaf3f2cff : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6fa4e729812d4717ba53a00b6b3cad37 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b23da89902ba486a8853f42d5e0f72a6 : SUCCESS in 19m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be9670614bec4f77b58bf1d4019a09b2 : SUCCESS in 41m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f11f5d0933dc4bc4b3dae4c404dde27f : SUCCESS in 39m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1ff37ce59ac4a4ab43621e083764ef2 : FAILURE in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8954cb8b4cd7469687cdd57b8b6234dd : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7530512eb8c4cac9f3ff04477b5a1a4 : SUCCESS in 18m 56s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/utils/__init__.py","accounts_in_message":[],"_revision_number":5},{"id":"12540d0623d4b662005dc1b9e63a17a21997060b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 21:17:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac56820f8b647b384b194bcafd13c83\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/400debfd93bd4afea0f0f919f0f4d956 : SUCCESS in 15m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01450dd47cd747f6b1ac109eb0c3f81f : SUCCESS in 1h 00m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/56446c2e5d7744fb903ceff32694a900 : SUCCESS in 23m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/849e0e0a7c3946d4b03fbdf2c03edec1 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":5},{"id":"6f5b1951b81109145f77ce6429cc46c7954a9848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-04-29 21:21:32.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6475052f54fa1fc8fa017172cb40fa7485573e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:36:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73b5a4891238463ea86b4c4ead81b5dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c7e92ea11434831a08af57435100f54 : SUCCESS in 7m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2f257ecfd8d46129b956fe93b0cb7d2 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9ddcef2382d46ed9e5715b574ea8f15 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d239fd97223841e187140a7248faaf40 : SUCCESS in 17m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb9b593ebbd14b4498acd39c51343028 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8448f1a18554b6c92f8f6c048672ef9 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b8f7e1de917405badf6b3dfa4dd7267 : FAILURE in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47b7eeb79cba469dbd0c9e82310faecb : FAILURE in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/41d450ad90ba4e3a98b64187b0ec1a21 : FAILURE in 22m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/492312a268ff4c81b1d92ef77f2d83eb : FAILURE in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36fcb12285284c9fb14bdcfd56506176 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5852a6dbf9fc469590a719ad6b2bdecd : SUCCESS in 23m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef627069a0f7449abb175df592b3dd41 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a88734e5f069478c8cb1cd872f8b3086 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8aa1832863ba4aab90674dbed355b599 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b85353803c04424a9ea312cfbda5d703 : SUCCESS in 19m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97ce263623ad40c0815c5f069924a01f : SUCCESS in 44m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c75cf337fdd45b78be6ed87cb2d80e5 : SUCCESS in 54m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8245c57dba4b28a815f857b1401ea5 : FAILURE in 21m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a78c291ad64702a525f9db300436b3 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d7577deabff46b988291d389063d1db : SUCCESS in 19m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file swift/common/utils/__init__.py","accounts_in_message":[],"_revision_number":6},{"id":"762828b0e3931e7254babcaa8817aca569191f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:37:48.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977b96f44d6a41b9a0bba0c58795be05\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19bbf83ca574492bbb9716165f9fb9b4 : FAILURE in 1h 14m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7292252fbe054d6893443775b51c5d8f : FAILURE in 17m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3738571cb075429b9e9655a2a3973e60 : SUCCESS in 59m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/10c1602f5eb74a678f314dfbe1fb5284 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":6},{"id":"15362f88e27a09909b6bbce58548a95b67ce78fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 21:40:41.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"467f90a4429c6da7014eb99fd0687f5d2b0da833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 21:44:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6bd96ff208610fee293d6aaf2561b2348bec240b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 21:48:25.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a7311e4743b943eaa6964fc679ea60a7007688b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 22:09:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9e4ad40a81b5e1687651f4b3a70e8eb97a8152e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 22:09:37.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"644934abb6de732a84c5d50d65e5057c2bb85ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:59:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91017f5172af463ebf49e49e8ef93cc4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82271ffeb4c248d2baca402dafb964d0 : SUCCESS in 49m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f13f6a85c893482394e9d4fbdb482d60 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f16c56f94a1c4798a3b3b9768812a6bd : SUCCESS in 36m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d630aa29f0b1475d9a5898653a746db7 : SUCCESS in 19m 27s","accounts_in_message":[],"_revision_number":9},{"id":"cf18cb5f2ab99a294e2dc10f08082bb703fd9515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 23:12:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d161799d6df3421286521230375f1269\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc5ab7805cc4fe8a17e6318d02b906a : SUCCESS in 6m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca70218f0d0544f98be1bc76be876136 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dffb48f76bee4e1bb55a7c6ca8a3b23c : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88ad463f4fb14db292ba72e29c8d295f : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5499db74e3de4226887ed2791d9fc7e2 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7adab373f0e94b3c9e02a39c19733960 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc0380db1ebb4af3838bc34b070e6d4b : SUCCESS in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8dc6dc05f07c498cbfc8d71b74eff7c9 : SUCCESS in 15m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/95414d48bb4248fb9d8cd2783f1c2704 : SUCCESS in 15m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2555c5986f16487793fd30a86393d96f : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83b4114bae1643c688528473a1fb484b : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2d3081a1571d48bca6e3aead59b3c5a4 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1e4cf6c65f344368c40504709bdd1f9 : SUCCESS in 18m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be8cb8082fc1410ca99eef8843fb6b48 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a73b366e8b7e4fa9844d2ed8d30e35cc : SUCCESS in 16m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa17518df5b14b8da52f8c1cf9bbad14 : SUCCESS in 16m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6cfc63fd4bf429a91b4994f7c00f5ad : SUCCESS in 56m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/490fdc2c391644d68ffca791f418ac9d : SUCCESS in 34m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6d39daccd34cf19073fedd8e214194 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe071f294234663b91d4d9617992b81 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19754a1899f74a16a8e210febc7eed74 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8eab2c0ec718f3d819a524088b50b5ea00385143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 23:12:27.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"aaa0bbe7d9ba90e56fb12f1dc3d7e1a4cbff84c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 00:29:34.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9647d652a5d44414b4d2a9627baad633\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76c794aa90c442c91a89b01137b3799 : SUCCESS in 6m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0e6f827c52444130be190ac9ee655b5d : SUCCESS in 8m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/34310e3b362247dc9b8966327fcacc71 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/861109214897424e9e5239421e69a1b6 : SUCCESS in 16m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f261791d9b94411a9ee1e4dee387c95a : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5ca29545210476284c0f85615be24dc : SUCCESS in 14m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc564d7b464f4a3aad6620f34f63d3f7 : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d476d9955a38400cb00563b0e6809b92 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d03a3d072fda47f59a5689cf7f79a469 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65a1db8d6423402bb47c3bcb18d9b8aa : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5541e70b47604a588642bb1025314950 : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3ce8d08a539f433e8366527e1ffbfbd7 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4234b315c5ae4f0691a799648fc5e04c : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b52738b1f8cb463e8a8258779b25f173 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/114432f468e346d9982f385d40f0e1b1 : SUCCESS in 7m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa24130abc6f41fcb3a5f24877a3b184 : SUCCESS in 58m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/315eecc994fa49c4b52a280982c48764 : SUCCESS in 54m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f20c2fd8384d5a9f32990e29b367d6 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bee28bd26d846c58e9ebd3c4ecbd573 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":9},{"id":"0e229636bdc95ac192f5e75de31144d0cdc8019f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 00:29:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"58df4d5049db1aff6b1761a2bb616b3ef461c341","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-04 00:31:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/effd52415b414ae3a64951005a61878f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f6077f291a40758fe5c7ae50d57161 : SUCCESS in 1m 09s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ebd9cb897a934d9d809adfdafc5d3484 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/79208cb52e7c464c9bce3a4f2d70912b : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"9da22bb5fedd900bbe004aa76fb163abfb769bd7","revisions":{"5e13de119d3039fb097e3089edeab5e4a3ce1ac1":{"kind":"REWORK","_number":1,"created":"2024-04-01 19:34:40.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/28/914828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/1"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e13de119d3039fb097e3089edeab5e4a3ce1ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e13de119d3039fb097e3089edeab5e4a3ce1ac1"}]},"branch":"refs/heads/master"},"7d6614bd60517e81a8e7a09c05774aa357df30f9":{"kind":"REWORK","_number":2,"created":"2024-04-04 03:23:09.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/28/914828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/2"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-04 03:22:36.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d6614bd60517e81a8e7a09c05774aa357df30f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d6614bd60517e81a8e7a09c05774aa357df30f9"}]},"branch":"refs/heads/master"},"348d19142eb144fd5aaae99db08bda27391abec8":{"kind":"REWORK","_number":3,"created":"2024-04-07 05:21:27.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/28/914828/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/3"}}},"commit":{"parents":[{"commit":"b3c421acbe7e1db5f57adfec855a4a77a25cd775","subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3c421acbe7e1db5f57adfec855a4a77a25cd775"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-07 05:20:58.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/348d19142eb144fd5aaae99db08bda27391abec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/348d19142eb144fd5aaae99db08bda27391abec8"}]},"branch":"refs/heads/master"},"85c73a3531a4b52ab5df478e25ce7df7fe1e1e7e":{"kind":"REWORK","_number":4,"created":"2024-04-11 19:41:20.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/28/914828/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/4"}}},"commit":{"parents":[{"commit":"37cdcef79c1087f2534c285bab33af37e66f60b7","subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37cdcef79c1087f2534c285bab33af37e66f60b7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-11 19:40:45.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85c73a3531a4b52ab5df478e25ce7df7fe1e1e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85c73a3531a4b52ab5df478e25ce7df7fe1e1e7e"}]},"branch":"refs/heads/master"},"2cd5fc46a675595fe924b6c1626d1a826616f2fe":{"kind":"REWORK","_number":5,"created":"2024-04-11 19:46:45.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/28/914828/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/5"}}},"commit":{"parents":[{"commit":"37cdcef79c1087f2534c285bab33af37e66f60b7","subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37cdcef79c1087f2534c285bab33af37e66f60b7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-11 19:43:46.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cd5fc46a675595fe924b6c1626d1a826616f2fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cd5fc46a675595fe924b6c1626d1a826616f2fe"}]},"branch":"refs/heads/master"},"4031ec4599650757fd764a9f221069b11916a254":{"kind":"REWORK","_number":6,"created":"2024-04-29 21:21:32.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/28/914828/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/6"}}},"commit":{"parents":[{"commit":"0bebadf20ceef1db47c36cd127099e21d0997fbb","subject":"Refactor utils - Move statsd client into it\u0027s own module - Move all logging functions into their own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bebadf20ceef1db47c36cd127099e21d0997fbb"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-04-29 19:59:09.000000000","tz":-240},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4031ec4599650757fd764a9f221069b11916a254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4031ec4599650757fd764a9f221069b11916a254"}]},"branch":"refs/heads/master"},"035e993407c134d2630a56e7a2a543a3a121637d":{"kind":"REWORK","_number":7,"created":"2024-05-03 21:40:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/28/914828/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/7"}}},"commit":{"parents":[{"commit":"ec3224cc18879c8130421e629134704b6562534f","subject":"Merge \"Clean up watchdog threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec3224cc18879c8130421e629134704b6562534f"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 21:40:31.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/035e993407c134d2630a56e7a2a543a3a121637d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/035e993407c134d2630a56e7a2a543a3a121637d"}]},"branch":"refs/heads/master"},"f1644f499e8a26e91d1114e76da7940a0dfabf3c":{"kind":"REWORK","_number":8,"created":"2024-05-03 21:44:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/28/914828/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/8"}}},"commit":{"parents":[{"commit":"ec3224cc18879c8130421e629134704b6562534f","subject":"Merge \"Clean up watchdog threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec3224cc18879c8130421e629134704b6562534f"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 21:44:39.000000000","tz":-420},"subject":"Move statsd tesing in its own module","message":"Move statsd tesing in its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1644f499e8a26e91d1114e76da7940a0dfabf3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1644f499e8a26e91d1114e76da7940a0dfabf3c"}]},"branch":"refs/heads/master"},"9da22bb5fedd900bbe004aa76fb163abfb769bd7":{"kind":"REWORK","_number":9,"created":"2024-05-03 22:09:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/28/914828/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/914828/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/914828/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/914828/9"}}},"commit":{"parents":[{"commit":"ec3224cc18879c8130421e629134704b6562534f","subject":"Merge \"Clean up watchdog threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec3224cc18879c8130421e629134704b6562534f"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-04-01 19:29:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 22:08:23.000000000","tz":-420},"subject":"Move statsd testing to its own module","message":"Move statsd testing to its own module\n\nChange-Id: I4b5b12a3b0288b696a39903264741bc862a94ad7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
