)]}'
{"id":"openstack%2Fswift~914943","triplet_id":"openstack%2Fswift~feature%2Fmpu~I4b3f525fdbe6c67132384901071d6824c2964011","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-04 18:01:19.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-04 18:01:19.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-04-04 15:49:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I4b3f525fdbe6c67132384901071d6824c2964011","subject":"Clarify behaviour of InternalClient.make_request params arg","status":"MERGED","created":"2024-04-03 08:30:31.000000000","updated":"2024-04-04 18:02:25.000000000","submitted":"2024-04-04 18:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914943-p-mpu-mw","meta_rev_id":"4ef5a13f37f2b8da800efcc250dbbad15c4a90da","_number":914943,"virtual_id_number":914943,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-04 18:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-04 15:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-04 15:49:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-03 08:58:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-03 09:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-03 15:11:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-04-03 18:46:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"347977d4ca233806783c8c438ede1ade1a31e8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-03 08:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c8915874b795e3cee5b8cc5f2e61224f368a6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 08:58:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96cf9e1fc79b4e2eb6e207dccbd8c4c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3d9863712e3407bb7c4ff6c3f617843 : SUCCESS in 18m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8815bdc60fcf474ca817904c115f5d4d : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6702ec5cabb3420088fffb997e9a318a : SUCCESS in 21m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2a7752a7a6f46549461f3d79ef95aa7 : SUCCESS in 21m 07s","accounts_in_message":[],"_revision_number":1},{"id":"994c923856c9a91158d03cc1d6af906f926c0073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 09:52:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83671071f57441a29230693a9af3d120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ee31117db049dfbe78390f23c7a874 : FAILURE in 6m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3461316cccf4f6fab67ea8fee5f84ca : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/724d78ce69884cdfbd9014fff4e1b786 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01af6e95c06745d7a906b7d59e2490d1 : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bf133a20b8b47fbba957b8358fceec4 : SUCCESS in 11m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/af699ccb06ae4a828d70d281f7da0dfa : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a44bacc5129346679bbbabb3e7566268 : SUCCESS in 21m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd20d8ba657493c8ea57e75c92fbbe8 : SUCCESS in 17m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c2f700bd6f489f97448733ae6fb9ae : SUCCESS in 15m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9c4f15c1a8d8448cac48c76454ebb2ae : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d66c43e51dc41639d736f6e17386b98 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f4fa5e4adcb4a36a1864b2ad1b6e79a : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8bf7feca2732441c9108c326036d20e6 : SUCCESS in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b7eb3a77ade42acad3c48d935b6cafb : SUCCESS in 23m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7a24ca5b4e164295ab14f4f51fdb4b62 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f9979975a17493baebae8273ae49e19 : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4e414cdc0954301ae4022ed9c30fa72 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb78519df137464a9ac7b350bdd0183c : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc426d47a6f94123bd2e47ca5ef26bd8 : SUCCESS in 7m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f24f28e01bd142b4a4e7ac4da6b13037 : SUCCESS in 56m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e8172920e1d442ea0ae61ed11349a8f : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5708aa790bed4a22a31e5e1d2daed54a : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4dd73f67c7bf4a3da696a5e8e96ef133 : SUCCESS in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64c04315df07cc15adddf0e7708e296761e184c1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-03 15:11:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0f0eee85449ed905dc3583a4e1a3c9a949bd44de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-03 15:47:38.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6864882ac9d2af38d85bc153422ffd2aee7e2389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 17:17:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/764062d149e04043a342b3cc2f753266\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9baf1ee53871403ba6f74352addad55c : SUCCESS in 14m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c17bbb97c2854cbd813ba43c744eb9f9 : SUCCESS in 1h 12m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/50ac09f5209b4305b81de5d2bf956bf5 : SUCCESS in 24m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e45ab13ab87f4d8287c70acac337a981 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"98952b7e2cd0bab817bcbaf409a76b5ce699bd59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-03 18:46:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"689791f760562a8ea0432bad4bf4efb65f4e5f64","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-03 18:49:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c904420145d4b27157932c444166a625ce014d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 19:44:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d98c18df77146b488a528655d0c05fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf47cc78f8f45d7998afe772e7ca838 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d84dbfb988c5419ca6c96a8c327d45ed : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6e622dbb7ee4d6298834a085929497f : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8a189cfa6024071b7ad9791144bbc3f : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3bf2836dd444111b08786440521ca18 : SUCCESS in 1h 04m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/769ab925ceaa4046822cc56d02e3d878 : SUCCESS in 1h 04m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52131cd7b0a64253b92e22a56ecab1ec : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5313134e184c468b84001eb64ab269e0 : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca8cdc49c33498cb0511a5f23cff50b : SUCCESS in 17m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/44e545504afa4df29be7c3a5c74f5f25 : SUCCESS in 18m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6d606f44fc6461f904bbe8727bcc323 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40a3aa297f2a40f4ab4ba984dfef657c : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/948adc2b25dc45f882c1b273f9714d00 : SUCCESS in 20m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f798516790864edf9b2726f263338f80 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ce30e8aabf14f1eb34f11157729fa54 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ade12e0bdff64e709ff6a2f3b8cbdf25 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f00e30a763f417693649f3156f09b8d : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9f87b04c0114a6bbf14a956ac939ee9 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4cae79f9335049a186591940530c4e8e : SUCCESS in 7m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e23598b6676d401183275e5b5a297f1f : SUCCESS in 1h 04m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f8f5277c3174b458902b0dacf7a924e : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e60f4110334fec88249af2060c7230 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbd98d299afc4de090b08206eafb1a9a : SUCCESS in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f032cfb0e2464c6560ca3fcafbe565ba736216b7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 13:22:26.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c58c3ca495710d1c8b6d75579f79e5ce81c0dde2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 13:22:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"049f67bae5ea5d456ed92bd7616609769682e9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 14:48:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f6f3f8e990a469993e03d5db8e0dea6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3fb93c3a734086840f8c4614c61eb7 : SUCCESS in 8m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2efd4eda7d64636a0f0adc071c8f4e5 : SUCCESS in 28m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/662e09fc71af43299e65fd10b659feb0 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7396c648e57643d09f50bbe1af22be5e : SUCCESS in 31m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44a876cf28ba4767865228789f07fa3f : SUCCESS in 12m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fe51ea82974e4448b4dd90e74cdf7b42 : POST_FAILURE in 16m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8e32b711dfa4cc0bb18c1ee0c12a4c2 : SUCCESS in 32m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/963ace9f2543400ca85acfa43b2b76f3 : SUCCESS in 26m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94f33ca10ff74d69955edfaa137ad505 : SUCCESS in 27m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1cc7102dc8284d0e92e1683b4e068993 : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a2e0ae29f814eb1b6a438c8d5a82264 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/feeef877db874e8dab653888f28ea4e6 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/862281e363ed40898b8942605f49fa2e : SUCCESS in 17m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47aa9a760d4041e8aa0eae4dc02afee1 : FAILURE in 25m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d8a3d44820e42158b62282fcdd9c85d : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bb491a0dbc8542a795dd115bc88592e2 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8766260d696d42e392456f9ad5e46dc4 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eaee97b7542c4883871529ceb42f7ab9 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84a20a9edb43446a8249244f666308ee : SUCCESS in 6m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5676a43c7264ef5829cc10a70f15c04 : SUCCESS in 1h 16m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cb2fa450e2a441499777765ccd230bb : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350abfc86df1452885250949b4bdd30d : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a10d1bfdba7470c80a65661e767dc69 : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63145bc0928401cfb100b37f3ded76d719d48979","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 14:55:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9469104d6f4020610fc4fd8bf34ae7126168aa53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 15:06:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405e841868bd4934bcdcf470b95d2b50\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dc5f24b88b54d86860b6b0403982b20 : SUCCESS in 17m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8774da9a21554188a7a706f37bac5f54 : SUCCESS in 1h 18m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e7818e734454ee584a8c9130351317c : POST_FAILURE in 22m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/97352c462f2340afab3d2f268f5e801c : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":3},{"id":"30cd6cea15b90772cd6e7add82f25e4b31353007","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-04 15:49:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1c1ef8879e84407ef68277b0af2207bb1a4671cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:25:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5560aedd45054627906be3e562d1ff7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be6358596844692b84f3474395235da : SUCCESS in 7m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f93cbc6db9d4cd4a89061b5a87bead5 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b986797893c4cbfbd6c6c6bffcc0293 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c06a8ffc235149e7af5157aaa7c6961b : SUCCESS in 15m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/671a57b167684341bed39ca8e041a4f0 : SUCCESS in 20m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1c51a47505cc42d5a70cac3018fec69c : SUCCESS in 14m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6e637b658934241af003b997d004229 : SUCCESS in 14m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1805cefdf06740d4a91a68891cc0b2e1 : SUCCESS in 20m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab7d81f5a51049c594f4a08ceb6f24a5 : SUCCESS in 17m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb720984cab34f5695f87d8b30caa4f5 : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/417f956088db41629699078a667be99d : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7304659dc1d4e1ea1eaaccfef0d9416 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a5e1219932842d0b4500dd146506707 : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3362d7f290b14280bb69423ea302c845 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b5da573d2684919bb994fad07f85fae : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0a3b7d945966492cbc175702ff53a1b2 : SUCCESS in 17m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b4df4e860e24a97a8984c9ee8d97dfc : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0555fb07b5b40f3a9ba987440c1c367 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1fbbdc599e4438589c2b38280fee11b : SUCCESS in 7m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e3edeea9b824a9e8d0f9c020d2d6140 : SUCCESS in 1h 09m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a501c5fc7341d398ad85a6870d4e6e : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fbc96e2cb08412eb28016e9255f4135 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ca7572a711b43469ca93cca1684ae7f : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d773ef7be7ba80f942ab0b6696709e8220e16a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 16:25:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f892969e93c43c1b9e2d79e52c6c00a674beda33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 18:01:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3413ca4d6840a88089cf8672f96ed6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906f3bea03d140168b6d727ad9e100f5 : SUCCESS in 7m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0d48ab7b83c64d139e54d2ba1677e44c : SUCCESS in 11m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/26e702b7886b46d18d50ed9d98818dfd : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/647b21afe47f45bda60c91b8ba4baa22 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/653524ef723d45e7b52c35ebafee92d3 : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ddd0a1cb039644a585c2609502268a9b : SUCCESS in 17m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4a1b918b80b149589d7ba6b22efdf651 : SUCCESS in 17m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c25a77d93934b7b9aef770f7b398211 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3248a6cac7b044169b8d6dd195d19335 : SUCCESS in 24m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fd4a23cce4841cbb708c62178084cff : SUCCESS in 19m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c42e07ae1ca464d95b4fcf75083c9e3 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/120d0604402b483aa0c7bef63a1e5695 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eaa25f479cd246ae8ddc578987994c32 : SUCCESS in 31m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c99ab8150d24d5095b1404427b0cf62 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0cc4a0f031934a01a7c37d963ec25ec4 : SUCCESS in 7m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/da722c5146d3442e95daa9a42cbe7555 : SUCCESS in 1h 29m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b46873ea7b7b4c95be67ae7ced9c10fd : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc84eeb995f444c38aeacd8c5c7fbddc : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a7eaea455d0cca47f70fe5b85d75461526b735b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 18:01:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4ef5a13f37f2b8da800efcc250dbbad15c4a90da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 18:02:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74dc8b0ef7ef4fba8184d8dd950466c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/692c2284fdd1420fbf3d1a311ebcdc3e : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3347dc8cc33b45e7a80171838e6ce2b8 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/49bc0f04c39a4a8e96e805f3dc3fec13 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f0e1713c8efda8d979d68a8c1e0e39351d701ae9","revisions":{"92499eaf0640ad73f3d997bf59b1c67ebc6c2045":{"kind":"REWORK","_number":1,"created":"2024-04-03 08:30:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/914943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/914943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/914943/1"}}},"commit":{"parents":[{"commit":"d61e4729366202f9627565eac82bc4b1088f436d","subject":"FakeInternalClient: move to test/unit/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d61e4729366202f9627565eac82bc4b1088f436d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:13:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:26:47.000000000","tz":60},"subject":"Clarify behaviour of InternalClient.make_request params arg","message":"Clarify behaviour of InternalClient.make_request params arg\n\nUpdate make_request docstring to clarify that if the params argument\nis used it will override any query_string component of the path\nargument.\n\nAdd a unit test to verify this.\n\nNo intended behavioural change.\n\nChange-Id: I4b3f525fdbe6c67132384901071d6824c2964011\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92499eaf0640ad73f3d997bf59b1c67ebc6c2045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92499eaf0640ad73f3d997bf59b1c67ebc6c2045"}]},"branch":"refs/heads/feature/mpu"},"ddeb814184297fe716af2f77eeba292989ba6ac1":{"kind":"REWORK","_number":2,"created":"2024-04-03 15:47:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/914943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/914943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/914943/2"}}},"commit":{"parents":[{"commit":"d61e4729366202f9627565eac82bc4b1088f436d","subject":"FakeInternalClient: move to test/unit/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d61e4729366202f9627565eac82bc4b1088f436d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:13:07.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-03 15:47:38.000000000","tz":0},"subject":"Clarify behaviour of InternalClient.make_request params arg","message":"Clarify behaviour of InternalClient.make_request params arg\n\nUpdate make_request docstring to clarify that if the params argument\nis used it will override any query_string component of the path\nargument.\n\nAdd a unit test to verify this.\n\nNo intended behavioural change.\n\nChange-Id: I4b3f525fdbe6c67132384901071d6824c2964011\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddeb814184297fe716af2f77eeba292989ba6ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddeb814184297fe716af2f77eeba292989ba6ac1"}]},"branch":"refs/heads/feature/mpu"},"f0e1713c8efda8d979d68a8c1e0e39351d701ae9":{"kind":"REWORK","_number":3,"created":"2024-04-04 13:22:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/914943/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/914943/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/914943/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/914943/3"}}},"commit":{"parents":[{"commit":"d61e4729366202f9627565eac82bc4b1088f436d","subject":"FakeInternalClient: move to test/unit/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d61e4729366202f9627565eac82bc4b1088f436d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-02 14:13:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-04 13:22:35.000000000","tz":60},"subject":"Clarify behaviour of InternalClient.make_request params arg","message":"Clarify behaviour of InternalClient.make_request params arg\n\nUpdate make_request docstring to clarify that if the params argument\nis used it will override any query_string component of the path\nargument.\n\nAdd a unit test to verify this.\n\nNo intended behavioural change.\n\nChange-Id: I4b3f525fdbe6c67132384901071d6824c2964011\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0e1713c8efda8d979d68a8c1e0e39351d701ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0e1713c8efda8d979d68a8c1e0e39351d701ae9"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
