)]}'
{"id":"openstack%2Fswift~915032","triplet_id":"openstack%2Fswift~master~I651ddd40e9115a56727096d4a3aa84589146308f","project":"openstack/swift","branch":"master","topic":"proxy-log-metrics2","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-19 13:29:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I651ddd40e9115a56727096d4a3aa84589146308f","subject":"test: assert behavior of proxy_logging metrics","status":"MERGED","created":"2024-04-03 23:50:29.000000000","updated":"2024-04-19 13:30:14.000000000","submitted":"2024-04-19 13:29:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915032-proxy-log-metrics2","meta_rev_id":"995b0e80c60c07ffd2b47e09c109a29d5c70016e","_number":915032,"virtual_id_number":915032,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-19 13:29:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-19 11:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-19 11:31:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-04 01:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 11:31:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f68ef192212b5d0ea782cd1d18379b6c8d1b5232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-03 23:50:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f86b622d068f2fe4ea5c8f372b6eda25762d370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 01:06:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ef19696908d4a0581d0a869e692d8c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e8db03c2bc4c4c961b468adfd2b558 : SUCCESS in 7m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a7fa6952f524b5e87d9c90a9d74e92d : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca06cec693114b17bb65d292ce7e4d7b : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/471ffc69093d4d87bc1b89475d04309f : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5337c077297d4615b58b7c5083fb7083 : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b0c112c84bb44a894810123ec015400 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8b02e54da98462baf15324dd763703b : SUCCESS in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b723b2efe7b43b1b48522f6a18dac55 : SUCCESS in 18m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2901c2da428d4f569d07e0b313a82c6a : SUCCESS in 18m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d3005eff5ba34aa3ad64b851afaaecdc : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25a96f50a8274a879c30f28506540a06 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a5b8a81791d54d68a2eafb5656336348 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b3bd3a775aaa4162bae0f01c1b415e31 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1478af1c6e8947e28fab615655c5c648 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a7d0a5ab3f14cd5852fbe1227e6303f : SUCCESS in 18m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/279a5dfda6174360be42c0225da166ce : SUCCESS in 18m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ff305e9a42f448e96dec653ba54852f : SUCCESS in 53m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adc98ae1f4ec4200a32ef4937d0539d6 : SUCCESS in 1h 03m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab9064eebd614ae9a207d687c958be69 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abda25a05b941ab86045be680c58802 : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e83a69c3bbc41d99e8230156468decd : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e638487df1c71d3f6ea6879d392c6846e86103f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 01:23:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fced26065c44d99aea98ff62abaa93\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f63d3e49170b4be888494fc0ec80645b : SUCCESS in 15m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/269135b37ece4e5b8719d6802696382f : SUCCESS in 22m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e310476f43d04e7fab3f267fe52c69d3 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6157bfc9710a4c64921a57c2d72e7774 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bd6fd7e7d66e9f4c434429853e72649d9fe1a378","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 11:31:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c7a5f2fa2cf597b798d8524bee7e9524f64e4bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 11:32:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"440a5642519b42563503cc303d90f701af73b249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:29:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b56beb09284409bb7b17a2f87f41714\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2fe5b36d57427094b54c7c9c25a0d9 : SUCCESS in 8m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/460496be7dcb447f9930fef13126db16 : SUCCESS in 9m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/da0eafdab66c496b9d16c6b5ebcd3393 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ebe21c258d74e6cba557648a4ea65e4 : SUCCESS in 22m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a87c052a4a04e088031771e524ac4cf : SUCCESS in 18m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/957d34b8e7264f7ba595876cc92ba0af : SUCCESS in 17m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/96b4fd816a85488fb64fecb2023794a9 : SUCCESS in 21m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/753c327ae04a4301b0b04248f759a926 : SUCCESS in 22m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76fbf76f121348ce95be91a8a308cd12 : SUCCESS in 21m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d0d883376064ddfb2b97aa47319515a : SUCCESS in 19m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6b427b2129d4d29acda33f124b0b5ea : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3b7fa02ec3044022978a07991b6bdc73 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d4e26e5718d9452eae74b5c5c22fab12 : SUCCESS in 20m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48033b9bd5ff4bba9b31dabc37df8c56 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/646614c2bb1a4440a818114fecd4c681 : SUCCESS in 7m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55a6555a29f24f3d8f1957e7ca52f97b : SUCCESS in 53m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd19905914bf4cf28fe49a35e3130e35 : SUCCESS in 45m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3ad693156d14732839932f06c13c55f : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa18d997688f418cac4bdac4264297e9 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5fbc74f40b8df9284e37a10e6c07e589f1a71f2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:29:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"995b0e80c60c07ffd2b47e09c109a29d5c70016e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:30:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9c4ce4a2404dce8694b0e3a2ccd225\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8175bad461544c6f8697bd051143793f : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1bd302e057c54e29afecd575fb754ab2 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6735bab86bef4a939067035759287a4e : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c449e78e481bd363c6a537b59aff3458624c01a","revisions":{"3c449e78e481bd363c6a537b59aff3458624c01a":{"kind":"REWORK","_number":1,"created":"2024-04-03 23:50:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/915032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/915032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/915032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/915032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/915032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/915032/1"}}},"commit":{"parents":[{"commit":"0e5aeb50457345801ad70aed0c367e1f9820e113","subject":"Merge \"s3api: Fix handling of non-ascii access keys\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aeb50457345801ad70aed0c367e1f9820e113"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-03 23:49:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-03 23:49:13.000000000","tz":-300},"subject":"test: assert behavior of proxy_logging metrics","message":"test: assert behavior of proxy_logging metrics\n\nChange-Id: I651ddd40e9115a56727096d4a3aa84589146308f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c449e78e481bd363c6a537b59aff3458624c01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c449e78e481bd363c6a537b59aff3458624c01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
