)]}'
{"id":"openstack%2Fswift~915287","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1fd1daf3bcbb76107cc428b00e13d4c9c22eb2f0","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","hashtags":[],"change_id":"I1fd1daf3bcbb76107cc428b00e13d4c9c22eb2f0","subject":"MPU middleware: support list MPU parts and list MPUs","status":"ABANDONED","created":"2024-04-08 17:14:45.000000000","updated":"2024-04-12 14:22:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"895846c5d05be0421729eb72ccb6fce6e634ab6b","_number":915287,"virtual_id_number":915287,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-08 18:20:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-08 18:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a09f7b85f7c8c498e2703d5ab09674de317c8a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-08 17:14:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b57126aed4958ddb176d1a85d16127969e13c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:20:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3f7b72d88c4daa842987488622c7db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4173ed67f7564a17b6427221463ae332 : SUCCESS in 9m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29afa8b22f98462aa25e1d126d35bdbf : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d07de81ca1b4e8185b0192714094b27 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85c11aa1b64244499f1ebedc9f194ae4 : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1d29e13cf1f4fea88a64e30612af447 : SUCCESS in 11m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7197902e60cc498dbd845a9e106aa6fc : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5521eeb4da1744e0bde1702744803572 : SUCCESS in 22m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c608bbdaa20c4d9d92600bf7862702a0 : SUCCESS in 19m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/30fdcb0cf33b4346b34ede1e6998508a : SUCCESS in 18m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d4d9403316a6411daa6930cbf268d220 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5c14100f30e4912ad2efc1599ecad03 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9331c43ba639425e98edebedb4573e0f : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5fabe8063f448eabfbf55f3073e5616 : SUCCESS in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa208fc5b6324f87852dc0482435bec6 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d5bfb15888e4c80bd8bde5176b2ea1e : SUCCESS in 15m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5e2c76cfff84ba08d9e17a021d0aa5d : SUCCESS in 17m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61d558ce49e54c1b981f94bbe8381c25 : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f30870b941f0443b95671ed838941fef : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/420890a61fd54ab991cd4fc30337f228 : SUCCESS in 6m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a380c68cb9644ed8d8ec9b47ec42f14 : SUCCESS in 59m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48e8812cc45247e5a7b3668e777a4753 : SUCCESS in 20m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df250d6e4c34e17ae5d1a8ebd03b5ea : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67b56770979044418fd01046ec759d85 : SUCCESS in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9702ece91b3500a5322aca8d228994c8fc036bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 18:47:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6623501ba59341259a984053ccbc66a0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1fd3a68c0c3472b8d511df2de7c3d70 : SUCCESS in 16m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6433a032e1af42c9afb87070b1be1730 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50f86b59cd94902b29184d7ab1bc49c : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1ccbb49588b4e79b38c78195344b353 : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"895846c5d05be0421729eb72ccb6fce6e634ab6b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:22:49.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ab917a6e517117d1594a08732a2cfcae3b52c18","revisions":{"5ab917a6e517117d1594a08732a2cfcae3b52c18":{"kind":"REWORK","_number":1,"created":"2024-04-08 17:14:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/915287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/915287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/915287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/915287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/915287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/915287/1"}}},"commit":{"parents":[{"commit":"e2cc09fec856219c8a295b9d1f5e0b905fb5162a","subject":"First cut limited functionality MPU middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2cc09fec856219c8a295b9d1f5e0b905fb5162a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-08 15:08:13.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-08 17:04:05.000000000","tz":60},"subject":"MPU middleware: support list MPU parts and list MPUs","message":"MPU middleware: support list MPU parts and list MPUs\n\nChange-Id: Iaf1ab9705a938d95fbe04e3346f96e47175d513e\n\nMPU middleware: support list mpus\n\nChange-Id: I1fd1daf3bcbb76107cc428b00e13d4c9c22eb2f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ab917a6e517117d1594a08732a2cfcae3b52c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ab917a6e517117d1594a08732a2cfcae3b52c18"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
