)]}'
{"id":"openstack%2Fswift~915485","triplet_id":"openstack%2Fswift~feature%2Fmpu~I3b43bec308f6c5fb583f99095d001cd5e0514b35","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","hashtags":[],"change_id":"I3b43bec308f6c5fb583f99095d001cd5e0514b35","subject":"mpu mw: add session state metadata","status":"ABANDONED","created":"2024-04-11 14:01:58.000000000","updated":"2024-04-12 14:20:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd0d2c1c9e5a6daaf776ae94c22a3c758221657b","_number":915485,"virtual_id_number":915485,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-11 15:50:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-11 15:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1a7a87d5bf61baf068f133a827615844a7b72a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-11 14:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88c3b30e43f8e21ddad1e132fc21d90ae82b236a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/274d5bdc30ec4a1f8f6a11fe2d116911\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913a8659af7b4ca3a8bff66e87c61564 : SUCCESS in 9m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/818d712b15d34b5195a23b201eb0c8d6 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54f5f9d68e4847b4bac1860335ff42c9 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b44a03fa5a654dfd8cd29b393eb26adf : SUCCESS in 19m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/725f92c1a2874787b021bfb595663593 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b00033b8aaf146a5af27f0699771e923 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d8ba6cd4c1b46249df559d10560d835 : SUCCESS in 21m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d226ca25f97e4694a310240f5a8832f9 : SUCCESS in 19m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/532441d7b3824a20b827617f8c7df7b2 : SUCCESS in 17m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e5e0de11be7d468f93939ef35119aa23 : SUCCESS in 21m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/757e26fe1dc0421d9566c12ba7113989 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d28c361ec774b46b45ebd446e412a2e : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d2a530453ff40349ae0990fd9009078 : SUCCESS in 17m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9813593b6053499d94bfdcf399988a3b : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bddbaa32d3924dc780eaf288eee4b27f : SUCCESS in 19m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8b848fc5e42741e1958e17c8f79de9e6 : SUCCESS in 20m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0202d974535488e885c8df1fb507518 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c27d8efc17644dc79e50f0f3f09397a9 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28c11a4e5f004792913d7a48c1ac7fa3 : SUCCESS in 7m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3eaa25b0c50447babae77a1e02d6275f : SUCCESS in 1h 06m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c33fd2d94a94e0d9c7581c10f0cec8c : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330005976cbb49099313979ec6ddf05f : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5715e208b634f55a3edfa8b25f05ba3 : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd68efeed1db9e80d13745bfc02058c1b10cf395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 16:12:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43df016526ab4f7ba01fc1fe4257f2b2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e14650073784e659eb53b2230fad510 : SUCCESS in 21m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a58bc7e611e8469a8e2490ae78649f34 : SUCCESS in 26m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6713a73d4934679bb8c6226fb42711f : SUCCESS in 26m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5167d492aaf64d24afca95a8f4a8e7eb : SUCCESS in 1h 16m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cd0d2c1c9e5a6daaf776ae94c22a3c758221657b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:20:05.000000000","message":"Abandoned\n\nsquashed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a16c6981b1f9319f6a3736febe0f2fdb74f24825","revisions":{"a16c6981b1f9319f6a3736febe0f2fdb74f24825":{"kind":"REWORK","_number":1,"created":"2024-04-11 14:01:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/915485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/915485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/915485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/915485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/915485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/915485/1"}}},"commit":{"parents":[{"commit":"36196452da8fb0f39d2d5f45f14e8e0cc76c80bc","subject":"mpu mw: transfer user meta from session to manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36196452da8fb0f39d2d5f45f14e8e0cc76c80bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-10 16:51:50.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 14:01:46.000000000","tz":60},"subject":"mpu mw: add session state metadata","message":"mpu mw: add session state metadata\n\nChange-Id: I3b43bec308f6c5fb583f99095d001cd5e0514b35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a16c6981b1f9319f6a3736febe0f2fdb74f24825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a16c6981b1f9319f6a3736febe0f2fdb74f24825"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
