)]}'
{"id":"openstack%2Fswift~915486","triplet_id":"openstack%2Fswift~feature%2Fmpu~I5fae2656e3ee594e8abf9532f7bb8c5af802a293","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-acl","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-25 13:17:20.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I5fae2656e3ee594e8abf9532f7bb8c5af802a293","subject":"Add abort upload support","status":"MERGED","created":"2024-04-11 14:01:58.000000000","updated":"2024-04-25 14:46:59.000000000","submitted":"2024-04-25 14:45:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"915486-p-mpu-acl","meta_rev_id":"8eeb94ad01338739f007f1f1c16fd52cf7655b22","_number":915486,"virtual_id_number":915486,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 14:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-25 13:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-25 13:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-11 14:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-11 15:17:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87c282cf8bd9c24d2a3d8c05cb03725ec1a4f1c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-11 14:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbf272543e8842391c5c412acec78dd527e49831","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 14:56:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cbac6dece743658a661400248f59a6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/693b6e09fb8141808d4c17a3f9102b9d : SUCCESS in 21m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4e0518447814215bbc2457ada78cc9c : SUCCESS in 21m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/65e9a785f8924aa093c4a5228d76549a : SUCCESS in 25m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/996a5da4410c4525856c102df3f56c6f : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"102ad802ea58cdffea65e716dce0893e96f6756e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 15:17:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/384d9a3067a64c4f8c60bf295068384d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a0903499d64fb6a3e8e5cd6d08e9a8 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/814022f7ae9f43c3b3c35503a6d7c11e : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac7bb55abf114d369c06299dbe25c09e : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ca951fa22a2418fa3ac65a845232877 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e322d5ac4b7b4d1e8b5f2268025dab2d : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02e20241cb5144fc89e250434b83f441 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5f35a011fac49b6814b522b1714390e : SUCCESS in 19m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87bfcbe43b0846649a6666f87c9c3b65 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3553ba2733ed4b54b349c3f35aa5daab : SUCCESS in 16m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/00e2458790304ee7a8abb848ca566dd4 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38eca1379d4f44609bd864d81e0ba501 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62985d1ccc3f4d3686b69cadd6ba3c3d : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a0e17326e97a4c73b5119161d1def516 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9af05878e3954374be0ba385e95ce1b5 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e9db34bbf79a4e5c9c52a4e0d44e0d5a : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/76cfc712d7cd4ff3b952f5fe54171fc3 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbbbcd2e9ad24aaaa5e71acdb9f5f65e : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c38d910889c4f70817b140f2932cbc0 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e1fed97ee7e45eb9ab91340645e9631 : SUCCESS in 7m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/687245b8d3d44287a9eb6166230b46f8 : SUCCESS in 55m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94d34880aa2a428a8d92124c7e77c8a4 : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f9f99d4a1841079883f13330c9706c : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e139aa416bd41868e03a470eeab723e : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3be57e7a7500fd2dc7c95ef549610ae1522eb447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-12 14:19:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"261035f8311bb8a0919b75b77f9024ea8b37d282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:23:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055b9c73a3be42cf939dad0cfb797503\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e014e487594a4ac59edb25ffe3d41e50 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/564943c217454a589cc50f15ca3f9161 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef584078798049bfa256e08e2c8d1d00 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f5f1234ef7f4171885aaf875f45329c : SUCCESS in 26m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a7b3ab35f904afeac4981a2db598442 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/143d6223df8440638ece2b9011873b66 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a688c7d9c3f493188aae25f3fd5889f : SUCCESS in 25m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd192415d1746eaa35b6878e28eb94a : SUCCESS in 20m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/39e8a2e8abfc4a64ba0f2f4137354250 : SUCCESS in 16m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1e246501b3a249bbad2f79f007b8d7cd : SUCCESS in 18m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66ce86078ea5458da963850330d7c7fb : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f8c237fa1bb44f4928e684ee21bccec : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6ee6b1fcd5b4cddab0f6317920f6d55 : SUCCESS in 19m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ddc22692fdb4b58a8609f159ad1d888 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/01fba49e25a649ba97c2ebdff836520c : SUCCESS in 21m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e5705e64a1d49dba1bdd41d2af0a8f4 : SUCCESS in 19m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9973918a94514018b900ecdd76abd434 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16b2b9f4921d46bbb23da984fb175e1b : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/087215858b4c4d72b6f0a98d153e6a25 : SUCCESS in 8m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/db7b312de39e423d8f1ae79a88350cee : SUCCESS in 59m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d735386b9cd4119804fc11c9c40a52f : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed684bce05f45b28a1e84eea8b67804 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/930ea59b29be466180701532cba118d9 : SUCCESS in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9898ae1b82d8c7bf4b613c7a0c6eceb5e386becb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 15:53:03.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613158e734094cc389c160a35666485b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d02ec13185894d9ca1c6afb797e97a4e : SUCCESS in 21m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8db22b576c8a414bb988d9b6b8c642ef : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/68daf035d23a4857a98584ea47d04a6d : SUCCESS in 30m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d680c5a722ce463fac5cd04d1f3bccf5 : SUCCESS in 26m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4579b7afc1126239263adf42bf8c5833920a157f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 14:57:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c929852664f23ba1db69d632a4358905f0823b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-19 14:59:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6685d9758fcf267c3b1d9311a5ed679f6110f12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:35:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/885df51c94534186b404507c575fa3ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5d70724c894670bd187cfa4f5bc0f9 : SUCCESS in 8m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4039f818fa844ff8e0a8527e70e2c45 : SUCCESS in 23m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67f9fca37c514098a875ee0a3dc94534 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c5482b9860d4526aeb4bf0ec1472375 : SUCCESS in 24m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00f941c3035b4c6d87b27680cf360e9e : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df1db5b1e1cc453f9947a84c7986a02e : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d7102855ddc4bcebfc0a634518b9ec4 : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4f4d6499264bdf9e910972ed938f31 : SUCCESS in 20m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3ea9edcda9849d9837ff042a48c57e7 : SUCCESS in 20m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3439fb1f3a8b4b87995aef7483ced49c : SUCCESS in 18m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63cb741896584866908b5b440348f374 : SUCCESS in 21m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0becceb9724d4b9e90ff8430426c70e6 : SUCCESS in 21m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2571d1fde9e04882b324da357766041f : SUCCESS in 21m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ee2791376cf04d29807d38848e3ed50b : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e884a6c3ef5e41d4b7755610ecc18ee3 : SUCCESS in 24m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/52f8bf30578148698bea468757d5d41c : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cfd733af066464e894cbd00381055a5 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd59496085c44a999dd07c08fc07dbd3 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75a39e375aff49b6859363044be78b00 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/281bf1f5821c4071bfb4cf389c860f89 : SUCCESS in 1h 01m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1de0d84064a347ae9b2aa782556fa14b : SUCCESS in 25m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e969e0aa059456d8c9a575f3443315c : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01156338a3d14e1f81b909cfed398150 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98f51df257e68e00a861f2f9782d1f147f4c34ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:41:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f9d4b44e0ad434bb7906dbc6e35bae1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1701f891a6524414bca99f6332370688 : SUCCESS in 1h 05m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3951c2c5d79e4589a5398d83f4402083 : SUCCESS in 1h 13m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c0bb1e3833464c8c722356a3a0e059 : SUCCESS in 1h 18m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbd13be35f044a01a9cf51027d96a305 : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":4},{"id":"b4d9be468c1cf15f9d356a7166a43910880ddd9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 18:34:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8ace86f847643e505ee8c7875240c5a3b2571d37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 19:33:25.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9508e85e951c41a4b367ffb6baefde8f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3f13158ebc349d1a1e444767b64d2f6 : SUCCESS in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/318748e7252a40c9a86eeb34dbf859dc : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeb3cae2168246b58e900a5da32a06c8 : FAILURE in 56m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a8128b101874ff2873260932feffc18 : FAILURE in 56m 19s","accounts_in_message":[],"_revision_number":5},{"id":"403e003617145edfe481e644f2d434c2b2c59988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-24 13:23:08.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"cfa0895278e38c3721f5418daf7d6ada47e6ca42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:34:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77de8c2c703c4c92b7a78166c9d7da4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870e30d14792467d8b9708314a33f1d1 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b104099a0e634fc29cd66ed0ad1dbce7 : FAILURE in 52m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/864a5bdbf7fb44eeaf1af4773f60c6cf : FAILURE in 1h 00m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b387b0bc0c074bb2a970b9b66790f93f : FAILURE in 54m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c20aa21cd664c8caf7eb35d5c3ba4c5 : SUCCESS in 12m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6e7efc379ed4428a31ecec808f0e913 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e07dba69fa9e43229c37adf85ff55d44 : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e5773898eaa466c9558d00ca873d8cc : SUCCESS in 16m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8808f1f7e2894c6a9d4958cf962cfe12 : SUCCESS in 18m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9d4e774a033745438de5c01e0e7184e3 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a0f96f5880943ff9cb5685fb22512f1 : FAILURE in 59m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/286323dbe6ea4ad6b4e063031522c40a : FAILURE in 58m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a88fe679e034c1ca2303cb7b3d1a9d2 : FAILURE in 58m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f39f3b326c114d7da4049b85b6b05fee : FAILURE in 56m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a55f671493ed47d0b8df7f92c39a1d21 : FAILURE in 53m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4bc125967654c5aab302e035bac47c4 : FAILURE in 53m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e5cb8bad2d6401b9987eafe579a9f75 : FAILURE in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a74654bf24d64cd69a65d77ed94af387 : FAILURE in 1h 00m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8ff29dc5a6e34652b6e4299365584f05 : FAILURE in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c259ec29c30400981af216f1bdb8f1c : SUCCESS in 59m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b4bf71e2aff41b79a4698b6d31e78a3 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e2284b18554539848a07da5d037292 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bd3e1b771b746429773a6903753a261 : SUCCESS in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4350aed3a0d6903b8566cf608e268e891086c687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 14:51:55.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265104612e924803afb2ed50d8355c43\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b75fe4b3ceae4c0b9b65c1ed1502ff94 : SUCCESS in 20m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5654cf2dda8440658062e848bec06bf5 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c2c25b67f674f089b263390a20311e8 : FAILURE in 1h 16m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7053918c5ae4bcc99163ff904c947c2 : FAILURE in 1h 19m 57s","accounts_in_message":[],"_revision_number":6},{"id":"73bbec62866d4a8d63ebdc596678ba2c08075240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 09:38:28.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"64be312a21adc50b8120ba13f0716a0e0502c846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:48:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7e22b9e29b04deba3d03da3cf703dc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9afef0fb9954c95a256636eb6839c79 : SUCCESS in 9m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e033fab273b4f71b4683f495d6c72bc : FAILURE in 57m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dc9d43ad22e46f0884ce9198612ca39 : FAILURE in 55m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95c93f01ee4a4b1dae3689889e7a3942 : FAILURE in 55m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56982749ab924b93b0e3232f2d515e70 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f264f3f57cec4657bc2b84c0787fc5ed : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87d0fa2c5c8543898b588f7e1d434074 : SUCCESS in 21m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a05882315e524a578cd0790be6ba604d : SUCCESS in 19m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/70f6f3d35b67429c98563a4c4be11fd7 : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/acb381f002b146aa98a51dbefc840098 : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cedf02a1369243879d519b250c47de65 : FAILURE in 55m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15500b97ba594132a4bce01d5a25c6c8 : FAILURE in 1h 00m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fabef038a23e417a8d531284b26706b9 : FAILURE in 57m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/29165e9aacd5409593424b1aa3933ffa : FAILURE in 50m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/42051cba38e24fa5b3f3baace073a094 : FAILURE in 50m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ce335b396fe4a7ba9ba0ccf5b94602c : FAILURE in 52m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef463cc6f62a4aa783f85aea88a5cd50 : FAILURE in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af2b7fa576e54171a4faae79983e83e7 : FAILURE in 59m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/031ec80eabe443fc9eff84dd54e1c248 : FAILURE in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/61ecefcea1d24252a7ff0070a7387516 : SUCCESS in 56m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ed3a2940d9243deb54b036fac063c6e : SUCCESS in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752f8bcab5794abc8aa6d0e56aa14389 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ccdf7fcc38624ec0bd996c241d0624ec : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7d572614c98c43c854b8c168d42b799380714b80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:53:19.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ba4ed4275744eeb47c74669ecf8fe8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1424dbfa879347e38586f5ee1edd8d8b : SUCCESS in 28m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c37e9c42859c4981aec4998ebdd0e394 : SUCCESS in 27m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5246be0ac3144fc686f9768598cd551f : FAILURE in 1h 11m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7b46da64bbb42e79b8591950f2798e6 : FAILURE in 1h 13m 13s","accounts_in_message":[],"_revision_number":7},{"id":"ada4112bdb330c4c8f8c129f8f4c8b3013d0e32b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 11:49:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"088bd2cef6522ea0d0db3c9dfde32e70fdcab8aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 12:14:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a39410942a34486b9784e5fed1aab74\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e198df970154fc682d9ba9023dbb449 : SUCCESS in 17m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d1732b2e2754f889e95dc47e55fb104 : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/67d85feb1cf9434c9aad9c8e18c08138 : SUCCESS in 20m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6457676ba8f42fd9af26bff11ba6d83 : SUCCESS in 23m 14s","accounts_in_message":[],"_revision_number":8},{"id":"cf63a8fb1e22d72cc2eff2a8e6b872ae21907970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:05:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd7cb9cb6cb499cb34ad87eab8f226c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfea2ede2a041dba687df8256992a77 : SUCCESS in 9m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cbad02bfda44bf8b3d6092219f42c33 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b62b2fcc57cc435293e13e6d4e06d03a : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca0ccb9c9bfd4fd1a38010fe575afef7 : SUCCESS in 23m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac9357d1084e42eab82e3617d9f158c4 : SUCCESS in 11m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09f60634231649788ac2207a509e0ee4 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6118a90ee3f647c0a6a63188c1f3f590 : SUCCESS in 22m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d82033619ffc45518dc99d42b7ed4668 : SUCCESS in 18m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/decbee01c96b478c96dac566698b4bb1 : SUCCESS in 16m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2de336dec8ad414f94a591cedf2c3fe0 : SUCCESS in 19m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bca872aa6cf4e2d9d89f582b37d8f6a : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ef0fd3936d14dec8c8c23bb90cefb0c : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2819949c43f14682bac03e66887dad5e : SUCCESS in 22m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e928e394a442495ab5b26edaf35ac236 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b3adb6bae6f14f2e96b72421cd224b73 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d048051aac85465491f2492fb755aa2f : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca6db8fd709a496b9567d6d524061e52 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e557d2164b3c4b28b134c181b14ddf06 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c31ff73ee98f441fb959c4b1c08fb7b1 : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aa175fac35b419abb6d47588da8ad2a : SUCCESS in 1h 08m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b07f13c0283b46dfa87795d9eba3ef01 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f341ea1af8c84e8c896cf4d11e7092ef : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ea1b8f5ecaa4350b2bdcc31a7cbc73b : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"57ea552b41794dee6caa099b9145461430481e54","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 13:17:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"31c0739cc73a81882b1a67bcc2ed7452616dac7e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 13:17:28.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"a1a9e3bafbcaf5a3a99554820d61541ddb184acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:17:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2d6187c5492022190bebca36b69a6f16339ee6b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:45:51.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14e14d8b34544e6b4ef3b2ea761c143\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2625518a73a145b5a11fc8309babdee2 : SUCCESS in 7m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d2703f05a0274a308ad3e7f427de3164 : SUCCESS in 9m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1cb1df9eb95446f8b0f3fd79dab3e39c : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eabce1fa10fd43acb3e2580559acce86 : SUCCESS in 21m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62bc48956e5747df8ee323e6ee1f004d : SUCCESS in 21m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5fcca65f9c8642818723e676968e3c94 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/037d018d57874d46b80b84aba2cfdaba : SUCCESS in 14m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee1462f1001b4e40bbe54f5dc39ce111 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cad31f4bb5644699abd58618aa0a0f8c : SUCCESS in 21m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c5b93d9a83c4e0ea78f0444f16f8993 : SUCCESS in 24m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eefd1d660a3d4a88a3552bbdf8d2b1e1 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cc6274be5e0a43aea1fe300967eab620 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e033fc078be346ddad97d01b56a7b3ad : SUCCESS in 20m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46f7db0997e3443e82577cbe76d98b0a : SUCCESS in 8m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6295e62bcd90421dbacea7309c5ae274 : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8865ce06f7fd45bab049faf957216664 : SUCCESS in 1h 03m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1870dba55c0342fe968b6f59cb1ffcb1 : SUCCESS in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd65c232ff8049f4839a8c42379b4b60 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":8},{"id":"6110e26126a62b53aec57ac695476bb254516d2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:45:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"8eeb94ad01338739f007f1f1c16fd52cf7655b22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 14:46:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6282bb44f77644c78fbd98e24938db6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5660489e9937494892bfa34ef558060a : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4274130d977b477ab0024ba4c5a0f813 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/162ee7879aac47a3b28050f385ee6b91 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ab5b5448925f97c1c79cddfab519d10db33913ca","revisions":{"9b2897b57f31d0ca6c4e831e6534b99073e2c4bf":{"kind":"REWORK","_number":1,"created":"2024-04-11 14:01:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/1"}}},"commit":{"parents":[{"commit":"a16c6981b1f9319f6a3736febe0f2fdb74f24825","subject":"mpu mw: add session state metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a16c6981b1f9319f6a3736febe0f2fdb74f24825"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 14:01:46.000000000","tz":60},"subject":"WIP: add abort upload support","message":"WIP: add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\nTOD: cleanup parts\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b2897b57f31d0ca6c4e831e6534b99073e2c4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b2897b57f31d0ca6c4e831e6534b99073e2c4bf"}]},"branch":"refs/heads/feature/mpu"},"4e766665f0e651d45c53edc42a093c308cb4c5ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-12 14:19:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/2"}}},"commit":{"parents":[{"commit":"e4e14ffa3214b5ae2badd6381e78b39e5a647234","subject":"Add first cut MPU MW async cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4e14ffa3214b5ae2badd6381e78b39e5a647234"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-12 14:13:59.000000000","tz":60},"subject":"WIP: add abort upload support","message":"WIP: add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\nTOD: cleanup parts\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e766665f0e651d45c53edc42a093c308cb4c5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e766665f0e651d45c53edc42a093c308cb4c5ac"}]},"branch":"refs/heads/feature/mpu"},"3119296e880b4742bbdb21f1b247c11640636d9d":{"kind":"REWORK","_number":3,"created":"2024-04-19 14:57:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/3"}}},"commit":{"parents":[{"commit":"9b83262ddb8ad8265c09b81994e00df07a137c66","subject":"Add first cut MPU MW async cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b83262ddb8ad8265c09b81994e00df07a137c66"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-19 14:44:41.000000000","tz":60},"subject":"WIP: add abort upload support","message":"WIP: add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3119296e880b4742bbdb21f1b247c11640636d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3119296e880b4742bbdb21f1b247c11640636d9d"}]},"branch":"refs/heads/feature/mpu"},"b9da30ab0db21df78cf1db270694d604136e98c8":{"kind":"REWORK","_number":4,"created":"2024-04-19 14:59:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/4"}}},"commit":{"parents":[{"commit":"9b83262ddb8ad8265c09b81994e00df07a137c66","subject":"Add first cut MPU MW async cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b83262ddb8ad8265c09b81994e00df07a137c66"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-19 14:58:19.000000000","tz":60},"subject":"WIP: add abort upload support","message":"WIP: add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9da30ab0db21df78cf1db270694d604136e98c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9da30ab0db21df78cf1db270694d604136e98c8"}]},"branch":"refs/heads/feature/mpu"},"8419a7a0907f9c013cd747e97e98b13ffbc29aee":{"kind":"REWORK","_number":5,"created":"2024-04-22 18:34:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/5"}}},"commit":{"parents":[{"commit":"9b83262ddb8ad8265c09b81994e00df07a137c66","subject":"Add first cut MPU MW async cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b83262ddb8ad8265c09b81994e00df07a137c66"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 14:03:03.000000000","tz":60},"subject":"Add abort upload support","message":"Add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8419a7a0907f9c013cd747e97e98b13ffbc29aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8419a7a0907f9c013cd747e97e98b13ffbc29aee"}]},"branch":"refs/heads/feature/mpu"},"aed3147e2a3ea8c306ee2c85418bd6d815866202":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-24 13:23:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/6"}}},"commit":{"parents":[{"commit":"35baf41057dd7d8b58ceff5300c3ab4d0a70e28b","subject":"Add first cut MPU MW async cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35baf41057dd7d8b58ceff5300c3ab4d0a70e28b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-23 09:09:31.000000000","tz":60},"subject":"Add abort upload support","message":"Add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aed3147e2a3ea8c306ee2c85418bd6d815866202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aed3147e2a3ea8c306ee2c85418bd6d815866202"}]},"branch":"refs/heads/feature/mpu"},"6ca3c99b9395def3668db86b10e7c44cc8704871":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-04-25 09:38:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/7"}}},"commit":{"parents":[{"commit":"799135100a81cf107d52af9e3fb2e174f5024f35","subject":"Merge \"Add functional test module\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/799135100a81cf107d52af9e3fb2e174f5024f35"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 09:33:12.000000000","tz":60},"subject":"Add abort upload support","message":"Add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ca3c99b9395def3668db86b10e7c44cc8704871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ca3c99b9395def3668db86b10e7c44cc8704871"}]},"branch":"refs/heads/feature/mpu"},"ab5b5448925f97c1c79cddfab519d10db33913ca":{"kind":"REWORK","_number":8,"created":"2024-04-25 11:49:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/86/915486/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/915486/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/915486/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/915486/8"}}},"commit":{"parents":[{"commit":"799135100a81cf107d52af9e3fb2e174f5024f35","subject":"Merge \"Add functional test module\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/799135100a81cf107d52af9e3fb2e174f5024f35"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-11 13:58:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 11:43:44.000000000","tz":60},"subject":"Add abort upload support","message":"Add abort upload support\n\nChange-Id: I5fae2656e3ee594e8abf9532f7bb8c5af802a293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5b5448925f97c1c79cddfab519d10db33913ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5b5448925f97c1c79cddfab519d10db33913ca"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
