)]}'
{"id":"openstack%2Fswift~915762","triplet_id":"openstack%2Fswift~master~I30fe900d00619c4df3fce45232571eaa357fa4f6","project":"openstack/swift","branch":"master","topic":"utilsreorg","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-01 19:51:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I30fe900d00619c4df3fce45232571eaa357fa4f6","subject":"wip: sq? utils refactor","status":"ABANDONED","created":"2024-04-13 14:00:17.000000000","updated":"2024-05-01 19:51:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00ca6ad3136645da3df9b58ffaf9992659897adb","_number":915762,"virtual_id_number":915762,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-13 15:13:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-13 15:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-13 15:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0a377280fe3b21d932789c98d5a04ee9ee14d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-13 14:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"796b6380e6cb654efab274aa5868ce845c5e88bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 15:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5edddba087764baf9496bb968e61a958\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/075840f3bbb043a0a2b25f4f31810779 : SUCCESS in 14m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad2883b855a846679492d1471086d240 : SUCCESS in 44m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17680080aca4905b345dbabce119284 : SUCCESS in 21m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/083bed3bba4c419396015563468c0221 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"45a770994dc39410b42917037974ef7ef7275acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 15:13:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad162371c7bc40519c74ae3adac2c411\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e82bc52436984297964d11b178b47637 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52e9b0c240648c6b12d8db0196d7621 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b21734450d24c2bb285940d47aeac56 : SUCCESS in 4m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/bbb6fc2cae7a4d2f8f1dd4d652e54daf : SUCCESS in 41m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1ac91ba88f94d70a3c8187f6c2c9386 : SUCCESS in 35m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95aaa15ba3ab41f98d0596a6dd380a05 : SUCCESS in 41m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/790d15ca126a44ceb657c5767c779aa1 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9c61121da8542d3b85c395ab6b99dd3 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da8aeebdef964023943f9fa482c73a67 : SUCCESS in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f67f06172d24bd5b93687acef6332fb : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f93981af9fb843d6adddecd4aa20d0a4 : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f847e909f18042cc820a0c65b1785257 : SUCCESS in 20m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9034cfe591344c02800c80b4f3852ceb : SUCCESS in 17m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e287607b8f5a4209ad692d88ccde09ad : SUCCESS in 17m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1c6282992bfc446093297f13cf7383ac : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e0d47bdf5a4e4acebe2c1834687efa24 : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d59231ff35d445a2aa22d4f27cf5c739 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7d3f29f32b5a48188c9c3df0efb57018 : SUCCESS in 18m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/03e913f89ca549cdb5b084314d7a5eab : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a822b7d4c1fd45a7b3e5ccc2e466c310 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2cfd48e91cc14bbca7b0b37c8041c521 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9fdb4c90dda4652afb7d9d521f69c56 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af0899d479404dab96096054555b2bc8 : SUCCESS in 12m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eabbc0f2a7ed469693a750af21c988b9 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddc125baf763459a9231b578f32abda1 : FAILURE in 58m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05fa47ad23204be4ae7716eefa88bdc7 : SUCCESS in 52m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4838546d65f47dcab3eb90092287701 : SUCCESS in 34m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d068ecdfec8946b5b86352e47bf46816 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d51140890c4eb29d543b6273ea6757 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/56a7be218538410c96f7e57e2a29cf4a : SUCCESS in 18m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8619d8e8e64c43ad83da39d4528c768b : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":1},{"id":"00ca6ad3136645da3df9b58ffaf9992659897adb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-01 19:51:15.000000000","message":"Abandoned\n\npretty sure this all got squashed into 914029: Refactor utils | https://review.opendev.org/c/openstack/swift/+/914029","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86a979827796ce6b54f378000f5f55640478180c","revisions":{"86a979827796ce6b54f378000f5f55640478180c":{"kind":"REWORK","_number":1,"created":"2024-04-13 14:00:17.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/62/915762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/915762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/915762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/915762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/915762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/915762/1"}}},"commit":{"parents":[{"commit":"d998bc81d45bfda7baaa85e5a6915b7b4d660dde","subject":"Refactor utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d998bc81d45bfda7baaa85e5a6915b7b4d660dde"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-13 03:13:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-13 04:47:40.000000000","tz":-300},"subject":"wip: sq? utils refactor","message":"wip: sq? utils refactor\n\nChange-Id: I30fe900d00619c4df3fce45232571eaa357fa4f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a979827796ce6b54f378000f5f55640478180c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a979827796ce6b54f378000f5f55640478180c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
